)]}'
{"id":"openstack%2Fkayobe~874498","triplet_id":"openstack%2Fkayobe~stable%2Fyoga~Iecab44969cea015b363ec6884ef6a7c9960a6b3f","project":"openstack/kayobe","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-02-21 12:35:39.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}}},"hashtags":[],"change_id":"Iecab44969cea015b363ec6884ef6a7c9960a6b3f","subject":"CI: Don\u0027t run Heat and Horizon in overcloud jobs","status":"MERGED","created":"2023-02-21 08:37:01.000000000","updated":"2023-02-21 15:17:36.000000000","submitted":"2023-02-21 15:16:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874498","cherry_pick_of_change":872981,"cherry_pick_of_patch_set":3,"meta_rev_id":"89c2319894e26fc769cf16df93f7e152e4488642","_number":874498,"virtual_id_number":874498,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-02-21 15:16:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2023-02-21 12:35:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-02-21 09:38:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2023-02-21 12:35:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-21 08:37:01.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2023-02-21 09:38:26.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2023-02-21 10:43:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b000ffbd8ce56c16260364030cb3cbc51f063211","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-02-21 08:37:01.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c88211bab6473c7e29e4a4dc4998c9edef1311e9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-02-21 09:38:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b0b83070cf4b145fcc3c30a66cf592e17369fdc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 10:43:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37b63b34f611438299172360494a4ad3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e19462c4c94d434c8f96e0fb8a3b0fe1 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85db640a7ef44f168c22938bebf6e55e : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c537c2baf4b14d838dbb5e0f0bb5c706 : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b130dd4e6ef94461bf347294ee9223a7 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cec45be657c4007985b2fce75410df4 : SUCCESS in 12m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/37272eebf3ff4c82b4a795cbc0b7489a : SUCCESS in 6m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f3818fa9f100492c913a073a1e3ba43e : SUCCESS in 16m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/01bbe88fddc04ac4bd8ea8e7c2408dfb : SUCCESS in 25m 03s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c73ba31c802246588b335a6e0053e516 : SUCCESS in 1h 09m 39s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/116902e18cb643d5a7afd9527245ebc8 : SUCCESS in 1h 04m 30s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/afacb6d322a643b9a9b06a399813e39f : SUCCESS in 1h 17m 56s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67f44f3afb0b4c179c9935369d7f9303 : SUCCESS in 1h 00m 22s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/2277964906d843f890f6a5f2a6c69711 : SUCCESS in 25m 36s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/e3975a981efa4d5b81a9c6c5971f3599 : SUCCESS in 23m 31s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5dcb7fd42a544a9ba1a2be7042e13a3 : SUCCESS in 16m 48s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/217032682ab349aa95c9c3ed2fa3b5e5 : SUCCESS in 17m 33s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e3aca3777478422cac25279000819549 : SUCCESS in 1h 16m 30s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/27178a85eac3408aa9dbea2530f73604 : SUCCESS in 2h 01m 31s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d22fd338dd2c437d93a13a011be44b4a : SUCCESS in 1h 26m 41s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d7762997b42d42b2a9d847ffb892942a : SUCCESS in 21m 17s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/56f235556ac8490ab065996fd04e7ee3 : SUCCESS in 20m 15s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91ad9dfa6f894f11b39c1b926c4ad4ff : SUCCESS in 17m 44s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/929587f260ee4e19bad1c94260a57781 : SUCCESS in 20m 15s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8f1ca2981a044d4fb79d54f8026117eb : SUCCESS in 54m 26s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e562661bedd4ac8b3c42531a126f242 : SUCCESS in 38m 57s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/dda7c5edcf544442bea67de8f8984130 : SUCCESS in 16m 36s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/5344ee1c1a08424abb989bbc1a66accb : SUCCESS in 38m 29s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85ce766d0f9d49639f10d02477932a97 : SUCCESS in 12m 57s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/770b4bc9df5b4192879e088cff01dc8d : SUCCESS in 15m 15s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/ce5b87c1d5db4413a82c34a0b68df77c : SUCCESS in 21m 35s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/7a08ec3f70444f08bb5be5e80b4dfd25 : SUCCESS in 18m 52s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/452f8b79a9134c428ccef04efd5dc52e : SUCCESS in 13m 43s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/23a5e995eb1743bab280a6533085303f : SUCCESS in 15m 41s","accounts_in_message":[],"_revision_number":1},{"id":"66ae60e489fa77fb81915c8bef5fcb4ab59b5ed3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-02-21 12:35:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ce5afdb776b7aefc5c66abb8edee354911f7c54f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 12:36:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"869acd69e8e011f4010a68a7611432a127e60c9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 15:16:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abe356f7ce1e43fbb634f25f366de893\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e94a7e3082ea4a7c86661d8baa5c28b9 : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/176f4bdf8792499db8621aecc18eb6ab : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8681749b75cc4af8ab3bfebf6ba9dcaf : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1640af59cad4f24a8481f1c9ef31b51 : SUCCESS in 9m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0fe6f3fa443147a194cac86bd67b5c14 : SUCCESS in 6m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3605de5a1695471aaeaa323445aaee9f : SUCCESS in 13m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ffdc08a345994be58fd557e5b54bc732 : SUCCESS in 19m 01s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a93504104b0b43b98dcc16f3e651c704 : SUCCESS in 1h 02m 51s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/9f7f7f680807441b8825023f78eb2cb6 : SUCCESS in 1h 00m 09s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/589adbef60574b46a523c9d792530823 : SUCCESS in 59m 43s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d1f08c5d23344db5b030bf936bef94fe : SUCCESS in 52m 15s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d7b240af9dd647b19d016c831107efce : SUCCESS in 23m 25s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/8b38ff09ae8442bdbf1a612e818c9387 : SUCCESS in 20m 14s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2cdd602c87a3460b9e4ec2428444e1d6 : SUCCESS in 15m 48s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9db9f99661a34892b5c66e4e6a6678d0 : SUCCESS in 16m 10s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f9eb6c47e3274cceab76d18e3a07df3e : SUCCESS in 1h 04m 22s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/04c5db1630dc4d8892bb166175689bd1 : SUCCESS in 2h 23m 23s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c92531c87b84125afe29d9a277430a6 : SUCCESS in 1h 07m 30s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0bfc3a900338482482ac979f213cc770 : SUCCESS in 22m 25s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/b2f35d4bc198482a934bdec665e45928 : SUCCESS in 18m 51s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7305b6c9125f49958980b064d202103a : SUCCESS in 14m 27s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38e7a2ca58724d288ec507b0f3a53db6 : SUCCESS in 18m 45s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/88ae3410f3524a16ae6378f95bc4be8c : SUCCESS in 44m 45s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f7b2cd4d2874dfab6631a6fb1a678db : SUCCESS in 27m 18s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/6b4862e3b5b74087a0cf58cb6f57407e : SUCCESS in 17m 01s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/db2440b7d8e64b3da7569490ed753292 : SUCCESS in 15m 22s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ae13bcb936f4fe893ef065662e87f2e : SUCCESS in 11m 06s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dfaca9dea30f486ea5758e4e2e90f99f : SUCCESS in 12m 35s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/ad2a1925c3784b4ca6f56611ace6a159 : SUCCESS in 20m 43s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/c3ca194d2dc2479bb6142d451c79fa52 : SUCCESS in 17m 02s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91cd729d0d1140f882a34f5be0511847 : SUCCESS in 14m 43s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e8d67196fdba4f67a4ca8f3ea9f864d3 : SUCCESS in 13m 17s","accounts_in_message":[],"_revision_number":1},{"id":"6de434bb9df95a2854515b1f375ec555f383938e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 15:16:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"89c2319894e26fc769cf16df93f7e152e4488642","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-21 15:17:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4f3c5ba33ad4336a9ec8274afd1e7fa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c29e6c27f7cc445fac0d58d3b1e3f822 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"834cf25febd3d990d5c4466d53ec1ac377bc0379","revisions":{"834cf25febd3d990d5c4466d53ec1ac377bc0379":{"kind":"REWORK","_number":1,"created":"2023-02-21 08:37:01.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/98/874498/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/98/874498/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/874498/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/874498/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/874498/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/98/874498/1"}}},"commit":{"parents":[{"commit":"6ee465288c0177388b802a4866470bf403fbb028","subject":"Merge \"Replace references to Docker Hub by Quay.io\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6ee465288c0177388b802a4866470bf403fbb028"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-07 16:45:53.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-02-21 08:37:01.000000000","tz":0},"subject":"CI: Don\u0027t run Heat and Horizon in overcloud jobs","message":"CI: Don\u0027t run Heat and Horizon in overcloud jobs\n\nNot only TLS jobs need that treatment, Rocky9/CentOS\nStream 9 jobs have the same issue - let\u0027s disable\nHeat and Horizon in all overcloud jobs.\n\nChange-Id: Iecab44969cea015b363ec6884ef6a7c9960a6b3f\n(cherry picked from commit 7dc506ef48e53be9127385516d6ce045de321950)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/834cf25febd3d990d5c4466d53ec1ac377bc0379"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/834cf25febd3d990d5c4466d53ec1ac377bc0379"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
