)]}'
{"id":"openstack%2Fkayobe~874738","triplet_id":"openstack%2Fkayobe~stable%2Fyoga~Idf5ee822b03ba40179803c981500a6bad37594bf","project":"openstack/kayobe","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-03-04 02:47:26.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-03-02 13:46:34.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-03-03 22:50:33.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-03-04 02:47:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idf5ee822b03ba40179803c981500a6bad37594bf","subject":"Add Rocky Linux 9 support","status":"MERGED","created":"2023-02-22 08:29:32.000000000","updated":"2023-03-04 02:50:20.000000000","submitted":"2023-03-04 02:47:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":45,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"874738","cherry_pick_of_change":855656,"cherry_pick_of_patch_set":77,"meta_rev_id":"073591ecf91e8731acce3bf9fe5fe001685d96a5","_number":874738,"virtual_id_number":874738,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-04 02:47:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-03-03 08:29:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-03-02 13:46:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-03-03 08:29:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-02-22 08:29:32.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2023-02-22 08:30:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-02-28 14:30:53.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2023-02-28 14:30:53.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"b62456cb7af233520d9c01c27c2850dd4856016c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-22 08:29:32.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* ansible/inventory/group_vars/all/bifrost\n* ansible/inventory/group_vars/all/dnf\n* ansible/inventory/group_vars/all/globals\n* ansible/inventory/group_vars/all/infra-vms\n* ansible/inventory/group_vars/all/ipa\n* ansible/inventory/group_vars/all/kolla\n* ansible/inventory/group_vars/all/seed-vm\n* doc/source/configuration/reference/kolla-ansible.rst\n* doc/source/configuration/reference/kolla.rst\n* doc/source/configuration/reference/os-distribution.rst\n* doc/source/installation.rst\n* doc/source/resources.rst\n* doc/source/support-matrix.rst\n* etc/kayobe/globals.yml\n* etc/kayobe/kolla.yml\n* playbooks/kayobe-seed-base/overrides.yml.j2\n* zuul.d/project.yaml","accounts_in_message":[],"_revision_number":1},{"id":"5941b5a90a45859c0b755c6488628db207705285","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 08:30:05.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kayobe on branch stable/yoga.  The error was:\n\n  while parsing a block collection\n  in \"openstack/kayobe/zuul.d/project.yaml@stable/yoga\", line 2, column 1\ndid not find expected \u0027-\u0027 indicator\n  in \"openstack/kayobe/zuul.d/project.yaml@stable/yoga\", line 14, column 1","accounts_in_message":[],"_revision_number":1},{"id":"ff67bf12b527b06d4d62eb9f78d962ce79bf5ac1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-22 08:42:30.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"608c30d93019b58565390e9c605064b6396004ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-22 10:53:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4d6d64a7676145438619ae80c120cbe3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c69347d0509464f8053f5601ef6f3ce : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/372fbcd32222446eb265a10312b243a4 : FAILURE in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f5ff0223e0f4475b82b5880c3319c07 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cccd7cb628da45c3bbbdfeef8131b858 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/999810dab9774f958275bff1350770bd : FAILURE in 8m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d16f3b82702f454691cbe8d00a42e7a6 : SUCCESS in 3m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/36e442d747604315b04b7ef8319dd295 : SUCCESS in 5m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/764c66d1eecb4623852b2e844a68cd79 : SUCCESS in 15m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/83719451c9a64df6bd54a8599d6b83ee : SUCCESS in 19m 22s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e6685ad73954459ea9ecd3077c785c85 : FAILURE in 19m 43s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/77129a3f6e0c4b4eab97cb0e5c1ba8bd : FAILURE in 20m 20s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/43e0e63e87f44374b20582628caf0a62 : FAILURE in 17m 10s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/f32e654eeeb943b6a20db1d7956e4c20 : FAILURE in 17m 39s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09af7b0c393d42ce8295a6991f90460e : SUCCESS in 51m 40s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c2667d073dd54d538aa7d8a3be8455a1 : SUCCESS in 58m 19s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b64735b3c2ce468281f3b5729412b5bf : FAILURE in 12m 49s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/71573c9a297f48a6b6d36eaa6cca36dd : SUCCESS in 21m 05s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/63d46dff3c81439fb5c505b245eb0cc9 : FAILURE in 12m 51s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b2733ff31e044ab6a642602f180e3416 : SUCCESS in 20m 16s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ae0b7833cf74f53865453dba110ca69 : SUCCESS in 16m 26s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/48288368c6654d168272fabd840e2c7f : SUCCESS in 15m 10s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/061b82d81b2148d0a5699d9ef6ba8148 : FAILURE in 24m 40s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/d9c27038dbed462e97d2c14126d1e1b4 : FAILURE in 22m 06s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1a38450e971547108f73cd30e55dbb58 : SUCCESS in 2h 04m 04s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3aead0a934c2493186550273314ebc2e : SUCCESS in 1h 20m 45s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a0519c25f6b9415c99dcff9f8144604d : FAILURE in 10m 00s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/f92392a7a34a474ebf7bebf59ae69cb0 : FAILURE in 8m 57s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/d2a7d53c557143f2905d686a7a78e677 : FAILURE in 9m 39s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a762d6ee7e14d9180e13d98d45adf1d : FAILURE in 7m 54s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f78e4f3c10f84b078cf83a46caad7001 : FAILURE in 8m 02s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/79c474eb9fea4ec8b595a65cbdd6d879 : SUCCESS in 34m 03s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5899d25b870842b7960bffb6fe81b841 : SUCCESS in 26m 44s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f85592c4d2a746b2a41b20a17ff8b684 : SUCCESS in 25m 22s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/8cdec750e37f48e6b2b8b4bba953bddd : SUCCESS in 16m 34s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/44eec5d6ae8a49efa715caf13250f3b7 : SUCCESS in 16m 10s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8cc62de71d77450387b1cd9d0da6435a : SUCCESS in 15m 30s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f386f75e48c6485fbe5bda8de671f114 : SUCCESS in 13m 10s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/fbe7f8bc4d4d4b7c8f3cc366ca0c4696 : SUCCESS in 18m 21s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/a5a673e39e01426f888926ef14eac85f : SUCCESS in 19m 46s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1d01972a1e2c448baddd18ce169b5b29 : SUCCESS in 18m 56s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7186b4470b354973ad32a6ff197734e8 : SUCCESS in 13m 56s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf134bc24fc3458894c39c0336f40cfc : SUCCESS in 13m 50s","accounts_in_message":[],"_revision_number":2},{"id":"bafd59f17fc763d57b065ec52133d2ed8aad9e6e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-23 14:59:21.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"5d3e8877f0333b2a6a1a89eda0b125ee332e76fb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-23 15:19:10.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"df7b2b9f48c30d43e6a4a66a7668eae0852f5d47","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-02-23 16:03:58.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"9a2db213fb0798481a16a85038f20a84903860cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-23 17:58:08.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3ce3ec6216014087b51e921ba497a0f7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3e9a4fe8f3d48a3953f427b06b8f59d : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c51054012c549dfa8ed31d73b875081 : SUCCESS in 9m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e21ece10a7142258a15123022bc83ba : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1fc90b9871754986a4adade455018768 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/180de53821974dec87c5e183554ff18f : SUCCESS in 9m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4707ec67e01b4795820e2839cb174108 : SUCCESS in 3m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4149137519774efcb31c9e18fc5d4e97 : FAILURE in 6m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ad70dab1e01c4cefb287e4cd6362a395 : SUCCESS in 18m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/17d0c313cf904970acf64c2859fc7de4 : SUCCESS in 19m 32s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7f5b3f2a003b45b09f4e4d6b292ca2b4 : FAILURE in 13m 07s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/ac9966ad220c459da7c4c9c702c8f6c9 : FAILURE in 11m 31s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/b01f6ac856964a5294c30988e2af18d8 : FAILURE in 9m 32s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/5b51ffc6e3684e39a2a13f536e1b867a : FAILURE in 12m 57s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d3c9d1cbb4945f8aeb3ad85c9718105 : FAILURE in 10m 20s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/75c8fa770a16473fa3a10b2134933115 : FAILURE in 10m 44s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d85ec46902fb4fe1bc7934137d0065c1 : FAILURE in 15m 19s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/c9f036d57f8a41f19ea7dd1dd0f734f2 : FAILURE in 11m 49s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/e7a80de2275a44e0a758e321d472b8af : FAILURE in 11m 26s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/480916c40d7845f1a09051f075428661 : FAILURE in 12m 33s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/63dc2f46d8ad40029295e2fea0d46b3a : FAILURE in 9m 50s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc682333cc144bd19bff0441504ed9ae : FAILURE in 11m 34s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c34f184aea88441eac48e6d8219547a0 : FAILURE in 13m 16s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/9343c7a31cdb4b9285119760f7dda98e : FAILURE in 12m 35s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/03ab7af622ef4a5597f73e6904310171 : SUCCESS in 1h 52m 07s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2100722ca3d94d86badab75c0d02461b : SUCCESS in 1h 10m 43s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/620c4b090b76468a87a983524ec1bb41 : FAILURE in 10m 38s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/2fcf7be9dcda4d3ea2c1dbe7eef89760 : FAILURE in 9m 43s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/142f6c95374b4c0c8c8cf50f1927f3c2 : FAILURE in 11m 12s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22179cb2e16b4a9080ee09d596c63f34 : FAILURE in 10m 47s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ecfa05030d994c2796639fc185b925a6 : FAILURE in 11m 10s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8ed9ccafd7c64d28bc06b22609afa082 : SUCCESS in 48m 02s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b977ac38f684ad0a055ed784404c507 : SUCCESS in 28m 21s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/3320ffbbdbfd472a8d76dc00f01fc068 : FAILURE in 12m 30s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/2a310621ab734a8fbd8e6445a5795d1a : FAILURE in 11m 05s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6a8ee4edc55843308b92bda74b56d613 : FAILURE in 11m 41s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e47c383e93df42cc9ae8159f2d3d8f00 : FAILURE in 9m 41s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5feb8af4ecad4a8eb1a038c69e15caa9 : FAILURE in 9m 05s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/d0bdd727b7f842e9a8c290fcc648ab79 : FAILURE in 12m 48s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/ff2418b5eafa4fe088e758239c2e3685 : FAILURE in 14m 50s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9d911600288241c59bf22f7bc68cf29e : FAILURE in 10m 18s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c926e447f2154da49672b2e1df175993 : FAILURE in 10m 27s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a41e913a664477eb3a30d9b1a395109 : FAILURE in 10m 17s","accounts_in_message":[],"_revision_number":5},{"id":"cf9281a73de69bc1cdec1700766dcbfe3d48ad87","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-02-24 10:01:41.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"3ff4e6e50be125a01afadae174b76d285f0a75dd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-02-24 10:20:34.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"3da2885183504488a43f44ee710b30f8865d9024","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-02-24 10:21:08.000000000","message":"Uploaded patch set 8: New patch set was added with same tree, parent tree, and commit message as Patch Set 7.","accounts_in_message":[],"_revision_number":8},{"id":"051579c5cc18ce09657a7543e2c794547acd2b03","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-02-24 10:49:34.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"9909d833ebf673864ce30cf7081b5b199ee10815","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-02-24 11:25:07.000000000","message":"Uploaded patch set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"3b24a2416820e786304efc2e4143f4cb78325ee7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 11:28:02.000000000","message":"Patch Set 10: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/874917 is needed.","accounts_in_message":[],"_revision_number":10},{"id":"9029253bfe9a799a74e9cc80d1d1d37b34fd62e0","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-02-24 11:31:44.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"365e54b6365da0c6f9a4c1af7fd7a293f94215c0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-02-24 12:59:56.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"269f3518450637d736ab3eefdaa97c5977fb728f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-02-24 13:29:52.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"c0fb14095857494fd2fe5e58f07b5e4ecf7428e1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-02-24 14:06:23.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"cc6955709a3f6993fa2d2cb217d715f43e1771e3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-02-24 14:32:21.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"7c3eda19e2645dc7039fa28e2b05a1f052b6e153","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-24 16:18:56.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ec539560fba448819169465cbf882678\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4ba1394346543a2b5dabc016a374a70 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7386e369e6d94411b992ed4898f57463 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9ac62e63be04492a8348759373baaa3 : SUCCESS in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6fb6bd46a540479d9b5b6ac04ebaaee4 : SUCCESS in 7m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2df8c827b29f4e6db219960818cf2c88 : SUCCESS in 8m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6687ffa15c7648788c8f347000e9e0dd : SUCCESS in 3m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7ff609fc712542d8aea78714f1f0f504 : SUCCESS in 6m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/441dbe73f1384cce86dccc0682311458 : SUCCESS in 14m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/80b8242c0945475f803f381714da69de : SUCCESS in 20m 14s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/62fc800cff924b3a861e1fe47d52e063 : FAILURE in 51m 57s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/be31eda984bc4555a6b485b703191c26 : SUCCESS in 1h 03m 12s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/bbf95f284e8f4d33b4656ebef2bd1db4 : FAILURE in 25m 43s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/bbf93a8269c6425d871295da86ce46f3 : SUCCESS in 1h 04m 24s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/133443374a5e4e64bef75b2e680fd9f5 : SUCCESS in 52m 43s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d92fc844f8b44db38e59a62c24ed05a4 : SUCCESS in 51m 24s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/28d794b5d8ce4561b0cc2ad9a1a45cff : FAILURE in 8m 29s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/5c9b8afa0aa94ceba778dfe80e6942b8 : SUCCESS in 30m 02s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/599c4310524649b0b79631523ee465bd : FAILURE in 11m 20s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/8282e8324719417ebb7270efcb695a19 : SUCCESS in 20m 52s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e91f58193dd64429a993d05e7b1f4502 : SUCCESS in 18m 41s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1a23b638b69a4fd993a02361d58ab2fd : SUCCESS in 20m 38s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/616dbc9ef0184f7aac4654482d8b19a9 : FAILURE in 54m 33s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a8c530d18e724831b199968eb6f39556 : SUCCESS in 1h 04m 55s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/671ef7e8232e4577b9e5d3631fe610aa : SUCCESS in 1h 44m 23s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c42413719a0e44a9975f1810a18af565 : SUCCESS in 1h 07m 07s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/bb3bff5bd8344dd09bb0ac94dc40f6fc : SUCCESS in 23m 11s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/9bdc2310dd774981a2fc91520439a5f1 : SUCCESS in 20m 02s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/08e0906dc07e4d1e950e074362e81a6d : SUCCESS in 19m 56s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a7c74df99efc49efa555af51ff4b790f : SUCCESS in 22m 23s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cc7d8b0e0c6143c2985168af9c4db812 : SUCCESS in 15m 08s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/66d54ea3c7234575b59bac3877f48194 : SUCCESS in 42m 52s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff103c6b926b46b08c9a706e491d7162 : SUCCESS in 28m 19s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/19a31ba635c44f3a9aeb1ae9d6a5378e : SUCCESS in 24m 20s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/e260670dd6854d6da830261429a605b3 : SUCCESS in 13m 38s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e1b909bdd4d04670a88f97766cc2d872 : SUCCESS in 16m 54s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ecf7d57bc6845c2b8532ebed70194bf : SUCCESS in 15m 22s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d03256d0fbe34fad914c90a417163087 : SUCCESS in 11m 17s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/e25ca9a0993c491b89c4a103b0326dde : SUCCESS in 21m 38s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/42c1a1028a7047978dc853b440403ad8 : SUCCESS in 17m 29s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9546919e52d3487ba6d54f0f7368a002 : SUCCESS in 14m 54s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a2f9cf2e8d5b43c9a35ac62ffae2b36a : SUCCESS in 16m 00s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ad8fdb0a0e94155bba71ca5ae2c8ef6 : SUCCESS in 12m 47s","accounts_in_message":[],"_revision_number":14},{"id":"3f64c97bac82905967be578267695f4021cb09e7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-02-27 11:13:07.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"6788506152d0aaf5d948340e3e97160c380d080b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-02-27 12:03:22.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"82f9ac457bd6ccc6733b54fec6c348d9f238bc9e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-02-27 13:37:37.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"c4a2d6f51ba853a13934a8361e66768fc4aaf813","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-02-27 13:56:04.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"1db11fc204fa57f05492cf4661cff6b068ae2bbc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-02-27 14:40:31.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"a3f6aea134d678fd9406ecddd2631a5eff6e470f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 16:46:31.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2543b23069104e47b78a5d85056dfef4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57d7704dd6cc455e93c028d1a9e13ecd : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52fe96681a7a4976b2ef91c04894d470 : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5372885329244f7e9a94544512778c41 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/83eb81c9dc7b4c51a56d40c4fff46cf2 : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4020ebb28095481ca610324ed11bb590 : SUCCESS in 10m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54c698f40232462082ee90cbae4daae6 : SUCCESS in 3m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ef3dda08455c4238a0aa37556fb3ce58 : SUCCESS in 5m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/00c1657db1024db881382f000fa5c3cd : SUCCESS in 13m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d4f8076979bd4f53b007676171d43c5d : SUCCESS in 18m 33s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4904b385f6a344bcb8c2245f924573f4 : SUCCESS in 1h 17m 44s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/80adf7c401d0433f84b9469b10142bc9 : SUCCESS in 1h 10m 58s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/05df52f72a784be28f47b4baa2c7ba32 : FAILURE in 29m 50s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/70be3a0d23d545b19197ee7db7a2ecd0 : SUCCESS in 1h 04m 44s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c96887fcaff9450695e60a8ef633b546 : SUCCESS in 52m 34s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9026b354ea7f41d4ae1882759da447e1 : SUCCESS in 55m 31s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c63a1d64d6a04dcb807b2479a9833902 : SUCCESS in 25m 25s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/f0d81a40eb5647d38d1396a826122b55 : SUCCESS in 20m 40s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/ae888249c1e84778ab7a7a114d3760a9 : SUCCESS in 20m 30s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/239cf933e7ca4a8e8f9e9aee19a46537 : SUCCESS in 19m 04s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6da04fb9f839480e8dafefb3924468f7 : SUCCESS in 16m 38s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/44f13a11c83b418490e8752bbfb8be03 : SUCCESS in 17m 05s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f3c630aea0964b1e947d4325a85d2211 : SUCCESS in 1h 00m 21s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/bf98adac333049f0863a94719c2658b4 : SUCCESS in 1h 03m 00s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bd4736cc65fb42e88563c0d2257caf06 : SUCCESS in 2h 00m 12s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/19b5bc16ad1c4597998f24af74669cdf : SUCCESS in 1h 21m 12s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/00b5d50f78f947d995b1df3da7942422 : SUCCESS in 17m 39s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/892d5abf05404b58abe38a6bae0c6e34 : SUCCESS in 18m 45s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/7f2ba1387b6c430392827af6bd782755 : SUCCESS in 20m 46s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/29af0720f8bd4ca9bc632d6bd06ce6b9 : SUCCESS in 15m 47s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/37e70f86ced645c6a9dc096dd262b7b6 : SUCCESS in 16m 42s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/323e37bf8c6c49a9be80155e5456363d : SUCCESS in 57m 16s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dde3b128f1754279b7fcb14ccb806116 : SUCCESS in 28m 04s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/acc836c4ead04742a971fc8306f5fad3 : SUCCESS in 17m 41s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/b2f73e3873bc49f18d210e80ac77a8da : SUCCESS in 16m 13s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d24d936f63634ccba5b17a03c319bbe7 : SUCCESS in 15m 44s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c6ab1af58424de9b907bc1c033078fe : SUCCESS in 13m 38s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/282ea61fbbc5469497914ae4e73f11fa : SUCCESS in 14m 13s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/de9734a6231a4890a6bd3fd3bb7572dd : SUCCESS in 18m 34s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/f2dd865525134f5cab324be0f3408b84 : SUCCESS in 20m 16s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7ae06351f1ea4774ba4133e0ec7e5cd2 : SUCCESS in 18m 45s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/673604a8d8074697af90354eaefbb078 : SUCCESS in 12m 35s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/854deebb7cbd415587a9861a183c1a50 : SUCCESS in 14m 08s","accounts_in_message":[],"_revision_number":19},{"id":"c5801c6b199b1c76a558bbd29302349bde51253c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-02-28 12:03:45.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"6646aa3e241b3d6ab4c887cd52bf8f10d12764ad","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-02-28 12:52:11.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"3bb3c68ac3750ce573d4170156d13bf9812e34ca","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-02-28 14:30:53.000000000","message":"Patch Set 21:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":21},{"id":"550da93987086003ce4d1f2de817c16bf4847ceb","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-02-28 14:32:01.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"0329144fe2350bfca2d1a15c6bf3bde703a04b4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-02-28 14:33:33.000000000","message":"Patch Set 22: Commit message was updated.","accounts_in_message":[],"_revision_number":22},{"id":"bef45684a636dbc57704ffef343355cecfdaed90","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-02-28 14:47:31.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"99456109a79a5f7e5f95cfe272b58d6b654a6d7a","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 15:45:24.000000000","message":"Patch Set 22:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f50902854f5d4292bc487c1f89e63950\n\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/1d7abaad01db4bc2824b777cfa0f05c7 : SUCCESS in 52m 55s (non-voting)\n- kayobe-seed-images-centos9s https://zuul.opendev.org/t/openstack/build/f837f74089d448469435445b2b41f039 : SUCCESS in 52m 03s (non-voting)\n- kayobe-seed-images-rocky8 https://zuul.opendev.org/t/openstack/build/3ecf9460f8f44ad78bd9e6396185541f : SUCCESS in 43m 04s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/4b51af454e314ade864c92066f0c29ae : SUCCESS in 46m 17s (non-voting)\n- kayobe-seed-images-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5b9b669b6e44829bb77000728e3656e : SUCCESS in 36m 28s (non-voting)\n- kayobe-seed-images-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9dd090acc8a44c7b90e1ce7d16a5e860 : SUCCESS in 35m 04s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"a3ecfa355c46ae0ef2c601c9b18411dc24589c44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 16:46:45.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5839632de2014e0481e63f372d191565\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3694f03da0fa4afbbf2352c97a84de52 : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d03ae639de74faea9c468dd75286d85 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c52cec3d769749a298c84808875f2bc9 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f53733d5d07b4283b7aef8573ccc622b : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a099bd005bbd4299a8a808cc3ef06e68 : SUCCESS in 9m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0cecb5d809ee48a996525db27df9c6e5 : SUCCESS in 2m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/28aa832d51a546418cc16fa4897eb2ae : SUCCESS in 6m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/415ef4e188fc48938c24f2f2bf82a2bf : SUCCESS in 15m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7b0eb10dc1af45d389068b3f70d609eb : SUCCESS in 19m 15s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a9fd523e91364043ae607751624e8488 : SUCCESS in 1h 08m 10s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/bd53c794c46149f0a5f4026307c558e8 : FAILURE in 47m 36s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/4cfc9e9720964032b146ae4cf0f323c2 : SUCCESS in 1h 07m 57s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/b4ba191ed94c4a76b422d724cc948f79 : SUCCESS in 1h 05m 25s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27c271d50e474bfe94447eae946c3413 : SUCCESS in 57m 21s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a451f79ea8e4e148d443b793e3d5ef0 : SUCCESS in 47m 16s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/aee99bc8d966424688ab5879faf06a5b : SUCCESS in 22m 23s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/a9a6191d711a4a5c890688fc2268d4fc : SUCCESS in 20m 18s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/5a18cf5a2dc34a419eb0b923cf24e7bd : SUCCESS in 22m 27s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/52124bc4ccc2421bb7b89d64b2b8d50c : SUCCESS in 21m 33s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ece45f251f7949fabf95aeed5bbe41dc : SUCCESS in 16m 46s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/af256365b34044038a719e418ef2b412 : SUCCESS in 16m 22s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/735010ee1d9347ffbe1d6c52dac59932 : SUCCESS in 1h 14m 58s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/42a5c71307b946dda294aea5dd2aba73 : SUCCESS in 1h 05m 54s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2348cb08600e4eae89aba3bb402b649a : SUCCESS in 2h 03m 34s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc24ef50e12d4558a50287e95e184bb2 : SUCCESS in 1h 12m 53s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/37148275c5a44761a7ff9439a5a7c306 : SUCCESS in 21m 57s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/09e53506ab4f4f919124f02fdcebf30a : SUCCESS in 21m 19s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3b8422067bf84cc6a4e2d9fb2446408d : SUCCESS in 20m 15s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/981fafc9fdfc472ab33f5e6836cbb70c : SUCCESS in 16m 43s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a9158d0356224971bd547981874b44df : SUCCESS in 15m 25s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/eaa9eb80f3754fc19eb03dc7297bac23 : SUCCESS in 38m 33s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/38feb5377abb44ac820c7543aab6f361 : SUCCESS in 33m 31s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/1234a6057c5c4bf98907938bbeca3ff5 : SUCCESS in 17m 59s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/6d240321e7f84d21ac8cfe8ebd87d488 : SUCCESS in 16m 35s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4456f8a4fdca4eb081443621c8fef7c7 : SUCCESS in 16m 53s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0bd254e2323a47e8b337d83dc285587b : SUCCESS in 16m 38s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/419e618f48c74566862d61588c7c518d : SUCCESS in 13m 02s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/7b88c25115f24981a8d1223aa47f6d34 : SUCCESS in 23m 34s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/e72c64f504ee49efb95c80a94e39964e : SUCCESS in 20m 50s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bafd189800594938a2950e1cc5b66df2 : SUCCESS in 16m 25s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68d64cbd51114a14a9f8be0ac5cc87a2 : SUCCESS in 15m 45s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a13d2dc6e28c47e1a3556bb7c63f2208 : SUCCESS in 16m 31s","accounts_in_message":[],"_revision_number":22},{"id":"ad5ffa7de4af7f96fb28fc192a0350c89ed30e46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-01 13:25:39.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"0164adb0ee20142ca6d2d350e036d3eadd677394","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-01 13:35:43.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"2a35f80a310eefceeb2bde46b4082e2b3cd33d50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-01 14:46:54.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"a630a3fd69ff727de6edc13a1c04b25a933713e3","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-01 14:59:12.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"9cd83686c20587e8374e69ad6e142dbbeab552f7","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 16:08:35.000000000","message":"Patch Set 24:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1947e46332a141358c973e5befc84679\n\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/67e9a3d41e9c4a6dbcc180632285c3d9 : SUCCESS in 44m 10s (non-voting)\n- kayobe-seed-images-centos9s https://zuul.opendev.org/t/openstack/build/f5622392bcd649cb8329f11e40447717 : SUCCESS in 43m 23s (non-voting)\n- kayobe-seed-images-rocky8 https://zuul.opendev.org/t/openstack/build/111a96c4c3d340a1bb2bb61b21337101 : SUCCESS in 44m 17s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/09930772452743f5b933d3d583d7075a : SUCCESS in 38m 25s (non-voting)\n- kayobe-seed-images-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1ab7b1e5fc1409e9fd41308493403bc : SUCCESS in 43m 21s (non-voting)\n- kayobe-seed-images-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/84cb36d6b82d4dc3b13a7bbe699ec69e : SUCCESS in 39m 04s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"aeb5ef4de81494d7a18ab0f1b2afd0b0f9b0602f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 16:44:29.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f185a07ebedb45e9b2886ac16b391855\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e6ca2a0966c7482ead49848cc3d8853e : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97085b54abaa46558ea9c707791d4d82 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51fd2122bca94652909282f0b31440f8 : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41335b6b9f924782b058bac91823de1c : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99722225caae465c8a016114948cb550 : SUCCESS in 9m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da43d7c55d0a40d39f7034dd78a1a160 : SUCCESS in 3m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6b353e18431c42eaaba0e73a0a0c0b0e : SUCCESS in 5m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/45576dbc6e4647dba1a20038aa37f870 : SUCCESS in 14m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c13b61eef685453998d3ebf4a92c301e : SUCCESS in 22m 57s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0ed85464f5404f39927d2826450732cc : SUCCESS in 1h 05m 16s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/7a76537cb9fc47b8bee3df7c80cb0afe : SUCCESS in 57m 27s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/aa3712ca0aa24112bb23540452914852 : SUCCESS in 1h 07m 22s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/58f24e5a867748f19f459138447a3ae1 : SUCCESS in 1h 05m 44s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cafdc9f352704290af0318acd7b614c9 : SUCCESS in 56m 35s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3212c4530e6e4e1f82694c74be0cfa47 : SUCCESS in 1h 02m 58s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/4093b2daf01f47b39734d2f3f6a223e9 : SUCCESS in 23m 44s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/c047acb8a1ad4b1aab1b253c4698cf45 : SUCCESS in 22m 10s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/091643c147cc437db4fed4a9b60e1386 : SUCCESS in 24m 35s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a03c5a3bf0fc424794d2a3c191db810f : SUCCESS in 19m 00s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a563599bb329420aa7511e655ab58f00 : SUCCESS in 17m 34s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/985e8fce5945472d8ec752f6e6ecdfea : SUCCESS in 17m 24s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/26db189185b24f82bca66e36e0bf1555 : SUCCESS in 1h 19m 35s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/335d0e2d553a40a9b2ac494e7806ca7e : SUCCESS in 1h 11m 35s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9571f3843f4748eeb2f7369ae7cf4266 : SUCCESS in 1h 47m 40s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02a3a3cacd2e48059e4d2d8401bf3268 : SUCCESS in 1h 18m 30s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/2d80e7bc20364b558b93a59f36839e61 : SUCCESS in 23m 24s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/f1778f9a9712490a920d4a14de83455e : SUCCESS in 20m 20s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/9f42fe8a66c6468798091edcabd9de18 : SUCCESS in 21m 39s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bf98c9e4b39646e5a69cf1c103cde276 : SUCCESS in 17m 24s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c0ab85ed0ad4749a0cf0365252765bd : SUCCESS in 19m 11s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b5807e366018409da99ce2808de4d53a : SUCCESS in 40m 24s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1696edc7f24483dbf76d8c3188c7d90 : SUCCESS in 30m 33s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/9133d59212b6473bae14c8de481d0a16 : SUCCESS in 20m 19s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/0e5b99cb8ef5484e81c8435ce6e8929c : SUCCESS in 18m 26s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/30cb2bc69f9d4a2783c78b317503d7b0 : SUCCESS in 17m 27s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c25a08069a904ca78da25caf96078e1c : SUCCESS in 14m 39s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e6e1bc72f1640eba5400715e88721b4 : SUCCESS in 15m 03s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/81b076f39a254fd38b1ca81558dd2fc1 : SUCCESS in 23m 51s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/86ed094b9c68423c8e0bd207299cd2e8 : SUCCESS in 16m 50s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0d32fd3df66f40888816dcb65a9c9062 : SUCCESS in 18m 35s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af1b21319dd842289bd56a4da12ee28a : SUCCESS in 13m 24s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e735d6a184d34a4ab3a6472eda1cc7eb : SUCCESS in 17m 26s","accounts_in_message":[],"_revision_number":24},{"id":"610ee9ee4acc0790b0c7a744bb520b2c4d60f91d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-02 11:08:01.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"ae42dc5ddbebdfbe821c9b00917340408e70eb21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-02 11:09:06.000000000","message":"Uploaded patch set 26.","accounts_in_message":[],"_revision_number":26},{"id":"04ffdda15c40953e9944c48cf397c949c78aaae0","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-02 11:11:29.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"dc755c2f5d6f1e3e62241c1353eb154f03e45f6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-02 11:17:51.000000000","message":"Uploaded patch set 27: Commit message was updated.","accounts_in_message":[],"_revision_number":27},{"id":"56d5cdaded103e8164518cd8a8c24d56e9ad5a8c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-03-02 12:04:33.000000000","message":"Patch Set 27: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":27},{"id":"40c4571584923b633899e8ff8a41d90c574a9f81","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-02 12:40:09.000000000","message":"Patch Set 27:\n\n(7 comments)","accounts_in_message":[],"_revision_number":27},{"id":"5947eceb7145c029b2bc399c96414a9eb19d1077","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-02 12:40:20.000000000","message":"Uploaded patch set 28.","accounts_in_message":[],"_revision_number":28},{"id":"8b4f55c6d15330d06a2cde6715a72e507aa7bd36","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-03-02 12:40:38.000000000","message":"Patch Set 27: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":27},{"id":"e54c8374ffd6b0da184367124b55af7365045e17","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-02 12:40:43.000000000","message":"Patch Set 28:\n\n(2 comments)","accounts_in_message":[],"_revision_number":28},{"id":"d15cf4ce1c52fb6ce3f59bf66c53de3524876dd1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-03-02 12:41:20.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"5105bace839d2ba1022808a7604e943d24475672","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-02 13:34:39.000000000","message":"Patch Set 28:\n\n(6 comments)","accounts_in_message":[],"_revision_number":28},{"id":"057f7f08bd1edd707e9f9326d08a9266cf73b11d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-02 13:34:49.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"8a49b7e0db675855df63cb97d10bd782f4af36ac","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-02 13:36:22.000000000","message":"Patch Set 29:\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"93f7b994e11195905e64e25f6c7bb8dfd3a3aa08","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-03-02 13:46:34.000000000","message":"Patch Set 29: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"0ef9bcd4d80350293594e8a1994600fa13e7257e","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-02 13:53:00.000000000","message":"Patch Set 29:\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"f14cb18cf2dc66c939d17120d3ee87b2f7415d1e","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 14:41:09.000000000","message":"Patch Set 29:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff3ec3889cf248f9baa2760095619b09\n\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/f97158e14bce40939b161c84082cf7c0 : SUCCESS in 56m 07s (non-voting)\n- kayobe-seed-images-centos9s https://zuul.opendev.org/t/openstack/build/35f2cddeda104b48b8e694198dcdeb70 : SUCCESS in 42m 31s (non-voting)\n- kayobe-seed-images-rocky8 https://zuul.opendev.org/t/openstack/build/b04de3f5c1ce4c00932c759c579769a0 : SUCCESS in 40m 59s (non-voting)\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/3eb7e4a2040b4214b2432e18367a9575 : SUCCESS in 39m 30s (non-voting)\n- kayobe-seed-images-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc93a0b283274d379e6e67f8de0b2765 : SUCCESS in 41m 19s (non-voting)\n- kayobe-seed-images-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0341a1f0a53b4607bf88660b0c03d5dd : SUCCESS in 38m 21s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"50c818a94fbbe267cb1b42252b2508c2bd1effe2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 15:36:41.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3e48b47ef9f94d138567e79804a7032c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6462fb9e7ccd4b819caf5b8220954102 : SUCCESS in 7m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9eb3dfdb3e94eb7b51141e661c3b061 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7644bb16ff604b518dfbd38a1e20983d : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ab0b0c6c258487aa5f799f0bb1eced8 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/922a762562c54d6e83d5deb7b8417e9e : SUCCESS in 10m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b97f5f6f23a4abab7c003780f19b3dc : SUCCESS in 3m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/eb44115320cb42e39e215e7f8a03fe0e : SUCCESS in 6m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/10dbd1cc583e4dfb9c8bae40e01e5731 : SUCCESS in 13m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/85c424be07c3432a986c49cfda11bb3d : SUCCESS in 18m 00s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/99fc05774d6440e583d816a5d08f82c1 : SUCCESS in 1h 25m 25s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/11e280be84604cf49a581678eb3ef591 : SUCCESS in 1h 04m 51s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/b1186bb7546649098f8f7eaed4fb5f08 : SUCCESS in 1h 06m 06s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/65683f953045467d8beec7773ec9a2dc : SUCCESS in 1h 07m 04s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fae66cd68624436bb07b48d5ea58d249 : SUCCESS in 56m 37s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2923e4803eb34df48d2eca0e8d26c398 : SUCCESS in 1h 12m 44s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0cdfd3a9707c40fc8b96f652661a4f0d : SUCCESS in 26m 41s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/cccadf23a43947ab85e1d4b18aaee124 : SUCCESS in 21m 19s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/f11ccaf645b744a1838dbb8e65a19c76 : SUCCESS in 19m 12s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/31d7ac390d7c4b9f95e743df4026b401 : SUCCESS in 20m 36s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08fec07e0ea5430ab3289e0303e53400 : SUCCESS in 15m 49s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dbb4ae345fbf4884b2889d983c2a1a8a : SUCCESS in 19m 06s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/20432081da884c93923d96d989520f0e : SUCCESS in 1h 14m 45s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/9d03d692bfe54ac4aec0688664c76510 : SUCCESS in 1h 07m 23s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/04d295edea9e44b885241847081c871c : SUCCESS in 1h 55m 32s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/322151d8899b47c9950a08b4c1599c0a : SUCCESS in 1h 18m 45s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/f042832be4df4323b9d5b8396aa28479 : SUCCESS in 21m 07s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/6feca00a8b76441fafdad1791d78f3ae : SUCCESS in 19m 30s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a1833d6c941a411ca75a6c151c91e297 : SUCCESS in 14m 57s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bfccd7d5abd44b7b8610de25412f4624 : SUCCESS in 16m 34s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8777263c5d344ad187ce35f6fb2cb249 : SUCCESS in 20m 43s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fa6e935e5c1b49f8a50893a647137848 : SUCCESS in 44m 21s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5854edbebd874a3aa4f9118ae4f71685 : SUCCESS in 32m 50s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ae94089de9cd4471b878c710064ab4c3 : SUCCESS in 16m 53s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/ad7eb1c78535416da627b0e0b3b1741e : SUCCESS in 17m 25s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f5294660a3514300b99eb102fe3ea8f1 : FAILURE in 12m 09s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e603c97b0e6942f799ad4fed5ffee179 : SUCCESS in 15m 44s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f5a479d040334f79bcd962df161b1646 : SUCCESS in 12m 57s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/84d6d17c25374bf49704a9c59e1d442f : SUCCESS in 21m 41s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/d997c7a52eaf4de7a7cec266c69df374 : SUCCESS in 17m 33s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3c5cf73693164b61b8e99d647b8e0658 : FAILURE in 10m 55s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2a6b52cb2474903b81243517979deb0 : SUCCESS in 15m 42s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7fc3b0f7ab5a4d739b532e6c965717be : SUCCESS in 14m 04s","accounts_in_message":[],"_revision_number":29},{"id":"b5974c776521ea0262c06a2a70806d74fad82689","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-02 15:37:15.000000000","message":"Patch Set 29:\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"371efc12985f0c951b32029795449d03d79639e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 18:30:33.000000000","message":"Patch Set 29:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b85cb7da0024e2ab76577262d4bf11a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f2200dc0cbe44e01b19730918d0311b7 : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e6da029f8a34db9b42b5d881cb19281 : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/784dfea1060f4be1a3417ba80c94e59d : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89231822ff664eb8b0a914a190877df2 : SUCCESS in 7m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed51ad53de7f4743a1ccfb19ea8d4e16 : SUCCESS in 11m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84cee5222609425abbcc54e0575aad4c : SUCCESS in 3m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/498211c8019f4ca89822b5d0389c18b5 : SUCCESS in 6m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7097751e2d55485eaa990b407e25311a : SUCCESS in 13m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/708684516a164e4e8b55d54cce5d791c : SUCCESS in 14m 30s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c8ab328d607d4a299a74464fc1c092c9 : SUCCESS in 1h 06m 42s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/f94344f8a9a24bfd9cb2aa49d2c4c089 : SUCCESS in 1h 54m 21s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/89384ef031014292bc46e6ef8234ce32 : SUCCESS in 1h 15m 57s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/296954bc5aab4969af68babcc021f657 : SUCCESS in 55m 49s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8cd9113229664d10b8e70f6e3b364613 : SUCCESS in 44m 56s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c9c9c4ea82c4495928851247e11549d : SUCCESS in 49m 42s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d0b9802bc3f1445f9bd6b14306dd9833 : SUCCESS in 24m 11s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/618368e2f1ad4975994aead1c0c5747f : SUCCESS in 24m 46s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/454b52345db84b329ee386596bdb541b : SUCCESS in 22m 50s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9928468e019145fb860413187c535ce3 : SUCCESS in 21m 10s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a949f05e5bd34de39f94c0bec7d80a2f : SUCCESS in 17m 44s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/946dc22715ca45fbab6d3194bbad5a2b : SUCCESS in 17m 48s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/bcfd1cc4018641b2ae710f5867df3bb7 : SUCCESS in 1h 30m 41s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/8e70269d2a0248828915523d9b9eaea8 : SUCCESS in 1h 08m 15s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/64f74c5eb45a415d91a7bd49bee2b1b4 : SUCCESS in 2h 00m 12s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e91b15b78da4d0daabe725fcf9bd7e8 : SUCCESS in 1h 15m 31s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3f78e571b5e7483b9d59d779b825eea5 : SUCCESS in 21m 27s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/5ca7c1e19a2d402092ffe0b8f86d9e62 : SUCCESS in 20m 03s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f5af85c332024a0b91163dbd3fe1f40d : SUCCESS in 16m 23s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00df74c189314b2697e98a49205b9713 : SUCCESS in 14m 51s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/166d018eb43a4af392a7e47761037235 : SUCCESS in 16m 08s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/55d585c953b8425593e8c6e6d9acb0ba : SUCCESS in 40m 29s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4731aa3aa3dd470d890974a141418ce4 : SUCCESS in 26m 29s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/3c2f2e93e322410c9b522dd22eb92dfd : SUCCESS in 18m 39s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/1fd9346dea8147238412a3365af1cfe1 : SUCCESS in 17m 38s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7daa6e5ff57b426fb8dadf220d5e99cc : FAILURE in 11m 25s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ceac39bb973645539856d3b86dae511d : SUCCESS in 14m 56s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b6531b49a18744f38d13da161ad69ba3 : SUCCESS in 15m 13s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/07fbe16102ce4443813055eb76427e54 : SUCCESS in 19m 45s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/93d1ea0e092e4ab2b3b90d1abb0e4683 : SUCCESS in 18m 53s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/20934edaa25f47329f5f88130302d63f : SUCCESS in 17m 44s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8586821a3bc94629bf411595d43eefa5 : SUCCESS in 13m 54s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1b1f719e7278407084afa218d5c7fe44 : SUCCESS in 13m 04s","accounts_in_message":[],"_revision_number":29},{"id":"969407be6b9deaf9785df2ac6c32b02a026d1ae9","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-02 19:40:53.000000000","message":"Patch Set 29:\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"e3cb232a5acdb4e993459980450316d3865db4a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-02 21:49:03.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b60d2600697a434eb672e5f24793ce65\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2145eda47d7341f7afc24a42e1c3584b : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e09a2eea879463d930cd83f5ae7c922 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de79efa3fb9b468db968a12f8d300b75 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ce673f357594c87ade7635bb0ab4a4e : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe0ff44c316e4da6a1596a2418c48f21 : SUCCESS in 8m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/024ffdf7fcdc48099998167bb960864f : SUCCESS in 3m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ca32edb336da472d94314511bbe9e427 : SUCCESS in 6m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/97999cd047724487bfd8c0c49952e174 : SUCCESS in 14m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0e19bbe2a6e343199a60cbf27e1bf987 : SUCCESS in 17m 46s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8b25a23d7e484c9a82cee04b4ae74abb : SUCCESS in 57m 42s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/99e9832487094e349d7950467ae1c30a : SUCCESS in 1h 59m 43s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/a6523d0f98aa45f7982779337c61f148 : SUCCESS in 1h 08m 59s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/4d3e889483ad4f8b8014c38c123e76ff : SUCCESS in 1h 09m 45s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7add40b46644f8592ea8a1bf367a3f1 : SUCCESS in 1h 01m 21s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/21033b5b61cb4e97b86c3bd3ffe7b8c2 : SUCCESS in 51m 17s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/acecf91654b3443d8d70d316fe00f21f : SUCCESS in 24m 56s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/2209b9842b794ab79d438e7722df8b00 : SUCCESS in 21m 16s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/a52b42368a7349aebe7681cda8b635ad : SUCCESS in 20m 09s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/7f70d6eaf19a4c8fbc5c55d4ecc774c5 : SUCCESS in 20m 14s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/376bdd878bfe401699b83b0b132b8eba : SUCCESS in 16m 46s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/313ccc35cc814a5890a5982a0acc1819 : SUCCESS in 16m 26s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/610733b61808472d8b5cedbc35412079 : SUCCESS in 59m 17s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/36996d7ed4924430890df87d2af4c1a7 : SUCCESS in 1h 05m 27s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d5e0408b00574ac1b1cf8f148b0d0830 : SUCCESS in 1h 53m 26s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89c23a9f7e1c43359ed8d9de9667ffff : SUCCESS in 1h 18m 11s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/6ffcce29d92c43adb9f0d6f3a1d8143f : SUCCESS in 23m 13s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/8fba5f810eea4c9fa8683dd14b9410b0 : SUCCESS in 18m 52s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/d2249c584e0c46bf8430b46150d9ca5a : SUCCESS in 21m 12s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1eda2478107c4d5a8ea4b70287727607 : SUCCESS in 18m 31s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d8defda1dc74ea996fd5bff80ce5991 : SUCCESS in 16m 17s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/92f7e900c7dd4add8888d160c53ad794 : SUCCESS in 42m 11s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c5e6abb9ec04407b7a944fcdb21fba8 : SUCCESS in 31m 27s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/a6517e0ad8c94e6aabb9f882676e4a03 : SUCCESS in 20m 03s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/9a9f1cc6c2004a28b5b10f44a2683f18 : SUCCESS in 17m 24s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/937f633310ca4f6cac4f84f5b2c802b1 : SUCCESS in 18m 27s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/941dc7b90864466fa2e6f27d9248cefa : SUCCESS in 15m 56s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b34d13cfbbef4089884d1c1eda4642b8 : SUCCESS in 14m 21s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/dbf1c46802e24c5fa5d6a1444a8721b2 : SUCCESS in 23m 34s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/6d9f61317b0c42f48b3630de8ed4625e : SUCCESS in 19m 32s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e5f06a5e7b3e4b2e973cab8d7ab19c79 : SUCCESS in 20m 21s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7724daf496b43e0acebc819c29276c2 : SUCCESS in 14m 15s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/86696ab5f786463f91917cf17f600395 : SUCCESS in 15m 31s","accounts_in_message":[],"_revision_number":29},{"id":"89afa39f6ab38ebcb3be128413a9eb648baf202b","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-02 22:00:55.000000000","message":"Patch Set 29:\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"ee78c8fd9be93f7d065c4f85e88551053517cfc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 00:15:21.000000000","message":"Patch Set 29:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e250282eb63044a4b78295c0c518cc39\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d93e6fa1039a4a788a55807e09bbc229 : SUCCESS in 7m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77c1c1bbfd254040856eb5d77f09d689 : SUCCESS in 11m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f2022ae0cbe84a4b8679cda8e509c1a1 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3cc584977a56420780c6b58cabea4772 : SUCCESS in 12m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea5530aecebe4c73a3f31c4a3952ea74 : SUCCESS in 9m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24e46b0779a144afb6cb94a25b091c49 : SUCCESS in 3m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6fd3568227db486ab30fa32ff71a4a28 : SUCCESS in 6m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/68f07c97904940b3bae644c92ac268b1 : SUCCESS in 12m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f518a9e607de4c19b1a6fdcac2300661 : SUCCESS in 25m 43s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/9aa2db0d4ee9475ea06a60a47677483e : SUCCESS in 59m 39s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/1106b57aea8944f295ad1928cdd3f17b : SUCCESS in 1h 06m 35s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/682f8e82ef7a418a925d56810976705c : SUCCESS in 57m 35s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/bbf39fb378f34b63af0abcd4317c25c2 : SUCCESS in 1h 37m 56s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0ce03b4f6e947a6b0a76828e5ac5f84 : SUCCESS in 46m 10s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/44c41678ea434684ab95a66df3a5125f : SUCCESS in 59m 05s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/04c413799bbe4cc089131df3e7c99a6e : SUCCESS in 24m 25s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/b00ea5c3965943d4834186fa0da37ff6 : SUCCESS in 51m 42s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/a3d21043a3fc471ca65d85cd01087e6c : SUCCESS in 17m 30s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/853a41ff82264f8eaecc5fe67a6546d4 : SUCCESS in 48m 56s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/efe19e0f21f1406894cca32f5cc6ab37 : SUCCESS in 12m 59s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad86c222f38e48858ac3a63d9d65beab : SUCCESS in 18m 25s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/69647e98aae1475d9c85e6b90847f824 : SUCCESS in 1h 07m 14s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/2a93c55fa8f3407cb263501f8e8b2bfc : SUCCESS in 48m 43s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1da174ab8b3a46f684171bbb77afa065 : SUCCESS in 2h 07m 41s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f8e283a78ee491bb1df4e64d64da782 : SUCCESS in 53m 40s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/fa24954223394e039ef135eaccf80c8b : SUCCESS in 25m 46s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/e7749fbcc5e443bfbb193228432c5624 : SUCCESS in 17m 23s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e353f6444736402787acd8ef50710b8f : SUCCESS in 19m 25s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ab10fae08df4712b129ad696b818f7e : SUCCESS in 16m 44s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a49766e070b477285c437782ed6a091 : SUCCESS in 15m 39s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/73144fc0a6f44df79286e84b25c746ab : SUCCESS in 44m 16s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0311ba67e294ae0b84c82a8fcb6c52b : SUCCESS in 32m 51s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/18455928ea284d2b82669e07ed23ea1d : SUCCESS in 14m 07s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/3f91ea9880e444e38e3a5f6577e71cd7 : SUCCESS in 18m 18s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5f6bda0331be49259264a3c7d6e4fd31 : SUCCESS in 17m 33s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f92da64a0705492890a1cdd485658f24 : SUCCESS in 14m 44s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f064e6581704117a4d737e23d9133de : SUCCESS in 12m 42s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/d3dc5f6b8d0148098648aa1a03975f26 : SUCCESS in 18m 41s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/ccdc74e4220a42cdb84f945237070fdd : SUCCESS in 16m 07s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/491c594123c6430ab48274ef6a1b5765 : SUCCESS in 19m 53s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/068f5a8742db473b81b9fc01faa97363 : SUCCESS in 12m 47s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c031dc29e2214679b6f7687738b5bed6 : SUCCESS in 12m 06s","accounts_in_message":[],"_revision_number":29},{"id":"8c6a2c0c66198d693b0c5fada5d66cbc5581d84d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-03 08:29:07.000000000","message":"Patch Set 29: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":29},{"id":"5b01d6ee970e9ebf18a5eb55da9cfe4dc7bb3e95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 08:29:43.000000000","message":"Patch Set 29: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":29},{"id":"91d4cf07e08e712701937cd4dc45e22971d2aef1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 10:24:48.000000000","message":"Patch Set 29: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2b85258c918b45a894855a8097f5378f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b2ca8f7d2a54027af993c4acbf9ce65 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4c5af0f3cba4e228508d9b5be6a4823 : SUCCESS in 6m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c2ac2a2e3ba4b008138805a97149f2d : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b9a22c507ab431d839a53a0dce01c20 : SUCCESS in 11m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/32c23ed8227f407c85c710e843631916 : SUCCESS in 3m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/780e877621be4b6db15912fe444b4c02 : SUCCESS in 12m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e2f3253352374181954a72a0ab3b5f78 : TIMED_OUT in 30m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a3426aa859b14b1bb912b58b81c0b2f8 : SUCCESS in 18m 53s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7845ec35ae53498d9687055338a72264 : SUCCESS in 1h 06m 49s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/41cf98779d15443ba5b6c9d4dcb39033 : SUCCESS in 1h 06m 32s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/89654195e7a34a9bb18bf7147a6ddf89 : SUCCESS in 1h 06m 07s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2647b2546df7451f81154d7ece62e57d : SUCCESS in 53m 51s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e91d02634c54728a0d762ee39243e63 : SUCCESS in 51m 21s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/092e3b3884cd48ab9e9c36e94e3fe999 : SUCCESS in 24m 15s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/b83a3170a01945e187ceca82ff32960a : SUCCESS in 21m 33s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/db00b9c50f5a4b3aa7e560ec7d28cacb : SUCCESS in 21m 15s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f4b8b403440942eab2f1fbf6499b228a : SUCCESS in 18m 29s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/81bb5a636ed34fe183c233f4f1b785c6 : SUCCESS in 18m 00s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/7a7e01bd769e4c76aa6791239a2e297b : SUCCESS in 1h 10m 41s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a326fff7016049a7a01e7276080dadd0 : SUCCESS in 1h 09m 51s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c8ded1bd4540405c8d649facc7551daf : SUCCESS in 1h 50m 41s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5508dd6dda954e1d85726507960da1e9 : SUCCESS in 1h 13m 54s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/bd1de8b579ce46afae34124654d6261e : SUCCESS in 20m 52s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/22d327f5bc7644b5b176dab21382d84a : SUCCESS in 19m 56s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/06b4b414dd5047e3996af6b1aa7d324f : SUCCESS in 20m 57s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/217f825ca0fd4b8cb748b7e3af4794be : SUCCESS in 16m 24s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e722ce8115ca4db8bb63ad115cd4de94 : SUCCESS in 18m 25s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3043aa3f48f94b66813131aea1b7514c : SUCCESS in 37m 57s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/28367e09260c40d5b4a7f050d32025c0 : SUCCESS in 30m 37s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f65942fede194c289537e68dd00c899e : SUCCESS in 17m 43s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/d58e82a7d2b34ba5b1672e56ece12e5c : SUCCESS in 15m 11s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8a765d257b244a97bf87e33bd863f842 : SUCCESS in 15m 15s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47d2993bd213407598ce51834bd348b7 : SUCCESS in 12m 36s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c7edf13e414c46e6a91b068bced14f40 : SUCCESS in 12m 05s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/3f44a7251a214ecebe7001b465ca826c : SUCCESS in 23m 38s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/a2f3583fe28444bf842ae8c677826a40 : SUCCESS in 17m 35s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6856705206684811812f9bb050c3d06a : SUCCESS in 15m 43s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33a51dcc44ff416681a2e3148692b380 : SUCCESS in 12m 25s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c74b4e9c823a4b8bae8f102a4a6fe61e : SUCCESS in 15m 41s","accounts_in_message":[],"_revision_number":29},{"id":"7e1e3a4d693b83802ddb2723c5c1964720ce9b3c","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-03 10:25:21.000000000","message":"Patch Set 29:\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"d78c02162fc902e1b038f6c0b54d9a69da283a18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 12:36:20.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88fd2d6953aa4dc7a698fb33119137a9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b19fa9b1f5ce4a4fa674a827878948b3 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9796a2ce787e4398ba5148348060ba99 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ab784edcbc84b34817b7f0d0ac5fbe7 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53e95906bafd40b8bcf2b321beea36a2 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/278e62971690498886cfbc497343b717 : SUCCESS in 8m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3451936636914c85a311a69aa0bcc450 : SUCCESS in 3m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b7865b545a0c4130a38159b821abaebf : SUCCESS in 5m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9ef28cf787d54f47a0a3b584ff7829df : SUCCESS in 12m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b5b605286ae04664ba788d8e3fc2c9ef : SUCCESS in 15m 15s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/129fbebd45b94c22b728864cb9616e4a : SUCCESS in 1h 08m 52s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/2fde074f07f24780ba41aa6a3330b660 : SUCCESS in 57m 18s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/3d922a7b8d72452cb8c1f2d14b641743 : SUCCESS in 57m 51s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/33c000ab4a534e33b055b1bc9ea301ae : SUCCESS in 1h 08m 30s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae1cf2e75c3c4320991b0bc66c681dc2 : SUCCESS in 59m 24s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9fc83927baf444b2b405cff64e399449 : SUCCESS in 49m 05s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ee8a46800d1e46cbb3cf8c313ca28ccb : SUCCESS in 24m 38s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/c1ddb2a5ceaf4a5797ed0b108f9ca63b : SUCCESS in 19m 16s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/bccd3195682d4bbc990fb3ea412243b6 : SUCCESS in 18m 28s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b3cb38ca5a5546a9a1c07d9ad064572d : SUCCESS in 20m 06s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5fee549997964a518a9cdd7ebf1a6e43 : SUCCESS in 14m 37s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/86a3601cc4fd42aaae393c2803a29256 : SUCCESS in 13m 53s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/cb94fb0b0e35414386d9ad32774610a7 : SUCCESS in 1h 13m 55s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/b09c6198e2324bb2ad50f28436e7ec13 : SUCCESS in 1h 08m 17s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/95847c717ef24e88aff1638b6507053d : SUCCESS in 2h 09m 37s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea2f722448c64bb1813a36312997a1ae : SUCCESS in 1h 06m 04s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/450d5de8d53942168786bc331da29212 : SUCCESS in 19m 20s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/f10df7fd42f24230b20464c80bd67fc0 : SUCCESS in 16m 48s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e4940ff9ca03469cb1b24052b8e9d0db : SUCCESS in 19m 43s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61d98a45960d43a6bd6af6dd3b358b65 : SUCCESS in 16m 40s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9187472da5954357bf3363723e7a1904 : SUCCESS in 14m 20s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/488681411f3c4897ad459bbe2bc86751 : SUCCESS in 41m 09s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/17610e8e0dc942edb72e3a98e998dd33 : SUCCESS in 26m 33s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/45afa36d00a04986ad0e34524d6c5443 : SUCCESS in 17m 39s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/7810ae1369384a739fd8dccac78a2838 : SUCCESS in 17m 53s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/49559a3a07d24fb3bb67e36f9214b10c : SUCCESS in 17m 02s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee4722ef152341d9ab47a9bcc3ef6117 : SUCCESS in 11m 28s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d639d86736a248f9919bba21748173e5 : SUCCESS in 13m 14s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/fd6ab9fa377f432b9a1719a4d5c4f709 : SUCCESS in 19m 16s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/e9a83a02067548ff9f0225b702f0ced9 : SUCCESS in 16m 52s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a40405b5100d4f268c282e2d70bba9bf : SUCCESS in 15m 31s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f1792d24d7e4575a5c18a9ddedf5573 : SUCCESS in 15m 47s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/478dcb9686144e7e8e949ff4a673e574 : SUCCESS in 14m 41s","accounts_in_message":[],"_revision_number":29},{"id":"9bc786db16c9f29007d91cd7d8316c53f974b82a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 12:36:52.000000000","message":"Patch Set 29: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":29},{"id":"a78b6bf8aeff3ebd3c521595d3bfb2530d67e01f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 13:48:11.000000000","message":"Patch Set 29: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1e57e8122dc547af98f5dc0636f6115a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9005e3f460d3407fba1c2bb92721b609 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c63548119304e10b801a9d3e2a73519 : SUCCESS in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/534fd4c545164aa18566730d73ecd621 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56f5be14bd184d65aabda96dcb6488f9 : SUCCESS in 9m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/458a47efa56a43089fc9665dd65e57c4 : SUCCESS in 4m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e4e82dcc7f824eef8fb081fb5ad8e316 : SUCCESS in 5m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7df7413814fd437eab452b5e33628176 : SUCCESS in 12m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/89ffe177e26544d5aab3597384dadd8f : SUCCESS in 17m 16s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5a1bef7a9dc64da694088ac170c51a96 : SUCCESS in 59m 56s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/cc9cd2aec20741c58eb172f2fb77280b : SUCCESS in 1h 02m 28s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/4ff3929effbb458588a591212a063f49 : SUCCESS in 54m 01s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09e5352038684942bfc52e5950757d08 : SUCCESS in 50m 25s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/266879560325430bae4fdc50f980a5c9 : SUCCESS in 43m 23s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d00ec5c3491f410da46edae30a0fa6a9 : SUCCESS in 22m 22s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/04ac866ee6c74ea696f233a18608a004 : SUCCESS in 19m 49s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6feb9d68d4f04b9ca3e043a01198fa93 : SUCCESS in 19m 12s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d8aa359230f443e81d58f20b50f6ec1 : SUCCESS in 20m 24s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4158e1088cb048dcbe9200ef91fd4103 : SUCCESS in 19m 12s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/af4a3cc08b6945d69a0fbc1c6fe44430 : SUCCESS in 1h 05m 57s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/abc5752236d9496da065bd85cb3acbad : SUCCESS in 1h 02m 24s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b820e4ae54a44c0f84a26280c9ec0091 : FAILURE in 1h 10m 04s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a52a3fa7a47a4d148e0f30d09c8019d5 : FAILURE in 1h 07m 12s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d6a927f2396147d09c51cb234ebac140 : SUCCESS in 24m 16s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/8fe35cea40874a1ca25b386731aee528 : SUCCESS in 19m 47s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/2ceb0a0fb86347b0a4bcef1ca658c2b0 : SUCCESS in 18m 45s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1631acd5799c436385c9344deacb9693 : SUCCESS in 16m 36s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d4a06b9be25e46ee8f20d437b03dd418 : SUCCESS in 17m 28s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ca1b6fd33daa42bea3ea848559eb0bfa : SUCCESS in 44m 08s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d2e46339febc4685a2ecd590c0827765 : SUCCESS in 27m 19s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/8fccbdee435c4627bf7fbadfb0d9df64 : SUCCESS in 16m 48s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/7a0b0053402f427294912f29293237ea : SUCCESS in 17m 59s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/03df7f0ef798422aaab5bcb63101fda8 : SUCCESS in 15m 00s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ecef47e0382e4340a5e171b4e64a7015 : SUCCESS in 14m 29s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/600efff296834114886d1689e8094cfc : SUCCESS in 11m 20s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/b10caf690a634effb8f7d84e87f75ed0 : SUCCESS in 18m 40s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/637ca3a73f4a41bca20590fb37eb3b7a : SUCCESS in 14m 46s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/435d848fef5d48679e879f347d74db60 : SUCCESS in 15m 02s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f78b2a095fa14951b04d13633022cf12 : SUCCESS in 12m 29s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7f8437fc055f408989a3c82942a6c007 : SUCCESS in 12m 53s","accounts_in_message":[],"_revision_number":29},{"id":"6eac0af77fbd2cf0f94bb4c422e964811a60776d","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-03 13:50:34.000000000","message":"Patch Set 29:\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"bd875e770c9e754431ad2357f6c1ca43aeb7d434","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 16:12:07.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a867bfc5b7a1400093bbab21f5039227\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3a9247673734f42a93e0558213d3be3 : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61af1607a7324b51abb78c5f606a8e5f : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/caebe5b986104eee8a8a40b86df8a2b3 : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f538a9f4e994dc7a1863044859bd3ee : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d729616f5184a4580107725d3fa6d97 : SUCCESS in 12m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b27514ae9b8449d1908f002083e0b1de : SUCCESS in 3m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c26ec9a9255a416ab85d4711cfe8287f : SUCCESS in 6m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c27eba29174c4f34973ae94a3534ec5c : SUCCESS in 19m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d09ce054a5ef46e78b4be6313fb188ae : SUCCESS in 18m 37s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/41c98a5881a6497baa7e9f2febeb69c5 : SUCCESS in 1h 05m 10s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/ea70edb2ee4645709340bae03176978d : SUCCESS in 1h 04m 10s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/c462a3824b824d1188c3616f8ad90041 : SUCCESS in 1h 07m 21s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/055d7e9ba05a4a9b83bb431a95615936 : SUCCESS in 56m 54s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e656e56a41f943abb25f00f86228ccce : SUCCESS in 54m 48s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b61e95d1a8b241d5bd220899a07a5e1d : SUCCESS in 1h 44m 06s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/01f700fb2a3d49f08dcae7ddc4cabe1c : SUCCESS in 25m 00s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/bce4b7bf664145a4b8e1ff6e957c0a31 : SUCCESS in 22m 55s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/d2cbd37fd60743a58040abfac68ea3ef : SUCCESS in 22m 11s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a5b92ce8e595461a9321af0a98474c20 : SUCCESS in 19m 11s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/209ce472265245ffbcc55b136a610794 : SUCCESS in 16m 42s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3aeee06c120d4aebb896c03438325d94 : SUCCESS in 19m 09s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/723219e7e821424c87ece51b41d8e1f3 : SUCCESS in 1h 14m 15s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/fb47165e8e7743f0813d4d3f50fdadd7 : SUCCESS in 1h 07m 04s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ca27f30d76ea4550a6bbd2ff6a1b7e84 : SUCCESS in 1h 59m 07s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c501bab11ba34edcab247b2e152dfd94 : SUCCESS in 1h 12m 51s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/f96766f22cc24c08a5cdd89da5b1111a : SUCCESS in 22m 28s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/b53d85761c0345cfb6d16e46ddee5953 : SUCCESS in 21m 36s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/9978e62c6c9c4f7ab58430fe88a2391a : SUCCESS in 19m 35s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/19a3b423e9a14b238f9f3f36d8a4159a : SUCCESS in 17m 51s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/25386343ab1945689a3cdb3b6c874155 : SUCCESS in 16m 37s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cca247f6f2d44081b9fa5846ab556828 : SUCCESS in 45m 13s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6d1bc2d03a045c28032951ccdf21ddf : FAILURE in 4m 33s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/bdfe3c2263da4d659f44b522402f3428 : SUCCESS in 18m 31s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/6897ef2ea4c64bb190c77303a376e1c0 : SUCCESS in 20m 50s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/449567dae7ef4cd29d6dbd23d0910fe6 : SUCCESS in 15m 29s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d126cda74bc84ea285d37bd66219ff2f : SUCCESS in 14m 36s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dddbde6f5e1a434f9db20805c99a2bf7 : SUCCESS in 13m 54s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/fc2303ff76314a3484ec4866b64e361e : SUCCESS in 21m 41s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/41cf2fcb930f4147adc9d01ff7930789 : SUCCESS in 24m 20s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/dd3ef294727549d0b7d1467bc35145d5 : SUCCESS in 17m 33s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa97f83fa448418f927afd1a4ff621a4 : SUCCESS in 13m 57s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/734d4d3ed87340d29ae3c2b60ebb4065 : SUCCESS in 19m 07s","accounts_in_message":[],"_revision_number":29},{"id":"5c103ef595166ef25e27bf46f95ce7ff2a84c971","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-03 16:15:45.000000000","message":"Patch Set 29:\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"04a135b77fca498dac562478498c882809656b74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 18:31:39.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c72083324b7741dca4bdd84d0c08e5d5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a2f8365e08d49018a16d2d47eb6a506 : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1857cd722b3a448ebc420f216bbb387f : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8105e8eb7c0429eaf1ebc9ac8d07c8b : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f07e04702ed438e9884a4efe300a9b2 : SUCCESS in 6m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1993e267a414724b5089228f299b2a1 : SUCCESS in 8m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce7c949a0ae5426f89216ecb910074ca : SUCCESS in 4m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fe41d4ecb17e48ba8121c38536642ce7 : SUCCESS in 5m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b9e00897a9014999bd10eb2524785c61 : SUCCESS in 13m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f8a4b983f38d439ba995945de108c6e1 : SUCCESS in 17m 59s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/cc335751c64e426984a02a460a3522b2 : SUCCESS in 1h 04m 49s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/811055c6d3b343068804126934034a11 : SUCCESS in 1h 50m 59s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/68ea586700444d34889fe97a81ed4533 : SUCCESS in 1h 07m 44s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/18b62059ddf945d88996a1d21523014e : SUCCESS in 1h 01m 54s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/adc6c41cf9b14307bc539070d70de0e4 : SUCCESS in 53m 09s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9e3abae77cda4fe9b7930efd5e662382 : SUCCESS in 51m 14s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/92a0f62626e0464893b56b7a85bc6746 : SUCCESS in 23m 12s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/1fb2a7bb8076491ebe4808191215f956 : SUCCESS in 24m 25s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/af891d6a2fc54eadb400f7e55288e24f : SUCCESS in 22m 29s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0c062f2713ce40219531e29708ef1bec : SUCCESS in 20m 27s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/737c5088c2024af48b17751e594c70aa : SUCCESS in 16m 45s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2476a45c18e34da7a79482e2819ac481 : SUCCESS in 15m 40s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6195f1dcb0a742958dc7838c701f2be7 : SUCCESS in 1h 09m 57s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/16d181dcd0e449738c9ef315e7951513 : SUCCESS in 1h 17m 29s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c00682a619cd4f14bd8910479d291dff : SUCCESS in 1h 58m 00s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a7005a14383f42cf9277b885310874a8 : SUCCESS in 1h 11m 00s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/28be7c64bb164a0b8a6aa5781973cb22 : SUCCESS in 34m 41s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/0866bd1822fc4c5f871a3e440ddc7031 : SUCCESS in 19m 28s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/6ddb8cfea6144fc1a15d4953e199f12f : SUCCESS in 20m 09s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b47156475a94b63a8c2917316101659 : SUCCESS in 19m 30s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/32afe9f1b1f9425dad6622e1249d9e28 : SUCCESS in 18m 34s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4d9b46a251a64bf5831c9a93fe23735c : SUCCESS in 42m 02s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a629fac5d74547709a89af4e638cf774 : SUCCESS in 33m 55s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/aefb952f8a054506a70f54ca0737d2da : SUCCESS in 29m 22s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/6fbb7098b48c4eaaa017568b0053fa03 : SUCCESS in 15m 40s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ebbe88fa8011417284100610533a529e : SUCCESS in 15m 38s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac90008610714b93a9dbf40f31953617 : SUCCESS in 12m 16s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/636e8b16ab204c22b3e9f7b7f969193e : SUCCESS in 14m 23s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/c85b74479c7b42f38b84b1b7036fb9a4 : SUCCESS in 19m 14s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/f8d76fe9eaf14c4aa5938120f14b1c06 : SUCCESS in 18m 56s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c26a4d639fa34263ac154793af1d1fee : SUCCESS in 17m 23s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8ce8ccfb9a14b74aae169955d587798 : SUCCESS in 16m 02s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b3f6711d8e444579bcfb794c8830e85a : SUCCESS in 15m 11s","accounts_in_message":[],"_revision_number":29},{"id":"f2977721d95f0f41189af1dbc8e510baad900249","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 18:32:10.000000000","message":"Patch Set 29: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":29},{"id":"c5edd7b855a46759269be3fd9a48837be926d0fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-03 20:54:18.000000000","message":"Patch Set 29: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/74a8de8177e1402bb9ad0e778b6ab60e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41060ff0d1dd4b36889a56af40ca5ebd : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92fce58b7c2240e491c8e36c547b8880 : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63964ef47d214e5392c50634271d1eff : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c77da894e35248edbc92ae674582505b : SUCCESS in 13m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4463c040b0a544cd80b069b84e7bf291 : SUCCESS in 3m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2d2ba8f844c34e70814a59442cf3feb3 : SUCCESS in 7m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b8d7c9d6f3dc4d95bca122d208cc9f5d : SUCCESS in 14m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bc288d5c80cb438eb89e0b5977642f6e : SUCCESS in 19m 50s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f80b030837b148918b42759978eeb012 : SUCCESS in 1h 05m 48s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/e8f2b15f133f4d4c81f1727cf1dc9686 : SUCCESS in 1h 15m 45s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a85235bcce0444a3b70a84a4da51a2e9 : TIMED_OUT in 2h 09m 31s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d87b5407e10a4318bd1eeb7b9348321c : SUCCESS in 1h 00m 49s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3efdb046de594cd3a0cb032a0fcf3d36 : SUCCESS in 1h 01m 52s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0a7b99c2b5104480abbdfb5fed09a468 : SUCCESS in 24m 57s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/8ae3aff127684f68a6116dfdec032dc8 : SUCCESS in 20m 50s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/aac60d7342754e8f884f39c4cde6b6bc : SUCCESS in 20m 08s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/935d944f07a344c9afbdb3ea39548ac5 : SUCCESS in 15m 35s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a9a7b4eb357848d7b6a03c020f957666 : SUCCESS in 17m 43s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5ad8aa967f1c45a395166ffe3ce63591 : SUCCESS in 1h 03m 34s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a2e0440789034e97a17e3542ac50906c : SUCCESS in 1h 04m 26s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b94aef7094514e6dbf535ee908f3d990 : SUCCESS in 2h 21m 33s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/26762dfdcadf4ee58b1559783c381843 : SUCCESS in 1h 11m 51s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3a622a34ea0e4e409565d7f422667915 : SUCCESS in 20m 46s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/c1a6a0925fa042d383b57a605a1b2ce8 : SUCCESS in 19m 49s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/7d7f04c999cc490da08162bdf3b3a391 : SUCCESS in 18m 32s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/486a6b8a90ca402cbf974ede69a8ddbf : SUCCESS in 18m 11s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3df997d69f194b65ba5a1f0d525793cd : SUCCESS in 20m 01s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5171fd30cee74260a0284ab66514d7d2 : SUCCESS in 39m 54s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51da6de9a85d48d389547025d3fa9c72 : SUCCESS in 29m 32s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/612676365fb640489f8515684cd0f98f : SUCCESS in 16m 18s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/7e1f5f9489e9438dacdd46839be6ea1b : SUCCESS in 16m 11s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/52b00e28f07d45e6b67d4800e1c88f0d : SUCCESS in 16m 52s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b884b5d2a4043ce854df8e2d41eeebc : SUCCESS in 15m 07s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5f3b48f708a74f0c93679cbe0e3a4321 : SUCCESS in 13m 18s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/bf4ce9a7f1b345fbb6638c83ccf8bd4a : SUCCESS in 20m 51s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/5a93843c3c0a45a58143ee9d44752693 : SUCCESS in 17m 00s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a9765022af414903b154c04d6b44dcfe : SUCCESS in 17m 14s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b7d6b66160c4862ad26daecd822a245 : SUCCESS in 14m 50s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1e0cfcaf93f941a1a4b2d7c4763e075e : SUCCESS in 13m 40s","accounts_in_message":[],"_revision_number":29},{"id":"ef2d9a58993329676fdf7722f3deff95657f4ac5","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-03 22:50:33.000000000","message":"Patch Set 29:\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"83d78a72cb0bc19e28ab0a7a6469003580f2f3ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-04 00:47:39.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56434b03a5584c0db61310774e7ee662\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51baa7d98e754c73a9a6bdc4a34e8196 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fec88ba348b443d8bee57b15f1638e3 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/869c1f0108714da591535e61cfdd0969 : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2844885990604c5d97ccb0874a00f6c8 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5badc4995984bd5b33018e7149ed4d9 : SUCCESS in 9m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d723bb71f447400d9b460af87f94c2a0 : SUCCESS in 2m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b57cfe348c954717a14c1740b0b66bc3 : SUCCESS in 7m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1328abcc9e8141d3ba10e3f8a8afe67a : SUCCESS in 14m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a3fee33c8c374aae81f88c806a233c80 : SUCCESS in 20m 56s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/654d78c689894287b8d442a11a77d3d0 : SUCCESS in 1h 03m 12s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/bd0c1300243743a9848cf5f92510bace : SUCCESS in 1h 05m 21s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/72c3bd7d78964357aad8e42e3dedfbb8 : SUCCESS in 1h 07m 13s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/4f3960c600824a05a0371032a1fafd0e : SUCCESS in 1h 06m 21s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e42d8d595c184adb888e11a62f113423 : SUCCESS in 51m 35s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0adcefb9003b4481aec2f71240701133 : SUCCESS in 40m 47s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/6688fd8f4f67465fb0d70f138a7d14b4 : SUCCESS in 23m 45s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/46721ba181374a0fa486cf98b31e7220 : SUCCESS in 23m 31s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/7954f6d879ab42529a1f13c849be6ac6 : SUCCESS in 21m 21s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/318d155e173049988769f0a4b13efbf4 : SUCCESS in 17m 03s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa61eeec29d74c6795b7c95d92e66471 : SUCCESS in 15m 26s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/636defbb02f641388238c356f79c95f5 : SUCCESS in 16m 51s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/2381ef45024747768b2b318e42b47bd8 : SUCCESS in 1h 10m 11s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/73f50e66cb50477eb59b04be6c67ffab : SUCCESS in 51m 25s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/77af6b0f0a414f529d80c2f622dd0148 : SUCCESS in 1h 51m 49s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87f4db439c064d91ae20595dd3395d09 : SUCCESS in 1h 19m 43s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/584d7ddb82704437a72ae9b9171a5d7c : SUCCESS in 24m 03s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/62dee81172994c7ca79dac818496ef65 : SUCCESS in 22m 50s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/d2966556883040e9985a9bbe4ff349c4 : SUCCESS in 22m 06s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31005625e6744fd9bdc30cc8499c2aba : SUCCESS in 18m 26s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f64b43e645c64c66baaf34c3bdf74843 : SUCCESS in 14m 31s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2979635b59f14b889961513f1d318459 : SUCCESS in 40m 24s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce1259ffc0a74cf78e4d122033c73418 : SUCCESS in 35m 06s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/536807f874dd415f85908aedd72fa1dd : SUCCESS in 17m 26s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/d23294e06f694068b820d12d21de738e : SUCCESS in 15m 40s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/30ea6d7a29f943178904783f630185df : SUCCESS in 15m 08s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/19bc65f2b4724714980d6bae4434b134 : SUCCESS in 20m 38s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1e8c3b7139374f56ab06bb073c07db4d : SUCCESS in 12m 02s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/ca3fb841093443569dc9a8c2c9f2afcd : SUCCESS in 22m 10s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/d3d9f94216514425ab1c1ff255bc8ae1 : SUCCESS in 19m 26s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6f9a8b8dc36046049fa821924f14b6b6 : SUCCESS in 16m 39s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/097b5e93f0054542978e93c92351397f : SUCCESS in 17m 02s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c7e2ded96bb14c7ab0491160d6c05cd5 : SUCCESS in 15m 02s","accounts_in_message":[],"_revision_number":29},{"id":"bc878f55738e7449c4fee9890064e64148557c0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-04 00:48:10.000000000","message":"Patch Set 29: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":29},{"id":"09f915dd50354fe82528a92ea663db3b0d7eb48f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-04 02:47:26.000000000","message":"Patch Set 29: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f463f6a4b2342bea06ea50eee13960b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e27474cdb52343eaa7ae98b701f52f11 : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63c46e882e5e44a7b77ebc431e2e2273 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e18dbef48a7d41bd92801c7ad7425310 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efe33ef5a74b4327bc242fb3b72c7d71 : SUCCESS in 10m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e05f59a77db4b46907ebd467d674d23 : SUCCESS in 3m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3f816796650e425d92200a6ece9a53d9 : SUCCESS in 6m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/57a494185c1a4423909f4f10d888f1b8 : SUCCESS in 12m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8efa7dc3daea4632a8af9d3e75e356d8 : SUCCESS in 15m 16s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/79957f026ae24bcaaa5a828057b4e1ba : SUCCESS in 1h 13m 43s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/70354f77185d4f6bbd6d859e5ad6ca66 : SUCCESS in 1h 08m 20s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/c9be9f816c384a10842b34d2e4b60cc9 : SUCCESS in 1h 06m 09s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12b62a48cb3b44d2b127d802191cd52c : SUCCESS in 51m 59s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b0150ab9e284a07931f2cafa710e3ee : SUCCESS in 58m 12s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7c01be40ec4d41f79f4cd23f76d09be9 : SUCCESS in 25m 15s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/b5b41d740a2046889c020e25fc4775e8 : SUCCESS in 22m 06s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a25c0ad6616845bbafbf19ceeed44e25 : SUCCESS in 23m 05s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9814354ae0704c2ebfce1efc49c89641 : SUCCESS in 16m 01s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f7ccef46d1143bb9733932234b01007 : SUCCESS in 23m 11s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/a6bdd37941d942d78f403f67c0443fc7 : SUCCESS in 1h 08m 17s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/7d44163a94b0481e858533bf704eef16 : SUCCESS in 1h 14m 16s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2094f961331647998ce8e0c2a0a61acf : SUCCESS in 1h 54m 39s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e25052d08524b5ebff52e2bc5002b1e : SUCCESS in 57m 09s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/acfdeae01dfb4bfe8a671836988ea720 : SUCCESS in 21m 02s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/791fb9b74d6b47fb93cb8f34547c9155 : SUCCESS in 23m 58s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f1517cd47c724c8f9a3dcc06ef9a49b7 : SUCCESS in 18m 43s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c0ec0c7fac6043dfaaecc04dffc6cc1f : SUCCESS in 18m 51s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a728882b95904cef83e124385e036e4c : SUCCESS in 14m 23s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/499f7d6d5c084a33b6d0146caafd8c28 : SUCCESS in 46m 58s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e523ef59531440a59c93473c46272b0e : SUCCESS in 32m 36s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/655c99a2c1d04b8eb249acbfba40a8f4 : SUCCESS in 19m 00s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/d963e33b71664182a7eb4e3ab04f7663 : SUCCESS in 16m 38s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c0c4f53fa25e4a95862f4b71462c4ecb : SUCCESS in 20m 39s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44184818bf98442dbaf541afad5bd9e6 : SUCCESS in 12m 50s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd3262d660b74c0f84bdaf03ce9c78cb : SUCCESS in 13m 26s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/f8d41e50f36c4ae68b49a043f034b1a7 : SUCCESS in 21m 29s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/c8f3f6b4b8ec4dd9acb8cd2710b0359b : SUCCESS in 18m 50s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a83bab0944384a41aed5220c1c3ea5dc : SUCCESS in 16m 56s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/115c39bf75694796b78edecd76904857 : SUCCESS in 11m 12s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df53e580c0644ca8b140fe08693dce13 : SUCCESS in 13m 53s","accounts_in_message":[],"_revision_number":29},{"id":"c9e9d6d6ba4d1002b04174ba4c5c14ae02cbe2e6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-04 02:47:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":29},{"id":"073591ecf91e8731acce3bf9fe5fe001685d96a5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-04 02:50:20.000000000","message":"Patch Set 29:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3da1d6366f048de9065765451bcff67\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c94beed889ad47e087876bd0d118a68a : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b0c7ceaa38a4bd692bcdebd993d4c38 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":29}],"current_revision_number":29,"current_revision":"a292e4bf2d01fcc2cca0a4c6ef6f125461aaeac6","revisions":{"8d3a0a945c357c082fa630abc25dc583d8058666":{"kind":"REWORK","_number":1,"created":"2023-02-22 08:29:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/38/874738/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/874738/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/874738/1"}}},"commit":{"parents":[{"commit":"834cf25febd3d990d5c4466d53ec1ac377bc0379","subject":"CI: Don\u0027t run Heat and Horizon in overcloud jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/834cf25febd3d990d5c4466d53ec1ac377bc0379"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2022-09-02 12:40:33.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-22 08:29:32.000000000","tz":0},"subject":"Move to Rocky Linux 9","message":"Move to Rocky Linux 9\n\nCentOS Stream 8 support has been dropped. Migration path will be present\nin Yoga release - as a followup change.\n\nMichaelRigart.interfaces does not support custom routes for\nNetworkManager yet. It has been disabled in CI for Rocky Linux 9\ntemporarily.\n\nNon-voting CentOS Stream 9 CI overcloud job is using RL9 container\nimages (as kolla CI is not building CS9 images anymore).\n\nChange-Id: Idf5ee822b03ba40179803c981500a6bad37594bf\n(cherry picked from commit 2f447f65453ff278fd53e5bf7975a5a092378b97)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8d3a0a945c357c082fa630abc25dc583d8058666"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8d3a0a945c357c082fa630abc25dc583d8058666"}]},"branch":"refs/heads/stable/yoga"},"5fbe377c54aa574b4edbab592077dc819a9113e9":{"kind":"REWORK","_number":2,"created":"2023-02-22 08:42:30.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/38/874738/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/874738/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/874738/2"}}},"commit":{"parents":[{"commit":"834cf25febd3d990d5c4466d53ec1ac377bc0379","subject":"CI: Don\u0027t run Heat and Horizon in overcloud jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/834cf25febd3d990d5c4466d53ec1ac377bc0379"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2022-09-02 12:40:33.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-22 08:42:27.000000000","tz":0},"subject":"Move to Rocky Linux 9","message":"Move to Rocky Linux 9\n\nCentOS Stream 8 support has been dropped. Migration path will be present\nin Yoga release - as a followup change.\n\nMichaelRigart.interfaces does not support custom routes for\nNetworkManager yet. It has been disabled in CI for Rocky Linux 9\ntemporarily.\n\nNon-voting CentOS Stream 9 CI overcloud job is using RL9 container\nimages (as kolla CI is not building CS9 images anymore).\n\nChange-Id: Idf5ee822b03ba40179803c981500a6bad37594bf\n(cherry picked from commit 2f447f65453ff278fd53e5bf7975a5a092378b97)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5fbe377c54aa574b4edbab592077dc819a9113e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5fbe377c54aa574b4edbab592077dc819a9113e9"}]},"branch":"refs/heads/stable/yoga"},"289ea54ebf03476f6f9ace612d02e34252811f92":{"kind":"REWORK","_number":3,"created":"2023-02-23 14:59:21.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/38/874738/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/874738/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/874738/3"}}},"commit":{"parents":[{"commit":"834cf25febd3d990d5c4466d53ec1ac377bc0379","subject":"CI: Don\u0027t run Heat and Horizon in overcloud jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/834cf25febd3d990d5c4466d53ec1ac377bc0379"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2022-09-02 12:40:33.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-23 14:59:20.000000000","tz":0},"subject":"Move to Rocky Linux 9","message":"Move to Rocky Linux 9\n\nCentOS Stream 8 support has been dropped. Migration path will be present\nin Yoga release - as a followup change.\n\nMichaelRigart.interfaces does not support custom routes for\nNetworkManager yet. It has been disabled in CI for Rocky Linux 9\ntemporarily.\n\nNon-voting CentOS Stream 9 CI overcloud job is using RL9 container\nimages (as kolla CI is not building CS9 images anymore).\n\nChange-Id: Idf5ee822b03ba40179803c981500a6bad37594bf\n(cherry picked from commit 2f447f65453ff278fd53e5bf7975a5a092378b97)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/289ea54ebf03476f6f9ace612d02e34252811f92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/289ea54ebf03476f6f9ace612d02e34252811f92"}]},"branch":"refs/heads/stable/yoga"},"4a0c2274a29e08e1a157cb32ffa054d0eb5f243b":{"kind":"REWORK","_number":4,"created":"2023-02-23 15:19:10.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/38/874738/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/874738/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/874738/4"}}},"commit":{"parents":[{"commit":"834cf25febd3d990d5c4466d53ec1ac377bc0379","subject":"CI: Don\u0027t run Heat and Horizon in overcloud jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/834cf25febd3d990d5c4466d53ec1ac377bc0379"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2022-09-02 12:40:33.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-23 15:19:07.000000000","tz":60},"subject":"Move to Rocky Linux 9","message":"Move to Rocky Linux 9\n\nCentOS Stream 8 support has been dropped. Migration path will be present\nin Yoga release - as a followup change.\n\nMichaelRigart.interfaces does not support custom routes for\nNetworkManager yet. It has been disabled in CI for Rocky Linux 9\ntemporarily.\n\nNon-voting CentOS Stream 9 CI overcloud job is using RL9 container\nimages (as kolla CI is not building CS9 images anymore).\n\nChange-Id: Idf5ee822b03ba40179803c981500a6bad37594bf\n(cherry picked from commit 2f447f65453ff278fd53e5bf7975a5a092378b97)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4a0c2274a29e08e1a157cb32ffa054d0eb5f243b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4a0c2274a29e08e1a157cb32ffa054d0eb5f243b"}]},"branch":"refs/heads/stable/yoga"},"9a39c9567102ffcb8d6c279350ecc3c0f1692413":{"kind":"REWORK","_number":5,"created":"2023-02-23 16:03:58.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/38/874738/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/874738/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/874738/5"}}},"commit":{"parents":[{"commit":"834cf25febd3d990d5c4466d53ec1ac377bc0379","subject":"CI: Don\u0027t run Heat and Horizon in overcloud jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/834cf25febd3d990d5c4466d53ec1ac377bc0379"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2022-09-02 12:40:33.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-02-23 16:03:52.000000000","tz":60},"subject":"Move to Rocky Linux 9","message":"Move to Rocky Linux 9\n\nCentOS Stream 8 support has been dropped. Migration path will be present\nin Yoga release - as a followup change.\n\nMichaelRigart.interfaces does not support custom routes for\nNetworkManager yet. It has been disabled in CI for Rocky Linux 9\ntemporarily.\n\nNon-voting CentOS Stream 9 CI overcloud job is using RL9 container\nimages (as kolla CI is not building CS9 images anymore).\n\nChange-Id: Idf5ee822b03ba40179803c981500a6bad37594bf\n(cherry picked from commit 2f447f65453ff278fd53e5bf7975a5a092378b97)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9a39c9567102ffcb8d6c279350ecc3c0f1692413"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9a39c9567102ffcb8d6c279350ecc3c0f1692413"}]},"branch":"refs/heads/stable/yoga"},"08da8a8fec57d66d0d6e789aeb3ee8bb5d72ae1c":{"kind":"REWORK","_number":6,"created":"2023-02-24 10:01:41.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/38/874738/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/874738/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/874738/6"}}},"commit":{"parents":[{"commit":"834cf25febd3d990d5c4466d53ec1ac377bc0379","subject":"CI: Don\u0027t run Heat and Horizon in overcloud jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/834cf25febd3d990d5c4466d53ec1ac377bc0379"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2022-09-02 12:40:33.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-02-24 10:01:40.000000000","tz":60},"subject":"Move to Rocky Linux 9","message":"Move to Rocky Linux 9\n\nCentOS Stream 8 support has been dropped. Migration path will be present\nin Yoga release - as a followup change.\n\nMichaelRigart.interfaces does not support custom routes for\nNetworkManager yet. It has been disabled in CI for Rocky Linux 9\ntemporarily.\n\nNon-voting CentOS Stream 9 CI overcloud job is using RL9 container\nimages (as kolla CI is not building CS9 images anymore).\n\nChange-Id: Idf5ee822b03ba40179803c981500a6bad37594bf\n(cherry picked from commit 2f447f65453ff278fd53e5bf7975a5a092378b97)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/08da8a8fec57d66d0d6e789aeb3ee8bb5d72ae1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/08da8a8fec57d66d0d6e789aeb3ee8bb5d72ae1c"}]},"branch":"refs/heads/stable/yoga"},"eaba4ffcc444aab690449ced3a9193e6ea9fb52e":{"kind":"REWORK","_number":7,"created":"2023-02-24 10:20:34.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/38/874738/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/874738/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/874738/7"}}},"commit":{"parents":[{"commit":"834cf25febd3d990d5c4466d53ec1ac377bc0379","subject":"CI: Don\u0027t run Heat and Horizon in overcloud jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/834cf25febd3d990d5c4466d53ec1ac377bc0379"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2022-09-02 12:40:33.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-02-24 10:20:29.000000000","tz":60},"subject":"Move to Rocky Linux 9","message":"Move to Rocky Linux 9\n\nCentOS Stream 8 support has been dropped. Migration path will be present\nin Yoga release - as a followup change.\n\nMichaelRigart.interfaces does not support custom routes for\nNetworkManager yet. It has been disabled in CI for Rocky Linux 9\ntemporarily.\n\nNon-voting CentOS Stream 9 CI overcloud job is using RL9 container\nimages (as kolla CI is not building CS9 images anymore).\n\nChange-Id: Idf5ee822b03ba40179803c981500a6bad37594bf\n(cherry picked from commit 2f447f65453ff278fd53e5bf7975a5a092378b97)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eaba4ffcc444aab690449ced3a9193e6ea9fb52e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eaba4ffcc444aab690449ced3a9193e6ea9fb52e"}]},"branch":"refs/heads/stable/yoga"},"78280b63befa4817603220318cd6fcf682331953":{"kind":"NO_CHANGE","_number":8,"created":"2023-02-24 10:21:08.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/38/874738/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/874738/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/874738/8"}}},"commit":{"parents":[{"commit":"834cf25febd3d990d5c4466d53ec1ac377bc0379","subject":"CI: Don\u0027t run Heat and Horizon in overcloud jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/834cf25febd3d990d5c4466d53ec1ac377bc0379"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2022-09-02 12:40:33.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-02-24 10:21:03.000000000","tz":60},"subject":"Move to Rocky Linux 9","message":"Move to Rocky Linux 9\n\nCentOS Stream 8 support has been dropped. Migration path will be present\nin Yoga release - as a followup change.\n\nMichaelRigart.interfaces does not support custom routes for\nNetworkManager yet. It has been disabled in CI for Rocky Linux 9\ntemporarily.\n\nNon-voting CentOS Stream 9 CI overcloud job is using RL9 container\nimages (as kolla CI is not building CS9 images anymore).\n\nChange-Id: Idf5ee822b03ba40179803c981500a6bad37594bf\n(cherry picked from commit 2f447f65453ff278fd53e5bf7975a5a092378b97)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/78280b63befa4817603220318cd6fcf682331953"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/78280b63befa4817603220318cd6fcf682331953"}]},"branch":"refs/heads/stable/yoga"},"04a34e1ef80005100e62779c31459aa5196191f1":{"kind":"REWORK","_number":9,"created":"2023-02-24 10:49:34.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/38/874738/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/874738/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/874738/9"}}},"commit":{"parents":[{"commit":"834cf25febd3d990d5c4466d53ec1ac377bc0379","subject":"CI: Don\u0027t run Heat and Horizon in overcloud jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/834cf25febd3d990d5c4466d53ec1ac377bc0379"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2022-09-02 12:40:33.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-02-24 10:49:27.000000000","tz":60},"subject":"Move to Rocky Linux 9","message":"Move to Rocky Linux 9\n\nCentOS Stream 8 support has been dropped. Migration path will be present\nin Yoga release - as a followup change.\n\nMichaelRigart.interfaces does not support custom routes for\nNetworkManager yet. It has been disabled in CI for Rocky Linux 9\ntemporarily.\n\nNon-voting CentOS Stream 9 CI overcloud job is using RL9 container\nimages (as kolla CI is not building CS9 images anymore).\n\nChange-Id: Idf5ee822b03ba40179803c981500a6bad37594bf\n(cherry picked from commit 2f447f65453ff278fd53e5bf7975a5a092378b97)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/04a34e1ef80005100e62779c31459aa5196191f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/04a34e1ef80005100e62779c31459aa5196191f1"}]},"branch":"refs/heads/stable/yoga"},"bddf762514e6f16592c9ca2d18b84cdcc4ee3bf6":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2023-02-24 11:25:07.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/38/874738/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/874738/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/874738/10"}}},"commit":{"parents":[{"commit":"834cf25febd3d990d5c4466d53ec1ac377bc0379","subject":"CI: Don\u0027t run Heat and Horizon in overcloud jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/834cf25febd3d990d5c4466d53ec1ac377bc0379"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2022-09-02 12:40:33.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-02-24 11:24:40.000000000","tz":60},"subject":"Move to Rocky Linux 9","message":"Move to Rocky Linux 9\n\nCentOS Stream 8 support has been dropped. Migration path will be present\nin Yoga release - as a followup change.\n\nMichaelRigart.interfaces does not support custom routes for\nNetworkManager yet. It has been disabled in CI for Rocky Linux 9\ntemporarily.\n\nNon-voting CentOS Stream 9 CI overcloud job is using RL9 container\nimages (as kolla CI is not building CS9 images anymore).\n\nDepends-on: https://review.opendev.org/c/openstack/kolla-ansible/+/874917\n\nChange-Id: Idf5ee822b03ba40179803c981500a6bad37594bf\n(cherry picked from commit 2f447f65453ff278fd53e5bf7975a5a092378b97)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bddf762514e6f16592c9ca2d18b84cdcc4ee3bf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bddf762514e6f16592c9ca2d18b84cdcc4ee3bf6"}]},"branch":"refs/heads/stable/yoga"},"f8c6aab7cdcffa7fb3ca1f9142acc0e892c9342e":{"kind":"REWORK","_number":11,"created":"2023-02-24 12:59:56.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/38/874738/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/874738/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/874738/11"}}},"commit":{"parents":[{"commit":"834cf25febd3d990d5c4466d53ec1ac377bc0379","subject":"CI: Don\u0027t run Heat and Horizon in overcloud jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/834cf25febd3d990d5c4466d53ec1ac377bc0379"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2022-09-02 12:40:33.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-02-24 12:59:52.000000000","tz":60},"subject":"Move to Rocky Linux 9","message":"Move to Rocky Linux 9\n\nCentOS Stream 8 support has been dropped. Migration path will be present\nin Yoga release - as a followup change.\n\nMichaelRigart.interfaces does not support custom routes for\nNetworkManager yet. It has been disabled in CI for Rocky Linux 9\ntemporarily.\n\nNon-voting CentOS Stream 9 CI overcloud job is using RL9 container\nimages (as kolla CI is not building CS9 images anymore).\n\nDepends-on: https://review.opendev.org/c/openstack/kolla-ansible/+/874917\n\nChange-Id: Idf5ee822b03ba40179803c981500a6bad37594bf\n(cherry picked from commit 2f447f65453ff278fd53e5bf7975a5a092378b97)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8c6aab7cdcffa7fb3ca1f9142acc0e892c9342e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8c6aab7cdcffa7fb3ca1f9142acc0e892c9342e"}]},"branch":"refs/heads/stable/yoga"},"0fe7e6124632fdae5c0e146d263c13d333c8cb2b":{"kind":"REWORK","_number":12,"created":"2023-02-24 13:29:52.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/38/874738/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/874738/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/874738/12"}}},"commit":{"parents":[{"commit":"834cf25febd3d990d5c4466d53ec1ac377bc0379","subject":"CI: Don\u0027t run Heat and Horizon in overcloud jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/834cf25febd3d990d5c4466d53ec1ac377bc0379"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2022-09-02 12:40:33.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-02-24 13:29:46.000000000","tz":60},"subject":"Move to Rocky Linux 9","message":"Move to Rocky Linux 9\n\nCentOS Stream 8 support has been dropped. Migration path will be present\nin Yoga release - as a followup change.\n\nMichaelRigart.interfaces does not support custom routes for\nNetworkManager yet. It has been disabled in CI for Rocky Linux 9\ntemporarily.\n\nNon-voting CentOS Stream 9 CI overcloud job is using RL9 container\nimages (as kolla CI is not building CS9 images anymore).\n\nDepends-on: https://review.opendev.org/c/openstack/kolla-ansible/+/874917\n\nChange-Id: Idf5ee822b03ba40179803c981500a6bad37594bf\n(cherry picked from commit 2f447f65453ff278fd53e5bf7975a5a092378b97)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0fe7e6124632fdae5c0e146d263c13d333c8cb2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0fe7e6124632fdae5c0e146d263c13d333c8cb2b"}]},"branch":"refs/heads/stable/yoga"},"7db5aae6303dfef573be8ce25ff9d664e92d4bbf":{"kind":"REWORK","_number":13,"created":"2023-02-24 14:06:23.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/38/874738/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/874738/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/874738/13"}}},"commit":{"parents":[{"commit":"834cf25febd3d990d5c4466d53ec1ac377bc0379","subject":"CI: Don\u0027t run Heat and Horizon in overcloud jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/834cf25febd3d990d5c4466d53ec1ac377bc0379"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2022-09-02 12:40:33.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-02-24 14:06:18.000000000","tz":60},"subject":"Move to Rocky Linux 9","message":"Move to Rocky Linux 9\n\nCentOS Stream 8 support has been dropped. Migration path will be present\nin Yoga release - as a followup change.\n\nMichaelRigart.interfaces does not support custom routes for\nNetworkManager yet. It has been disabled in CI for Rocky Linux 9\ntemporarily.\n\nNon-voting CentOS Stream 9 CI overcloud job is using RL9 container\nimages (as kolla CI is not building CS9 images anymore).\n\nDepends-on: https://review.opendev.org/c/openstack/kolla-ansible/+/874917\n\nChange-Id: Idf5ee822b03ba40179803c981500a6bad37594bf\n(cherry picked from commit 2f447f65453ff278fd53e5bf7975a5a092378b97)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7db5aae6303dfef573be8ce25ff9d664e92d4bbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7db5aae6303dfef573be8ce25ff9d664e92d4bbf"}]},"branch":"refs/heads/stable/yoga"},"7729b3a8c8ccdc7b3c77fabd49e0aa8df4e52abf":{"kind":"REWORK","_number":14,"created":"2023-02-24 14:32:21.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/38/874738/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/874738/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/874738/14"}}},"commit":{"parents":[{"commit":"834cf25febd3d990d5c4466d53ec1ac377bc0379","subject":"CI: Don\u0027t run Heat and Horizon in overcloud jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/834cf25febd3d990d5c4466d53ec1ac377bc0379"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2022-09-02 12:40:33.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-02-24 14:32:17.000000000","tz":60},"subject":"Move to Rocky Linux 9","message":"Move to Rocky Linux 9\n\nCentOS Stream 8 support has been dropped. Migration path will be present\nin Yoga release - as a followup change.\n\nMichaelRigart.interfaces does not support custom routes for\nNetworkManager yet. It has been disabled in CI for Rocky Linux 9\ntemporarily.\n\nNon-voting CentOS Stream 9 CI overcloud job is using RL9 container\nimages (as kolla CI is not building CS9 images anymore).\n\nDepends-on: https://review.opendev.org/c/openstack/kolla-ansible/+/874917\n\nChange-Id: Idf5ee822b03ba40179803c981500a6bad37594bf\n(cherry picked from commit 2f447f65453ff278fd53e5bf7975a5a092378b97)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7729b3a8c8ccdc7b3c77fabd49e0aa8df4e52abf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7729b3a8c8ccdc7b3c77fabd49e0aa8df4e52abf"}]},"branch":"refs/heads/stable/yoga"},"470f92c5fa3d33d599c00861c7ac7c19309a76c5":{"kind":"REWORK","_number":15,"created":"2023-02-27 11:13:07.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/38/874738/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/874738/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/874738/15"}}},"commit":{"parents":[{"commit":"834cf25febd3d990d5c4466d53ec1ac377bc0379","subject":"CI: Don\u0027t run Heat and Horizon in overcloud jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/834cf25febd3d990d5c4466d53ec1ac377bc0379"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2022-09-02 12:40:33.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-02-27 11:13:05.000000000","tz":0},"subject":"Move to Rocky Linux 9","message":"Move to Rocky Linux 9\n\nCentOS Stream 8 support has been dropped. Migration path will be present\nin Yoga release - as a followup change.\n\nMichaelRigart.interfaces does not support custom routes for\nNetworkManager yet. It has been disabled in CI for Rocky Linux 9\ntemporarily.\n\nNon-voting CentOS Stream 9 CI overcloud job is using RL9 container\nimages (as kolla CI is not building CS9 images anymore).\n\nDepends-on: https://review.opendev.org/c/openstack/kolla-ansible/+/874917\n\nChange-Id: Idf5ee822b03ba40179803c981500a6bad37594bf\n(cherry picked from commit 2f447f65453ff278fd53e5bf7975a5a092378b97)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/470f92c5fa3d33d599c00861c7ac7c19309a76c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/470f92c5fa3d33d599c00861c7ac7c19309a76c5"}]},"branch":"refs/heads/stable/yoga"},"8ee62c03d91ebb0e59c2575ed6413ddde8eadb34":{"kind":"REWORK","_number":16,"created":"2023-02-27 12:03:22.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/38/874738/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/874738/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/874738/16"}}},"commit":{"parents":[{"commit":"834cf25febd3d990d5c4466d53ec1ac377bc0379","subject":"CI: Don\u0027t run Heat and Horizon in overcloud jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/834cf25febd3d990d5c4466d53ec1ac377bc0379"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2022-09-02 12:40:33.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-02-27 12:03:18.000000000","tz":60},"subject":"Move to Rocky Linux 9","message":"Move to Rocky Linux 9\n\nCentOS Stream 8 support has been dropped. Migration path will be present\nin Yoga release - as a followup change.\n\nMichaelRigart.interfaces does not support custom routes for\nNetworkManager yet. It has been disabled in CI for Rocky Linux 9\ntemporarily.\n\nNon-voting CentOS Stream 9 CI overcloud job is using RL9 container\nimages (as kolla CI is not building CS9 images anymore).\n\nDepends-on: https://review.opendev.org/c/openstack/kolla-ansible/+/874917\n\nChange-Id: Idf5ee822b03ba40179803c981500a6bad37594bf\n(cherry picked from commit 2f447f65453ff278fd53e5bf7975a5a092378b97)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8ee62c03d91ebb0e59c2575ed6413ddde8eadb34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8ee62c03d91ebb0e59c2575ed6413ddde8eadb34"}]},"branch":"refs/heads/stable/yoga"},"b19271045e909cad055ae5bdb4041ce7de5e04f8":{"kind":"REWORK","_number":17,"created":"2023-02-27 13:37:37.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/38/874738/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/874738/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/874738/17"}}},"commit":{"parents":[{"commit":"834cf25febd3d990d5c4466d53ec1ac377bc0379","subject":"CI: Don\u0027t run Heat and Horizon in overcloud jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/834cf25febd3d990d5c4466d53ec1ac377bc0379"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2022-09-02 12:40:33.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-02-27 13:37:30.000000000","tz":60},"subject":"Move to Rocky Linux 9","message":"Move to Rocky Linux 9\n\nCentOS Stream 8 support has been dropped. Migration path will be present\nin Yoga release - as a followup change.\n\nMichaelRigart.interfaces does not support custom routes for\nNetworkManager yet. It has been disabled in CI for Rocky Linux 9\ntemporarily.\n\nNon-voting CentOS Stream 9 CI overcloud job is using RL9 container\nimages (as kolla CI is not building CS9 images anymore).\n\nDepends-on: https://review.opendev.org/c/openstack/kolla-ansible/+/874917\n\nChange-Id: Idf5ee822b03ba40179803c981500a6bad37594bf\n(cherry picked from commit 2f447f65453ff278fd53e5bf7975a5a092378b97)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b19271045e909cad055ae5bdb4041ce7de5e04f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b19271045e909cad055ae5bdb4041ce7de5e04f8"}]},"branch":"refs/heads/stable/yoga"},"d201d8e293e25d93ecc1893898c78e5079b5e33d":{"kind":"REWORK","_number":18,"created":"2023-02-27 13:56:04.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/38/874738/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/874738/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/874738/18"}}},"commit":{"parents":[{"commit":"834cf25febd3d990d5c4466d53ec1ac377bc0379","subject":"CI: Don\u0027t run Heat and Horizon in overcloud jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/834cf25febd3d990d5c4466d53ec1ac377bc0379"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2022-09-02 12:40:33.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-02-27 13:56:00.000000000","tz":60},"subject":"Move to Rocky Linux 9","message":"Move to Rocky Linux 9\n\nCentOS Stream 8 support has been dropped. Migration path will be present\nin Yoga release - as a followup change.\n\nMichaelRigart.interfaces does not support custom routes for\nNetworkManager yet. It has been disabled in CI for Rocky Linux 9\ntemporarily.\n\nNon-voting CentOS Stream 9 CI overcloud job is using RL9 container\nimages (as kolla CI is not building CS9 images anymore).\n\nDepends-on: https://review.opendev.org/c/openstack/kolla-ansible/+/874917\n\nChange-Id: Idf5ee822b03ba40179803c981500a6bad37594bf\n(cherry picked from commit 2f447f65453ff278fd53e5bf7975a5a092378b97)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d201d8e293e25d93ecc1893898c78e5079b5e33d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d201d8e293e25d93ecc1893898c78e5079b5e33d"}]},"branch":"refs/heads/stable/yoga"},"c03db79f7c192a201490aa33b874b20ed5005dd1":{"kind":"REWORK","_number":19,"created":"2023-02-27 14:40:31.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/38/874738/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/874738/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/874738/19"}}},"commit":{"parents":[{"commit":"834cf25febd3d990d5c4466d53ec1ac377bc0379","subject":"CI: Don\u0027t run Heat and Horizon in overcloud jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/834cf25febd3d990d5c4466d53ec1ac377bc0379"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2022-09-02 12:40:33.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-02-27 14:40:22.000000000","tz":0},"subject":"Move to Rocky Linux 9","message":"Move to Rocky Linux 9\n\nCentOS Stream 8 support has been dropped. Migration path will be present\nin Yoga release - as a followup change.\n\nMichaelRigart.interfaces does not support custom routes for\nNetworkManager yet. It has been disabled in CI for Rocky Linux 9\ntemporarily.\n\nNon-voting CentOS Stream 9 CI overcloud job is using RL9 container\nimages (as kolla CI is not building CS9 images anymore).\n\nDepends-on: https://review.opendev.org/c/openstack/kolla-ansible/+/874917\n\nChange-Id: Idf5ee822b03ba40179803c981500a6bad37594bf\n(cherry picked from commit 2f447f65453ff278fd53e5bf7975a5a092378b97)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c03db79f7c192a201490aa33b874b20ed5005dd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c03db79f7c192a201490aa33b874b20ed5005dd1"}]},"branch":"refs/heads/stable/yoga"},"dd234fdca80040d1a5c5be49cbd6fbdb5bfd8558":{"kind":"REWORK","_number":20,"created":"2023-02-28 12:03:45.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/38/874738/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/874738/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/874738/20"}}},"commit":{"parents":[{"commit":"834cf25febd3d990d5c4466d53ec1ac377bc0379","subject":"CI: Don\u0027t run Heat and Horizon in overcloud jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/834cf25febd3d990d5c4466d53ec1ac377bc0379"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2022-09-02 12:40:33.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-02-28 12:03:22.000000000","tz":60},"subject":"Move to Rocky Linux 9","message":"Move to Rocky Linux 9\n\nCentOS Stream 8 support has been dropped. Migration path will be present\nin Yoga release - as a followup change.\n\nMichaelRigart.interfaces does not support custom routes for\nNetworkManager yet. It has been disabled in CI for Rocky Linux 9\ntemporarily.\n\nNon-voting CentOS Stream 9 CI overcloud job is using RL9 container\nimages (as kolla CI is not building CS9 images anymore).\n\nDepends-on: https://review.opendev.org/c/openstack/kolla-ansible/+/874917\n\nChange-Id: Idf5ee822b03ba40179803c981500a6bad37594bf\n(cherry picked from commit 2f447f65453ff278fd53e5bf7975a5a092378b97)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dd234fdca80040d1a5c5be49cbd6fbdb5bfd8558"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dd234fdca80040d1a5c5be49cbd6fbdb5bfd8558"}]},"branch":"refs/heads/stable/yoga"},"961730970764d9b62472359fd281f1b4295bf349":{"kind":"REWORK","_number":21,"created":"2023-02-28 12:52:11.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/38/874738/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/874738/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/874738/21"}}},"commit":{"parents":[{"commit":"834cf25febd3d990d5c4466d53ec1ac377bc0379","subject":"CI: Don\u0027t run Heat and Horizon in overcloud jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/834cf25febd3d990d5c4466d53ec1ac377bc0379"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2022-09-02 12:40:33.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-02-28 12:51:04.000000000","tz":60},"subject":"Move to Rocky Linux 9","message":"Move to Rocky Linux 9\n\nCentOS Stream 8 support has been dropped. Migration path will be present\nin Yoga release - as a followup change.\n\nMichaelRigart.interfaces does not support custom routes for\nNetworkManager yet. It has been disabled in CI for Rocky Linux 9\ntemporarily.\n\nNon-voting CentOS Stream 9 CI overcloud job is using RL9 container\nimages (as kolla CI is not building CS9 images anymore).\n\nDepends-on: https://review.opendev.org/c/openstack/kolla-ansible/+/874917\n\nChange-Id: Idf5ee822b03ba40179803c981500a6bad37594bf\n(cherry picked from commit 2f447f65453ff278fd53e5bf7975a5a092378b97)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/961730970764d9b62472359fd281f1b4295bf349"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/961730970764d9b62472359fd281f1b4295bf349"}]},"branch":"refs/heads/stable/yoga"},"70ef0f12feb28ea31fba9fddcb6f9a84cc46e1dd":{"kind":"NO_CODE_CHANGE","_number":22,"created":"2023-02-28 14:33:33.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/38/874738/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/874738/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/874738/22"}}},"commit":{"parents":[{"commit":"834cf25febd3d990d5c4466d53ec1ac377bc0379","subject":"CI: Don\u0027t run Heat and Horizon in overcloud jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/834cf25febd3d990d5c4466d53ec1ac377bc0379"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2022-09-02 12:40:33.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-02-28 14:33:33.000000000","tz":0},"subject":"Move to Rocky Linux 9","message":"Move to Rocky Linux 9\n\nCentOS Stream 8 support has been dropped. Migration path will be present\nin Yoga release - as a followup change.\n\nMichaelRigart.interfaces does not support custom routes for\nNetworkManager yet. It has been disabled in CI for Rocky Linux 9\ntemporarily.\n\nNon-voting CentOS Stream 9 CI overcloud job is using RL9 container\nimages (as kolla CI is not building CS9 images anymore).\n\nChange-Id: Idf5ee822b03ba40179803c981500a6bad37594bf\n(cherry picked from commit 2f447f65453ff278fd53e5bf7975a5a092378b97)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/70ef0f12feb28ea31fba9fddcb6f9a84cc46e1dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/70ef0f12feb28ea31fba9fddcb6f9a84cc46e1dd"}]},"branch":"refs/heads/stable/yoga","description":"Edit commit message"},"d75eb216bfe3be1ea35b12807393a8e81d9f7af2":{"kind":"REWORK","_number":23,"created":"2023-03-01 13:25:39.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/38/874738/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/874738/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/874738/23"}}},"commit":{"parents":[{"commit":"834cf25febd3d990d5c4466d53ec1ac377bc0379","subject":"CI: Don\u0027t run Heat and Horizon in overcloud jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/834cf25febd3d990d5c4466d53ec1ac377bc0379"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2022-09-02 12:40:33.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-01 13:25:33.000000000","tz":60},"subject":"Move to Rocky Linux 9","message":"Move to Rocky Linux 9\n\nCentOS Stream 8 support has been dropped. Migration path will be present\nin Yoga release - as a followup change.\n\nMichaelRigart.interfaces does not support custom routes for\nNetworkManager yet. It has been disabled in CI for Rocky Linux 9\ntemporarily.\n\nNon-voting CentOS Stream 9 CI overcloud job is using RL9 container\nimages (as kolla CI is not building CS9 images anymore).\n\nDepends-on: https://review.opendev.org/c/openstack/kolla-ansible/+/874917\n\nChange-Id: Idf5ee822b03ba40179803c981500a6bad37594bf\n(cherry picked from commit 2f447f65453ff278fd53e5bf7975a5a092378b97)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d75eb216bfe3be1ea35b12807393a8e81d9f7af2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d75eb216bfe3be1ea35b12807393a8e81d9f7af2"}]},"branch":"refs/heads/stable/yoga"},"bd867cf80b5427aa1d98de2c74dc87aa05347463":{"kind":"REWORK","_number":24,"created":"2023-03-01 14:46:54.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/38/874738/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/874738/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/874738/24"}}},"commit":{"parents":[{"commit":"834cf25febd3d990d5c4466d53ec1ac377bc0379","subject":"CI: Don\u0027t run Heat and Horizon in overcloud jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/834cf25febd3d990d5c4466d53ec1ac377bc0379"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2022-09-02 12:40:33.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-01 14:44:58.000000000","tz":60},"subject":"Move to Rocky Linux 9","message":"Move to Rocky Linux 9\n\nCentOS Stream 8 support has been dropped. Migration path will be present\nin Yoga release - as a followup change.\n\nMichaelRigart.interfaces does not support custom routes for\nNetworkManager yet. It has been disabled in CI for Rocky Linux 9\ntemporarily.\n\nNon-voting CentOS Stream 9 CI overcloud job is using RL9 container\nimages (as kolla CI is not building CS9 images anymore).\n\nDepends-on: https://review.opendev.org/c/openstack/kolla-ansible/+/874917\n\nChange-Id: Idf5ee822b03ba40179803c981500a6bad37594bf\n(cherry picked from commit 2f447f65453ff278fd53e5bf7975a5a092378b97)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bd867cf80b5427aa1d98de2c74dc87aa05347463"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bd867cf80b5427aa1d98de2c74dc87aa05347463"}]},"branch":"refs/heads/stable/yoga"},"a9208e7a227ffdf886930d8e21505522e8e00a7e":{"kind":"REWORK","_number":25,"created":"2023-03-02 11:08:01.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/38/874738/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/874738/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/874738/25"}}},"commit":{"parents":[{"commit":"834cf25febd3d990d5c4466d53ec1ac377bc0379","subject":"CI: Don\u0027t run Heat and Horizon in overcloud jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/834cf25febd3d990d5c4466d53ec1ac377bc0379"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2022-09-02 12:40:33.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-02 11:07:53.000000000","tz":60},"subject":"Move to Rocky Linux 9","message":"Move to Rocky Linux 9\n\nCentOS Stream 8 support has been dropped. Migration path will be present\nin Yoga release - as a followup change.\n\nMichaelRigart.interfaces does not support custom routes for\nNetworkManager yet. It has been disabled in CI for Rocky Linux 9\ntemporarily.\n\nNon-voting CentOS Stream 9 CI overcloud job is using RL9 container\nimages (as kolla CI is not building CS9 images anymore).\n\nDepends-on: https://review.opendev.org/c/openstack/kolla-ansible/+/874917\n\nChange-Id: Idf5ee822b03ba40179803c981500a6bad37594bf\n(cherry picked from commit 2f447f65453ff278fd53e5bf7975a5a092378b97)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a9208e7a227ffdf886930d8e21505522e8e00a7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a9208e7a227ffdf886930d8e21505522e8e00a7e"}]},"branch":"refs/heads/stable/yoga"},"4af8ba8cf4dc58032e74bd24a4f55dbc4605260b":{"kind":"REWORK","_number":26,"created":"2023-03-02 11:09:06.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/38/874738/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/874738/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/874738/26"}}},"commit":{"parents":[{"commit":"834cf25febd3d990d5c4466d53ec1ac377bc0379","subject":"CI: Don\u0027t run Heat and Horizon in overcloud jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/834cf25febd3d990d5c4466d53ec1ac377bc0379"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2022-09-02 12:40:33.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-02 11:09:01.000000000","tz":60},"subject":"Move to Rocky Linux 9","message":"Move to Rocky Linux 9\n\nCentOS Stream 8 support has been dropped. Migration path will be present\nin Yoga release - as a followup change.\n\nMichaelRigart.interfaces does not support custom routes for\nNetworkManager yet. It has been disabled in CI for Rocky Linux 9\ntemporarily.\n\nNon-voting CentOS Stream 9 CI overcloud job is using RL9 container\nimages (as kolla CI is not building CS9 images anymore).\n\nDepends-on: https://review.opendev.org/c/openstack/kolla-ansible/+/874917\n\nChange-Id: Idf5ee822b03ba40179803c981500a6bad37594bf\n(cherry picked from commit 2f447f65453ff278fd53e5bf7975a5a092378b97)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4af8ba8cf4dc58032e74bd24a4f55dbc4605260b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4af8ba8cf4dc58032e74bd24a4f55dbc4605260b"}]},"branch":"refs/heads/stable/yoga"},"07c3f18f9e017c98986d6e27cc9c7c7744430625":{"kind":"NO_CODE_CHANGE","_number":27,"created":"2023-03-02 11:17:51.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/38/874738/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/874738/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/874738/27"}}},"commit":{"parents":[{"commit":"834cf25febd3d990d5c4466d53ec1ac377bc0379","subject":"CI: Don\u0027t run Heat and Horizon in overcloud jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/834cf25febd3d990d5c4466d53ec1ac377bc0379"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2022-09-02 12:40:33.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-02 11:16:51.000000000","tz":60},"subject":"Add Rocky Linux 9 support","message":"Add Rocky Linux 9 support\n\nMichaelRigart.interfaces does not support custom routes for\nNetworkManager yet. It has been disabled in CI for Rocky Linux 9\ntemporarily.\n\nNon-voting CentOS Stream 9 CI overcloud job is using RL9 container\nimages (as kolla CI is not building CS9 images anymore).\n\nDepends-on: https://review.opendev.org/c/openstack/kolla-ansible/+/874917\n\nChange-Id: Idf5ee822b03ba40179803c981500a6bad37594bf\n(cherry picked from commit 2f447f65453ff278fd53e5bf7975a5a092378b97)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/07c3f18f9e017c98986d6e27cc9c7c7744430625"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/07c3f18f9e017c98986d6e27cc9c7c7744430625"}]},"branch":"refs/heads/stable/yoga"},"25579a46bf1608583226478c308c6fbc32f581a4":{"kind":"REWORK","_number":28,"created":"2023-03-02 12:40:20.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/38/874738/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/874738/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/874738/28"}}},"commit":{"parents":[{"commit":"834cf25febd3d990d5c4466d53ec1ac377bc0379","subject":"CI: Don\u0027t run Heat and Horizon in overcloud jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/834cf25febd3d990d5c4466d53ec1ac377bc0379"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2022-09-02 12:40:33.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-02 12:40:15.000000000","tz":60},"subject":"Add Rocky Linux 9 support","message":"Add Rocky Linux 9 support\n\nMichaelRigart.interfaces does not support custom routes for\nNetworkManager yet. It has been disabled in CI for Rocky Linux 9\ntemporarily.\n\nNon-voting CentOS Stream 9 CI overcloud job is using RL9 container\nimages (as kolla CI is not building CS9 images anymore).\n\nDepends-on: https://review.opendev.org/c/openstack/kolla-ansible/+/874917\n\nChange-Id: Idf5ee822b03ba40179803c981500a6bad37594bf\n(cherry picked from commit 2f447f65453ff278fd53e5bf7975a5a092378b97)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/25579a46bf1608583226478c308c6fbc32f581a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/25579a46bf1608583226478c308c6fbc32f581a4"}]},"branch":"refs/heads/stable/yoga"},"a292e4bf2d01fcc2cca0a4c6ef6f125461aaeac6":{"kind":"REWORK","_number":29,"created":"2023-03-02 13:34:49.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/38/874738/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/874738/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/874738/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/874738/29"}}},"commit":{"parents":[{"commit":"834cf25febd3d990d5c4466d53ec1ac377bc0379","subject":"CI: Don\u0027t run Heat and Horizon in overcloud jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/834cf25febd3d990d5c4466d53ec1ac377bc0379"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2022-09-02 12:40:33.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-02 13:34:45.000000000","tz":60},"subject":"Add Rocky Linux 9 support","message":"Add Rocky Linux 9 support\n\nMichaelRigart.interfaces does not support custom routes for\nNetworkManager yet. It has been disabled in CI for Rocky Linux 9\ntemporarily.\n\nNon-voting CentOS Stream 9 CI overcloud job is using RL9 container\nimages (as kolla CI is not building CS9 images anymore).\n\nDepends-on: https://review.opendev.org/c/openstack/kolla-ansible/+/874917\n\nChange-Id: Idf5ee822b03ba40179803c981500a6bad37594bf\n(cherry picked from commit 2f447f65453ff278fd53e5bf7975a5a092378b97)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a292e4bf2d01fcc2cca0a4c6ef6f125461aaeac6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a292e4bf2d01fcc2cca0a4c6ef6f125461aaeac6"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
