)]}'
{"id":"openstack%2Fkayobe~876462","triplet_id":"openstack%2Fkayobe~stable%2Fzed~I1edafbb41a26587a5ef794b3b9886fdf189a0a1a","project":"openstack/kayobe","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-03-30 13:54:53.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-03-30 16:03:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1edafbb41a26587a5ef794b3b9886fdf189a0a1a","subject":"CI: Disable disk image build in seed-upgrade jobs","status":"MERGED","created":"2023-03-06 16:49:39.000000000","updated":"2023-03-30 16:04:51.000000000","submitted":"2023-03-30 16:03:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"876462","cherry_pick_of_change":876325,"cherry_pick_of_patch_set":4,"meta_rev_id":"9a58e74f9708b9f5483fbd76acfdc9fdc48d23a7","_number":876462,"virtual_id_number":876462,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-30 16:03:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-03-30 13:54:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-03-08 09:27:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-03-30 13:54:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-06 17:56:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-08 09:27:22.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-03-21 08:53:27.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"bc9cf1c3c494a65000a98da017b07f8490ee107c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-03-06 16:49:39.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"00d7fe4611afe9d25c24bc7f82e6ad0d84f3d6ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-06 17:56:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0bd63cc96db42cf8a316f078d57a913\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/87360f44b9814d708f4d1236f653e42a : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b42eb7529294e9da9fee33af587599b : SUCCESS in 3m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ab774416df6476eb98677d1d7ce1a91 : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5618a70a5bcc4a3697c1570ed6334242 : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dc69d30221ae470c88d58995d18aed87 : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d9edac3a57547108d8854ba9e588d88 : SUCCESS in 6m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e21c9498bd524175852701e7b565f5a9 : SUCCESS in 3m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/76b4527cf3e84ec29b7ad0cda46c44a5 : SUCCESS in 9m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d8c4135353e74223a22a06aa2fa500ba : SUCCESS in 15m 54s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a4f3c75403a7473ba0d0b6a59e58f2d2 : SUCCESS in 1h 05m 01s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/1a6f8b40615140e5bc263681777de47e : SUCCESS in 49m 29s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b3d068a25621477082b6879d3dccfc5b : SUCCESS in 42m 18s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/6e8522085a1c4ca5b9aef67b435c8ccd : SUCCESS in 49m 42s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/8c68fe460dce4cc8969d5118380e1b12 : SUCCESS in 17m 52s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/3118b6c3c12d46349ec70b7b53e51ea5 : SUCCESS in 14m 22s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/393261c1e32d495cb051c2b0b296cceb : SUCCESS in 13m 40s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/883ac93258324ff495f455a4067e15c7 : SUCCESS in 1h 05m 17s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4da9d951f5494029b9d9303497852c7a : SUCCESS in 19m 53s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4daeedfa6cbb410f9cc4f9418205b7ba : SUCCESS in 12m 03s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/a895e6bfd8e746fa829d234dad7e79c8 : SUCCESS in 32m 21s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4cafb1a773f2400c88aabb5d9df2ad43 : SUCCESS in 22m 32s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3359b16923b541c09a3a388e8163a938 : SUCCESS in 12m 11s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/034d7ef240d34ddabbdb9a948726bba5 : SUCCESS in 8m 45s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/776474967d284d9984b73d0407fc68ae : SUCCESS in 15m 37s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/691140af23124a8a973379c713ff62fe : SUCCESS in 13m 57s","accounts_in_message":[],"_revision_number":1},{"id":"8275bd507f914732d02f7a7bbd72e42fe86f9da2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-03-08 09:27:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b9b7eeb591be767492fd7515f55b44b6dc545f05","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-30 13:54:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9628c852f4e88fe06eef38775798843f4c65f9ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 13:55:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"180b0a7aea0e2225052e16eb4cdbb96ddf54e1a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 16:03:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e3cb721a94342e88848d2ed343d8fed\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13a9bdaad89a4d328aa3848763d4f3a1 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3173ceb68d1b4484b916ca93191fec01 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/958eeadc5efe448fa6bbb8ce8ded0238 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc4f6e3281354a3cafbd6d1e69864037 : SUCCESS in 7m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a3b3661fa462499d9394ccfa056133ec : SUCCESS in 4m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3fc6729faf9a4abd9f290989e9ff0612 : SUCCESS in 9m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/19da7c54872c49eb8a6276c7ac507972 : SUCCESS in 16m 43s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/182efd01a52948e7b2c0a1a7d1700d66 : SUCCESS in 1h 01m 19s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/744be4c1f0044e8d910d78798091c937 : SUCCESS in 43m 14s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/96a50305d2fb48fb91ee1beec489260c : SUCCESS in 1h 05m 35s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/455123501dac466d86b433120b03b5f7 : SUCCESS in 18m 13s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7751fbc1ccb24966be17eef5a353f948 : SUCCESS in 14m 07s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34095d5b8e5a4ef5916231f319eed78d : SUCCESS in 1h 22m 46s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/795f6b5d0efd4d58b3b6e897360a0bf9 : SUCCESS in 24m 18s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/100c270b8d5248d2981c1faf0ead5197 : SUCCESS in 15m 21s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/07d647adfccc4abf81b54b05455867bc : SUCCESS in 28m 01s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0c9c6549d5554f00b3fd623ddd8ab331 : SUCCESS in 14m 01s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e73f14010ef04620a8480e9f378dd1e5 : SUCCESS in 12m 23s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c5c8f67cb2b443d1968c3c62ee8ba38d : SUCCESS in 17m 58s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9e252a817e2f49f8a99bbdb1fcbf8289 : SUCCESS in 15m 05s","accounts_in_message":[],"_revision_number":1},{"id":"3a686cdce3bf807ca8ac38ed0f030203b3dfb28f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 16:03:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9a58e74f9708b9f5483fbd76acfdc9fdc48d23a7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 16:04:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/331827d651704a19866695726ed155a4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0220c49bb3484c71b59fddbf0aa68b35 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c53d0956039c29461c0ae56ae38d89df15e2c998","revisions":{"c53d0956039c29461c0ae56ae38d89df15e2c998":{"kind":"REWORK","_number":1,"created":"2023-03-06 16:49:39.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/62/876462/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/62/876462/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/876462/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/876462/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/876462/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/62/876462/1"}}},"commit":{"parents":[{"commit":"ee33d3a487ad24ae2709eb415d649df4cda20e0e","subject":"CI: Don\u0027t run Heat and Horizon in overcloud jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ee33d3a487ad24ae2709eb415d649df4cda20e0e"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-03-03 17:21:07.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-03-06 16:49:39.000000000","tz":0},"subject":"CI: Disable disk image build in seed-upgrade jobs","message":"CI: Disable disk image build in seed-upgrade jobs\n\nThis build takes time and can fail due to lack of disk space. It got\nenabled when we changed overcloud_dib_build_host_images to true.\n\nAlso fix bifrost overrides which was wrongly applied, we need to use\ndib.yml instead of bifrost.yml, like in kayobe-seed-base.\n\nChange-Id: I1edafbb41a26587a5ef794b3b9886fdf189a0a1a\n(cherry picked from commit 77ca0771958713804d2c65b8e33824a6573132fe)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c53d0956039c29461c0ae56ae38d89df15e2c998"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c53d0956039c29461c0ae56ae38d89df15e2c998"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
