)]}'
{"id":"openstack%2Fkayobe~876689","triplet_id":"openstack%2Fkayobe~stable%2Fyoga~I1edafbb41a26587a5ef794b3b9886fdf189a0a1a","project":"openstack/kayobe","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-03-30 13:55:12.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-03-30 16:03:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1edafbb41a26587a5ef794b3b9886fdf189a0a1a","subject":"CI: Disable disk image build in seed-upgrade jobs","status":"MERGED","created":"2023-03-07 21:48:26.000000000","updated":"2023-03-30 16:05:31.000000000","submitted":"2023-03-30 16:03:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"876689","cherry_pick_of_change":876325,"cherry_pick_of_patch_set":4,"meta_rev_id":"b3f2e6b0f5ff2b091544c445a0c74cc5015e3ee4","_number":876689,"virtual_id_number":876689,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-30 16:03:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-03-30 13:55:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-03-08 09:27:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-03-30 13:55:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-07 23:34:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-08 09:27:11.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-03-21 08:53:34.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"b20ac0b5e4232ca0344c6e513f4d1cfd03631f79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-03-07 21:48:26.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9f7a568f61872d41c8555327ec39bb4c8e0d7780","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 23:34:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92ba6a59b67d47b3847b31416929013a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2a460dd1903420ca823e170dc218e86 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e9b83f0de3d440088accc30c9404fb4 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf96aa6a980148e19835d85d67b18895 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b5553c829bb4bee9cd9fecf2397dd4c : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4293718a3d3541b89a3c264dbbc608ee : SUCCESS in 9m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2ca3789ef8ef4c31a1464e210e7466ee : SUCCESS in 6m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2698f341f1cb4d8d8551139c16830181 : SUCCESS in 10m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a7d6c0940f8b4d4e99d2edeb10020569 : SUCCESS in 20m 09s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7cf44d6723a34fb8a18ebc686493b190 : SUCCESS in 1h 04m 58s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/d96a22ee83b6447098ec420f2c980dde : SUCCESS in 1h 45m 14s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/5eda3d01542c42f19184e1a44e516331 : SUCCESS in 1h 11m 29s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/6e27790604624379a7196e9db4310ce4 : SUCCESS in 1h 06m 32s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61440f43786c4297b9f93cfb854b5133 : SUCCESS in 47m 10s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/735bb4cb08ed421d891e30afc6b5b040 : SUCCESS in 47m 03s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/10e45534e932443b8ff874c7c25faf54 : SUCCESS in 21m 31s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/3e3ad51648144a87bb97da4b2c29601b : SUCCESS in 21m 35s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/8be7a058d85443979f78c35f2f23690a : SUCCESS in 24m 35s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/7cf26fc7f51f41c4964bebe2730fdd9f : SUCCESS in 19m 19s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cbe67d37b5cb4b39923232fee6ab359a : SUCCESS in 17m 51s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e89a268018cb47cca9ba8befb6959e0a : SUCCESS in 16m 54s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/614f1a471a28409896c55b41b400fc8a : SUCCESS in 1h 22m 59s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/b97e21b4aa49432fa065e9c320e884f7 : SUCCESS in 1h 12m 07s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/79ff7ae1870042a3b7a82fef34cfc623 : SUCCESS in 1h 36m 31s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47671717bd134a099616499ffac1aec3 : SUCCESS in 1h 23m 43s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/1e461d4fcbc04515bd49e6de52880912 : SUCCESS in 19m 23s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/d683fa370ae24a119d4cad241defc551 : SUCCESS in 20m 53s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/26af077b410441c98d79e1f7f6a769f8 : SUCCESS in 22m 59s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9c0e5b227d04a16ba7c5a490a465ac7 : SUCCESS in 15m 33s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f4dc8bd025514c21a55e47290f545329 : SUCCESS in 17m 27s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/820357d1b29642e189f4456f96354634 : SUCCESS in 26m 07s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/366029c5f2d14c55bef3e1332608e62a : SUCCESS in 23m 45s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/80df6a8ea1a64a5b9e0f7ca8a5686496 : SUCCESS in 19m 36s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/9fe907414ab44338a19ee7c1405a9969 : SUCCESS in 16m 49s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/38ccce601af54ec3b78febcb02f69357 : SUCCESS in 19m 03s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e3cfbc18e7eb4d8c92f58b3650ac06e7 : SUCCESS in 12m 12s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f6c26d92b704fe59e534808cf260c43 : SUCCESS in 17m 32s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/0b82124ea95242b8b5b3c05bae9451e0 : SUCCESS in 22m 53s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/51d2f563e4ca4134bb76ba78a07125d7 : SUCCESS in 20m 34s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/05e4eb74f6ca498eb57ed27b5afd9d1d : SUCCESS in 18m 50s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3434280f3d62443ca825b7145316aee8 : SUCCESS in 18m 13s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7330a914dd7848eba8b55075526e7ce6 : SUCCESS in 14m 52s","accounts_in_message":[],"_revision_number":1},{"id":"38ab3c8af7f96ae4932f5bfc51bd67a6315d13a4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-03-08 07:44:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f0487d78f465aae243600fdcbc7a4e7142a56eb5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-03-08 09:27:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"27c8a8a4d034917506eaa4a19bc6836180e43de3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-30 13:55:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6343d2f2c62c9909ccdecda53141b37d3b14b703","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 13:55:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3ff7d6c7a69c6dcc770a533271860c4395ad8351","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 16:03:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afa56b7f5a7543899229c85c46c25bba\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a3584c1df9f4540a6e4975eceecb750 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa19990eda434fc09a19f9090710865a : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2312b08389e240aa909cb47e6ff58a6c : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b41965c9cfb45c08f610aecb9bb38be : SUCCESS in 9m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4e3957f03133418bb196db1eeb795c4f : SUCCESS in 6m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8fc9f351d88747e8b26bdfbcf3031cc1 : SUCCESS in 13m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/67731268910f4456aaf40f7e0e7d7ab5 : SUCCESS in 15m 33s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c43feda024734f4aaa49caacaf0a3246 : SUCCESS in 1h 09m 51s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/afc3552254cb40fbb809e0ceeb6edbdc : SUCCESS in 1h 06m 39s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/25ca7d207bdb4afea658694223bc1bd0 : SUCCESS in 1h 06m 43s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6dae634bd02948858b7eb4e56f291553 : SUCCESS in 59m 53s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ecd17f87c1d14903baea948238720a31 : SUCCESS in 49m 23s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/1477188712a0425aa7895e969b6c4ed7 : SUCCESS in 21m 52s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/55fc39be172c47f397138cc494eaba4b : SUCCESS in 21m 37s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/47f5cfdad64748c49600eb526d4d7299 : SUCCESS in 21m 18s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc0810bfab0f4f779e0d844a3ae2be9b : SUCCESS in 14m 31s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/98aa78ca947c4c8eb0e8ad650343989a : SUCCESS in 16m 48s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/fa2cbfed9b9c42c9a05e023417724e22 : SUCCESS in 1h 11m 19s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/7d922950ef4949ba82618cf10f45e00b : SUCCESS in 1h 06m 04s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2676843b46334ecbbd1b0654de2e4f39 : SUCCESS in 1h 58m 27s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e6805296498d4bd8a44e24cf9ec13bed : SUCCESS in 1h 10m 54s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/68dc0dc381f14d6f841b5f9594f9fa52 : SUCCESS in 21m 28s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/71bec6799e6b4141877ac4b663c746e9 : SUCCESS in 19m 28s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/5cd93eb5f31d4872bfaea6b3b319dfd3 : SUCCESS in 20m 04s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5690db1074e6430f8ff7eb02da416b96 : SUCCESS in 18m 37s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a556335efb564149a56585ce232c2b69 : SUCCESS in 17m 13s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4ad7363bf6fa4fd58350559c9389f312 : SUCCESS in 28m 05s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d71db1780864ded9b7ff4f080aa2b88 : SUCCESS in 25m 49s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f23aef5e5a5e41d5b8d74b5338962742 : SUCCESS in 16m 33s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/fea42dc5ca404cdabaea875665663885 : SUCCESS in 17m 00s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0fa074b958064d3f945b17d02306fc29 : SUCCESS in 14m 30s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1205ffc5538d4d038c785714ca153b38 : SUCCESS in 13m 36s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b16089271ad42b88c91d2e081d3ff7f : SUCCESS in 12m 40s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/da1527a929004accb6cc6fa8bed5a60a : SUCCESS in 19m 42s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/684e36303a3e4ae3879b622b6700b43f : SUCCESS in 23m 32s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c8b3736a7a3e41f982230b997d691310 : SUCCESS in 18m 01s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a032d2740944c778a5e3ecacacc452d : SUCCESS in 16m 07s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d651ca3019d942388ba7cfb8d4805e71 : SUCCESS in 14m 04s","accounts_in_message":[],"_revision_number":1},{"id":"efa70474bc52622ed9581556f7012fba5e000a44","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 16:03:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b3f2e6b0f5ff2b091544c445a0c74cc5015e3ee4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-30 16:05:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37f52a56df614d5388d9049a2b63f016\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/385bf0b8170d4438be8301abc682fa29 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7ffbd4e32db9ca29ef67e330740e035203bdaf64","revisions":{"7ffbd4e32db9ca29ef67e330740e035203bdaf64":{"kind":"REWORK","_number":1,"created":"2023-03-07 21:48:26.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/89/876689/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/89/876689/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/876689/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/876689/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/89/876689/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/89/876689/1"}}},"commit":{"parents":[{"commit":"5b35bbcb6a07c3eeb6d813a981bbb2823a89e898","subject":"Merge \"CI: Jammy followup - move os_release\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5b35bbcb6a07c3eeb6d813a981bbb2823a89e898"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-03-03 17:21:07.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-03-07 21:48:26.000000000","tz":0},"subject":"CI: Disable disk image build in seed-upgrade jobs","message":"CI: Disable disk image build in seed-upgrade jobs\n\nFix bifrost overrides which was wrongly applied, we need to use dib.yml instead of bifrost.yml, like in kayobe-seed-base.\n\nThe original commit also disables overcloud_dib_build_host_images. This\nchange is kept but is not relevant for Yoga.\n\nChange-Id: I1edafbb41a26587a5ef794b3b9886fdf189a0a1a\n(cherry picked from commit 77ca0771958713804d2c65b8e33824a6573132fe)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7ffbd4e32db9ca29ef67e330740e035203bdaf64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7ffbd4e32db9ca29ef67e330740e035203bdaf64"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
