)]}'
{"id":"openstack%2Fkayobe~876757","triplet_id":"openstack%2Fkayobe~master~Ie0d9bf895f6714cbb8b0bd31a008eb388f4c51c2","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-03-07 22:00:26.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-03-07 22:00:26.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-03-07 20:36:04.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"Ie0d9bf895f6714cbb8b0bd31a008eb388f4c51c2","subject":"use OS python to create kolla-ansible venv","status":"MERGED","created":"2023-03-07 15:42:25.000000000","updated":"2023-03-07 22:01:30.000000000","submitted":"2023-03-07 22:00:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"876757","meta_rev_id":"406c1cc9327652f65a8ed1d7b7aeb33e49d287d9","_number":876757,"virtual_id_number":876757,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-07 22:00:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-03-07 15:46:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-07 20:36:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-07 20:36:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-07 15:46:59.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-03-07 16:13:03.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2023-03-07 16:13:03.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-03-07 17:33:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e176cc2d04b2c3246877710cae7031ec967189de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-07 15:42:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f67a7b3c0a1f7de0813c72d2be448929bc55b4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-07 15:45:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cb7a3e0903c1fc76cb533149cecc671536e25583","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-07 15:46:59.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bf7f36e8bb913e6d802313880c8ba29cf6595cac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 17:33:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63c96f035b6b481bba276427345f9404\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf1033af856247d3b292e83ea65ddecd : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b86f6c3b89d4bb7846470c09b17e418 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e31cc96ce3547fcb4470c299f56c67a : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/03a6c3f8919e477a969d00d9ff65d6a5 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57fa2b9866734fa9ad0e88ef011c3e73 : SUCCESS in 7m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82022b8ffb34411199f455a4f2efd59a : SUCCESS in 3m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1913a56e298b434699f9f4209bb2deac : SUCCESS in 3m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/522b1686efcb4d7886687e2a3095bf71 : SUCCESS in 12m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3ef9ee4a6cc2469d9f6a320d34302201 : SUCCESS in 16m 53s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/041f439a577b4bc3a1eb7a6a0c478165 : SUCCESS in 53m 13s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/6402936955404f5d9a087cc19affc2d1 : SUCCESS in 1h 00m 42s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/004d1288bfd34fe390e1f3388b88b31e : SUCCESS in 51m 13s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e265c964f3b848c7b20a1c76b3c7d65d : SUCCESS in 59m 56s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d6c9ccaf824d4159981599f886b26ca0 : SUCCESS in 16m 28s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/00117e16ed7d4d36b3c0f699cd7e512e : SUCCESS in 17m 08s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1db19f31301f47819aa844f2059c6075 : SUCCESS in 15m 44s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/182b33b3e46a4f3fabb127f7d74f126a : SUCCESS in 1h 31m 55s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/481ef1deea6e4c51863049e659a73770 : SUCCESS in 1h 06m 11s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/09ee2abc94c24b6282a415ce5a0b6b0c : SUCCESS in 16m 53s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea52954c943a4b1f9ab7964772b2921d : SUCCESS in 14m 25s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/c1361a0288064da18005bffd71d2cf4b : SUCCESS in 34m 19s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6cb9c4b500214a50a9897e6c71a200fc : SUCCESS in 27m 27s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/397334476f584f5b8556c56c56069484 : SUCCESS in 29m 40s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d1d114ad37f6428485818a627a3023a2 : SUCCESS in 12m 47s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a8de47ffb0a14f788adabc1c588af642 : SUCCESS in 10m 16s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b8f875fae1b840eda6227e7e257aed94 : SUCCESS in 17m 42s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c381137ba7ad4502a8ce7a38ff303eb5 : SUCCESS in 13m 40s","accounts_in_message":[],"_revision_number":2},{"id":"09cddb685a7fcca9073609c6f6a945d45aa82f9b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-03-07 20:36:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2dfcf5681b7839b500559ef3d0a5a6d1618e1876","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 20:36:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cff1cb7189171ed21519b0dcdbe07d12ba82e61d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 22:00:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a80d14a7490b429eb8108869c6cb7ec4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c673a0153cdd4efc9b76659a7b28c401 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77966ec6117f4099911c2cd30d82bfaf : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/627d29c1729742f598ee6138bad8750b : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b734642cba57493cab0203f3baf96f95 : SUCCESS in 7m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/352909c45e774c7cbacfc533f404303b : SUCCESS in 3m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3a7377e4c8c34abbad583b3aff6f56cf : SUCCESS in 4m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6f2f8b669cef4c319dc59bb59a5aa458 : SUCCESS in 17m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5540a891d8da4007a58b247138b7813b : SUCCESS in 14m 37s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/1c45aaf9b258499dbdc1f67f4d6b524d : SUCCESS in 40m 04s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2ae363fa62554f3ca5e6a18705f8e87d : SUCCESS in 48m 45s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/d189c1a2beac457197225cf7a85c886c : SUCCESS in 1h 08m 31s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d09b015228ec49198d0ab9fe518cc2fe : SUCCESS in 13m 20s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39d4e5bcad7b41e68a0aea01aed040d6 : SUCCESS in 15m 11s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/408ebdaff68644bc93387c12f843797c : SUCCESS in 1h 08m 41s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4f9f62ca066b4a59bae8d156242c03a9 : SUCCESS in 1h 18m 47s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/66646b4859924ed7bc8d9d9ab3bbfa2d : SUCCESS in 20m 08s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/93520c3bacfb45bcb122a565619efa2f : SUCCESS in 10m 37s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d56b3207d9ce42feb7b0352fba06f89b : SUCCESS in 30m 55s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/71e67a7acfea4571a435f73c0f100ad7 : SUCCESS in 23m 33s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d26f495d7b05456e87e42e0c49627146 : SUCCESS in 30m 30s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1a44f022b21044219d6eaa9bd5e6a99b : SUCCESS in 11m 35s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b4ed34a0dddc486594bff25af322d7e1 : SUCCESS in 17m 45s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd4526c452ad4600922a07ee3392d5ed : SUCCESS in 15m 29s","accounts_in_message":[],"_revision_number":2},{"id":"4d6e221b6664c542f7db7c9475ba32672fadff5c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 22:00:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"406c1cc9327652f65a8ed1d7b7aeb33e49d287d9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 22:01:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3eff0d99d4954d1884cfbff61b727073\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb36757d7d4e45ad9c2564dd948ed1d1 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00fee715935f44679bd9a386b7d827ad : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"068581456fd0928cd71c3e1222e63461ff82f7e5","revisions":{"4f45d7a362844373075a436247570a216e70a4af":{"kind":"REWORK","_number":1,"created":"2023-03-07 15:42:25.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/57/876757/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/57/876757/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/57/876757/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/57/876757/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/57/876757/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/57/876757/1"}}},"commit":{"parents":[{"commit":"77ca0771958713804d2c65b8e33824a6573132fe","subject":"CI: Disable disk image build in seed-upgrade jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/77ca0771958713804d2c65b8e33824a6573132fe"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-07 15:20:15.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-07 15:42:17.000000000","tz":60},"subject":"use OS python to create kolla-ansible venv","message":"use OS python to create kolla-ansible venv\n\nCurrently kayobe creates kolla-ansible venv using kayobe venv\npython3. There are corner cases when creation k-a venv fails\nwhile using kayobe venv created with python 3.6 buggy setuptools/\nvirtualenv command. Using OS python to create venv solves those\ncorner cases issues and preserves compatibilty.\n\nStory: 2010634\nTask: 47599\n\nChange-Id: Ie0d9bf895f6714cbb8b0bd31a008eb388f4c51c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4f45d7a362844373075a436247570a216e70a4af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4f45d7a362844373075a436247570a216e70a4af"}]},"branch":"refs/heads/master"},"068581456fd0928cd71c3e1222e63461ff82f7e5":{"kind":"REWORK","_number":2,"created":"2023-03-07 15:45:07.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/57/876757/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/57/876757/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/57/876757/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/57/876757/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/57/876757/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/57/876757/2"}}},"commit":{"parents":[{"commit":"77ca0771958713804d2c65b8e33824a6573132fe","subject":"CI: Disable disk image build in seed-upgrade jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/77ca0771958713804d2c65b8e33824a6573132fe"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-07 15:20:15.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-07 15:45:02.000000000","tz":60},"subject":"use OS python to create kolla-ansible venv","message":"use OS python to create kolla-ansible venv\n\nCurrently kayobe creates kolla-ansible venv using kayobe venv\npython3. There are corner cases when creation k-a venv fails\nwhile using kayobe venv created with python 3.6 buggy setuptools/\nvirtualenv command. Using OS python to create venv solves those\ncorner cases issues and preserves compatibilty.\n\nStory: 2010634\nTask: 47599\n\nChange-Id: Ie0d9bf895f6714cbb8b0bd31a008eb388f4c51c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/068581456fd0928cd71c3e1222e63461ff82f7e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/068581456fd0928cd71c3e1222e63461ff82f7e5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
