)]}'
{"id":"openstack%2Fkayobe~876790","triplet_id":"openstack%2Fkayobe~stable%2Fzed~Ie0d9bf895f6714cbb8b0bd31a008eb388f4c51c2","project":"openstack/kayobe","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-03-08 09:21:36.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-03-08 11:51:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie0d9bf895f6714cbb8b0bd31a008eb388f4c51c2","subject":"use OS python to create kolla-ansible venv","status":"MERGED","created":"2023-03-07 22:02:26.000000000","updated":"2023-03-08 11:53:03.000000000","submitted":"2023-03-08 11:51:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"876790","cherry_pick_of_change":876757,"cherry_pick_of_patch_set":2,"meta_rev_id":"6d17e1bd5587a7eb8f3ca39b340f8b8f320c456a","_number":876790,"virtual_id_number":876790,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-08 11:51:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2023-03-08 09:21:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-08 07:44:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2023-03-08 09:21:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-07 23:22:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-08 07:44:56.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2023-03-08 08:00:29.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"c9be9344dc69bdc689a6b37619d923d865c2fa41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-07 22:02:26.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"30bc5d43bfe86018db062752def8cc2cacc28f88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 23:22:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d66006fa62c141caac229db8c9eded7b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2d2898d21d54e779d63b2b3c1caea6b : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa7fd97b87894d58a68b6204eb05cda1 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0fa1f2147adc42e4ac6143813c75e9fb : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1433ce29560e467c8b46fc8d2a644d48 : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1c39a4be7cde4e8f99f22bddc0524b03 : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db532ea58804406f83025ceefbdc5af4 : SUCCESS in 7m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82899cbd813441a7bcac12d97bd08cf7 : SUCCESS in 3m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/59f8a981744c439f81ea21665613ecb0 : SUCCESS in 4m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/82878650d8b84c33b6b29de7e05c9580 : SUCCESS in 10m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/43c03bd9e85c46dca9999692c521c6bf : SUCCESS in 15m 24s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/b1ccf6e0b94f4faeaa4ca370a4568fe7 : SUCCESS in 59m 14s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/2687cb46a00b461db51dec6a708c7d36 : SUCCESS in 51m 38s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/45db8ce7d5894f9eaced9e068d8c0fb9 : SUCCESS in 41m 02s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/d6065a45aa1542c7ac445bb25817e419 : SUCCESS in 1h 08m 17s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4f269c53e3f34d80ac113cad5d37c775 : SUCCESS in 16m 27s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/3744dc02ed704b44a0a8882806c39036 : SUCCESS in 17m 25s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/772c3b341018448f95673d5ccc67babd : SUCCESS in 13m 44s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8db5aad17b845d1beb7007e478377fd : SUCCESS in 1h 10m 34s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e908c8655dab42069e4dad831b49b029 : SUCCESS in 17m 27s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/505a01e4e6b346a79aa480849bccacf2 : SUCCESS in 18m 47s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/c7c66efe60f44281847aeaf3b894c446 : SUCCESS in 27m 20s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e649ae0e686043b48925595f833a29d5 : SUCCESS in 36m 17s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f8896abcefa44c8dba3bbf55b8697767 : SUCCESS in 13m 11s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/68ef9035bb21409c828e79d64326f400 : SUCCESS in 10m 55s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/73b9d097f3e442c1a4ca54f7697ff260 : SUCCESS in 19m 44s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ed46b59458442e78d5ced87e443c538 : SUCCESS in 12m 16s","accounts_in_message":[],"_revision_number":1},{"id":"ae853883bdc6292562b846e0c1e5247f78df6501","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-03-08 07:44:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"28d4a8b545edc660ff7ab147b72a035f8bd6f331","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-03-08 09:21:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"869d098d120d1800cf4255d4f7dd996e527d277e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 10:03:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4411e309319d34e423e13b7a38561aeab846b7e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 11:51:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53549134f53d4c64b5a1fabab2f2f31e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b27695b34e6483191f21253e33c3f98 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e93ad53602ee426ea15133dc98455f40 : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a12bae3501c84fef9c94e599590ff8f4 : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/619c414481104ac49aa2411de2e04497 : SUCCESS in 7m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64d034c56a514084aa3ba413ff9b4de0 : SUCCESS in 3m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d29776f287c3452b9092a4e5f7b919a8 : SUCCESS in 3m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8caffdd0ac83429b831efc22b14bf279 : SUCCESS in 11m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6c06909efa974766bc51e5f28110a371 : SUCCESS in 16m 04s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a5ea579753da464ba189646c0a5366f0 : SUCCESS in 48m 22s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/79f26d39530f407a90ec9db6acb70b05 : SUCCESS in 43m 29s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/acf5d1f06072466aaa8761899aecfed0 : SUCCESS in 53m 30s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a2fe502b66264189a52641610aab97ac : SUCCESS in 17m 34s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/198707e48eea46b1a4837b168f960964 : SUCCESS in 14m 29s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f887beba8acd439d861deb5d3dbf9fcc : SUCCESS in 1h 20m 21s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3f9e9f75677343c6b9a77ccc0cf7715d : SUCCESS in 16m 02s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a259d86e3cc4982bdaccfdeb6dabeba : SUCCESS in 14m 18s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f21203fa448b4a4e9a7b7c1be573f3f6 : SUCCESS in 29m 26s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/07175de9c5754c429e42a0d606f3d88c : SUCCESS in 12m 11s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e57ebdc2b354302822ac1b41e04721c : SUCCESS in 10m 29s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ea1b12ec48434e429d1e6ee472ce4d23 : SUCCESS in 15m 36s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c490794f7e04dd18008d2bfc12e7b0b : SUCCESS in 14m 29s","accounts_in_message":[],"_revision_number":1},{"id":"14d90831dda261ab25a9c7de7aed9048eefacb4e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 11:51:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6d17e1bd5587a7eb8f3ca39b340f8b8f320c456a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 11:53:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08a62a6ebc204321906c241470610329\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37e4e570ff2541c2a9f5269bcc7aa304 : SUCCESS in 37s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98e8dc37a35748b9a63583404588bdfd : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7c4307c90209a62908abfd69099c1c1683416fc5","revisions":{"7c4307c90209a62908abfd69099c1c1683416fc5":{"kind":"REWORK","_number":1,"created":"2023-03-07 22:02:26.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/90/876790/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/90/876790/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/876790/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/876790/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/876790/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/90/876790/1"}}},"commit":{"parents":[{"commit":"ee33d3a487ad24ae2709eb415d649df4cda20e0e","subject":"CI: Don\u0027t run Heat and Horizon in overcloud jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ee33d3a487ad24ae2709eb415d649df4cda20e0e"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-07 15:20:15.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-07 22:02:26.000000000","tz":0},"subject":"use OS python to create kolla-ansible venv","message":"use OS python to create kolla-ansible venv\n\nCurrently kayobe creates kolla-ansible venv using kayobe venv\npython3. There are corner cases when creation k-a venv fails\nwhile using kayobe venv created with python 3.6 buggy setuptools/\nvirtualenv command. Using OS python to create venv solves those\ncorner cases issues and preserves compatibilty.\n\nStory: 2010634\nTask: 47599\n\nChange-Id: Ie0d9bf895f6714cbb8b0bd31a008eb388f4c51c2\n(cherry picked from commit 068581456fd0928cd71c3e1222e63461ff82f7e5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7c4307c90209a62908abfd69099c1c1683416fc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7c4307c90209a62908abfd69099c1c1683416fc5"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
