)]}'
{"id":"openstack%2Fkayobe~876791","triplet_id":"openstack%2Fkayobe~stable%2Fyoga~Ie0d9bf895f6714cbb8b0bd31a008eb388f4c51c2","project":"openstack/kayobe","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-03-08 09:22:04.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-03-08 16:11:50.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"Ie0d9bf895f6714cbb8b0bd31a008eb388f4c51c2","subject":"use OS python to create kolla-ansible venv","status":"MERGED","created":"2023-03-07 22:02:44.000000000","updated":"2023-03-08 21:02:13.000000000","submitted":"2023-03-08 21:01:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"876791","cherry_pick_of_change":876757,"cherry_pick_of_patch_set":2,"meta_rev_id":"335b732d1dd0a5c5c892900845ddc5ad8d7c3882","_number":876791,"virtual_id_number":876791,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-08 21:01:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2023-03-08 09:22:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-08 07:45:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2023-03-08 09:22:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-08 00:27:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-08 07:45:00.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2023-03-08 08:00:37.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"9770bef6e0224f640eca9813f1da041a89ad2ea9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-07 22:02:44.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1a0b1bfc406bb7b29d3fa555228a126b593c87a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 00:27:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b0c2386ac13413891f33920b0680cdb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/573863f44dd84d33860f6b5b74842ecf : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/594ed061dd51493d865907ed30cba9ed : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3dd8446f16ee4269be8bc48749ff3257 : SUCCESS in 6m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1f0e8a4c5e384ddcbd60704a4431ebab : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d24aac2c3e342a68802a46b52a85e2b : SUCCESS in 7m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe67a1b789954fd09465496f1f2f9b64 : SUCCESS in 3m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f2ae3cf84b454b78b686f5b8f70a6565 : SUCCESS in 5m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c450bc3f44d54db0967140ee6fffb3c2 : SUCCESS in 13m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7cb3d8a12d8845d59ce63da6daf61789 : SUCCESS in 19m 36s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/46eb4319fba647e7add23b00462cab60 : SUCCESS in 1h 11m 12s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/c95509b4e61649b08a077968de1c5e16 : SUCCESS in 1h 17m 08s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/ae78e62e09be4f51ba5ed02fe0dee743 : SUCCESS in 1h 15m 27s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e5d37d8b4080468d947b111b6d12611e : SUCCESS in 1h 01m 02s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b982ebd7b34f41f582be9a0fb8d7e957 : SUCCESS in 56m 45s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b0ac13bbaf744b9aaf432b7e199c8649 : SUCCESS in 43m 28s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d42c55172ac54caca33d11a3e05c031f : SUCCESS in 24m 14s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/2ac40d62c895434581c1b8c0b0b92f3f : SUCCESS in 25m 25s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/4d4cba44985646a998a2e60dceccc049 : SUCCESS in 19m 47s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/378809369bdb4d5780c12833e8c68ee9 : SUCCESS in 23m 11s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1ca9cbf4d8a43cebe5a0c24e7ca59f2 : SUCCESS in 17m 17s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ba04229cd6d9438f8be00b5059673183 : SUCCESS in 14m 57s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/4b797401a8bd4c9e8c2c65a4b99cd1f5 : SUCCESS in 1h 12m 54s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/10006910c6fc49549014867e665c1752 : SUCCESS in 52m 41s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c523f4014f7048d78fb838bb42548f12 : SUCCESS in 2h 17m 21s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f31c2e37f1e64e8fb8ff2611e9401fa8 : SUCCESS in 1h 15m 47s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/1b643b74d20249c39fda2e397847bcfb : SUCCESS in 21m 48s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/5a57651a86e5484292855896891e06cb : SUCCESS in 16m 40s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/5cedc22b9794417cbc4b2593920c19d2 : SUCCESS in 18m 17s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/acff2b8080d849e6989060ebb0716e20 : SUCCESS in 15m 59s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4fde44aa2c914622bd17d5caf1c2a48d : SUCCESS in 13m 41s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d08e2788c42145719ad04db4b09dae88 : SUCCESS in 39m 10s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/699b9a32584941dbaad2a952fb38e885 : SUCCESS in 30m 55s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/bba5199bb7e740848c6cb234c67a7b17 : SUCCESS in 18m 58s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/d1c20fd22b304836906f9bec1eaaf147 : SUCCESS in 13m 42s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5cbd30f3aaca47489be95f23370aefd7 : SUCCESS in 21m 10s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec1f233e5b5447b2821ef6c63737ec9b : SUCCESS in 12m 04s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3aa0c1f30eb240b492853142dfc298f5 : SUCCESS in 13m 03s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/c547b6e528e54a6789f82562cb33d9d1 : SUCCESS in 16m 41s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/f7e1eaffafad43a39bdaaeb7e142c48d : SUCCESS in 20m 56s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d751782b39294a10bbba3beb025510d2 : SUCCESS in 17m 33s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aae55431c8324d33ac01fe504f45de0a : SUCCESS in 15m 24s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3861d22ffe7a44148f8801e216b1e56d : SUCCESS in 13m 05s","accounts_in_message":[],"_revision_number":1},{"id":"5a0aaeaa2c36996a3d34a7c099b974a629d0e1ba","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-03-08 07:45:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9131d77fd20f403de816edf013b6c01c541c2d6c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-03-08 09:22:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c81d028e8704a1c9cdd10a8f3bef7a99cd149677","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 10:03:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"df6ba7f1502eaf21c050a7fa8bcce3a2609a883b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 12:14:34.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6c33ba9652c14dbdbf9f32989f5fd288\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4461a1bbd4a04fde909bd3699d0611ec : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3805aafa3c284dd59c5e5f566b50614a : SUCCESS in 6m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/117a7f0a8035444082264e05fa09b4f9 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd833791dc634115afbfbfc75a1d3ca6 : SUCCESS in 11m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/08bb817e46b242899c8730d814934a2e : SUCCESS in 3m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/75d653bcbbee4df2b206ad3542d96a0c : SUCCESS in 5m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/aee4a72eea9e4e17a684698b0c6aadd8 : SUCCESS in 15m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/16d8fa906d644b9a8377821a805b0838 : SUCCESS in 17m 07s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/fc26bb2f9add48f5b53c2eced8bc815c : SUCCESS in 1h 20m 09s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/cab4b9a19a6e4b63a6753454f95d02b3 : SUCCESS in 1h 00m 44s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ddc5b2cd96da44c895e5e5134664ea58 : SUCCESS in 1h 02m 06s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5d66397f349426091f00b09ed91da76 : SUCCESS in 56m 55s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/80ceb5fe7f50450e8e0a02af3d9ce9eb : SUCCESS in 46m 23s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a5d04a631a49468cbafeee5dbc3a228b : SUCCESS in 23m 49s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/56c9e10355774d8887f8dc760d59c6c1 : SUCCESS in 20m 10s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c0b4bf2219b04db1b5befc557d765d6b : SUCCESS in 48m 49s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c81c3e840d5841bb98105e462cac940f : SUCCESS in 19m 02s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/acd59e3068264dadb7d95b2bf4f30a6e : SUCCESS in 20m 43s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/bd74be8b137145c795247c87f22fcea4 : SUCCESS in 1h 15m 35s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/7ebbd5dcce9141fb98142d2475b05f5e : FAILURE in 48m 11s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/805fb695643d4f97bb7bfa94a9f62a53 : SUCCESS in 1h 53m 33s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/681e989884674abfa49e67170cc2ab3e : SUCCESS in 1h 17m 34s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/5ab941667424479f9c4775b2d65a118d : SUCCESS in 20m 49s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/2a4d70059ec2434aa5a65ce11b741c9f : SUCCESS in 24m 52s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/6516514b207048e5aeb2f7c628f5d899 : SUCCESS in 20m 00s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8147419b54ca47cda7fa56155d5a7911 : SUCCESS in 17m 18s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c4e2a37f11044664b76d50bb53532b41 : SUCCESS in 17m 42s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e47437afe8284a72a981f3a194761d87 : SUCCESS in 36m 50s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1297a554da3b4a68a28f8aca3f6c12bd : SUCCESS in 25m 33s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/66742554cda945c4a6289abda3ebffb9 : SUCCESS in 19m 02s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/d9b4a362672c46569507c6fb2ea37411 : SUCCESS in 16m 40s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/43188c3e79784fd4a406e3a62ffa0786 : SUCCESS in 18m 21s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db0dc30f726d4b7c8b24ca033fd871b7 : SUCCESS in 35m 53s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b6ecfc0dfe7479990cd67c02c69dd4d : SUCCESS in 13m 27s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/9ca19b4fb3f94ac6a20119488f5b5334 : SUCCESS in 18m 25s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/2bba5bd531df4483ba6d5c9d1bbea2f7 : SUCCESS in 19m 49s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/cababee33a96460293ca92ca986f2921 : SUCCESS in 21m 12s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc7eac28f17f4784921b4c0a688e9b71 : SUCCESS in 11m 37s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52e88fac75044a9c84d3b71f7168f684 : SUCCESS in 16m 14s","accounts_in_message":[],"_revision_number":1},{"id":"2c0b70de5d04e5c2b72187c020d967d4635733e1","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-08 12:15:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ddfe571290a925683f22053237dd2581881e1e13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 14:07:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a3e378dc6654637802fa67f19b0bfeb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d4e0d4bcd9041a88414c34b347a17e1 : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ded309a1adf94d6db13efa5c2d2314fb : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f83f2e6d8fb349e293dce4c2869e54e4 : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0080cc22b8c741559804ec65180397d7 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/632664f24b5c40bda35b7547f3e083af : SUCCESS in 11m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/214cd207baf24dbeb4c053e2060e8dcc : SUCCESS in 3m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a9b0cbe921b04b6abbb7343cff631337 : SUCCESS in 5m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6d608d8df80646f38ea899d3cdd1b6b0 : SUCCESS in 11m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9eface3ae64a4ad8b77c39b5c94ae281 : SUCCESS in 18m 50s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/31f4f2858ab64d7eaf751f2ca289b552 : SUCCESS in 55m 13s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/9ebf3537b2b6475cb8c58851742aa7eb : SUCCESS in 1h 05m 42s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/75c662f754b3426696507642cf03b200 : SUCCESS in 54m 10s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/dcc2a115dec04681a4d22b87d3cb0e5c : SUCCESS in 1h 05m 15s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/deb792d581b944f5bb560775f246912c : SUCCESS in 53m 52s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3dde8e8ce4804dc1bfdc7be9be22c187 : SUCCESS in 45m 54s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/6b1c2d3a13164cb7ba10309a4c5587ea : SUCCESS in 23m 23s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/40e584aff0a3469a833b18ef9f89b0b5 : SUCCESS in 22m 27s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/199d7969413a419e946640a24e6880b9 : SUCCESS in 23m 19s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ff5a678edb4347c2bacf7ffc03d3ec05 : SUCCESS in 22m 45s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e383dc5af6284224a40d7f467259cc80 : SUCCESS in 15m 10s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2daa6b55737c4285a8138e1b71adcb86 : SUCCESS in 15m 36s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/45d3f8e2141d434eb8dd2690d0b14f6a : SUCCESS in 1h 10m 48s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/15b1d853045f442cb7163a4f47da6f90 : SUCCESS in 1h 09m 49s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/377a0ddee32e474ab88facfa35c4f07b : SUCCESS in 1h 45m 33s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/227cdd90d4214ef6bcd54b166a6d5e1b : SUCCESS in 1h 31m 30s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/f5864a45102846ed99cc72abb1856010 : SUCCESS in 22m 13s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/9f909fc54df645a9a67c3841043ce77d : SUCCESS in 19m 02s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/afecad37b05446d3a964e5aff839b4cb : SUCCESS in 19m 39s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f7bb6e22316441ebc52f45068fbc434 : SUCCESS in 15m 44s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/acf2a8bfa3f04cbd80af2112c54e5c32 : SUCCESS in 17m 35s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/da4ab9b13362492aac88f7f8806baa6b : TIMED_OUT in 1h 32m 36s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53e70cbd2a284a2b9a084f0bcb22248a : SUCCESS in 31m 04s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/355e899d3444483f86df07ab4c090505 : SUCCESS in 17m 39s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/2083dc48179d4456ad3d221a9e9e13bf : SUCCESS in 16m 26s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c323b5defbe8404faf682b85bd3249f6 : SUCCESS in 15m 49s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2acbc13c94c44431bc3eb0fa19e75085 : SUCCESS in 12m 42s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/751486e2658948ab94e1e830f91c615f : SUCCESS in 11m 39s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/4c165f1693824c8fa17a70cabb76dafa : SUCCESS in 20m 15s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/b1d1fd7fb0114e3a8d648864538fc260 : SUCCESS in 17m 20s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9287569ccff042259194b9aa220bacae : SUCCESS in 17m 21s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7528f20996c04cd6ad3e13cbb1dac0c2 : SUCCESS in 13m 17s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d7f22786db2e4d8eab53c1f3b007ef0a : SUCCESS in 15m 41s","accounts_in_message":[],"_revision_number":1},{"id":"9853b371e28cbb4aaaa6d223ec44215e693ca170","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-08 14:07:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b752fb4663146feae66282588bf090c510ea0fd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 16:05:37.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d27de3c6b80542c480e30391a2dd7a47\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d89358bb537f4908bae548f7d1ff7595 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a9f857d905e494c9ff703b80351a67c : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/790459aa79374a29aec23964b2555e81 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29cbeb73600347bcb715931614c58dbb : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bade72f911ab44c7a6a56efb0168108d : SUCCESS in 9m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7d7fed7d7cc48cc87dc04cc70a7821f : SUCCESS in 2m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9451f318d2fe483f86d1cb564971ba70 : SUCCESS in 6m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ca92777f0d72401a9069502c8f803e85 : SUCCESS in 11m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/12ecbdcd05dd476894809237c171e8b7 : SUCCESS in 16m 47s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5e7008ef0e8a4d4a99226983434c7a84 : SUCCESS in 56m 13s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/cad20890e90e4144a4a879ecd3dafeb7 : SUCCESS in 51m 58s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/ac92e7db7ff04cc484688ab0a9e8e7e2 : SUCCESS in 1h 11m 31s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/f5b870ad92e04e5b8076720a19c83be8 : FAILURE in 47m 36s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b30dc81b0f24fe997275beeb13a813d : SUCCESS in 44m 58s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c56955123ad4726a487661804cd134f : SUCCESS in 58m 51s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c8166eed45fa41b8abc40aaecc6b8fb6 : SUCCESS in 24m 23s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/a0c1da94736e4f169b89dc1cfcbe9b17 : SUCCESS in 18m 50s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/f6304eac65d849c0967a3d38ceb1bc0e : SUCCESS in 17m 50s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/86d0b3c0e6e74ba4871415fd63f3a5df : SUCCESS in 21m 10s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab9a6e42c76d4249a707906c132d8a73 : SUCCESS in 17m 07s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/846e22d2305c4cce8847c99bbbeaa463 : SUCCESS in 16m 43s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/a5baf0d9f0dc42638aa7a4dd18b473c0 : SUCCESS in 1h 17m 14s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/8745912a751a4f4f934d012c9bab9bdb : SUCCESS in 1h 17m 30s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b195d57450804553aa9edf5ef4ea7e70 : SUCCESS in 1h 53m 12s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ca268e894fb48a7a7ccba4a420eea18 : SUCCESS in 1h 33m 54s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0246af244ea54ad0b74203d985fedea9 : SUCCESS in 24m 34s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/092cc571fb9c4863909a20086bfd1013 : SUCCESS in 21m 44s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/853e57da3a6846c29fe548eb2c89303a : SUCCESS in 19m 41s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9145f60d955c41caa4732fb7827c2484 : SUCCESS in 17m 54s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e19e6b129d24f179539e18e46cbe44a : SUCCESS in 19m 33s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bd57e281d3524996aaaf8803acf68a18 : SUCCESS in 40m 34s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f90d720463b4c92b845d3eb041a0d13 : SUCCESS in 38m 12s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/24facdc8a70441b99749870cde3e68d2 : SUCCESS in 18m 43s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/d6db9ec1e309435197d55d77b6f89bea : SUCCESS in 16m 17s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e9d87f89109942158af99bf945978248 : SUCCESS in 18m 00s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66e5c2acb6af4c0ea5d99717c959dc29 : SUCCESS in 16m 13s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7f97d9cba69d4fcdac423f46fbea02d9 : SUCCESS in 13m 43s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/9635ac4307c84151ac2b2ed3c55924c4 : SUCCESS in 19m 11s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/df129d5e80e14e8c8ad685a9199b0fd8 : SUCCESS in 23m 54s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e2c1aff87f8941589c72a4c40fa94ab9 : SUCCESS in 17m 55s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a2f30c19ea32436a83c23c34e8abcbea : SUCCESS in 14m 56s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/80dbb5b413b44fcdb11e88d54012d0ad : SUCCESS in 17m 20s","accounts_in_message":[],"_revision_number":1},{"id":"8f63845438946e9525c8aea53e61355f0c8168eb","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-08 16:11:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f6dca46120560ef875f86b212b90d35a2b0d0884","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 18:22:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de58a06767114d0da330b2b9fe4292ea\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a756457d18b54c808539b71b0d895bac : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e813613b7e6481e9254ef9a2b43c081 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2facdc4cbfc4996929ea98993fca062 : SUCCESS in 6m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f2ee598ebee433d8490ee11e245bb1e : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6618e9e17fc745feb288cab8e355cb9e : SUCCESS in 8m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3688252555ac4cd2a448169c898c20f9 : SUCCESS in 3m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/be3f8a26ba224305a48c1e1bba017325 : SUCCESS in 5m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2d59f6aba9794d00ba9a8c27ee1f0a0d : SUCCESS in 13m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7e810924a7f34a969b9fb941bf7c3ee9 : SUCCESS in 19m 49s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/cca73991ec54457294eb5b8a68298455 : SUCCESS in 1h 04m 45s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/363d85f4e2454969b2fa9d6e89ffcffc : SUCCESS in 1h 09m 39s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/b43ef470684349bea047deb316e49317 : SUCCESS in 1h 06m 46s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/321075cc67ec4f7d9baeb8d1d9baadac : SUCCESS in 56m 15s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e1df9df0b344250aa46a25d6752b36b : SUCCESS in 53m 20s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7d39de7b6ca64085bd35d207b41435ac : SUCCESS in 52m 39s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/56a08c8f49934a6aa2aabc1465d3fc50 : SUCCESS in 25m 15s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/3fddb5253b844c599b915f73c14fcfb8 : SUCCESS in 21m 03s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/0a06027f790a43a4a8a1bea8f6d81029 : SUCCESS in 21m 47s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/fe57346955b24c84afa9345b96744fdb : SUCCESS in 20m 25s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d9a82e2f66c496aa8cff6f4aa2d2cb5 : SUCCESS in 14m 50s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a3e0e21dc6749a79357c4bcc4aa61f0 : SUCCESS in 16m 29s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d43d37c24ace42b096621999925d610d : SUCCESS in 1h 09m 08s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/4a6bd7f2793c44d58e7d5ac183db6117 : SUCCESS in 1h 10m 51s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6b153a0b60ad441db4d6c83b763ecd54 : SUCCESS in 1h 43m 39s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f561fd976ce14f35be7503b4998e84ef : SUCCESS in 1h 10m 23s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/33a7c9c1ac7e4cdfb0eaa945062abfdd : SUCCESS in 22m 31s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/15cb9b5f417243509850535af567dc32 : SUCCESS in 19m 37s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/6fa70adc37ad46bb857bcdf65cd33afc : SUCCESS in 19m 39s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5946ea513c494d2ba5afdaccab30a5e0 : SUCCESS in 16m 44s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fda669cd41f24b9ca789145499e02424 : SUCCESS in 19m 36s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/33b14ac395534eab9dc2738757d686df : SUCCESS in 41m 52s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4b694045fda44c5ae53fdb9a99138d9 : SUCCESS in 33m 47s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/9dce072c5bfb4a67a90da6a07b1c983a : SUCCESS in 18m 35s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/58ef90e6f8b3433d9a6862e70443c4b2 : SUCCESS in 16m 04s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5d6043437eb4444e95f8a7c3aa3cde5c : SUCCESS in 15m 29s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f7f330338eb4ff8a2bb362745aeef6d : SUCCESS in 13m 37s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/150bfd86e55b4db6b05eb62e73e13031 : SUCCESS in 13m 18s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/9d9ff404bdea4a54826e9ca1176257d6 : SUCCESS in 20m 48s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/6440ad4b1a8b485da1630244a3ca6bfa : SUCCESS in 18m 14s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1bc3880030ef441c97124e4ffc429885 : SUCCESS in 17m 26s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6deb6731850446369cd022d8d2b45189 : SUCCESS in 14m 29s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4fc5a1c4c2d84c9092f35fa3af9e2b16 : SUCCESS in 14m 41s","accounts_in_message":[],"_revision_number":1},{"id":"abfeb13bc9e9a5d70515b79ba347e6f396cf11ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 18:23:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3e5c2bbad875ef0159a83fb08a8c2c8284c68205","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 21:01:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd27ee326d3c4258a3ea7c2c227a5a23\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa9ae0e863c640cfab1292855ef21bea : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4311449f3f2f48a9892d11e83b32e647 : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2f24f33d33442fbbf9f96c11135338a : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fe07fdca9a9493cbe1bb8f3fc1829d8 : SUCCESS in 10m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c94642dadab145d69d8c314b709b871a : SUCCESS in 6m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4346dd9b155e41e4b7087740e803b457 : SUCCESS in 6m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9e44a46bef3b40988fc246b8a8d22b68 : SUCCESS in 11m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/04dc1722b6494eb18dffa27b0e292c90 : SUCCESS in 17m 47s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/329617008bcc43ab85461c69802b63f3 : SUCCESS in 1h 04m 53s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/3c65983969244eb19878c114c9f3c2b8 : SUCCESS in 1h 04m 25s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a2b3dd34f96441339fe62bdce84e28cb : SUCCESS in 1h 00m 39s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24551c84c5dd4533868eddbc03de37d0 : SUCCESS in 50m 52s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2552670c62424a07a49cb67dd152ccc7 : SUCCESS in 56m 16s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e9aafd7fc4324a5b9da70540446214a0 : SUCCESS in 26m 02s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/63c0386b451543dd9febeb03e82f1b6a : SUCCESS in 18m 16s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e817720d284846acb6c09261257df849 : SUCCESS in 21m 52s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d2b00d9183b4762aaef4a0336740e33 : SUCCESS in 14m 38s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7757c5618d9d43c3968fd2586a13c734 : SUCCESS in 18m 43s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/439155ed29de460ab334619495f41000 : SUCCESS in 58m 24s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c62980c3215647699e8cca6eb82f1a27 : SUCCESS in 1h 02m 58s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fb177449c61447e9aec1f1ff4cb970fb : SUCCESS in 1h 54m 04s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da0142fba9744c7f908eb1e206263ac8 : SUCCESS in 59m 28s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/fc32891876484404bf73935401758f91 : SUCCESS in 24m 07s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/8da32a3807144a2d901c307ee3621c54 : SUCCESS in 20m 07s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/9d346f20f02543c5ad44c0a58d310fc7 : SUCCESS in 21m 22s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ad52137a8604bc0b135b970a4216b95 : SUCCESS in 15m 51s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e176985b71940c98bf2390605c135d8 : SUCCESS in 16m 25s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0affbf95430a4d1f8c258021783c6934 : SUCCESS in 43m 43s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f95ca901743413db7bfa92706eccc1b : SUCCESS in 26m 48s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/3685a172513142c580cc895ad56335d3 : SUCCESS in 19m 03s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/70e5f5616de5409db30fd372afebca40 : SUCCESS in 13m 48s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/08acc53faef8417eac225a70614fe61a : SUCCESS in 13m 58s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6dee79f078e347cfb2422b459497927a : SUCCESS in 14m 47s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67612d6d312d4153a02567c9f34757ce : SUCCESS in 15m 12s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/ca04e428d24244d2a873a25e87ea5d42 : SUCCESS in 17m 22s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/b8c4ea4a382c412899227b0532b74674 : SUCCESS in 19m 59s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6ac2881741ae44f7b35ad98388264b3d : SUCCESS in 16m 02s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b2abcc3544d48359c1ed3f53e663e01 : SUCCESS in 19m 10s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1122b9a1e577438b9ade8f55554ed831 : SUCCESS in 16m 59s","accounts_in_message":[],"_revision_number":1},{"id":"2542de3f0fd8facd0cac361ee2d9eb7f17a365fb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 21:01:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"335b732d1dd0a5c5c892900845ddc5ad8d7c3882","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 21:02:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3835b382aa4543b2b06985ce8979afa4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31e7d09366f44fb9baf6397cf40c4c1e : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d816d91aeca44f37905bc0032d80d604 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5a92bd778be8709977baa8ecc20937cc4a140360","revisions":{"5a92bd778be8709977baa8ecc20937cc4a140360":{"kind":"REWORK","_number":1,"created":"2023-03-07 22:02:44.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/91/876791/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/91/876791/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/876791/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/876791/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/876791/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/91/876791/1"}}},"commit":{"parents":[{"commit":"5b35bbcb6a07c3eeb6d813a981bbb2823a89e898","subject":"Merge \"CI: Jammy followup - move os_release\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5b35bbcb6a07c3eeb6d813a981bbb2823a89e898"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-07 15:20:15.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-07 22:02:44.000000000","tz":0},"subject":"use OS python to create kolla-ansible venv","message":"use OS python to create kolla-ansible venv\n\nCurrently kayobe creates kolla-ansible venv using kayobe venv\npython3. There are corner cases when creation k-a venv fails\nwhile using kayobe venv created with python 3.6 buggy setuptools/\nvirtualenv command. Using OS python to create venv solves those\ncorner cases issues and preserves compatibilty.\n\nStory: 2010634\nTask: 47599\n\nChange-Id: Ie0d9bf895f6714cbb8b0bd31a008eb388f4c51c2\n(cherry picked from commit 068581456fd0928cd71c3e1222e63461ff82f7e5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5a92bd778be8709977baa8ecc20937cc4a140360"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5a92bd778be8709977baa8ecc20937cc4a140360"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
