)]}'
{"id":"openstack%2Fkayobe~876807","triplet_id":"openstack%2Fkayobe~stable%2Fzed~I0f7a3dcb021f726693725f070d9a2ff80d84fe0e","project":"openstack/kayobe","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-03-09 09:55:41.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-03-09 17:12:35.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I0f7a3dcb021f726693725f070d9a2ff80d84fe0e","subject":"CI: get repo directories from nodes for troubleshooting","status":"MERGED","created":"2023-03-08 16:22:57.000000000","updated":"2023-03-09 20:40:35.000000000","submitted":"2023-03-09 20:39:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"876807","cherry_pick_of_change":876831,"cherry_pick_of_patch_set":1,"meta_rev_id":"3171a46e2dc15d4a58fd409178337a1edf96c80f","_number":876807,"virtual_id_number":876807,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-09 20:39:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2023-03-09 09:55:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2023-03-09 15:57:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-09 11:46:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2023-03-08 16:34:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-09 11:46:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-08 17:50:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-09 09:54:43.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-03-09 11:46:39.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"bc358d4019b75457706a2261c666394e1196b4df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-08 16:22:57.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1820afc572c6487e90c977363c27a98a98468820","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-08 16:34:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"52b325dc24447c0bb7b9759d6839cb0fd25bd50d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 17:50:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4002b5dffed64b7b8e4b3c66c25b7bb3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f24fc079f51b43f39fd4eee14c350fca : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4878bfc562c4c95bc5d253b3092b90d : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/225938d6afa14d03932684a82d472b7a : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ffd96d9a131a4f618ed90d978a8a3441 : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/11541c6f9c494d269fbeb14953f14181 : SUCCESS in 3m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da2a01cd657b408bb40c992fcbe39ec6 : SUCCESS in 6m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7b1ed608d48346718c679fb11eb214b3 : SUCCESS in 4m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b11f8b2ec2db4d7087a54ed182d51eda : SUCCESS in 12m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c7f0d331278e42ad84df5f31a8eec1e4 : SUCCESS in 14m 45s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/4d124382edd742c29efeac6f566e2ab2 : SUCCESS in 47m 33s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/1a3215ee76f345c1aea3095cd71e6000 : SUCCESS in 48m 09s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f525f2ed19054bb193cd3523fe939946 : SUCCESS in 53m 34s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/8dc31df740044fb393b4c30fede6801a : SUCCESS in 1h 00m 59s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/df49a33e029d4c19b57e19c93e3150b4 : SUCCESS in 16m 09s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/29d9ff2f43254a32ab7d07df033cab44 : SUCCESS in 19m 47s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/48036edfb26347688edca2ab19c50516 : SUCCESS in 14m 15s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f15adac3883d4153a9b369977f699f5c : SUCCESS in 1h 12m 18s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/aecd933eac9740dc93bc3dd64af9f102 : SUCCESS in 17m 10s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9ba03b1f1b94a44a13f35e369e2a325 : SUCCESS in 14m 32s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/8932caefa0ad48339ef0226c4c3f956f : SUCCESS in 32m 17s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/edb576934178483ea4ecf9c006bd6d32 : SUCCESS in 35m 09s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ad77bbcefaba45689c2f77e27a1a8c9c : SUCCESS in 13m 03s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2e2a6f0d1f0c4210933b0d73f497e3a2 : SUCCESS in 11m 45s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9b7f149606bd491684a7d0a15f4a72c6 : SUCCESS in 17m 11s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c33eabed82fe44a8945458ae8efb7ed0 : SUCCESS in 15m 47s","accounts_in_message":[],"_revision_number":1},{"id":"c25df1e896a24d276acdd04954e219eb7a815255","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 17:51:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8ddf64151d8072460dd8e452d34699a8c654eb65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 19:47:44.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/16d8903d824f4f9faedb607d69bf254e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f962e2b88e8640fb835898c678093d50 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d01e75fdbc5426d87bc2dfde218dae4 : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/105aab5b848c441b989e248bdb27757a : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad1f167c5d6a4a139c865d74e135204f : SUCCESS in 6m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/855868a9a244416ba768984e79268006 : SUCCESS in 3m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e9fc256485cb47cc8668cca5560a1da1 : SUCCESS in 9m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bdf35092f95a4966914c8517117b648e : SUCCESS in 14m 31s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e64562e4dfe544fc9af1f45a31065b6c : SUCCESS in 55m 20s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/04e534c7d04d48b0b6a94ad47e5f94ac : FAILURE in 35m 25s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/f988b48c782f46198dff56be5e890c39 : SUCCESS in 1h 38m 45s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/688b9aa8b0fa42179835f67ce1eaa08c : SUCCESS in 12m 48s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d13cc7a716e4eb29c11c7518bb18455 : SUCCESS in 15m 55s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a823138d85dc4079b31a52fb1e3b8d70 : SUCCESS in 1h 09m 09s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/2286a87456f64f73bfb644975d79f772 : SUCCESS in 17m 32s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ff64996169b4e3594faeb12de759893 : SUCCESS in 15m 02s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec3f587985f645fab1eaf488a3375a2b : FAILURE in 28m 40s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/885ecf05a9b34aa684094588465b5bb7 : SUCCESS in 16m 06s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/587d9a0c7c484f7a8bd9656f695cbfac : SUCCESS in 11m 41s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/09c2675afcff4427b6a5dc3db457db4f : SUCCESS in 27m 01s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5823cd357b014bc787845c1f62c957a9 : SUCCESS in 14m 38s","accounts_in_message":[],"_revision_number":1},{"id":"40f4ac75b1d9ff3c6e3ff2784458837886c45415","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-08 22:16:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1890a210d90464c12194f5afd3236dbb1e273d62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 23:42:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39ae01b6989e460fb915cfcc5f46d694\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/acaf0bf4261343898b7d341843254d85 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97794423c56547aabcbb31660d8406a0 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83d1c8b540b5416ab30fe0bffd9524c2 : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/284e43b272dc42b3a0786bbb98aace05 : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/07a4ee01430b45a68cac5d70e6c9e3dc : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91c46dc23c184c968c8387ac0b4498d6 : SUCCESS in 7m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a2d21884a5e9446b8a6645dd750e7162 : SUCCESS in 3m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a0fc4156b2d24cb485bbf275a7ef7869 : SUCCESS in 11m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f21213a3c94c4be7bacb90a758adc8e2 : SUCCESS in 15m 50s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/2b9cbb2158ee4667887a0d8c47a33c9d : SUCCESS in 53m 07s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/6063839e26a34227b762e4047cbba2d6 : SUCCESS in 54m 19s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb3b3decb57747dfa39c204f22dbab8c : SUCCESS in 48m 21s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/34c45b100ff6455594e781f8475005e6 : SUCCESS in 1h 19m 38s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/94b3689bc5e04d4284998f15dd7e2a73 : SUCCESS in 17m 07s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/8a28379444cc450283b538b046ff142b : SUCCESS in 16m 51s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b5a4ade3e79401fb6749840a5c51bcd : SUCCESS in 14m 40s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54b9f2866e6144bc92939e4fcc496022 : SUCCESS in 1h 11m 29s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/d3ee9054a25c459391eb28a21bbbdacf : SUCCESS in 17m 09s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ccd153ac5c9445cd924915b916539703 : SUCCESS in 13m 20s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/d357c493d0884ed4b0e4adee25773962 : SUCCESS in 28m 45s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a4d410759ef47cab97ce00e6ded644d : SUCCESS in 29m 17s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c8f0847df48949938976d0f381026a94 : SUCCESS in 14m 15s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/85c71ea5b9bd4f1ab91c00e1d3def857 : SUCCESS in 9m 31s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b7beaeb3f53642498d9cc2f0c26ec2b0 : SUCCESS in 16m 10s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d5657d6a25124159b77bdd44dd638552 : SUCCESS in 15m 26s","accounts_in_message":[],"_revision_number":1},{"id":"627eb3e825076399149496f680d7556e1a15cb7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 23:42:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c3f8eafb47d3f410ed25d5d1ece2a3ef6368c4cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 02:46:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d6de8617fff409db066e94936f3f0b4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2255b29210fb4964b59f8dccdf73951a : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a4890ce24754f78a8624ca09b8f1448 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/17f3692d3f37473982d6db4be241926a : SUCCESS in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69a51cd1b8a243c98d87f929fa0d0bed : SUCCESS in 8m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3cd51e0bccef4e8ea053ed680f7e53c8 : SUCCESS in 3m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0a5c7eba22e64654b6bee912eeab2c71 : SUCCESS in 15m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/190209a52c1e4f0f943b477870fce24a : SUCCESS in 15m 21s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/f8b20787375c44f9ba4ac5cd9cde84cf : SUCCESS in 57m 22s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/373a8da941b64348ab2b3f0f1d199266 : SUCCESS in 49m 14s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/403d012cc1394524bb8b8a5d69f3be0f : SUCCESS in 1h 00m 31s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/cace9cbdb92a42ab9841f993d7f43c8c : SUCCESS in 15m 37s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/00104ab4bff14cf7beec2d466dd80aeb : SUCCESS in 14m 13s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e50dc6d5d6e41a79f65264ff41ae53e : SUCCESS in 1h 08m 47s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4cd6b9d03ed8462fb8ed9d00eda3f2bc : SUCCESS in 15m 53s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7886f9d43513422cacf72b5b99376c44 : SUCCESS in 12m 22s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df2711a995a44340b3199159d89cb5cf : SUCCESS in 32m 38s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/df65b12e2d1d458fa7abe0dd36259fa0 : SUCCESS in 12m 46s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fba7d62ca85c4cf480660b1d63bf5ead : SUCCESS in 10m 12s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/69d241e3345246eea08a062e4e03e135 : SUCCESS in 13m 58s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66e44a2b05244cae8f9e8f440c621a64 : SUCCESS in 11m 46s","accounts_in_message":[],"_revision_number":1},{"id":"09c20d238e139de2be18c74876ad4aaa6448987f","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-09 07:57:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0b724f768bb622c0159291329fc2cac00cbc4462","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-09 07:57:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5ec3316fd82f23b627dc447324871684c6c71e91","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 07:57:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"33d3501dcf68d9fc559748153fd637cce45e2cfe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 09:17:26.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/65190d23271f4976aa6f72e1e9936833\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2007c30f5e6348e3b97a1e77d95aa2b3 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/513cd58de4004345857fa12e5e60a197 : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2dd166b1c61d4d53a0f317e040d98ab9 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab8c4b3757e842dc8c6b12932abfadd2 : SUCCESS in 7m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a7af28adbeeb4a87b6775c33748754e7 : SUCCESS in 4m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cc47ec1318854c44a5f727695b5db945 : SUCCESS in 16m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/64e6f3f4c1da493ea7ed8c06d00757cc : SUCCESS in 14m 35s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/508d91d217744a64a8c7933328e93323 : SUCCESS in 52m 59s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c9628cfb2b174b99bf0f88589f9f074d : FAILURE in 37m 17s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/fc5afaeba95c4b969a68c457bfae28c4 : SUCCESS in 1h 01m 05s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6b20e102f5004634982b10f3d0dfc8fa : SUCCESS in 17m 08s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/597c25d549904f71a720e677af68426f : SUCCESS in 23m 55s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec0ab0a8b78f43e58ffb32da70b8dab2 : FAILURE in 52m 47s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/749ac64954cf4c7092d92cd50ae5a342 : SUCCESS in 17m 30s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07975e05bf0d4867867f82fcd2dcbc6b : SUCCESS in 23m 58s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bec751159f424f7184a0f6d15886aa3a : SUCCESS in 36m 16s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0bd66a4689ce48099bd06330be67cde1 : SUCCESS in 12m 18s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e762be3854a402ea496ba9ff22cec0b : SUCCESS in 20m 23s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/00ccf0b32cce49b7b1ea969b85441030 : SUCCESS in 17m 33s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/93db087dbfb24a6192c38642200a3267 : SUCCESS in 22m 21s","accounts_in_message":[],"_revision_number":1},{"id":"880b226d1045659e5897115fd3bd40acde9b3fca","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-09 09:18:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5dcef08ad11be2e6f9e0985e8a194caddd5ba122","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-03-09 09:55:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8c4b494c6526c3fc43276c677b0536ce9696ed30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 10:29:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4c61de8b5054709a62ba83d545a2106\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a65965facffe468082b92ddd381920da : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/697e8c7767ab474cb91c1efc633f3f54 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02012203f6ed493cad407e43f86a7c02 : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8193377c51224a809e6fbafedf3acb6d : SUCCESS in 5m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/619d5da3c80d44e4aa3cc0e3eb2700c2 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d21fab3571dc462eb5fda2d13ece2ee8 : SUCCESS in 8m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b7f1118824cd475387360583cbba233d : SUCCESS in 5m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/805c80d406dd45b18c0f896f69cfb4f6 : SUCCESS in 12m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4eca5097c22942428c129a33f00a24ff : SUCCESS in 16m 48s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/0399fc21af224299b4fbce3f22caee7f : SUCCESS in 51m 46s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/29459b24e2134fb0aa85873c8fc72fd9 : SUCCESS in 45m 14s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/120f7ddd7273468c813a0818fbd49815 : SUCCESS in 49m 09s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/01d7a895971a4a119fc7055bbd54cb2b : SUCCESS in 57m 30s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5e195facea05462cbcba87620be7ee40 : SUCCESS in 18m 04s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/217da02d45db413bac4149665406e05f : SUCCESS in 15m 42s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e2de93ba7804d65b5c04ce01f7dd73a : SUCCESS in 13m 13s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e893867b1804fd7afd5ea41e29183bf : SUCCESS in 59m 48s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/9745cc65268b4af4acd52faa950fdc8e : SUCCESS in 16m 12s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/142d0622ebfc4510a0baa915466b6fbf : SUCCESS in 12m 26s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/0c6865a0f95b4b0cb620003b79fe37e2 : SUCCESS in 31m 05s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ff63e2ed486412692708fe89ddc7333 : SUCCESS in 33m 40s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e28a4318cddb4d38852857bbaa1f6c49 : SUCCESS in 12m 09s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/34f7d54d4b2b40a79ade08c4ec28477d : SUCCESS in 11m 21s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e579c22b8c394456bd573a50e2e7b05e : SUCCESS in 13m 31s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/382b1b86fbd14c8798a879c78f41760e : SUCCESS in 13m 33s","accounts_in_message":[],"_revision_number":1},{"id":"4c3c28fb0d51fd3542a4cb3334c2a36d345494d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 10:29:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a0df24d41e7d840bc05124e0a11a0f9868eca7ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 11:43:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f0a350854444c53a9b4d59d9ffe98c8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8da937fb0b784a4f8d2a13072270f4de : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e99760629dd249e6af3166de3d0a8adb : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97b9ac42ef77477e8269365da1931acf : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c60366ff6f814e7ea3bf1aa8fed00825 : SUCCESS in 6m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1e3f6891668c47f8bc15e74b3aefedfb : SUCCESS in 4m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bec41e8dd37b4647bf111614e2153582 : SUCCESS in 11m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4b5e601ef6a04f4784b7c1bae78e242e : SUCCESS in 16m 37s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/98c8a62ee73a4ff9b77abc231c8540e1 : SUCCESS in 51m 01s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f2d338fa936544c4bac608ecf3f37ec2 : SUCCESS in 51m 34s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/18dbff667cf948fcbbef26ffaca3c68d : SUCCESS in 53m 06s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d8e35623f6dc40ddb89d773b2ca0def1 : SUCCESS in 17m 00s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38eeceb87b614f2da28d9821d2aee1cf : SUCCESS in 15m 03s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a4ad9386c9a41fea955e018c21bc573 : SUCCESS in 1h 12m 26s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/57344e2024f24d01af4dc1acc6965fd3 : SUCCESS in 18m 33s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3816b46062c44113a4c4bd2e21988a5c : SUCCESS in 16m 31s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e258390a231c4a05894984fbff61d1e2 : SUCCESS in 32m 42s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fab6b7e674164ab2989a34e049af2e91 : SUCCESS in 12m 26s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2bf40560af27427d99c154a4a6099c60 : SUCCESS in 10m 46s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5ae153efe36d4508b7a991c7db7f8548 : SUCCESS in 18m 31s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b74d0e805f1d4b3f94c4287fea4ed18e : SUCCESS in 15m 23s","accounts_in_message":[],"_revision_number":1},{"id":"580c3e11e97ab88b9a6fd28bf834df842c9387bf","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-03-09 11:46:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f79db3ee04f19fd5d895de8e26e89086f109fffa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 11:47:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"080d4b38c3a849e408391988bf169c01d46c48cf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 14:58:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/793670ed47ff453297e88e6ca1ea3414\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/265de9dba23d465ea53fb0e1dfd94e74 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4165ab334b774dfaa47b827a8d9fa5a8 : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c763505dfbd4470b835695ba52afa8ef : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c246e827055c409797aab70a27e17a4e : SUCCESS in 6m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3c4cbdc7a1464ca0ba3040ebc4283bc7 : SUCCESS in 4m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/adedd730720343d8a96db9502ad17292 : SUCCESS in 10m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a5b38688e9734846aed20dc23ea9cd33 : SUCCESS in 15m 49s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/81d0e3e5eaf44923835908b0b8d04b1b : SUCCESS in 57m 46s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a540eafcb8744abb3dc1f9482fdad42 : SUCCESS in 51m 06s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/b639405ebfca417988910a44d74a6670 : SUCCESS in 1h 03m 33s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/abea9f55b8c04c3baa0c9e73fd10a73d : SUCCESS in 15m 46s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/04a845d425f9421a9860e82be9df655f : SUCCESS in 15m 43s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/930f01b377074a7c9a158423599a52cf : SUCCESS in 1h 08m 30s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c0e395b63a734c40a21e2065dc52bc4c : SUCCESS in 16m 10s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d4062c8cbed7420e906f0c917a4a6c9d : SUCCESS in 16m 24s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e656ab1a53b4c4f8460d6c5fabacecc : SUCCESS in 32m 12s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/923085a4ea504cc286024fcdf5509c1c : SUCCESS in 11m 57s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4fa3ed7e58e34a748d5c6180d35f4cff : SUCCESS in 9m 47s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4ee376ef52e24d199bc4cc3b145e8af3 : SUCCESS in 16m 21s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bfd7f0d359684b8dbc39833d97752fc9 : SUCCESS in 13m 18s","accounts_in_message":[],"_revision_number":1},{"id":"c418aaed0e8a2957dae0e356a7f29be4015511d3","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-09 15:57:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"19caf824107a3b705a41cee5debb10028af2fd67","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-09 15:57:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"618385e305c19ccd31808ae9f0c091b59f2a96d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 15:57:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e71aae0b2accc5cdf593e1252dc650a43ab5e993","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 17:02:29.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b4b3354d455d4c9db871929b1c61a94b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93a2f032d871436d91590565d6dd1766 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff844e35098446b2ade2ffe2d3cc2a95 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f34db1ebe7a945519dc27079626ffe7d : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4bad5d1899b4f28a76f2b553f416044 : SUCCESS in 7m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e3f40db8624b40a489ace604484aae6e : SUCCESS in 4m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/84b6e9031d34463e807dd94d9731a5a2 : SUCCESS in 11m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dc8809d489e042c091623b90ad70850f : SUCCESS in 15m 34s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/bcf403ffb9d545c7bd6cfbdec9153686 : FAILURE in 17m 57s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee5b7fd0f41245b5a59e07ff728d711b : FAILURE in 20m 35s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ebbe2e6b281b4bd58d9a14883e13ba69 : SUCCESS in 1h 00m 19s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d59f2daaeac54521bbfb33327d3fc59b : SUCCESS in 16m 46s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ebf855dadd2942999425c5fdb4698e07 : SUCCESS in 17m 14s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0668b8cec83741babf454a1668f12bc8 : FAILURE in 42m 26s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/68cef6d533d44fa1ad0be77d3f99082e : SUCCESS in 17m 20s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2989db62112a4c14bab507739821397b : SUCCESS in 20m 32s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54ada366112c44dc8b7fd87366b87b49 : SUCCESS in 31m 33s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7ee6f4a1441344b1bc0e66eef8d2483a : SUCCESS in 14m 29s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ec06241f302b448495142ddddd356e69 : SUCCESS in 14m 26s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7bf0359970744f32a7c9fd7dd3fd24be : SUCCESS in 17m 01s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd9601a8ed194980b643d9eafa6e30ed : SUCCESS in 16m 42s","accounts_in_message":[],"_revision_number":1},{"id":"0303e4aba3e0993dc08efcf5d472cf03cec73090","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-09 17:12:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cbfcaf53bd204f285678ba372e68e8ef7bb76f13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 19:24:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b47138b4b0c74e92a6edba9a3742941c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7c2c6d2acd842108aa11fe1c07f8ab8 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50eb43d21f2e4ce5b1fab5a74cd81839 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e4ee392531340ae979894c132cbe189 : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a4134e564284028bba3effb9fefd475 : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/31285d192a32418a9af2ff7a1cec37af : SUCCESS in 3m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b9ecbc71d3a44f1b96026dae7b7ebf6 : SUCCESS in 5m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/528b557736244698b9358ee83a518db6 : SUCCESS in 4m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/37be576693cd4f9b9c2c0290ad7f11da : SUCCESS in 11m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cb34df2312124ef89e6f26daac18618c : SUCCESS in 16m 57s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d9b02ca9011b421baf4ba2489321fbf4 : SUCCESS in 59m 37s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/7ee68d03008f4be1bc2dce71a638178f : SUCCESS in 55m 08s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/463c63c63a1a48e290e3a9d3484e8b8f : SUCCESS in 46m 47s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c34ce83a41ca4cf8a03eae63a405a806 : SUCCESS in 1h 07m 29s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/95c0fdbab44c4ebfab3b76333302883b : SUCCESS in 18m 26s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/ae0042a6c3f54e06bea8b2c06a321516 : SUCCESS in 17m 58s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/72f26eee4a7c49deb4923ceae320b5ae : SUCCESS in 14m 02s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e5d9587c8254923a0ec1bcdd79d6add : SUCCESS in 2h 04m 35s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f01b0db7c39140fcaa4bd93057c86f56 : SUCCESS in 16m 31s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e334026fd7a74ff3bde585921de7623a : SUCCESS in 13m 20s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/7a5ca694cc40443fb8f98e38a9cb6a89 : SUCCESS in 30m 03s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9965979fdf6442d6ba83dc7f67dfc098 : SUCCESS in 31m 15s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a13ee78ac9ae44afac6ecb349bcef18e : SUCCESS in 11m 47s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce5ee8431d4f437caf66320262b190a5 : SUCCESS in 10m 39s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/862c2f6abeaf4a1eb75374b668eeba7f : SUCCESS in 16m 51s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/334c805b5f344677a5070a99b119af98 : SUCCESS in 12m 55s","accounts_in_message":[],"_revision_number":1},{"id":"122ba0c4870a79a7fafde06c607976dfe017e8ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 19:24:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"431c342cfb52413c227e5fec7bc7cf60087927d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 20:39:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbbd05a0509741ba9eb4ac8d450f60b2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90dc6929056f46759adac0005713d52b : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90d740ad78b94f9d906e3874ee20d1e0 : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b04a91f93c34f2e80003e115549ac46 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d9e250ae5244af5a123b7b255a2242d : SUCCESS in 8m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/af5c9bbb455640c18d4b1ca466c21165 : SUCCESS in 4m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/870c4406a84041d9b5fdce46cc0b016e : SUCCESS in 10m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c28a4801bef9457895a97674e0aed408 : SUCCESS in 14m 12s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d76a2336018044e8a9bfe16d8e53e999 : SUCCESS in 53m 27s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd2bf707d61f48c88644e79e3d20f333 : SUCCESS in 48m 22s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/1819b3aa2b104296beaa9367320eaf50 : SUCCESS in 56m 55s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f072ffa94f8c4839982274f1ea3dccd5 : SUCCESS in 14m 52s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c4500e01b6e34200b3f0f07451ec4954 : SUCCESS in 14m 57s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb6d3c0f514348b194b416a99a0ea409 : SUCCESS in 1h 10m 31s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b0c00bbf48b8408c9257ede37f9d688e : SUCCESS in 17m 56s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/26754eb7413946f5881a42a00e256c1a : SUCCESS in 14m 10s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aaf2df09de24445d95fbb49e911162ae : SUCCESS in 29m 01s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2880212c4cd24e7ba39d5da303aa6bdd : SUCCESS in 12m 21s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4a090dea0543499299698b93a53065ef : SUCCESS in 11m 18s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c7a885db463543f6b21c814f010c1377 : SUCCESS in 15m 43s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dad5dfc9e7e8415185b7dc89e7a96c69 : SUCCESS in 13m 37s","accounts_in_message":[],"_revision_number":1},{"id":"0c6c19519929885836031213d7e5411c08c60abc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 20:39:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3171a46e2dc15d4a58fd409178337a1edf96c80f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 20:40:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c4a7a6a707149b3a23d1f81ee9cb57d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8fbcce93eb84acbad9db7c69684c870 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c9b115ac3c4d4393fa6d9bf19393d8f42c18af83","revisions":{"c9b115ac3c4d4393fa6d9bf19393d8f42c18af83":{"kind":"REWORK","_number":1,"created":"2023-03-08 16:22:57.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/07/876807/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/07/876807/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/876807/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/876807/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/876807/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/07/876807/1"}}},"commit":{"parents":[{"commit":"7c4307c90209a62908abfd69099c1c1683416fc5","subject":"use OS python to create kolla-ansible venv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7c4307c90209a62908abfd69099c1c1683416fc5"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-08 10:28:54.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-08 16:22:57.000000000","tz":0},"subject":"CI: get repo directories from nodes for troubleshooting","message":"CI: get repo directories from nodes for troubleshooting\n\nChange-Id: I0f7a3dcb021f726693725f070d9a2ff80d84fe0e\n(cherry picked from commit f9fee1f17fcc0be3884860fc57154c1e80968c26)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c9b115ac3c4d4393fa6d9bf19393d8f42c18af83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c9b115ac3c4d4393fa6d9bf19393d8f42c18af83"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
