)]}'
{"id":"openstack%2Fkayobe~876847","triplet_id":"openstack%2Fkayobe~master~I8bf75ce44b9ec1fdd477fcddf2deebf2929d9fc1","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"25600":{"account":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"last_update":"2023-03-08 16:53:59.000000000","reason":"\u003cGERRIT_ACCOUNT_25600\u003e replied on the change","reason_account":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"}},"35264":{"account":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"last_update":"2023-03-08 14:19:17.000000000","reason":"removed on reply"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-03-08 12:35:28.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-03-08 22:00:03.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-03-08 20:37:41.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I8bf75ce44b9ec1fdd477fcddf2deebf2929d9fc1","subject":"Fixes repositories files name in Rocky Linux 9","status":"MERGED","created":"2023-03-08 11:59:41.000000000","updated":"2023-03-08 22:01:01.000000000","submitted":"2023-03-08 22:00:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"876847","meta_rev_id":"0ca074004c7e359a52d7acb0972818d90d437d78","_number":876847,"virtual_id_number":876847,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-08 22:00:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"all":[{"value":1,"date":"2023-03-08 14:19:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},{"value":1,"date":"2023-03-08 16:53:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":2,"date":"2023-03-08 17:24:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-08 20:37:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-08 20:37:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-08 12:12:16.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"state":"REVIEWER"},{"updated":"2023-03-08 12:12:16.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2023-03-08 12:12:16.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"state":"REVIEWER"},{"updated":"2023-03-08 12:12:16.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-03-08 14:18:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"078690767232f7ec79f830ae902ad48ba140e429","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-08 11:59:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dc356263208f0ee8ecd03e58b97023ae1b74b928","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-08 12:05:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7303e309602f96f23b6bfd976e3453e6c262841a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-08 12:32:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"80551ff7df4ecd2c3538714fc10da611a5d9ef0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-08 12:34:31.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"abb875fc88cd899e452b8f1c73f9afbed897ede2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-03-08 12:35:28.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"8e5599df1893553ca744a4ebe15339fce9ee909a","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2023-03-08 14:14:25.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6f08b6de0ad6b4ae1ba18b0622e918bc6d6239a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 14:18:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c14e11fdaff4bcdbbc99ac19539fdcc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6599814dea54eb79d896012e9975e74 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc02a014d59b4ef9b6e3c468bff2d97e : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6cbb066db52a4635be3ed8a2a4b31e1d : SUCCESS in 6m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a4616f3f73ef436bbaea3cf04e9d64f3 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1148775f18f24c7b9bc3006c3d73291b : SUCCESS in 7m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2fccb6fd4d064b13bcfb94e08ba4f518 : SUCCESS in 3m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d5bfc02f8c7846ada6af3692a7a7772d : SUCCESS in 3m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4e5bf05a77f84fa3972695b905f35446 : SUCCESS in 10m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7e5a5db3c83c4b2d8aa4c6037221626d : SUCCESS in 15m 24s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/b324b662846b44c8bec9258c6be4285d : SUCCESS in 47m 01s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/4fcb96a5c71145a4ab40d59c2a4d96f3 : SUCCESS in 56m 50s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/297a515f127f485584eecbfe51705e9a : SUCCESS in 53m 12s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/1a049fccd5004e848b1f9a654fe9bfa6 : SUCCESS in 1h 01m 13s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/96937ec6dc474999bdaab38eaae9eb64 : SUCCESS in 15m 34s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/5fcb68302a3f4c5887c504a47a6149ff : SUCCESS in 17m 34s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9bee9bfe4dcb488abf6d2233dca28c8b : SUCCESS in 15m 06s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/77026c65dd25441a920339afbbfb78cb : SUCCESS in 1h 38m 40s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9a9b54827c3407a8b4895c95ba48ac1 : SUCCESS in 1h 11m 25s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c2a5822d4586468da41e258cacfe78b3 : SUCCESS in 17m 00s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f6c1fb43b7a41b0ad02471ae0cbe3f8 : SUCCESS in 14m 06s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/b27b2efccf9547468200da3d7a6de417 : SUCCESS in 26m 16s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ed39a0ddf3fd4fe0b5a6b668a07fa2e1 : SUCCESS in 28m 16s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/643fe6a2a4834fdca17fb67ab5747366 : SUCCESS in 21m 11s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4be7615384a94e4d9e7e2ba45cb9eddd : SUCCESS in 12m 33s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f850a53925d447ddbe96ecffafdc3bf8 : SUCCESS in 11m 27s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2fc3a16caa3247b3a5b1f274cee01b63 : SUCCESS in 14m 49s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7069148914da4a11b0d4479c8e065f2c : SUCCESS in 13m 29s","accounts_in_message":[],"_revision_number":4},{"id":"0ac6cbe5754047ed3123b61cfabe749cb288dd8c","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-08 14:18:28.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"04ead85d0b285c287d77d6cc2b2f3d4502971b90","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2023-03-08 14:19:17.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"70c57f0ba03995c2cb057dc22c26ca032debac77","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2023-03-08 16:53:59.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6235c21b0366a48c3049141ea63f49a2478f17e2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-03-08 17:24:20.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"17de088cf904a8b711fced49669510947ea7647e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-03-08 20:37:41.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"366a2afcc484b4e99f5a97134c887fdb7646750d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 20:38:06.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"2fccb4c5f7c2e0049af8c34ce98d8190edc16df7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 22:00:02.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58591d65619946fab2b003819d063c4c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a1bb9ee36594ac3ae3b79d7e4678be5 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b51cbe17d59e4da2a3bf5efa3ce56cea : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/080daaf05b6244808e1b450eff52ae08 : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c5a21807b484184995b6587826ed158 : SUCCESS in 7m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e47c6220e484f7cab5b369882732f62 : SUCCESS in 3m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9b6b3f4b8e3b4d95ad815a6f2d64cf90 : SUCCESS in 3m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8b5be5f91c3a4a06ba834851a14d7be4 : SUCCESS in 11m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/82c38f27dcc04dd2b8b4e95bcb1ed68e : SUCCESS in 13m 30s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/9e20c1b3558341019565c95b753f812e : SUCCESS in 48m 07s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eee89355a79f4db8a9e71caadeb1efe8 : SUCCESS in 45m 22s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/75eb32ce23fe40e58fc029513dbd76a2 : SUCCESS in 51m 02s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/24a720c20c7247ccbf03293972b03223 : SUCCESS in 14m 36s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f256dd4cf3fd42fda50f116810d013ef : SUCCESS in 14m 12s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3c6d41d31f8c4cb09ac8a27da4ee4851 : SUCCESS in 1h 18m 58s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1ff006ce3e0b47fdbfcf96c866c40c40 : SUCCESS in 1h 03m 44s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a0fed0294b6f4e80ac4c7174b5540690 : SUCCESS in 22m 17s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bcb84027ed02446d811bb9c5676c7fe2 : SUCCESS in 14m 08s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/92ac2a562c87482f935d15e4f104c6b4 : SUCCESS in 23m 13s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a074fc3c7814495ab54b6ceb767090c : SUCCESS in 22m 36s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e2972f82040448a1b0c80d798b881d91 : SUCCESS in 13m 03s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5b8fa1efb60d4b05bca4a162d06d9f45 : SUCCESS in 13m 23s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ffa4ca8c9e784ee88dbb4e753e4b372e : SUCCESS in 13m 03s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/973f3152ca2046aaa1d0e2b7644de983 : SUCCESS in 13m 15s","accounts_in_message":[],"_revision_number":4},{"id":"80d42d0781d77ee512dc55b45530cc22e887d9f7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 22:00:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"0ca074004c7e359a52d7acb0972818d90d437d78","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 22:01:01.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67946b88e8f342e79949c8720524829d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5140849c0cd54a6caa66a3c341678bc5 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/968f27f45c034761a3461b8cbfffc9e2 : SUCCESS in 32s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"e0ff9a0283bd870df72ea7dec62d55ef0b8ed972","revisions":{"9e70dbb1a69620b11b42979f16f8e8c55f8397b0":{"kind":"REWORK","_number":1,"created":"2023-03-08 11:59:41.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/47/876847/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/47/876847/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/876847/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/876847/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/876847/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/47/876847/1"}}},"commit":{"parents":[{"commit":"f9fee1f17fcc0be3884860fc57154c1e80968c26","subject":"CI: get repo directories from nodes for troubleshooting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f9fee1f17fcc0be3884860fc57154c1e80968c26"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-08 11:57:02.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-08 11:57:02.000000000","tz":60},"subject":"Fixes repositories files name in Rocky Linux 9","message":"Fixes repositories files name in Rocky Linux 9\n\nDistributions moved to lowercase names with RHEL 9 release.\nThis patch also removes previously created uppercase files.\n\nStory: 2010637\nTask: 47604\nChange-Id: I8bf75ce44b9ec1fdd477fcddf2deebf2929d9fc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9e70dbb1a69620b11b42979f16f8e8c55f8397b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9e70dbb1a69620b11b42979f16f8e8c55f8397b0"}]},"branch":"refs/heads/master"},"dd3402810c11b4e47897faed316c45123105f4fe":{"kind":"REWORK","_number":2,"created":"2023-03-08 12:05:57.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/47/876847/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/47/876847/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/876847/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/876847/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/876847/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/47/876847/2"}}},"commit":{"parents":[{"commit":"f9fee1f17fcc0be3884860fc57154c1e80968c26","subject":"CI: get repo directories from nodes for troubleshooting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f9fee1f17fcc0be3884860fc57154c1e80968c26"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-08 11:57:02.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-08 12:05:41.000000000","tz":0},"subject":"Fixes repositories files name in Rocky Linux 9","message":"Fixes repositories files name in Rocky Linux 9\n\nDistributions moved to lowercase names with RHEL 9 release.\nThis patch also removes previously created uppercase files.\n\nStory: 2010637\nTask: 47604\nChange-Id: I8bf75ce44b9ec1fdd477fcddf2deebf2929d9fc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dd3402810c11b4e47897faed316c45123105f4fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dd3402810c11b4e47897faed316c45123105f4fe"}]},"branch":"refs/heads/master"},"2cd4ced298d5aa28cf8b61b3676247f9cb5f16e1":{"kind":"REWORK","_number":3,"created":"2023-03-08 12:32:40.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/47/876847/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/47/876847/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/876847/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/876847/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/876847/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/47/876847/3"}}},"commit":{"parents":[{"commit":"f9fee1f17fcc0be3884860fc57154c1e80968c26","subject":"CI: get repo directories from nodes for troubleshooting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f9fee1f17fcc0be3884860fc57154c1e80968c26"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-08 11:57:02.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-08 12:32:30.000000000","tz":60},"subject":"Fixes repositories files name in Rocky Linux 9","message":"Fixes repositories files name in Rocky Linux 9\n\nDistributions moved to lowercase names with RHEL 9 release.\nThis patch also removes previously created uppercase files.\n\nStory: 2010637\nTask: 47604\nChange-Id: I8bf75ce44b9ec1fdd477fcddf2deebf2929d9fc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2cd4ced298d5aa28cf8b61b3676247f9cb5f16e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2cd4ced298d5aa28cf8b61b3676247f9cb5f16e1"}]},"branch":"refs/heads/master"},"e0ff9a0283bd870df72ea7dec62d55ef0b8ed972":{"kind":"REWORK","_number":4,"created":"2023-03-08 12:34:31.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/47/876847/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/47/876847/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/876847/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/876847/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/876847/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/47/876847/4"}}},"commit":{"parents":[{"commit":"f9fee1f17fcc0be3884860fc57154c1e80968c26","subject":"CI: get repo directories from nodes for troubleshooting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f9fee1f17fcc0be3884860fc57154c1e80968c26"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-08 11:57:02.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-08 12:34:11.000000000","tz":0},"subject":"Fixes repositories files name in Rocky Linux 9","message":"Fixes repositories files name in Rocky Linux 9\n\nDistributions moved to lowercase names with RHEL 9 release.\nThis patch also removes previously created uppercase files.\n\nStory: 2010637\nTask: 47604\nChange-Id: I8bf75ce44b9ec1fdd477fcddf2deebf2929d9fc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e0ff9a0283bd870df72ea7dec62d55ef0b8ed972"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e0ff9a0283bd870df72ea7dec62d55ef0b8ed972"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
