)]}'
{"id":"openstack%2Fkayobe~876910","triplet_id":"openstack%2Fkayobe~stable%2Fzed~I8bf75ce44b9ec1fdd477fcddf2deebf2929d9fc1","project":"openstack/kayobe","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-03-09 19:12:29.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I8bf75ce44b9ec1fdd477fcddf2deebf2929d9fc1","subject":"Fixes repositories files name in Rocky Linux 9","status":"MERGED","created":"2023-03-08 22:17:29.000000000","updated":"2023-03-09 21:41:21.000000000","submitted":"2023-03-09 21:40:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"876910","cherry_pick_of_change":876847,"cherry_pick_of_patch_set":4,"meta_rev_id":"5ab437570588f39b2aa64838cfce8d7bc8a73666","_number":876910,"virtual_id_number":876910,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-09 21:40:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2023-03-09 09:00:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2023-03-09 09:06:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2023-03-09 09:06:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-08 23:21:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-09 09:00:24.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"d218bf7f14f96b377b6f5444446c4a46529c2066","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-08 22:17:29.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"43b0d49501ba68b8df81a279ba8c95a1490af53f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-08 22:18:13.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"fa668aa6c968d3ca5d1b3361c87fe1f14fd35d8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 23:21:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/515b38dea2554203a9147cef9435493f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/177c696b73ce4ab5b31af7b788329aa7 : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed895cf4edfd4741b47681ea9a99a7d6 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/446d6d9e9fa349c3b7b4f9d743bdeac8 : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4393405e2ae648ab8e96eb716f5649a1 : SUCCESS in 3m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/45db205610994fd5898e531573074061 : SUCCESS in 3m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6099269599a64c49a9f29bd92539cce8 : SUCCESS in 7m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4faa207459cc4d1aa130914c953b2069 : SUCCESS in 3m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/661ff92222424974aed56b640e50eb92 : SUCCESS in 3m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/72d570238775488c81745df97ccf806e : SUCCESS in 8m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d52f3832c859437581c3dcdf66b9f0cd : SUCCESS in 15m 52s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/410bfc9592f84c7288477cd5c7dd9483 : SUCCESS in 59m 52s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/cf04c63adf7541e5b70aec6541173d74 : SUCCESS in 41m 51s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/214fed999650462c9cf4b600f9a266d6 : SUCCESS in 41m 09s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/d0f12aab6fc9449fb5e699475350147a : SUCCESS in 1h 02m 19s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5178b7fc3f414b19b21065668f477c46 : SUCCESS in 17m 34s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/0c3661d6ee6448a49f6188a48a012d03 : SUCCESS in 13m 16s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/418f64104be346cab99e83f8068962a3 : SUCCESS in 13m 40s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25578462ebc348a5b327390360b44d47 : SUCCESS in 57m 49s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/37248d3095c04bf4a09117f698385246 : SUCCESS in 17m 38s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf25d14d67fb4632994882c283cd108c : SUCCESS in 12m 27s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/a7ca77d9a74f4de3a27ff200bdb8169f : SUCCESS in 32m 05s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d57bca88e2e44e479d151ada4f312047 : SUCCESS in 30m 16s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/62e5411aad4b4838bcfee36ee435e34d : SUCCESS in 16m 26s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7e62ddeea604c8e9818228583056395 : SUCCESS in 10m 00s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/70143c3413e349559a8da8514d73a5a2 : SUCCESS in 17m 42s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0540121f4b9b4b108ddddcec5d0dd89b : SUCCESS in 10m 34s","accounts_in_message":[],"_revision_number":2},{"id":"5bb50ea218c1c52dc2ca7673b651584bcba9323a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-03-09 09:00:24.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b4767cb09652f93da2a38c5a00b9dc674c43e7ae","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-09 09:06:35.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"494074e43f4da38a5c5c3e808c1a4a5572989ddd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 09:17:28.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/876807 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"377f3dc1f56619d23fe7495e7fd6d59846ed0652","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-09 11:54:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"694629a8293d8f20ab0f41620f5aed3de7cb872a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 13:15:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/492294856ca54cce8911c75526d80282\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a1e1eac2392543f98a87bc6cd8c5c838 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c4a829b452047229c9044c77c3380f5 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6714012302e74884ad9b86b524e4c60d : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b6145025fec49a1b754bf6a9867172a : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3e088be42d8944138937f5a97fa557a5 : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74d06b980d10449daeb3bf621117d94a : SUCCESS in 6m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53060329dcae421aae4fb554dccd50b0 : SUCCESS in 3m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9d0351b910fc4806aa05cd761aae8beb : SUCCESS in 4m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5b2d566454e2459fae24e87f5a875ed0 : SUCCESS in 11m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/06efeafdee31400c92bb9387389fbe2d : SUCCESS in 15m 45s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/b34ac709d0994248a87b0c18bd6b95a8 : SUCCESS in 53m 19s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/d4a0d92f127c465daedc8a4c87122d52 : SUCCESS in 57m 57s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/14c6c2b0c6504d1b8d5ed9feb0772d4a : SUCCESS in 46m 58s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/23dd6367f6a54af080a25e9eb9c967ca : SUCCESS in 58m 30s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ccde0856b75c4e15848bd0596056eddc : SUCCESS in 19m 04s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/92e6f05f429a46e78c94e819e315ff2d : SUCCESS in 18m 37s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d654e964211d4594a7d10f89a59f806e : SUCCESS in 13m 54s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/deb6a5eeb31749f8847eed9a476ab10a : SUCCESS in 1h 16m 00s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ba78f8f97f7d4da9b7115e58f7f46cae : SUCCESS in 17m 05s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2a76db55e591443dae6cf64c1bc8dac0 : SUCCESS in 16m 17s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/143eea5986a1489fa7d40db598fdbe12 : SUCCESS in 22m 40s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b6ac76710ee4f3ab5daaf1d639163eb : SUCCESS in 29m 10s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d423b72ae69b4848a9b0020174683ca3 : SUCCESS in 14m 20s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a9eaaf1f659b4fb3846fadfa42a46033 : SUCCESS in 10m 52s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/afc6dfaf2e364ffe8e622c324e63f861 : SUCCESS in 12m 46s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2a114629179e4ceca37de7c60c1d20cc : SUCCESS in 14m 34s","accounts_in_message":[],"_revision_number":2},{"id":"a8f07f299f8534f454130b86bd105bd28ca606a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 13:16:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e046405cf9289bba9427f8a75cead40fe39a5114","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 14:58:40.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/876807 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"03490271ce8691336d336940c7722df0bbf36d4b","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-09 15:58:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f7ade40405c9d1e4c8c8fc242eb4c47bd24a50c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 17:06:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b6b950cdfefb4b10b8357ba78ce7f1fc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6cc26ceafb2043c881fa192829acaac6 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41cda8f03c0945aabfd859a683fff5eb : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b037d58ce374d8db3fab3f5636ff376 : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4131d9ee2b784b73a81894ad91e562a0 : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9818a3075ee444dcacf38a451bcc034d : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73072c57c5dd4c47818195d01e241069 : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db2c4059a4b8431393f585b32ac41f77 : SUCCESS in 3m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/450a67eaac08439c819fbb31baa0ee71 : SUCCESS in 3m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/48ce68e35ac54216aa045497715f95c3 : SUCCESS in 12m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2560102729ed4aa1bfa7dcd8af3b7683 : SUCCESS in 14m 20s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e0f54213ede24dd29f2491a12bee7e4c : FAILURE in 48m 44s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/c6d76931c19a4737a643b55654b5abb5 : SUCCESS in 57m 23s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/202c55fb389d419bb1e4b9631535b3fc : FAILURE in 18m 53s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/b5fa62f4432741078e34917c0dbe0d1c : SUCCESS in 57m 07s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4df861af02d94a73bcaa2d7c67ad27c4 : SUCCESS in 20m 48s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/a71e35b5ef59409da09a0f046ca25d4a : POST_FAILURE in 18m 58s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/93fc219e50264fc1a5098d07e54ca107 : SUCCESS in 19m 07s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a134fd9834b9406589d9d1daa4b009f6 : FAILURE in 43m 38s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/63fc934a878d4459999ffcadba5eac2a : SUCCESS in 20m 10s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/14f7e120cbaa46909383427a982a4a13 : SUCCESS in 17m 09s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/f9a5954a9dc04bbb85dbb79097301a93 : SUCCESS in 33m 39s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ea2c31b6cae41bdae35035c5b0d7818 : SUCCESS in 30m 29s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bed2f741393b42009b6880eb9bc4d664 : SUCCESS in 12m 09s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2593cb91be8420dbb29a41896a63813 : SUCCESS in 10m 28s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2d76009eaa7c4071aa481e9aa629484c : SUCCESS in 16m 11s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c6bc444e199418aaba34df72fd0fb98 : SUCCESS in 16m 14s","accounts_in_message":[],"_revision_number":2},{"id":"57c278d12eb4687c61d940c709b770763b710f79","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-09 17:12:00.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"908d8060866fce746cfa501853751630b91e308d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 18:13:45.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8f5bc4fa94146939994ce30db5fc9b5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ef0cc64e10a4ceaac9fbba02abfb20d : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7446de59fac4d73baa9acb713f1495b : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b93547faebc4d268b0a1771ada1f719 : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/954e92a33c0e40a38e653b18645d4ae8 : SUCCESS in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aa8d34da00d84eda92d2fe43f6da6f5a : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91d775dbe5954c2d984a17b12f64b927 : SUCCESS in 7m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84ac1992b4274a29a5f8e64c1f3996ae : SUCCESS in 3m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8e18529a53a64e4ea987d6ca6bca4d29 : SUCCESS in 3m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f832f60e32104405906c49554fa38c48 : SUCCESS in 10m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c367a7fc52dc42ecb4797455ba8cbf8d : SUCCESS in 14m 13s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/4bd5055573684a11bdb9cb6167f39d95 : SUCCESS in 54m 01s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/2d8b236df44242cb9a301502dde7b180 : SUCCESS in 54m 31s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4631174ed193413fa3b5bc6317d48363 : FAILURE in 16m 32s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/afcfb60d932045438ba857c93e2693f4 : SUCCESS in 59m 42s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0870ffe2ece54fc7961859962ab1040f : SUCCESS in 17m 22s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/afce8f37338847089e0d2cbc83f07f20 : SUCCESS in 18m 50s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/615474678bbf4507a40a83177344034b : SUCCESS in 17m 03s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d99272b0df37440cb602fe9e87e1094e : FAILURE in 20m 45s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/d680b23368574234af4679f33bb12bfb : SUCCESS in 15m 40s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e76ecf559d4a4d238be74460970dbbac : SUCCESS in 16m 37s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/5665356a4d194974b616d0937d3669a4 : SUCCESS in 31m 04s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e3ed52f4f1644f7fbb502330e4582329 : SUCCESS in 32m 16s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bee41306751846f4a55d0c760844b7b9 : SUCCESS in 12m 08s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5495e70929f0499ab9d282b407f49984 : SUCCESS in 9m 39s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/23a74f023f484948a69ce890ee57868a : SUCCESS in 16m 07s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d4f6d43621e14771a49b04659027282a : SUCCESS in 21m 58s","accounts_in_message":[],"_revision_number":2},{"id":"e41e88f3dc4023bddd3ed0a3cd2cab4cf3d28bc3","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-09 19:12:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9572ff986f0b33c468db589c5e421f2d4e379c0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 20:26:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/475588ef3d014d05a35ec826890a04c3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92f495099d4f4cda9ab403f564b81b8e : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6929c6d3a26d4560b8196acf00dbce5f : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cdafef3c80e7425aab33bb3fa44226dd : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d1fc40534ee845ef831eba69f561a69a : SUCCESS in 5m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/842879ea04004fb4b41c0c91b4b4c6bf : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1d425b91a684e0e923871adad6fff5f : SUCCESS in 7m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a4ebe8de53d4c78a636aa6deeda890f : SUCCESS in 3m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/89efbd35ccc54b76811e5254b553623d : SUCCESS in 4m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/db9e172a6b3e4acd8b1af465ed3f57f2 : SUCCESS in 12m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4235c764933d4fb5a39191525978ff04 : SUCCESS in 19m 01s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/66c7cc0ae288444a8803e1acb5283dde : SUCCESS in 58m 20s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/3edb9e2c250449bfb5ecc7fbd8406966 : SUCCESS in 58m 27s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a5b59674a48434e9a9d2c0431f8809a : SUCCESS in 54m 55s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/18c5009adde1437ca29ebb499775f0db : SUCCESS in 57m 10s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d340f333f3944ac88c3b840ffb6aa309 : SUCCESS in 18m 15s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/9301c87ba8c945c6aeee983607de3da2 : SUCCESS in 20m 09s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/285972d58ca54a0696974d8ad8127192 : SUCCESS in 16m 06s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dedeff48af6648c3bb9f7244cbe9fe1b : SUCCESS in 1h 03m 12s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/80a19fac0ef04405b17bcc974f127967 : SUCCESS in 17m 10s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad180479e87f4b7f8b296d8eb649ba41 : SUCCESS in 15m 06s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/97ca6cc0fa53438aa1b304e2eacf48f1 : SUCCESS in 35m 02s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2700bb140b9f4c069e437c041bf7c406 : SUCCESS in 33m 29s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/eb0ad19f649b4fdab345e34b92039703 : SUCCESS in 13m 00s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/27889753cd4c4881b2c0a0823e9fe9a1 : SUCCESS in 10m 56s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/905d7d72e68c4509a3d2ca000c801364 : SUCCESS in 15m 45s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/60f8112cadf94373891ce76b69167287 : SUCCESS in 15m 26s","accounts_in_message":[],"_revision_number":2},{"id":"2a207331d9bb02809830be65fbb3313337166f03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 20:26:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5282d582017e04b4251c2f0ca7a6132a16ea97df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 21:40:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/446c6dd8023a439aba335acfd064ee86\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f2a9e1971e04ac9a4063ad775883559 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1566efd88f24775861ff3a5cad76107 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b0143b34db6d4c7abc3c2ce9f17d04f9 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e42f3b50167849e387031d5f8273ecbf : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d1297da8f8e4778802ea819463b31ac : SUCCESS in 3m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7e95cd2688374ebeaf29b9ae943289db : SUCCESS in 4m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b60976c953de4292897d36e2b8224d47 : SUCCESS in 11m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b7aa7fdca54c4403b3c53ca3a025b655 : SUCCESS in 16m 15s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/4973c442def941898b2c92dfb01ab7e1 : SUCCESS in 52m 44s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b10922cd8b474fcd8c10c1c6e3f8cf90 : SUCCESS in 49m 51s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/d044bc0ab45f4a39a7922e4d2bbf6f9a : SUCCESS in 56m 19s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/8a8545d5f87f44a8bc52e1aa2b86a37a : SUCCESS in 17m 01s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1dc38b97f43b4b57a0f3c5b1e6430d78 : SUCCESS in 13m 47s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35431c783a194283a31a99911896bbed : SUCCESS in 1h 11m 16s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/15379c0cd33d43afb71957a1edf4638f : SUCCESS in 17m 03s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b75cc4ccac6f48569b860119634fdb2f : SUCCESS in 15m 06s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/509cf08198d34e2093be45c3eda77567 : SUCCESS in 27m 33s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2ed5a2a676d343299e6d7ce54b6774c9 : SUCCESS in 11m 34s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8aa83287a8c4fb0a26f14502b360833 : SUCCESS in 11m 25s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5ccd42f541bb472a844cdb19f632495e : SUCCESS in 14m 11s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/03272b2305004411976c29c6a0cdd8c8 : SUCCESS in 14m 45s","accounts_in_message":[],"_revision_number":2},{"id":"306d063cbe89bcbd5eb41248cd810b9f660f260b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 21:40:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5ab437570588f39b2aa64838cfce8d7bc8a73666","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 21:41:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/910978077da242eeaad37f132bbb7526\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b280b4f861c345d788a33a992d1fd726 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/173f829ab9bc412aabb73d8b748368be : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b31fac7d3d1817dc7dea68991b08fbc22c6e15e7","revisions":{"ac02da36f8980c67126cbee4bb0d62a69fe1b206":{"kind":"REWORK","_number":1,"created":"2023-03-08 22:17:29.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/10/876910/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/10/876910/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/876910/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/876910/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/876910/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/10/876910/1"}}},"commit":{"parents":[{"commit":"7c4307c90209a62908abfd69099c1c1683416fc5","subject":"use OS python to create kolla-ansible venv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7c4307c90209a62908abfd69099c1c1683416fc5"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-08 11:57:02.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-08 22:17:29.000000000","tz":0},"subject":"Fixes repositories files name in Rocky Linux 9","message":"Fixes repositories files name in Rocky Linux 9\n\nDistributions moved to lowercase names with RHEL 9 release.\nThis patch also removes previously created uppercase files.\n\nStory: 2010637\nTask: 47604\nChange-Id: I8bf75ce44b9ec1fdd477fcddf2deebf2929d9fc1\n(cherry picked from commit e0ff9a0283bd870df72ea7dec62d55ef0b8ed972)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ac02da36f8980c67126cbee4bb0d62a69fe1b206"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ac02da36f8980c67126cbee4bb0d62a69fe1b206"}]},"branch":"refs/heads/stable/zed"},"b31fac7d3d1817dc7dea68991b08fbc22c6e15e7":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-03-08 22:18:13.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/10/876910/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/10/876910/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/876910/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/876910/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/876910/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/10/876910/2"}}},"commit":{"parents":[{"commit":"c9b115ac3c4d4393fa6d9bf19393d8f42c18af83","subject":"CI: get repo directories from nodes for troubleshooting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c9b115ac3c4d4393fa6d9bf19393d8f42c18af83"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-08 11:57:02.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-08 22:18:13.000000000","tz":0},"subject":"Fixes repositories files name in Rocky Linux 9","message":"Fixes repositories files name in Rocky Linux 9\n\nDistributions moved to lowercase names with RHEL 9 release.\nThis patch also removes previously created uppercase files.\n\nStory: 2010637\nTask: 47604\nChange-Id: I8bf75ce44b9ec1fdd477fcddf2deebf2929d9fc1\n(cherry picked from commit e0ff9a0283bd870df72ea7dec62d55ef0b8ed972)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b31fac7d3d1817dc7dea68991b08fbc22c6e15e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b31fac7d3d1817dc7dea68991b08fbc22c6e15e7"}]},"branch":"refs/heads/stable/zed","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
