)]}'
{"id":"openstack%2Fkayobe~876911","triplet_id":"openstack%2Fkayobe~stable%2Fyoga~I8bf75ce44b9ec1fdd477fcddf2deebf2929d9fc1","project":"openstack/kayobe","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-03-09 14:06:56.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-03-13 10:05:04.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-03-13 10:05:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8bf75ce44b9ec1fdd477fcddf2deebf2929d9fc1","subject":"Fixes repositories files name in Rocky Linux 9","status":"MERGED","created":"2023-03-08 22:34:32.000000000","updated":"2023-03-13 10:06:07.000000000","submitted":"2023-03-13 10:05:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"876911","cherry_pick_of_change":876847,"cherry_pick_of_patch_set":4,"meta_rev_id":"46a2161a0d1684026a84694b62752140fca45915","_number":876911,"virtual_id_number":876911,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-13 10:05:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-03-13 08:06:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-03-09 23:18:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-03-13 08:06:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-09 00:27:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-09 09:40:01.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2023-03-09 09:40:01.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-03-13 08:06:10.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"e86e0f3a758a912e9f87ee19958ac98697d0044f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-08 22:34:32.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* ansible/roles/dnf/templates/CentOS-Stream-Extras.repo.j2","accounts_in_message":[],"_revision_number":1},{"id":"dd58d8efca2807e9079fe176c7810310a0609827","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-08 22:35:05.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"1fa389817654a24ac42b187f244a173d867cba9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 00:27:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/61dcc9b644244161a95e795667ac5845\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9277a17df6dc4ac8b18ae8e08666e2f4 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abac2a48fb2a49299b7134295e2c2d36 : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/897bfbf5661b4214b12013b256b8fa17 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a953e64398e54a3d89414d7448f81ecd : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ea2ee9324524cb99d8b5af8cacb7258 : SUCCESS in 10m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8df260d0ecb646ea80c68a072c04d8b6 : SUCCESS in 2m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e456ecf3c3a24170afb819c6472274c8 : SUCCESS in 6m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7220480b01754b5dbdd15fdf2dc49bc8 : SUCCESS in 10m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8dde956e8f7646ddb7427457d24a1a5b : SUCCESS in 17m 08s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5c0636d100dc453fa60f010c6961fa9e : SUCCESS in 1h 00m 32s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/904150c80e0c4a1ab5a2a6328351f8c8 : SUCCESS in 1h 02m 47s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/89ca1e23f4fe4379b5779c8870c1e2db : SUCCESS in 1h 02m 57s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/7e838bc7749f43eb81e1b98bb25c3179 : SUCCESS in 1h 01m 02s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b9b63bb2a9e422bb1afe61347122ad8 : SUCCESS in 53m 23s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4fb9faeeb243440e87a85a4f799f43ab : SUCCESS in 51m 10s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/620e65351a4142b8bf1961f3383398dc : FAILURE in 11m 40s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/7e2229ba156142a79afaa6c695ad3ff5 : SUCCESS in 21m 20s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/07aad382a5824bbabf07bcef39ab3584 : SUCCESS in 20m 01s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/32f773322cdc43d69a65a9ca059c8184 : SUCCESS in 19m 02s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5419fde9c93e4f08a1067e0f89bde617 : SUCCESS in 15m 05s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/310d77843dd442999a50b85b53646f25 : SUCCESS in 15m 33s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/0d4704a5e67e49de919f65c9e284f9a9 : SUCCESS in 1h 08m 57s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/933b3c668c164b328983c52f54715d28 : SUCCESS in 1h 02m 51s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b433c06d8eca439692933415bd6a7916 : SUCCESS in 1h 47m 04s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c900a71c55904bbc8dea9d2b093b2046 : SUCCESS in 1h 08m 56s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/10483e9d5f2744bda9ebc2b66f521a3c : SUCCESS in 20m 18s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/093d1f2ba4594ebca9ad2a9dd98c8d0e : SUCCESS in 18m 48s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/21823fc1877a4ca0b404dbc3367b69ce : SUCCESS in 17m 59s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33e6ee9628ca401c98e0be131a0363d1 : SUCCESS in 15m 56s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e46ec898e7c9441bb398e32b082a130f : SUCCESS in 18m 12s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3d346678c52e459881e1880ddfcf9d6e : SUCCESS in 39m 20s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/88f59ec09904412cad77ce02e2073abb : SUCCESS in 33m 31s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/97c4e7b731da41d9b3bbbafbffca5462 : SUCCESS in 14m 21s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/63d92f85b89e46a887beff014011906b : SUCCESS in 15m 09s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7bc4970b0dbe4f0380f9c2a0664407e9 : SUCCESS in 16m 55s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5659a02a46f412c919d324956d275c2 : SUCCESS in 14m 12s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8db73b1d0a6b4471a56417fef2d08069 : SUCCESS in 11m 34s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/35195971357549f29e4203b6199d44b2 : SUCCESS in 19m 48s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/c774f20ea2f7494ebf6fb40fe3eda671 : SUCCESS in 19m 38s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/682370809e2348b0af42f2bfbfc73235 : SUCCESS in 12m 52s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/384c283c598147ec830c39b60661aabf : SUCCESS in 15m 12s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b43b027a7ec6459583186ba45e63d2c9 : SUCCESS in 11m 33s","accounts_in_message":[],"_revision_number":2},{"id":"9eeb0c0e2eb11679dccb1567c7618a700ea2c544","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-09 09:36:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f3362f21cbf15195dd28eb1489a21ef97114711d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-09 09:37:42.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b8aa5cc99a8581e3321a8a260b23dc894a3dcb33","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-09 09:40:01.000000000","message":"Patch Set 4:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"34c2db13b6514fa133c1713cd15ced1dce36e29d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 11:34:22.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/94a18581b3eb4d77afaeb856df510148\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/98f38e8c7bda45ec9fa87d3f5f585c4d : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3b574c2c1d348bb9cb3a6140085fd00 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2dc4bde8149f4e2ca1a939436b93f531 : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e866452137454210898e9785ad2a1a07 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5be395da4b5c4202bcf5c1335e787742 : SUCCESS in 9m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64057d853ace4935995a9ca48e8c2002 : SUCCESS in 3m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6773506f79c74cc592e1329534000aa3 : SUCCESS in 6m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/11e1e328773242ec95de72e5fa1a8b8e : SUCCESS in 12m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3233bf7896a94bf89b53dc265baebbe3 : SUCCESS in 18m 12s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/cfa46843830d4863a83e57b14878c0a1 : SUCCESS in 53m 31s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/7485f11eaea9416dabea8f68d69839bd : SUCCESS in 55m 11s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/ad539ae057c14a32b3f0e8fa1440f5de : SUCCESS in 1h 02m 12s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/b4a519e50edc4b219b30de62050bf145 : SUCCESS in 53m 28s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78190c35665e434689adf3a12cb20319 : SUCCESS in 50m 15s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dbb2d8ddb479471bb2a0f8f756123244 : SUCCESS in 44m 55s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/32409b6e51cd4ba9abf2c0f2b5f56c35 : SUCCESS in 25m 18s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/4d85d5b5f6f54c6b934c9a040cb378b1 : SUCCESS in 20m 36s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/12d45471dc0e46549a24694b16333af8 : SUCCESS in 18m 27s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/7909485e01e84b2cb537a6b87853e36c : SUCCESS in 14m 06s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c1c09b7f0444c1191ed667adea22cee : SUCCESS in 14m 57s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1d3df5cb8aeb4b7489af0b49d66de06d : SUCCESS in 16m 39s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6d5f366d4eb047c9aa0e0f5ad38dcd93 : SUCCESS in 56m 56s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/fb09147526774ef7ac7969eb6c7d431c : SUCCESS in 57m 12s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e1e2c92a69fb4c769ba15c7a76548399 : SUCCESS in 1h 47m 42s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/627a3fb0be804887b8ab4bfadeced402 : FAILURE in 36m 25s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/112aba5ffbc34c38adf167ea321a3261 : SUCCESS in 20m 48s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/e0dc02889f6c48bd940845976a1638f2 : SUCCESS in 18m 13s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/38e5806dc83f454faaf9a1e7804ff978 : SUCCESS in 18m 54s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33492f4aaabd43fb904e579d4562eae6 : SUCCESS in 14m 42s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/50a12e9fe91a46e9aff303d5b46ca6c2 : SUCCESS in 17m 33s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d6ce0181dcf2489584372a8ce5daf90a : SUCCESS in 40m 39s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b164a4c212a844bfb0c1052c286208db : SUCCESS in 26m 43s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/a73f4dce20a14de6bc853158b40763a7 : SUCCESS in 14m 45s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/bf3e4725144e497e88c97e13b1cd9707 : SUCCESS in 20m 25s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/dbd2df608586482cbf2bf912a8084df4 : SUCCESS in 13m 50s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c0d1ae41181b4652bebe00e3e2425847 : SUCCESS in 14m 18s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b874207446f24f23aab92e2fe2ecd318 : SUCCESS in 13m 03s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/0d35ae5ac75543699ba5b1d69a1eeb39 : SUCCESS in 19m 19s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/49e22f4512874dfa8ecc7f5278b320e6 : SUCCESS in 20m 28s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1c6b68ea23994b929783a9b83ad18eff : SUCCESS in 16m 47s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03eceac1ee814957bc4e516d8a12bf5e : SUCCESS in 15m 49s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/770ee6177ac44ec39a5444e805aa4400 : SUCCESS in 15m 31s","accounts_in_message":[],"_revision_number":4},{"id":"959f260bd6acb47e9c41d1fd75240420950d1195","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-09 11:35:33.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6bc6d3b9a5b43d6d498b8bd6492654de26faa168","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-09 13:20:00.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/320d3c83b8a84d66b860fec834696681\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e582afdeee0a4577a5ad34bd247be562 : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e039b3d9b02b46308428654ef319613c : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d479d05450144adbffd84c42e366aee : SUCCESS in 7m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66a6212205ab4b4dae29faac3939e0ef : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dadc8c94222840c3b177abb511ef40a2 : SUCCESS in 9m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db658f351b5a47deb69fb5f45311daf5 : SUCCESS in 3m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2248b40e015147e3a23ceedc468de72f : SUCCESS in 5m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f173431f6ce34032a106b1dd8df92d40 : SUCCESS in 12m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1535493290114db7a217e7a7c671ca55 : SUCCESS in 17m 32s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/26b2d1522fed4f6b9588b625586aaab8 : SUCCESS in 1h 03m 49s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/1c829bcffd6449559bab30f19fb7355e : SUCCESS in 57m 59s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/5805b95ec87a4d5a84982318cd73defa : SUCCESS in 1h 11m 52s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e1215b97d05b4123acbc6c4846ebfa75 : SUCCESS in 58m 08s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cda457870e4546bea1d2e3f0196dc799 : SUCCESS in 53m 05s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/434a1e2364154cb4932d03aa3c3a2340 : SUCCESS in 52m 53s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c45b7c5ecd644ad2a3e5de3ef8e51986 : SUCCESS in 23m 38s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/3f109badc6e1453591c87b8437f89dee : SUCCESS in 20m 59s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/64fb258b7a7f4f948b6cdb34f045e7f2 : SUCCESS in 23m 54s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/69843fffcec04b389e397ff703916d62 : SUCCESS in 20m 01s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/835dcbc5dee04d2293f31a70dda201a8 : SUCCESS in 18m 34s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ccdddadf5d348b3916ec131c9337bd3 : SUCCESS in 19m 05s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d9f18d95e8fd4a55bd84e826d95f1cd1 : SUCCESS in 1h 12m 09s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/3824e6374bfc4ab3bbd488e56a276074 : SUCCESS in 1h 19m 28s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e0d75d1c728c4659b35767e267b6d1d4 : SUCCESS in 1h 42m 19s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f73858e95d6429c9aa0ff9d8a2bd249 : SUCCESS in 1h 18m 57s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/de75f14d2fd747c38578fcb4ea9da293 : SUCCESS in 21m 38s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/bcaa941d452a4511bc8a5dcb29abdd91 : SUCCESS in 18m 55s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/928c803e66ff4d1e88cdf764176cfc36 : SUCCESS in 17m 54s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4797c1b4867a48d4827cb324d5dfb188 : SUCCESS in 19m 00s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/207a7e56c8ab46d4aa47a94514a1db06 : SUCCESS in 18m 51s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1dbedc4926794ee987700adfcb4a9eb9 : SUCCESS in 39m 53s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca383f8615a442bd943d323aac408d65 : SUCCESS in 30m 38s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2c88d07a8e0441dabde80546c5deda7b : SUCCESS in 19m 02s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/454503d7d19e4a20af42eab3ac4442bf : SUCCESS in 19m 03s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/79946219f4cd4b7da05c2b5e6340e528 : SUCCESS in 16m 06s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/71c6c7b84ab44e05a600f4f6776b6337 : SUCCESS in 13m 50s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/90df87f26a6c418782df5e81c089b03f : SUCCESS in 11m 18s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/906058da14f542e3bd24ce764edf5c47 : SUCCESS in 16m 51s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/72f1861af0f14f09b299bb582cf629e7 : SUCCESS in 24m 53s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/14ac4f27e4e44a168a6f888c8e71ecd6 : SUCCESS in 18m 08s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36a048e646d94a508aec08c7b73be07f : SUCCESS in 14m 50s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e0e3035c02241f8be04c4334feae285 : SUCCESS in 17m 29s","accounts_in_message":[],"_revision_number":4},{"id":"611c205406b6072afd345330ee8b9f8d7264f71b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-03-09 14:06:56.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"1dc5d57bb3f5559ed2903874438c7281c43c8bec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-09 23:18:53.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"2c60c84314c890107e7ae5a7ddd944d21819b07a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 00:38:24.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/992f4abc44314ce0a5760933241c9777\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8edf347fdab840b5b51b3d2ac7cf183c : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4eff2182611749d8aa8cfd2b17d25021 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/855df110cb3044e49e8f04f34a2ed0c5 : SUCCESS in 6m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9335259da7c64c14be372041e8ab96c0 : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae36d0a5c2a54e2697c518ce5f03e699 : SUCCESS in 9m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c5aaf54fcf4f4c528a1a00e93195fa59 : SUCCESS in 4m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a4cb7ec0c4864a2a9cf7741975ee1e32 : SUCCESS in 5m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1b1ffc88a1f745fd9c2176c35f351e18 : SUCCESS in 14m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/673bc07510334143b53f5c4a8cd092c8 : SUCCESS in 16m 42s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e221b3259d754f468d8d273f802d45f5 : FAILURE in 43m 08s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/5d3f60fc977542f6933fbd416c9bf168 : SUCCESS in 51m 56s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/b11b4034f37a4500aebea76b0e7a5d27 : SUCCESS in 1h 01m 41s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/0bea3d4bc1654316a2ce9f1b34cd5355 : SUCCESS in 54m 00s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/06d8771e323447fa971dd17e8aaa53e0 : SUCCESS in 50m 42s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bd1944b3b508430b899ce5c5d9ff3fc9 : FAILURE in 39m 53s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7fbd67ba67324b3e9249d2c9fa304cbd : SUCCESS in 21m 00s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/33d0bcec7536421fb89b72d29dba92ab : SUCCESS in 20m 47s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/75bb42562b3d4b2d8e5ec49b0562c638 : SUCCESS in 24m 38s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/36c7d180649b489e8fe368455b02295d : SUCCESS in 19m 45s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f0ad83163c94a8792a9aa2195d3a6eb : SUCCESS in 18m 07s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c42680027f2841fdb4f43018ce6b942a : SUCCESS in 15m 15s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d7b93bd796e949b1acd0f2de0acbf752 : FAILURE in 53m 03s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/83ec765ff42d4561ba174990dc456b49 : SUCCESS in 1h 01m 48s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fc12e8fa5888414c8974255a083fca57 : FAILURE in 22m 33s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02591749345f49e38306dc0be3a8287d : FAILURE in 39m 10s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d59d1fc38c414242989a2b5e8ce1e704 : SUCCESS in 20m 04s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/c60492e276104fdca1b13407a2e8cb19 : SUCCESS in 23m 28s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/d3a01894a7fd4a17a3e5c6a8793d29d5 : SUCCESS in 19m 17s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca54511e52db436b976208a28a273784 : SUCCESS in 15m 30s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a8f2d8ff1adc477fb5d980f8b2b4e498 : SUCCESS in 17m 17s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a39c13678415450e87340c6631a2241d : SUCCESS in 37m 00s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e73467359d414e47a65d73eb70d15c4d : SUCCESS in 27m 27s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/fc83dbbb562c44bcb4620dad7e9914cc : SUCCESS in 19m 03s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/07a0785ee4ed4354af5392e162077eb6 : SUCCESS in 16m 21s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d9a4a4a7e9e6468da5893d1e30be9aac : SUCCESS in 14m 45s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/058786c2ab6c48318331ffb62aa31eff : SUCCESS in 13m 56s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/45a85025b7124cef919aaa1e75ab713d : SUCCESS in 19m 38s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/5d74d4b452fd4b4a9ece185f78cd456b : SUCCESS in 18m 46s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/dfd3eaf8cff94cf485bf07d7e2e08a07 : SUCCESS in 13m 44s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b9dbf3afd3b44affa9fbe163b0c0c609 : SUCCESS in 16m 40s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/81c518c3ccc74056847a74f6380a646e : SUCCESS in 14m 24s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e0e4582429214f6189310c427f221f10 : SUCCESS in 13m 49s","accounts_in_message":[],"_revision_number":5},{"id":"7a40e6ae0ba50e6b6c1043db814fc5cf9ee224f2","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-10 09:35:39.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6ba5e7b8d2162ad2270d1d2ae3f45332b354d20c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 11:49:07.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d11662c887140f99b2acd1a135a2625\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2912a9b644b64e9eb9c0f47d48a7860d : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e84294953c3442d58cdeda51ea731cc0 : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa1674a6901f4e7b9953fc732e2a5fe7 : SUCCESS in 6m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7ad8f1394c04c05834393f223567210 : SUCCESS in 7m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1705a48956814edb8a84f31e81c1a463 : SUCCESS in 9m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b51bc083f5e949239daabfe671124cd3 : SUCCESS in 3m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f0fb2fb8b0434ca8b56205a1a24f4716 : SUCCESS in 6m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2779881af1ae438581f4668013b426ed : SUCCESS in 12m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/93a218073ad2404a9c69ba47834bbcf7 : SUCCESS in 15m 43s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7b75d5eb1ca44c9b907c47c8b7c54ab7 : SUCCESS in 1h 11m 20s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/906303882ab64b22b32f57e573e8ec3a : SUCCESS in 1h 14m 03s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/b897ac2a876b4c4f8b329accf5c8782b : SUCCESS in 1h 14m 10s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/33764ef6cf774520b834924d048d88fe : SUCCESS in 1h 09m 15s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92768f5db0d64bd0804bb3ba412402be : SUCCESS in 1h 33m 27s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b1bf7c3a4687433a9499c3dbbe3d0ebf : SUCCESS in 58m 29s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/46d0efb92c64448ba597a62f6b801bfc : SUCCESS in 23m 20s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/1afe226de97b433fbd6387ffa1bb131f : SUCCESS in 22m 15s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/29eb32c518f643d4a74c8e87534a5454 : SUCCESS in 22m 08s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/fd64da49925942edb7cf56e1eef78d48 : SUCCESS in 21m 58s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ccd25ff53394ae78a2efebe04500011 : SUCCESS in 17m 07s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c2229b907eb7470884f1df2f9c3abf8b : SUCCESS in 17m 47s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/4e8c7e2473274640a2891fe409a28634 : SUCCESS in 1h 11m 30s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/9b03a203664d4f24b958dab65af793ea : SUCCESS in 1h 05m 56s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/059c4f89b1f94917b7911c2aea43e588 : SUCCESS in 2h 01m 16s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/326dbf5768cc45f9b9d242e87048fe29 : SUCCESS in 1h 24m 11s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7719c253a4a54e27b923aaa621029411 : SUCCESS in 23m 55s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/1805e9d9ad504c248df4bd026f323d3d : SUCCESS in 21m 09s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/32434587f4ee4d299216d80b426dc7fc : SUCCESS in 21m 39s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c70199a644da477f9ed46ad376510ad7 : SUCCESS in 16m 18s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef410d7c8f014be9a312a18b688addc3 : SUCCESS in 27m 39s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9359f94279d54c66a6a66533b870fa56 : SUCCESS in 47m 07s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/886c6063a3754afa8cf7837459ecd192 : SUCCESS in 32m 50s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/9d4eddd5dfdc4c16aeb3efbbc954f517 : SUCCESS in 17m 35s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/a47889c4093d4da7a67bb678d0807aeb : SUCCESS in 17m 41s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7be5e455d75241989c9f85e949fc407a : SUCCESS in 18m 03s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a13e0864ed64479f96259d739e7c13d3 : SUCCESS in 12m 24s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/27b3e42ab38842d79a90088da9e7d909 : SUCCESS in 14m 24s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/095329999d3c49599381a78e0618c7a7 : SUCCESS in 19m 37s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/e15a5e578d00423a962b72d8cd8ef6dd : SUCCESS in 20m 12s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a5dc0802d77d4c63957cf51265821ec2 : SUCCESS in 18m 10s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d83747c888c40dc99b3ff9c817ddc79 : SUCCESS in 14m 20s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a5d01ba331694ba186c6e6c3070e8fba : SUCCESS in 15m 36s","accounts_in_message":[],"_revision_number":5},{"id":"d083f5b8ecc3c21e5d7481531cfcb8cc514c9c18","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-13 08:06:10.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"644aba087b3e1f1fb0f7947d3325f2be389431d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 08:06:37.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"84ebd1b37cbd914bb35433e581852b1676ba4b96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 10:05:03.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41ce1955aa00429fa79981ee823ca9cd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88505263fea24082813d43340cfe8894 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/400d4dbf38fc4af8be20f9a932b3fa40 : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d08017cbad44059bcce8644f24458fb : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/811f889f7f1e450f800a12f70d7dedad : SUCCESS in 8m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1839542b40ff4a5bbb6fd44cf1950f9d : SUCCESS in 3m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cde0b39c229e4fa6885eaa64a6af3c04 : SUCCESS in 5m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d3997ded30b44f90acf091288eecd0bc : SUCCESS in 15m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a455b4ee350e4b07aeba69696f603f8f : SUCCESS in 16m 38s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/10492a60b3ce4394bcc4a68093227da5 : SUCCESS in 1h 12m 33s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/ee6daa50e8d748e983f296dd450b2d19 : SUCCESS in 1h 01m 31s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/95f93306c24d42d6afb7770c9f41174b : SUCCESS in 1h 01m 25s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7188cb2cd444db19d104b5353df24bb : SUCCESS in 49m 28s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e0ae9737020f4a3ca2feb04411768c70 : SUCCESS in 56m 03s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d853a0edd30f40a8bb4c03a4eb123b76 : SUCCESS in 25m 14s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/13c33cd84ff044efb72e192285129cac : SUCCESS in 23m 22s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/38f1702acec0455eb3ccdd6a1286d787 : SUCCESS in 18m 25s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e939979f4a314bcb85b6a5506822a930 : SUCCESS in 16m 08s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/84827e1251b3449ba9907af66dab782c : SUCCESS in 20m 13s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/3a1751f9388544c7a343e6247185d235 : SUCCESS in 59m 30s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/74d890d75e254bc88ea54cb69d912571 : SUCCESS in 1h 04m 08s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8c70ad4de5304413aef74ef1cfafb534 : SUCCESS in 1h 53m 55s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/19563f5d3b4b4aad93f812a96451965e : SUCCESS in 1h 01m 31s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e18038c4afe748148c14e1b8d25e709c : SUCCESS in 16m 43s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/3a19621b3b464f7b95d3cfc9ad8a5661 : SUCCESS in 18m 50s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/38f76b9bd5334d8d9f5480f252d1052a : SUCCESS in 20m 19s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d532d398e57b41ea867232882e34599d : SUCCESS in 15m 39s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f8546a1dce2c45b0b496ba06457060c0 : SUCCESS in 20m 05s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4d11f2a568d648fb92a78a93c49f8c95 : SUCCESS in 34m 35s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ffc84c0dc0a14d63829503d86a56a99b : SUCCESS in 35m 54s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f17bdfbd1b354e84aec0cb5719fb4b32 : SUCCESS in 17m 55s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/7ad1adbf19fd485298140c81d84e13ea : SUCCESS in 16m 46s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/27d9f25014e0436b9a29812bc14969ee : SUCCESS in 15m 51s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4737fde3d73b46698f6c9dec68ebce6d : SUCCESS in 11m 05s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b21e4131cde454497a47087fbe28d67 : SUCCESS in 12m 50s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/bfc67087626b4e0b8009bd4382b39080 : SUCCESS in 23m 00s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/207f4c4139d84bf8b5ec9e118bc39de0 : SUCCESS in 20m 00s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9445226156e5402a976777ca41a0499b : SUCCESS in 19m 33s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c802074e639c4d829abcfe388cb9853e : SUCCESS in 14m 23s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/724b5d0902864414b5c7660e29b535df : SUCCESS in 24m 14s","accounts_in_message":[],"_revision_number":5},{"id":"d1dbd4cc953b2f14c8aa50b1dc4fc5f13a21d402","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 10:05:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"46a2161a0d1684026a84694b62752140fca45915","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-13 10:06:07.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/504e04a9e66c46e1a74b572554a85812\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e6cb963955b4d51ade219e94e139de1 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/864420a5500c4b83bea4af141998fe1f : SUCCESS in 44s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"c5be87367b1a63758ab5c15b1616ed966b18da1f","revisions":{"e8bbeb9dfe2b2f21df444e8205d9783fae35f9a3":{"kind":"REWORK","_number":1,"created":"2023-03-08 22:34:32.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/11/876911/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/11/876911/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/876911/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/876911/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/876911/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/11/876911/1"}}},"commit":{"parents":[{"commit":"5a92bd778be8709977baa8ecc20937cc4a140360","subject":"use OS python to create kolla-ansible venv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5a92bd778be8709977baa8ecc20937cc4a140360"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-08 11:57:02.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-08 22:34:32.000000000","tz":0},"subject":"Fixes repositories files name in Rocky Linux 9","message":"Fixes repositories files name in Rocky Linux 9\n\nDistributions moved to lowercase names with RHEL 9 release.\nThis patch also removes previously created uppercase files.\n\nStory: 2010637\nTask: 47604\nChange-Id: I8bf75ce44b9ec1fdd477fcddf2deebf2929d9fc1\n(cherry picked from commit e0ff9a0283bd870df72ea7dec62d55ef0b8ed972)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e8bbeb9dfe2b2f21df444e8205d9783fae35f9a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e8bbeb9dfe2b2f21df444e8205d9783fae35f9a3"}]},"branch":"refs/heads/stable/yoga"},"0cf6944f80ef12cccf47eb3f54ecbdd1f3dc2b32":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-03-08 22:35:05.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/11/876911/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/11/876911/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/876911/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/876911/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/876911/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/11/876911/2"}}},"commit":{"parents":[{"commit":"9c696fd516e496583c562680fd41e0f096ce61bc","subject":"CI: get repo directories from nodes for troubleshooting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9c696fd516e496583c562680fd41e0f096ce61bc"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-08 11:57:02.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-08 22:35:05.000000000","tz":0},"subject":"Fixes repositories files name in Rocky Linux 9","message":"Fixes repositories files name in Rocky Linux 9\n\nDistributions moved to lowercase names with RHEL 9 release.\nThis patch also removes previously created uppercase files.\n\nStory: 2010637\nTask: 47604\nChange-Id: I8bf75ce44b9ec1fdd477fcddf2deebf2929d9fc1\n(cherry picked from commit e0ff9a0283bd870df72ea7dec62d55ef0b8ed972)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0cf6944f80ef12cccf47eb3f54ecbdd1f3dc2b32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0cf6944f80ef12cccf47eb3f54ecbdd1f3dc2b32"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"},"f8e8fbad087a917d34c962664cb67a0154937bdb":{"kind":"REWORK","_number":3,"created":"2023-03-09 09:36:19.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/11/876911/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/11/876911/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/876911/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/876911/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/876911/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/11/876911/3"}}},"commit":{"parents":[{"commit":"9c696fd516e496583c562680fd41e0f096ce61bc","subject":"CI: get repo directories from nodes for troubleshooting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9c696fd516e496583c562680fd41e0f096ce61bc"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-08 11:57:02.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-09 09:36:04.000000000","tz":60},"subject":"Fixes repositories files name in Rocky Linux 9","message":"Fixes repositories files name in Rocky Linux 9\n\nDistributions moved to lowercase names with RHEL 9 release.\nThis patch also removes previously created uppercase files.\n\nStory: 2010637\nTask: 47604\nChange-Id: I8bf75ce44b9ec1fdd477fcddf2deebf2929d9fc1\n(cherry picked from commit e0ff9a0283bd870df72ea7dec62d55ef0b8ed972)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8e8fbad087a917d34c962664cb67a0154937bdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8e8fbad087a917d34c962664cb67a0154937bdb"}]},"branch":"refs/heads/stable/yoga"},"4b2a6afd2e5c64aae7e5e3dd9366adb34668632b":{"kind":"REWORK","_number":4,"created":"2023-03-09 09:37:42.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/11/876911/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/11/876911/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/876911/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/876911/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/876911/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/11/876911/4"}}},"commit":{"parents":[{"commit":"9c696fd516e496583c562680fd41e0f096ce61bc","subject":"CI: get repo directories from nodes for troubleshooting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9c696fd516e496583c562680fd41e0f096ce61bc"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-08 11:57:02.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-09 09:37:33.000000000","tz":0},"subject":"Fixes repositories files name in Rocky Linux 9","message":"Fixes repositories files name in Rocky Linux 9\n\nDistributions moved to lowercase names with RHEL 9 release.\nThis patch also removes previously created uppercase files.\n\nStory: 2010637\nTask: 47604\nChange-Id: I8bf75ce44b9ec1fdd477fcddf2deebf2929d9fc1\n(cherry picked from commit e0ff9a0283bd870df72ea7dec62d55ef0b8ed972)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4b2a6afd2e5c64aae7e5e3dd9366adb34668632b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4b2a6afd2e5c64aae7e5e3dd9366adb34668632b"}]},"branch":"refs/heads/stable/yoga"},"c5be87367b1a63758ab5c15b1616ed966b18da1f":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2023-03-09 23:18:53.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/11/876911/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/11/876911/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/876911/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/876911/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/876911/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/11/876911/5"}}},"commit":{"parents":[{"commit":"5864191fe9e6cc2851daf7252fa4bedd1e2d2280","subject":"CI: get repo directories from nodes for troubleshooting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5864191fe9e6cc2851daf7252fa4bedd1e2d2280"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-08 11:57:02.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-09 23:18:53.000000000","tz":0},"subject":"Fixes repositories files name in Rocky Linux 9","message":"Fixes repositories files name in Rocky Linux 9\n\nDistributions moved to lowercase names with RHEL 9 release.\nThis patch also removes previously created uppercase files.\n\nStory: 2010637\nTask: 47604\nChange-Id: I8bf75ce44b9ec1fdd477fcddf2deebf2929d9fc1\n(cherry picked from commit e0ff9a0283bd870df72ea7dec62d55ef0b8ed972)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c5be87367b1a63758ab5c15b1616ed966b18da1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c5be87367b1a63758ab5c15b1616ed966b18da1f"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
