)]}'
{"id":"openstack%2Fkayobe~877556","triplet_id":"openstack%2Fkayobe~stable%2Fyoga~I78504d5efb649377395f8d21c5087b2318b2f46a","project":"openstack/kayobe","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-03-16 08:07:18.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-03-16 11:15:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I78504d5efb649377395f8d21c5087b2318b2f46a","subject":"Rename Rocky Linux 8 templates to avoid conflicts","status":"MERGED","created":"2023-03-15 21:25:11.000000000","updated":"2023-03-16 11:16:44.000000000","submitted":"2023-03-16 11:15:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877556","meta_rev_id":"65f658ab7a4a4275e5c4a13800e709de275d2fc5","_number":877556,"virtual_id_number":877556,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-16 11:15:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2023-03-16 09:03:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2023-03-16 08:07:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2023-03-16 09:03:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-15 22:33:03.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-03-15 23:43:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-16 09:03:40.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"6bdb2fd5431a32acf3eb220576295cb160eb25e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-03-15 21:25:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cd06b216e1d4822981df841a3150991167a3413d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-15 23:43:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d16447f702543059e8bf625ed3f1843\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b85f46881f7f4598b1a81c2c82d825a2 : SUCCESS in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0acd63872ad94504941d5539577f2935 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/faf2e2398ebe456dbcc469b61a6a9ec8 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9862efa7909b49ecae0ea91345593efd : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c707d965e024bbf9133754e8bed6ff3 : SUCCESS in 9m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b8c4fe9413254dc6ba0aa15ae24cd2ad : SUCCESS in 6m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/aab959c77c8649cc8536c65f21f95c63 : SUCCESS in 11m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cbd5c3b79bc54307aa7fd0465f3d7f3d : SUCCESS in 17m 21s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c35cfa91c3034c1d986d4480e1062057 : SUCCESS in 53m 32s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/7350961ed2424aa09d3b6f785e95464d : SUCCESS in 1h 12m 05s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/a8466c796fb2464fac02a53cc88f8fa2 : SUCCESS in 1h 14m 43s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/9d9043b917f744899dd87c8d9d4026e7 : SUCCESS in 1h 03m 16s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c1d989800813477d90e31614e05168d8 : SUCCESS in 46m 34s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0a36bafc71447b69a6ff7090244cc72 : SUCCESS in 58m 38s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7a23a829208c44fe9a3adba38995f43d : SUCCESS in 25m 13s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/ce4f79aaa3a048cfa87ebbd73283d2c3 : SUCCESS in 21m 39s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/936ce22f9ba94a7691a8e7edb9db3dc3 : SUCCESS in 21m 13s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/afac3892f78548328ff5ac33b038253a : SUCCESS in 20m 37s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c882912127ce48e496dce7e8171e5b68 : SUCCESS in 15m 05s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0bcb00946824ea7ab5bbbfd933b1ad0 : SUCCESS in 18m 39s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/dbd612186e194b4e9635e6600722a3b7 : SUCCESS in 1h 14m 51s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/6755f46a9ed145639a83533f777bc374 : SUCCESS in 52m 49s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/73037eb9fa6c4bb7a45d76179b1fa513 : SUCCESS in 2h 12m 34s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f496269e7f943cd8bd71c6898d1b356 : SUCCESS in 1h 02m 07s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/64f1998920794728a41fb4c293edd1f9 : SUCCESS in 23m 10s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/f384f726c53341c39354334c7bdbae35 : SUCCESS in 20m 22s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/376d00d5c5df4e83bce06f2733e22dc8 : SUCCESS in 17m 06s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5670698a7c3413da0d201f7551e60d0 : SUCCESS in 19m 38s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/871347a670954af8a3db32f4bacaf727 : SUCCESS in 13m 34s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c20197da0ddc44038a5a074472dec117 : SUCCESS in 41m 01s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c1b19df6661342998931d8e7d0798edb : SUCCESS in 27m 17s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/1e7fa028abaa49a48f3e1892ff6a2acf : SUCCESS in 20m 12s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/ef9df5d3c89e4c82aad41a7844b8c6e5 : SUCCESS in 16m 32s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ab8c7b10f3174782b52c96179cf2da97 : SUCCESS in 14m 45s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8840d2a101743618d01f4ee1918237c : SUCCESS in 13m 24s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c3818d09eed84f008856f63f52a26a8f : SUCCESS in 13m 57s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/07c322aeccbd4f918a0adf432ce31a41 : SUCCESS in 21m 44s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/c64194dcd0a14c349b16ff4f0e77f6d4 : SUCCESS in 18m 19s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1c71666fcb4b4d158bfde33813897b8b : SUCCESS in 15m 02s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6028a19c8bba49dd9a398852fac88672 : SUCCESS in 12m 53s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a355205396744149b8f769f81445d150 : SUCCESS in 14m 42s","accounts_in_message":[],"_revision_number":1},{"id":"3d6f5daea223fb82a4217fba7c7734463f872228","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-16 08:07:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f81819350614eb39de4c3de28a524a2675e714ab","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-03-16 09:03:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5fa5ed9dc275d6da3646131a529a232df9d43fef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 09:04:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"349571cd808f48fde9281d9d523316c52a415e92","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 11:15:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da6a633f9467496da941e795eb1f6b2a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60b54a6870ee4a738bdd58c27dfda431 : SUCCESS in 6m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/231b7e4229e9457f9946d644325ef243 : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c67d8cd88a1f49ceb4019e2e1bb35793 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bb744dcaa8f4fd28aec9b6e87f8fc7f : SUCCESS in 8m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/006b93b8b614423da995b8b045ef7aae : SUCCESS in 6m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3d22f0e293e649058f59d4ae0644072f : SUCCESS in 12m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bdf2959477294a2dad8ca64264cb08d5 : SUCCESS in 16m 22s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/38ffd1fd835d4e7e8cc28a72e8c44de4 : SUCCESS in 1h 10m 39s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/c7f59477169445239006c6001f42eb33 : SUCCESS in 57m 53s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/99cbe94f09614087835cc6f85d043cb6 : SUCCESS in 1h 03m 19s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5bf76636701411591c1e67a0c3984cc : SUCCESS in 50m 05s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f82271e79bf546f68b8a7a6f54978453 : SUCCESS in 53m 56s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/5a07bba192ce4c9cbf17c7dc08004f2c : SUCCESS in 25m 59s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/9911fa122d8040c894f75394cca25510 : SUCCESS in 20m 12s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c9915302c83142d281d43b1768539fbc : SUCCESS in 18m 22s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1670e881d7a341b482e6f1dc60f55925 : SUCCESS in 21m 14s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4a30ff8f17234e329d356e70b609d50b : SUCCESS in 17m 15s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/46493782188c42b3997e8ff4b7b53154 : SUCCESS in 1h 14m 11s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e0e816539f054e34b796b7b4b6a6e5f3 : SUCCESS in 1h 13m 57s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/31607f3fadd74e4ba78ab004b0eeeed2 : SUCCESS in 1h 56m 16s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fbfe27377ef442128e8a42fbdc73fa28 : SUCCESS in 1h 16m 11s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/af295cd9bf2a4f62980cb3593ffbb83a : SUCCESS in 23m 12s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/5a1011dc476f44c3b2dcf2aee8f0a5a4 : SUCCESS in 23m 17s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c1bcf2799397412fbf7091b22e8aea92 : SUCCESS in 16m 00s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e28848631214b94969fd88d44e23ff4 : SUCCESS in 22m 18s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0c5567d5fa14e2885072675287fa4f3 : SUCCESS in 19m 06s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/75e2678c611d46ac953ef2806d242d86 : SUCCESS in 36m 49s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab05be4777c1423a9c9691493aec72dd : SUCCESS in 30m 55s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/8b5a4ac8fd9540f4b6e17687f9a44672 : SUCCESS in 18m 43s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/1d6ec0c162ea4fa0bc4f489d8f29c1a8 : SUCCESS in 14m 19s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3fe986fa7e7240d79f5d6fd05ba9c54d : SUCCESS in 17m 09s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cff85abe919349d8a6938b6cb2adca8c : SUCCESS in 15m 44s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/64dd68b227f84d19baf7729950846719 : SUCCESS in 11m 30s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/e29c461a0eb44702bc889bf06e3d7311 : SUCCESS in 21m 28s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/59b86e8b1cbe47bb8915ffd07a26defe : SUCCESS in 20m 21s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ec410c6aede548eeaa5b09f6c1202a2c : SUCCESS in 14m 33s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2a39ad92ad0458391043a476e495c83 : SUCCESS in 17m 10s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d992884fbba547a7a207bbe07ab1f8fe : SUCCESS in 13m 24s","accounts_in_message":[],"_revision_number":1},{"id":"a24883d24b47162b59540fdfc14ee4760959f086","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 11:15:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"65f658ab7a4a4275e5c4a13800e709de275d2fc5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 11:16:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/733eaaa109354c7ab902de162da2cedf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/950372aa0547426786161e8d7b3e8973 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a106e60c89d2d1759bcfef834b3a90cb9f8c356a","revisions":{"a106e60c89d2d1759bcfef834b3a90cb9f8c356a":{"kind":"REWORK","_number":1,"created":"2023-03-15 21:25:11.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/56/877556/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/56/877556/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/56/877556/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/56/877556/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/56/877556/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/56/877556/1"}}},"commit":{"parents":[{"commit":"b951fbc647016a67473e9118f0ff84cdc2399772","subject":"Run handlers before configuring veth interfaces only on Rocky Linux 9","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b951fbc647016a67473e9118f0ff84cdc2399772"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-03-15 21:24:54.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-03-15 21:24:54.000000000","tz":60},"subject":"Rename Rocky Linux 8 templates to avoid conflicts","message":"Rename Rocky Linux 8 templates to avoid conflicts\n\nThe backport of change I8bf75ce44b9ec1fdd477fcddf2deebf2929d9fc1 to\nstable/yoga [1] supports repo templates for both RL8 and RL9. This\nresulted in two files differing only by case: Rocky-Extras.repo.j2 and\nrocky-extras.repo.j2. This causes issues on case-insensitive file\nsystem since Git confuses the two files.\n\nFix by renaming templates to Rocky8-* to avoid conflicts and adapt the\ntask accordingly.\n\n[1] https://review.opendev.org/c/openstack/kayobe/+/876911\n\nChange-Id: I78504d5efb649377395f8d21c5087b2318b2f46a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a106e60c89d2d1759bcfef834b3a90cb9f8c356a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a106e60c89d2d1759bcfef834b3a90cb9f8c356a"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
