)]}'
{"id":"openstack%2Fkayobe~877637","triplet_id":"openstack%2Fkayobe~stable%2Fxena~Id7e4bd7f199ad9dcb21d5db082e7a187cb310df9","project":"openstack/kayobe","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-04-12 14:00:25.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-04-12 17:19:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id7e4bd7f199ad9dcb21d5db082e7a187cb310df9","subject":"ironic: switch to kernel_append_params","status":"MERGED","created":"2023-03-16 18:30:52.000000000","updated":"2023-04-12 17:20:55.000000000","submitted":"2023-04-12 17:19:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877637","cherry_pick_of_change":834357,"cherry_pick_of_patch_set":1,"meta_rev_id":"6c17af553098e815c11d055137a2a9d48fa3d859","_number":877637,"virtual_id_number":877637,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-12 17:19:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-04-12 14:00:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-04-12 13:51:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-04-12 14:00:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-16 18:30:52.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2023-03-16 20:37:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-04 20:34:42.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-04-12 13:51:31.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"b96cff0c295ab02ed866077c92d68b5e9252c3d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-03-16 18:30:52.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"4ef4ebb96062a03c078a6c2ba2876be1a47057b8","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-03-16 18:36:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d47cef0ed1602131f109dad804efaa9f64f2f6e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 20:37:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/592807351f1d41a1ac2a94874c47ac2d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e0e0e30acec4df591d5305c4920c596 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46dd875ccdb844cb87e722b23869f75a : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0e1bb6e363f4289bdbaf9f80c6be0fc : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/512b42ad99af489280f6ae2f91119057 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa3539ebd91e450486b8647c1f8b239e : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffa84d98068b4f2a8c245bf728263f7d : SUCCESS in 8m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c16bd6c6cbdb4973bfd43c196021735a : SUCCESS in 5m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/09eedda8fed54821baea1c869300fb22 : SUCCESS in 12m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cd0ed2a3d3ea42b0aaeec6d1d14dc7a0 : SUCCESS in 19m 23s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/cb96f6c4235945fa8ceccf8400af8e0b : SUCCESS in 53m 45s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/f001ad4012e84ef5af98e9896d361652 : SUCCESS in 1h 15m 02s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fcc05c3086c34508878d1162e3b9943e : SUCCESS in 50m 33s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/2018777f2f47428c88e96e3394d4826d : SUCCESS in 1h 11m 10s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e3940c00750b45e1900c943fca7416c4 : SUCCESS in 22m 37s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/70311718cc1f4393bf06b851cd3f01b6 : SUCCESS in 25m 41s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/68aa245e5256445199d8af5b1de23c6f : SUCCESS in 13m 10s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b268773dd1294ff8a1baa38326403966 : SUCCESS in 2h 01m 39s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd3666c406fe4228b888aca9790a56ac : SUCCESS in 1h 08m 07s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/9e212820b9dc42c4a8aa5d8bbf0f0aef : SUCCESS in 20m 45s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/a10d47c5c59947acb9d19a39567c9489 : SUCCESS in 17m 07s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2be30b8d6e8c4bef8a51fde81ac109ae : SUCCESS in 15m 13s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/6062c6503f8745828b669f1d5f923d22 : SUCCESS in 36m 49s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0a20dbe8f43d4ff79c85c83f6e44668e : SUCCESS in 30m 31s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16819f67fe344fb5b1069dbb0267acea : SUCCESS in 21m 32s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2966256d6cbf49cd92912a3bdedc2c46 : SUCCESS in 19m 11s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/f52b0df539c843e399359bfb81fdfaba : SUCCESS in 17m 26s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c10ef4fa0f14c59804ec0de557f6f13 : SUCCESS in 12m 49s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/0e5f84d73da44d25a5f55447cbd941b3 : SUCCESS in 21m 07s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/f556e8e0516c4bd281fbb4388e079db8 : SUCCESS in 21m 11s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d520e0c17424aababcb5bcb2c05cdc2 : SUCCESS in 13m 59s","accounts_in_message":[],"_revision_number":1},{"id":"b7100a3f604983b430374adf35dcf6d385dded71","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-12 13:51:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fe70b7b681dcd693515ffe6d82e29df4ed0c3180","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-12 14:00:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"895c36aa1f3527417b0ac55f8a72b3b94d177473","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 14:00:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"68a1532d1dbc8632b5c01f685abc417857d1842f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 17:19:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4d1d38a8ab34ab89e5e5e1a29a2af33\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5001a1941b4401bbbde51b531793061 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5186d3034c0a4fc0a127be27248ec5fa : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c0164c5d9fb4a90af22ae4e43c2732b : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d4384ae2e4045acab285fa2411704c0 : SUCCESS in 8m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1508fafe3d9543a4964ef28686e4f1c2 : SUCCESS in 5m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/642046cc0ecf422eb66ae86441ed3259 : SUCCESS in 12m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a5bf668444b74cfda9def72f34c22f75 : SUCCESS in 18m 24s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b4593bb933bb4c43a98e99721efebdeb : SUCCESS in 1h 06m 20s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/363aa691fdae4f7f8a191291aedc2f67 : SUCCESS in 1h 13m 34s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a783a0d86dab42dcad81f33dd5f3e7ae : SUCCESS in 56m 17s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/70c2c94fc45642ebb4f5b4bd8f9ee784 : SUCCESS in 1h 09m 53s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/aabb945cdc4c49b681db57a4a2771594 : SUCCESS in 26m 31s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/543d5e7d9e95415dba8d4bc17916ea6e : SUCCESS in 22m 06s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c634a71521f432fbb9377378f05b3ee : SUCCESS in 16m 03s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6a6e2b598fe242a6a1f10e93471d1c44 : SUCCESS in 1h 57m 06s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/631d0f8ef4f847528bb300b5882b989d : SUCCESS in 1h 10m 50s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/055d52815cbd49ff92b6d8625acd11e8 : SUCCESS in 21m 48s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/4b2a65c7f3fc431d95f1c8f64374c029 : SUCCESS in 21m 46s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc854c33a01745aab44eeaf903b5487a : SUCCESS in 15m 56s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5db4cd6cf5f64b32b001da503cc649d3 : SUCCESS in 33m 10s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a8cf363a9734bcbaf61841bab5f2a61 : SUCCESS in 23m 31s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/1580dd02a0f6493dbe9331c61f153812 : SUCCESS in 21m 47s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/5b045eb18ca44b58aa21934ec3129954 : SUCCESS in 17m 44s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/76cf5cb23488465b9bef81b2a7643536 : SUCCESS in 11m 21s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/8420291d024241179bebe360f90acfcb : SUCCESS in 19m 47s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/1418a5d44ac6492ca157f124090dab18 : SUCCESS in 19m 25s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/82e61beb70b14ccda94ba3a9cbb15481 : SUCCESS in 16m 36s","accounts_in_message":[],"_revision_number":1},{"id":"3f2a87d8c52888e04ad3c33f77c9f7dbad8a442e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 17:19:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6c17af553098e815c11d055137a2a9d48fa3d859","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-12 17:20:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e322087406d64d889ef9adb149ed0dc2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f2b3e168e0a4a258a13e43d4178236f : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3cb796147e6a7731ca7a7538e1a1e33c725d7a79","revisions":{"3cb796147e6a7731ca7a7538e1a1e33c725d7a79":{"kind":"REWORK","_number":1,"created":"2023-03-16 18:30:52.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/37/877637/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/37/877637/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/877637/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/877637/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/877637/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/37/877637/1"}}},"commit":{"parents":[{"commit":"56af77a25ed3952b114e69c26fa934ba0efbce41","subject":"Fix Rocky 8 cloud image URL","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/56af77a25ed3952b114e69c26fa934ba0efbce41"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-03-18 20:41:23.000000000","tz":0},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-03-16 18:30:52.000000000","tz":0},"subject":"ironic: switch to kernel_append_params","message":"ironic: switch to kernel_append_params\n\nSince I4fb42d376636dc363cd86950ed37de4a3d28df73, kolla-ansible sets\nkernel_append_params in ironic.conf, instead of the deprecated\npxe_append_params. Make the same change in the Kayobe ironic.conf\ntemplate, to avoid getting shadowed by the kolla-ansible default.\n\nThis also fixes the overcloud TLS job, which started failing because it\nlost the ipa-insecure kernel option, making the Ironic API callback\ninvalid.\n\nChange-Id: Id7e4bd7f199ad9dcb21d5db082e7a187cb310df9\n(cherry picked from commit 994af643e67b1efc9368e38997aaa7c412eabf3a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3cb796147e6a7731ca7a7538e1a1e33c725d7a79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3cb796147e6a7731ca7a7538e1a1e33c725d7a79"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
