)]}'
{"id":"openstack%2Fkayobe~877649","triplet_id":"openstack%2Fkayobe~stable%2Fyoga~I6af90d02e619d85b25ad364dc17d411008fd792e","project":"openstack/kayobe","branch":"stable/yoga","topic":"elasticsearch-opensearch-stable/yoga","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-03-23 12:25:58.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-03-23 12:25:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6af90d02e619d85b25ad364dc17d411008fd792e","subject":"Support OpenSearch configuration","status":"MERGED","created":"2023-03-17 09:18:55.000000000","updated":"2023-03-23 12:26:57.000000000","submitted":"2023-03-23 12:25:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877649-elasticsearch-opensearch-stable/yoga","cherry_pick_of_change":866940,"cherry_pick_of_patch_set":2,"meta_rev_id":"fee970e79e985130dd33ed786f6790fc83a75aa8","_number":877649,"virtual_id_number":877649,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-23 12:25:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2023-03-23 08:57:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-23 09:15:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-23 09:15:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-17 09:18:55.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2023-03-17 12:12:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-21 16:09:10.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-03-23 09:15:40.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"8a1681ded28813e0b144bf364a6c3f074e68dfcd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-17 09:18:55.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* ansible/inventory/group_vars/all/kolla","accounts_in_message":[],"_revision_number":1},{"id":"f6b6782d5409c52a12e9c20065d54ec5b357579a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-17 09:19:59.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"74d67f1738454e3650ad444752d032fae907d6e7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-17 09:21:25.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"33b80966c23953a2f4125aa2ce6981420dc8387b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-17 09:22:05.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"54a8cdd6d6e87316cc9bf275f3e31a98830d67f6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-17 09:22:34.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"7aa62cc3df7b84c9c3f2409a61df120c884e155c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-17 09:24:44.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"8690ebe3fbbc0cf446389191cfd1bb36a69293b3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-17 09:25:26.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"90456af8ad69e6462451d06753c6a4c651ff759d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-17 09:26:06.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"edda7bdc301e915f56d2653b150e9d76b0b0cb6f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-17 09:26:43.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"526adbf352942e6fedd6fc7e4c6bbf24431ef57c","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-17 09:27:13.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":9},{"id":"a1d12306f5500eff17fdb6d6c9a799cbab272259","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-17 09:48:32.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"307d14b3727616b5b4e1b04aa6eeef35739c56d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-17 09:51:11.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"e46d8c9057afdc47cba252f4e4cab51ea84f980a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 12:12:56.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4324889f49a74a2fb9c53815fb161b21\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01b257244a20411ea2f91f56b23ea37f : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d4cda72711a4e96b9eb211a272ec027 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cde6a473fb674df8ac58d7d77bc3c237 : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/22b12de71fbf495e94922933a0ea800e : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebdbc84d993543da9329e6616041b2ee : SUCCESS in 9m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c462254e93248babc7f8dc5dda65577 : SUCCESS in 3m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8925caea392146e5b1bbc8005853e8da : SUCCESS in 6m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/484374719f6f45ee816ee850c0c678cf : SUCCESS in 15m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b5f7f8a2f4904d659e13e66c27b72b7d : SUCCESS in 19m 46s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ab3b0aaa926f4592bb94317ba9c9a4fb : SUCCESS in 56m 18s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/106df682ef894098821361f9234a86f1 : SUCCESS in 1h 16m 18s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/508e28cd20814caa864158ce5d4cfc29 : SUCCESS in 1h 00m 16s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/c85c1ba479b24622a2e877ef73959a3b : SUCCESS in 55m 37s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9dd00f36fbfe423c8132bf2c1e13da75 : SUCCESS in 54m 35s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f2c515cd2864f53b38c1f0bf092e579 : SUCCESS in 58m 59s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ea04c8657f50494d959bf23ee2c323e1 : SUCCESS in 26m 19s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/26a2216bb7154bbb8ef922447c49204e : SUCCESS in 20m 17s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/7ef531cf50d94381898864f360573b96 : SUCCESS in 21m 31s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9c611375b6d84f3da283f50b715c9036 : SUCCESS in 21m 55s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f02c2d63a2b24106a37d70d97417a585 : SUCCESS in 21m 57s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b31eb0ddb47c495c9d61070527fec0b2 : SUCCESS in 15m 04s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/78315026ca624469af2472e11e12a644 : SUCCESS in 1h 06m 09s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/0c259b88b0bf4d67b25ec19474199eaf : SUCCESS in 1h 07m 16s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0848461cb0cd4cc4a2a74f22a4b92bc0 : SUCCESS in 2h 17m 21s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e85977af8834917866f35b6cab90412 : SUCCESS in 1h 16m 34s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/1f69603200e4401f9d477e24aaf5d71f : SUCCESS in 19m 34s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/7251e784301841998ba43a03b1d3a50b : SUCCESS in 21m 24s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f5177ab5b87741068050bbd5c2881219 : SUCCESS in 18m 43s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f17f5cee6d8476e9bb1ec8b4284f720 : SUCCESS in 18m 04s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb6c58d7b2f441668db47b2f6e1bc059 : SUCCESS in 16m 08s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3cf24f5fb79543eaa2ff630900fd6903 : SUCCESS in 36m 41s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6a3e658829c4765a3b5b95df6e4edac : SUCCESS in 27m 43s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/297c19a79cae4c788835482aacc3c2ea : SUCCESS in 17m 54s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/3c696abda43e49839547bea82752fd3c : SUCCESS in 17m 17s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/51d590e6569f4235ae26b931dfa1b25a : SUCCESS in 15m 41s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56b15ab0e472496b9f2893d44f74324b : SUCCESS in 12m 40s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/098d209d731242d6a785840d8efa4d37 : SUCCESS in 15m 00s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/1765b599cff24bf2914b093d52b8ba15 : SUCCESS in 20m 08s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/23327540dcf14423af969d19ac8f8fda : SUCCESS in 22m 27s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/20480dba25d64f0ea590791cbd40459e : SUCCESS in 18m 18s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fcf00426467c4897aedb49b9b499f464 : SUCCESS in 16m 54s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/592f6d47affe4186b579abf911ca4ee4 : SUCCESS in 16m 11s","accounts_in_message":[],"_revision_number":11},{"id":"1dc95af9d31766a9ba289befc2cb73dc2e3ab632","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-21 15:19:27.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"477f1d388fbe56806a40892f02384571c0b331be","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-03-21 16:09:10.000000000","message":"Patch Set 12: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"d245383a5bc4e21aeb420e32895c268a0ff861e6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-21 16:16:02.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"be54a3b519d61ff4949d492d73080ec7a6dc23e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-21 16:18:57.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"19278fa87c103531b159909f6539150e648ab7c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-21 16:20:07.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"8b0d46d0ec84718b26ff2b0ffdd625be4a034240","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 18:42:26.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6dd548f6d3e942bbb398e5d7f6c60d2a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe3a5d5db42a48bf98cb1e1a0589b9aa : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83559b6d092e4bbdb1c6f9c5ded8637f : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7f2e509a87f4bdc9332066c9fabf01d : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b499b3c18008435faf76fc98f45879ca : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cc4db5407c84ef097c33c7c1dc585e4 : SUCCESS in 8m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ae82bddcc9543a4b3ebaec217157263 : SUCCESS in 3m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1333546e23e44c69955b3377086cf956 : SUCCESS in 6m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0207cc283865490fba441da9a2519883 : SUCCESS in 16m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3c37dcc57fb847709cc60b048dec1789 : SUCCESS in 17m 01s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/2f0694f5acdb4efeb9929e6340601f8c : SUCCESS in 1h 22m 20s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/12c8697a02954d448f648683d28dbf08 : SUCCESS in 1h 01m 36s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/ba0d8a8aa30a463f8b3f1486c05cbb59 : SUCCESS in 1h 02m 11s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e008a1db6e1b4b09aef26c1fda0133ae : SUCCESS in 1h 06m 17s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e2b70ca10ea4a57bdb1cb37b2bd7509 : SUCCESS in 1h 00m 43s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2075629d6f9a441789db1f3653e33344 : SUCCESS in 52m 01s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e6762e62582844a7923fbe86d1a2d110 : SUCCESS in 26m 44s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/be4a6764546a4a38ba15332a1ef4c54f : SUCCESS in 21m 41s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/9ee93f5a8aef4f0e9ef02c902f3afe68 : SUCCESS in 25m 04s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e83e9dd826714f1b8adf78fc28cf2f26 : SUCCESS in 21m 42s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/86e8bd5d5a314b6aa73bd39316a26a67 : SUCCESS in 18m 48s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eb06c6e243714c96abafc8ea889f1cc3 : SUCCESS in 17m 23s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/286bec47ab204287a43b611a44b2cdd4 : SUCCESS in 1h 16m 51s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/cddb0d48685c408791397007d189710e : SUCCESS in 1h 04m 30s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/db92234d2d4846269560003ec5de9457 : SUCCESS in 2h 07m 25s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5563855736cf4a2f906a31cab6652ef0 : SUCCESS in 1h 09m 39s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/15e248be5d864079aa93202b322c0f6a : SUCCESS in 24m 47s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/e540769ef9bc403492199116a5156bb3 : SUCCESS in 19m 28s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/7ea14cadb0ba45938a3ca2c48315394a : SUCCESS in 24m 09s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d006be96584487faff34cbe20cc063c : SUCCESS in 17m 22s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f3143e5251e74119a432f66ed6abe9a5 : SUCCESS in 22m 52s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/eadf40c3dc9248b097cf0b42774ada15 : SUCCESS in 33m 29s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ed6f1b2ceb0448d8c919f6f95ea6d08 : SUCCESS in 35m 35s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e0350fc26dbe4f98b6bdf718a200ecd6 : SUCCESS in 17m 16s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/208bb103b3e64d29a05421662edc6e7c : SUCCESS in 17m 56s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f41465997d0a43a48834a54ea9283450 : SUCCESS in 16m 50s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/45f0c9875d194bd380e041cbe2dfcf12 : SUCCESS in 14m 35s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a7a17f8ec044701b826fa1d6e797185 : SUCCESS in 13m 50s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/b24f1abfa90d491a840e9a03683e9188 : SUCCESS in 23m 14s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/120d8b557a7c4e218a9655a41ebd4aa5 : SUCCESS in 21m 03s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8af1ee4fcd9d46169f022d3e5becf426 : SUCCESS in 20m 20s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca90cb8dcb574481bdaf388db2df2404 : SUCCESS in 14m 20s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e830a9aa4c964b6db967ec6a86c67a74 : SUCCESS in 14m 07s","accounts_in_message":[],"_revision_number":14},{"id":"925c0035837cf3ad7b141d0b353bf339953a5764","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-03-22 09:16:51.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"324b1f1a67d788456dc35deef8228b0580d3585a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-03-22 09:18:38.000000000","message":"Patch Set 14: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"08dadd37606e2ae16ab6e59b2cc558cdafb59509","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-22 10:54:03.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"d69e7018e14c907f6e110d4601bdcf1b6156ae70","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-22 10:54:31.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"e14337b1496ebf828fa93b5564ca757ad68da333","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-22 10:54:44.000000000","message":"Patch Set 15:\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"09556ce261fe1592e62ac98154687956d8a5c695","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 12:18:09.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3a5a2cf428fc45b6831f7ba5d5c1876c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1076fd4013e481e9a2b80d23c204131 : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f4fdba1b54548e192dc4e2324ebaca7 : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3ec4326434443b2bb5d15e9c0378d62 : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a12c39169e8f4f46a1c085b6dae56b5c : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67e8340da9f04d278b89abda154e241f : SUCCESS in 10m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a6330719ed746839cc3a764fde03c0c : SUCCESS in 2m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/617d6c957e374659bd3e80a31b479387 : SUCCESS in 5m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fb9e9ec871954b008ea2fa5bbd69e879 : SUCCESS in 13m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2484104724994943beda4f45421adf26 : SUCCESS in 18m 23s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/33c6c25d7c1a48b0abb354fc036f1e8a : RETRY_LIMIT in 7m 28s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/84b7eccba062481aa1660a2382d399b7 : RETRY_LIMIT in 9m 20s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/8913a1aa629a481ebb78ff3cc68ea329 : RETRY_LIMIT in 6m 26s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/1c309720e1fd4b76b143c71c42efc710 : RETRY_LIMIT in 7m 55s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9bc118807e9d45cc8e0de2fb2a39fa20 : RETRY_LIMIT in 5m 49s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7857beb1ffdb42e3b517ab7171ec36ca : RETRY_LIMIT in 7m 21s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0653ad23d7c54cf0be8d1bcff5a2c295 : RETRY_LIMIT in 6m 50s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/cfbb14f15cea497ba8791016ff937fbd : RETRY_LIMIT in 6m 35s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/9740f3cd502247ffa6cd1569f8712100 : RETRY_LIMIT in 5m 49s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e3e6f51f3a53465182f8e7aff3ed10e1 : RETRY_LIMIT in 7m 59s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d03402faeb141a49fe7686b4b462a69 : RETRY_LIMIT in 6m 00s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/578489abb46e4dc192a6134f586ce47f : RETRY_LIMIT in 5m 20s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/4fc7a0131d274298a15e839521f0f558 : RETRY_LIMIT in 8m 46s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/68b4b7c50160458f904015b581cf82fc : RETRY_LIMIT in 7m 46s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1adf3f14b5d6455abb0e013e178ab1c0 : FAILURE in 1h 19m 08s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/470830744a954cf88a14c0002720df35 : FAILURE in 56m 20s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7c6da6f0893c4f2abffc02c65adb2c9d : FAILURE in 9m 27s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/af13404815f6432498db1c29a7a873b2 : FAILURE in 9m 56s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/020f706bdd97432398c36bd5c3910b1d : FAILURE in 9m 01s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/818f2d4345a94fa19880a074b594494c : FAILURE in 7m 56s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1427172e7d2944e59a88d386f07c735f : FAILURE in 9m 21s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/689c253397ca4fe8b505a8be9de0ece1 : FAILURE in 31m 19s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02eae5af1a7e4dfa89b7732b7f85231e : FAILURE in 25m 48s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/0f16d4bf08b747d5b5566d45f24b816d : FAILURE in 10m 16s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/3f09c2d1b244416e8033c8f1ce4704a9 : FAILURE in 9m 27s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ea648d9bb6fb4d32baafb9ec5d14e2b3 : FAILURE in 9m 07s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cae5c81740294f48a9288712af64efc4 : FAILURE in 7m 47s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/960718aedd14489e997a7d70a6ed8b4e : FAILURE in 8m 09s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/2e111603689e43878efd389497016af3 : FAILURE in 9m 39s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/5c891154138a48349b41bb2ffe4b4eb0 : FAILURE in 8m 09s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c089501afbae470c877208f0174d21f4 : FAILURE in 8m 41s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a680b01f4b94447491342c3a44fef6bf : FAILURE in 7m 57s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d1ab0e1b66a44218996bf3984e6f3856 : FAILURE in 7m 38s","accounts_in_message":[],"_revision_number":15},{"id":"bd0fe4a5d16d2a54625b05d07221e3a4ea64c82f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-03-22 12:43:31.000000000","message":"Patch Set 16: Published edit on patch set 15.","accounts_in_message":[],"_revision_number":16},{"id":"33c2c12f7e212a6471fe035036a805d113dae96d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 14:59:53.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8013c7eb22de4d9f9ac33d1f03e323ea\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d318e12ce11499382a5a734c6553251 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/545501b944db4366bce52b503a3091ad : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2a1d18e67694987a6b3950c3ade8a34 : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f1106b313c349678e805d760c18af38 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdcc16103b7f40a3884f9433fa44ba80 : SUCCESS in 9m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/740f20a4d9b94d6b879a9db9e6ab932c : SUCCESS in 3m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a74d5c75945047818dfd8c9a042d64fc : SUCCESS in 6m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dca72bf72833417faedb2f9a56a334cb : SUCCESS in 13m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e62a4aaa312744f8b96e3d200abc0cec : SUCCESS in 16m 28s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8911a76244e049fa8fd57bc559bbaf68 : SUCCESS in 1h 03m 14s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/4250fe8934bf423ea5bd38a3f5e441bd : SUCCESS in 1h 09m 02s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/d2844726e4e1467ebd6406e9099ba7d9 : SUCCESS in 1h 01m 42s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e104f0bc200e437181fe71f34183d144 : SUCCESS in 58m 06s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7a6a7bcaf4848849389814761ca4589 : SUCCESS in 50m 50s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/138897241e2d4cd782451d244a4e6e0a : SUCCESS in 50m 57s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/343d16970a6446a1b069c1c781dbd2da : SUCCESS in 27m 00s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/7811d67b70a04fd899a354f710e40b35 : SUCCESS in 21m 01s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/337ab56e4c5e4b59a8c4280cbec03208 : SUCCESS in 19m 58s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4ce134206197491a822bf28186eb5f96 : SUCCESS in 20m 14s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eda9a50a02484b80b7073b00f9b870bb : SUCCESS in 14m 23s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/029ce0bc8031491f8e448ebe09f91eb6 : SUCCESS in 15m 27s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c674bebb43bc4be0aec3718970f3dbcb : SUCCESS in 1h 08m 49s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/20b260ceadd044ad920be0e7647bffbc : SUCCESS in 1h 06m 47s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bfbabf2d0f5044d684df77ca307918fd : SUCCESS in 2h 09m 00s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b50d0bbf5a154c3b91a16c60d2624716 : SUCCESS in 1h 13m 31s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/98fb8ed428f74eefaceba549e6a8873e : SUCCESS in 21m 06s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/710a4b258fc64c0c8720f0867812dd75 : SUCCESS in 22m 14s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3b93ae78666e4568a0491a099e5fcc11 : SUCCESS in 17m 29s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd329314218748a2b037bd2dcd50238f : SUCCESS in 22m 44s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b5cbe2e2b534e149d57f48ab442d75a : SUCCESS in 18m 45s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ca032b955e314db4ac2462319c64461b : SUCCESS in 39m 49s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94a960abdd9f4378bbb5ff3c4233143a : SUCCESS in 35m 39s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/a9bec9cf78a8413ea2b46b166173ba82 : SUCCESS in 16m 36s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/56e32619894e4e09b50778aa2489337b : SUCCESS in 17m 29s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/55c9771da52a45cea0b017185a767b0d : SUCCESS in 16m 08s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11f0b30cfb6d4904ad07ff3f5d6cab86 : SUCCESS in 12m 48s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9283574e93744ebea24b0ee4d7c076bc : SUCCESS in 10m 58s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/0b48ff69b63449d798f42bfe0dbc8892 : SUCCESS in 24m 25s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/bf25e3404fb448b28a322c609686da11 : SUCCESS in 20m 27s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d4cdd573955349418b8774dffb7933e6 : SUCCESS in 17m 57s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a5c5589b58d45e99042f3f11cd37c90 : SUCCESS in 18m 14s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e991f160ec304037b46cf6c35341284e : SUCCESS in 15m 34s","accounts_in_message":[],"_revision_number":16},{"id":"53d15dafb969353091811d47eecd5807b2dbfaa7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-03-23 08:57:40.000000000","message":"Patch Set 17: Published edit on patch set 16.","accounts_in_message":[],"_revision_number":17},{"id":"6f28414578d59b76358c4cc1fb4daf89542683a2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-03-23 08:57:58.000000000","message":"Patch Set 17: Code-Review+2","accounts_in_message":[],"_revision_number":17},{"id":"b1cfb96ed3fcdc891315bd7b0b4f9c1fb3f130db","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-23 09:15:40.000000000","message":"Patch Set 17: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":17},{"id":"90eba34b370560f4730ec5de89adcb8badfdd3b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 10:48:00.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bddb818e202a4b00abdeda0f9ea24ad2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e6c812a2b4249abaddebd7ae078d469 : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60dbd9f6a0d64875b1d7eb991dce5c34 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1f608d5fe3645e1b4acb35d5498d4ad : SUCCESS in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37c888d961e145fca798faef99dd8f3b : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4822fb05243d40389c4c7f122a35dd33 : SUCCESS in 11m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fefdb10d09cd4b53be1acfb38adda4ce : SUCCESS in 3m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/544988baa2c6471380ebc36a76efc6d7 : SUCCESS in 6m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8ca72c5e2a1048b88a1634b60690040c : SUCCESS in 16m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bab295a5714a4339acb05cb5480ba639 : SUCCESS in 18m 26s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ebd1bae4d56c498fa14d298635af002d : SUCCESS in 1h 20m 15s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/adb9992b27cd4ac7b7e43c8039bf98f4 : SUCCESS in 1h 04m 17s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/9c1ac9785120492e8982aba72d333688 : SUCCESS in 1h 17m 06s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/55b626c99471447d8f608b4cee9b37df : SUCCESS in 1h 02m 01s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/48e2fc6c8c444c65ac35b897fed02b1b : SUCCESS in 1h 05m 44s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ad7c9d4df5f4bcfac14d3a6b1cb72b9 : SUCCESS in 1h 02m 20s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/447ffe641c524ba4bc582ff8e4a95c94 : SUCCESS in 26m 10s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/5d23843dceec4b6d8825d9d59dd76d52 : SUCCESS in 24m 16s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/e4138d76d686443793613b8bd24897f9 : SUCCESS in 23m 03s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0c0928999c48426e9e1a55de8bdcd86d : SUCCESS in 19m 17s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52b5ea8956434d13adfb5bfa67fa7a17 : SUCCESS in 23m 00s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cbf8a89ccf16403782ed63a98a508f0b : SUCCESS in 16m 49s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/50004d8a158b491c92b3af5cd56c4798 : SUCCESS in 1h 23m 25s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/01e07785a6c242f8b029f8ff8719c412 : SUCCESS in 1h 10m 07s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cc177007b73042a0a2347b9c4293fa22 : SUCCESS in 1h 38m 25s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b3ed60da7e34fab8dd789c1eee5c811 : SUCCESS in 1h 26m 55s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/32db20d233954680a4c0eab20c276a02 : SUCCESS in 23m 10s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/0ea8b1240b5b4a89a3d5ec506382d5d2 : SUCCESS in 29m 35s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/2ee69c0a367843b7bb7bf56b158f1f90 : SUCCESS in 23m 21s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c3d82d94446746709f5715f262252a41 : SUCCESS in 20m 48s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e2ab06eae2a846f3b1785cea7c8f9ae9 : SUCCESS in 25m 51s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bdf3bf5208204de6b9579754a31dfb3e : SUCCESS in 36m 56s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ac441303c4841a19b29677969a8ac12 : SUCCESS in 32m 51s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5da770c41680430fae062e6be644b9fb : SUCCESS in 21m 53s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/35bdebbd5e2e4ee4acebb908d56e6efc : SUCCESS in 17m 35s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/27364a03fc6c410596ef8799bf162a7c : SUCCESS in 16m 21s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/662861448bf04f9cb2301735d85e07bc : SUCCESS in 17m 09s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59f63d20f1674ec9ac74d648bf1ca987 : SUCCESS in 16m 54s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/cdbc5a5472b44de08b8152395a785596 : SUCCESS in 23m 38s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/f745fded01fc4609846a0c5c96056ede : SUCCESS in 21m 34s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/87e6513c1c0f47a681f0aa3111804a80 : SUCCESS in 20m 17s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3d5fcdab5b3460f98d3c9f17190d1d2 : SUCCESS in 17m 41s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/200ee7feb422488f94893ee2ba64458a : SUCCESS in 17m 23s","accounts_in_message":[],"_revision_number":17},{"id":"60047ff42a762e174c5d17dd8ea01a6d9237e4ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 10:48:27.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"f83053dcecf81813081841c11ca6a85a45647a79","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 12:25:57.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84eceb7ae2df4bf9bbc19c1a8abb0c74\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5fc44dbc19a4a8f905864c7db279bed : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d168dce4d1ab47a0b4b4291f3dd92b52 : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9da06d3898ef4aa9b179cbf3382e5829 : SUCCESS in 6m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad82249dcf874cf8b9ccf73f9b900a62 : SUCCESS in 8m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f5ae4f3f30f4f87b54372e749076a09 : SUCCESS in 3m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4b3ccf3f53084324b2e5b10b5cda5316 : SUCCESS in 5m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2876312423984474bec2bb2397c2e81d : SUCCESS in 12m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e4d1d22704be4358891c726c1c188ba4 : SUCCESS in 22m 26s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/79013fe230274b88bd0a70eb197cd240 : SUCCESS in 1h 02m 03s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/f7887f7faa7740829c8f368753cdaa42 : SUCCESS in 1h 04m 07s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/1a28edeaf85543529b1a379ba485cfee : SUCCESS in 46m 57s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4633b05303df4faea02eb9247479ac89 : SUCCESS in 51m 06s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/68d968f20c7e4923b1c7e3c274a4a850 : SUCCESS in 54m 19s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/674b02e589fa4d589735713957f713db : SUCCESS in 22m 54s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/5ab2f86b35d346dc80a913553d69f20e : SUCCESS in 18m 50s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9b4a2e3de9214a8aa3b943d5f04782a7 : SUCCESS in 22m 37s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/170d6cb1c5e24f91a8e5501831b44926 : SUCCESS in 15m 46s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/725b0180e30242ea81e32b698db833d3 : SUCCESS in 16m 59s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/370b3e5f9e9a488b8f6be16eb4b2c5bc : SUCCESS in 1h 01m 07s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/b5306fceb30447ee8eef9156c86ae6f6 : SUCCESS in 1h 17m 07s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6ce2c3aedc1a4d3ea800b538640d3263 : SUCCESS in 1h 31m 36s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd95420929874f3b9bb96585e6a9a46b : SUCCESS in 1h 12m 22s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/05c622d294814c55826d6b1880b8b561 : SUCCESS in 21m 45s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/38d7bc755e4d493e9e972786d3f90198 : SUCCESS in 19m 08s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f041cc2cc2704c4d8073921d6564c485 : SUCCESS in 14m 13s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7a3b5a6413847aeaf854aaf7d3817c0 : SUCCESS in 17m 22s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/571ff87b9155487f95ce7d79075b4b1b : SUCCESS in 15m 57s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ecdf3aa96e42473bb0e22315ae162d9c : SUCCESS in 35m 22s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bcc0ae7e6c0e4da6b6611752e004f34d : SUCCESS in 25m 47s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/6226376a2ed049da805915479a29e681 : SUCCESS in 17m 39s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/473399e55a83405789d4fd1d2c964139 : SUCCESS in 16m 48s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/198dd89bf55d405da9f956b0fcf9caca : SUCCESS in 15m 32s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/158a46085df84f22a83e7b908eaf3855 : SUCCESS in 10m 21s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b58b4127b534ac5a94d989a7c50715a : SUCCESS in 14m 06s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/c5c0f908b89849199a655e716dcc97f7 : SUCCESS in 22m 00s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/bc22d92733784e5194755bc837d52ae3 : SUCCESS in 16m 01s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a2987a069ffe4b61998f25f48e6114d9 : SUCCESS in 18m 40s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/187534c8aba14342911cd1dca62dfc2b : SUCCESS in 15m 55s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe4c6f70b81e47c59f5b148ca2740d66 : SUCCESS in 15m 41s","accounts_in_message":[],"_revision_number":17},{"id":"457e0f982119e926e8aa1a13ec9abe5237107253","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 12:25:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":17},{"id":"fee970e79e985130dd33ed786f6790fc83a75aa8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 12:26:57.000000000","message":"Patch Set 17:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1b0db7081734c71b1a1fcbf9f0a6879\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a538ba01a0cb40579906bedce76eef50 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25ccf4c48cc84953805376a4c0c85a06 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"d0482b658c15cdcdf078deb1c9173157704d26dc","revisions":{"c05befd6e72252cf9ffb2c7b57d5f27d57a7a8f8":{"kind":"REWORK","_number":1,"created":"2023-03-17 09:18:55.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/49/877649/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/877649/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/877649/1"}}},"commit":{"parents":[{"commit":"a106e60c89d2d1759bcfef834b3a90cb9f8c356a","subject":"Rename Rocky Linux 8 templates to avoid conflicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a106e60c89d2d1759bcfef834b3a90cb9f8c356a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-07 22:08:04.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-17 09:18:55.000000000","tz":0},"subject":"Support OpenSearch configuration","message":"Support OpenSearch configuration\n\nChange-Id: I6af90d02e619d85b25ad364dc17d411008fd792e\n(cherry picked from commit 12ab39568e029843f800bd0e24bf507a7ea9d364)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c05befd6e72252cf9ffb2c7b57d5f27d57a7a8f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c05befd6e72252cf9ffb2c7b57d5f27d57a7a8f8"}]},"branch":"refs/heads/stable/yoga"},"2e0502762355d129692f151095aec6d08d3c4af0":{"kind":"REWORK","_number":2,"created":"2023-03-17 09:19:59.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/49/877649/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/877649/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/877649/2"}}},"commit":{"parents":[{"commit":"a106e60c89d2d1759bcfef834b3a90cb9f8c356a","subject":"Rename Rocky Linux 8 templates to avoid conflicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a106e60c89d2d1759bcfef834b3a90cb9f8c356a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-07 22:08:04.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-17 09:19:58.000000000","tz":0},"subject":"Support OpenSearch configuration","message":"Support OpenSearch configuration\n\nChange-Id: I6af90d02e619d85b25ad364dc17d411008fd792e\n(cherry picked from commit 12ab39568e029843f800bd0e24bf507a7ea9d364)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2e0502762355d129692f151095aec6d08d3c4af0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2e0502762355d129692f151095aec6d08d3c4af0"}]},"branch":"refs/heads/stable/yoga"},"70735192b3716ad5fc2af26008491b1bacfe71ce":{"kind":"REWORK","_number":3,"created":"2023-03-17 09:21:25.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/49/877649/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/877649/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/877649/3"}}},"commit":{"parents":[{"commit":"a106e60c89d2d1759bcfef834b3a90cb9f8c356a","subject":"Rename Rocky Linux 8 templates to avoid conflicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a106e60c89d2d1759bcfef834b3a90cb9f8c356a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-07 22:08:04.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-17 09:21:25.000000000","tz":0},"subject":"Support OpenSearch configuration","message":"Support OpenSearch configuration\n\nChange-Id: I6af90d02e619d85b25ad364dc17d411008fd792e\n(cherry picked from commit 12ab39568e029843f800bd0e24bf507a7ea9d364)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/70735192b3716ad5fc2af26008491b1bacfe71ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/70735192b3716ad5fc2af26008491b1bacfe71ce"}]},"branch":"refs/heads/stable/yoga"},"368284515a1e45b1ad38816fc56effacde56fe86":{"kind":"REWORK","_number":4,"created":"2023-03-17 09:22:05.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/49/877649/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/877649/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/877649/4"}}},"commit":{"parents":[{"commit":"a106e60c89d2d1759bcfef834b3a90cb9f8c356a","subject":"Rename Rocky Linux 8 templates to avoid conflicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a106e60c89d2d1759bcfef834b3a90cb9f8c356a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-07 22:08:04.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-17 09:22:03.000000000","tz":0},"subject":"Support OpenSearch configuration","message":"Support OpenSearch configuration\n\nChange-Id: I6af90d02e619d85b25ad364dc17d411008fd792e\n(cherry picked from commit 12ab39568e029843f800bd0e24bf507a7ea9d364)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/368284515a1e45b1ad38816fc56effacde56fe86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/368284515a1e45b1ad38816fc56effacde56fe86"}]},"branch":"refs/heads/stable/yoga"},"3dbc71b473fd76c4d682167ed27485fc4cbb2b27":{"kind":"REWORK","_number":5,"created":"2023-03-17 09:22:34.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/49/877649/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/877649/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/877649/5"}}},"commit":{"parents":[{"commit":"a106e60c89d2d1759bcfef834b3a90cb9f8c356a","subject":"Rename Rocky Linux 8 templates to avoid conflicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a106e60c89d2d1759bcfef834b3a90cb9f8c356a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-07 22:08:04.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-17 09:22:34.000000000","tz":0},"subject":"Support OpenSearch configuration","message":"Support OpenSearch configuration\n\nChange-Id: I6af90d02e619d85b25ad364dc17d411008fd792e\n(cherry picked from commit 12ab39568e029843f800bd0e24bf507a7ea9d364)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3dbc71b473fd76c4d682167ed27485fc4cbb2b27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3dbc71b473fd76c4d682167ed27485fc4cbb2b27"}]},"branch":"refs/heads/stable/yoga"},"ab53faa93107854ebe129ad46abb67363eae7d1b":{"kind":"REWORK","_number":6,"created":"2023-03-17 09:24:44.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/49/877649/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/877649/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/877649/6"}}},"commit":{"parents":[{"commit":"a106e60c89d2d1759bcfef834b3a90cb9f8c356a","subject":"Rename Rocky Linux 8 templates to avoid conflicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a106e60c89d2d1759bcfef834b3a90cb9f8c356a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-07 22:08:04.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-17 09:24:44.000000000","tz":0},"subject":"Support OpenSearch configuration","message":"Support OpenSearch configuration\n\nChange-Id: I6af90d02e619d85b25ad364dc17d411008fd792e\n(cherry picked from commit 12ab39568e029843f800bd0e24bf507a7ea9d364)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ab53faa93107854ebe129ad46abb67363eae7d1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ab53faa93107854ebe129ad46abb67363eae7d1b"}]},"branch":"refs/heads/stable/yoga"},"2fe773710098fc3a660d71a01d482b3845147b75":{"kind":"REWORK","_number":7,"created":"2023-03-17 09:25:26.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/49/877649/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/877649/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/877649/7"}}},"commit":{"parents":[{"commit":"a106e60c89d2d1759bcfef834b3a90cb9f8c356a","subject":"Rename Rocky Linux 8 templates to avoid conflicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a106e60c89d2d1759bcfef834b3a90cb9f8c356a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-07 22:08:04.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-17 09:25:26.000000000","tz":0},"subject":"Support OpenSearch configuration","message":"Support OpenSearch configuration\n\nChange-Id: I6af90d02e619d85b25ad364dc17d411008fd792e\n(cherry picked from commit 12ab39568e029843f800bd0e24bf507a7ea9d364)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2fe773710098fc3a660d71a01d482b3845147b75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2fe773710098fc3a660d71a01d482b3845147b75"}]},"branch":"refs/heads/stable/yoga"},"631c4ed3f627bf9bf007cee0e1ca4e44822d70ac":{"kind":"REWORK","_number":8,"created":"2023-03-17 09:26:06.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/49/877649/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/877649/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/877649/8"}}},"commit":{"parents":[{"commit":"a106e60c89d2d1759bcfef834b3a90cb9f8c356a","subject":"Rename Rocky Linux 8 templates to avoid conflicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a106e60c89d2d1759bcfef834b3a90cb9f8c356a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-07 22:08:04.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-17 09:26:06.000000000","tz":0},"subject":"Support OpenSearch configuration","message":"Support OpenSearch configuration\n\nChange-Id: I6af90d02e619d85b25ad364dc17d411008fd792e\n(cherry picked from commit 12ab39568e029843f800bd0e24bf507a7ea9d364)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/631c4ed3f627bf9bf007cee0e1ca4e44822d70ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/631c4ed3f627bf9bf007cee0e1ca4e44822d70ac"}]},"branch":"refs/heads/stable/yoga"},"accd1785f98e109bc6ac5e61ff3751877381e49a":{"kind":"REWORK","_number":9,"created":"2023-03-17 09:26:43.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/49/877649/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/877649/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/877649/9"}}},"commit":{"parents":[{"commit":"a106e60c89d2d1759bcfef834b3a90cb9f8c356a","subject":"Rename Rocky Linux 8 templates to avoid conflicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a106e60c89d2d1759bcfef834b3a90cb9f8c356a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-07 22:08:04.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-17 09:26:43.000000000","tz":0},"subject":"Support OpenSearch configuration","message":"Support OpenSearch configuration\n\nChange-Id: I6af90d02e619d85b25ad364dc17d411008fd792e\n(cherry picked from commit 12ab39568e029843f800bd0e24bf507a7ea9d364)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/accd1785f98e109bc6ac5e61ff3751877381e49a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/accd1785f98e109bc6ac5e61ff3751877381e49a"}]},"branch":"refs/heads/stable/yoga"},"727251387a8166733dfbc82f35acb066b5636a1f":{"kind":"REWORK","_number":10,"created":"2023-03-17 09:48:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/49/877649/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/877649/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/877649/10"}}},"commit":{"parents":[{"commit":"a106e60c89d2d1759bcfef834b3a90cb9f8c356a","subject":"Rename Rocky Linux 8 templates to avoid conflicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a106e60c89d2d1759bcfef834b3a90cb9f8c356a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-07 22:08:04.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-17 09:48:32.000000000","tz":0},"subject":"Support OpenSearch configuration","message":"Support OpenSearch configuration\n\nChange-Id: I6af90d02e619d85b25ad364dc17d411008fd792e\n(cherry picked from commit 12ab39568e029843f800bd0e24bf507a7ea9d364)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/727251387a8166733dfbc82f35acb066b5636a1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/727251387a8166733dfbc82f35acb066b5636a1f"}]},"branch":"refs/heads/stable/yoga"},"e3ad7e6e5cfb608d4ad83462f150108c256717bd":{"kind":"REWORK","_number":11,"created":"2023-03-17 09:51:11.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/49/877649/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/877649/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/877649/11"}}},"commit":{"parents":[{"commit":"a106e60c89d2d1759bcfef834b3a90cb9f8c356a","subject":"Rename Rocky Linux 8 templates to avoid conflicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a106e60c89d2d1759bcfef834b3a90cb9f8c356a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-07 22:08:04.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-17 09:51:10.000000000","tz":0},"subject":"Support OpenSearch configuration","message":"Support OpenSearch configuration\n\nChange-Id: I6af90d02e619d85b25ad364dc17d411008fd792e\n(cherry picked from commit 12ab39568e029843f800bd0e24bf507a7ea9d364)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e3ad7e6e5cfb608d4ad83462f150108c256717bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e3ad7e6e5cfb608d4ad83462f150108c256717bd"}]},"branch":"refs/heads/stable/yoga"},"5a130e23baac9815123e060d263aa183ba56355d":{"kind":"REWORK","_number":12,"created":"2023-03-21 15:19:27.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/49/877649/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/877649/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/877649/12"}}},"commit":{"parents":[{"commit":"a106e60c89d2d1759bcfef834b3a90cb9f8c356a","subject":"Rename Rocky Linux 8 templates to avoid conflicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a106e60c89d2d1759bcfef834b3a90cb9f8c356a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-07 22:08:04.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-21 15:19:22.000000000","tz":0},"subject":"Support OpenSearch configuration","message":"Support OpenSearch configuration\n\nChange-Id: I6af90d02e619d85b25ad364dc17d411008fd792e\n(cherry picked from commit 12ab39568e029843f800bd0e24bf507a7ea9d364)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5a130e23baac9815123e060d263aa183ba56355d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5a130e23baac9815123e060d263aa183ba56355d"}]},"branch":"refs/heads/stable/yoga"},"d00a29f8e9031173aa888b06020ffb33c9b50914":{"kind":"REWORK","_number":13,"created":"2023-03-21 16:18:57.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/49/877649/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/877649/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/877649/13"}}},"commit":{"parents":[{"commit":"a106e60c89d2d1759bcfef834b3a90cb9f8c356a","subject":"Rename Rocky Linux 8 templates to avoid conflicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a106e60c89d2d1759bcfef834b3a90cb9f8c356a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-07 22:08:04.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-21 16:18:57.000000000","tz":0},"subject":"Support OpenSearch configuration","message":"Support OpenSearch configuration\n\nChange-Id: I6af90d02e619d85b25ad364dc17d411008fd792e\n(cherry picked from commit 12ab39568e029843f800bd0e24bf507a7ea9d364)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d00a29f8e9031173aa888b06020ffb33c9b50914"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d00a29f8e9031173aa888b06020ffb33c9b50914"}]},"branch":"refs/heads/stable/yoga"},"947ea921fd0d8563202980a7e11b81f75db15922":{"kind":"REWORK","_number":14,"created":"2023-03-21 16:20:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/49/877649/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/877649/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/877649/14"}}},"commit":{"parents":[{"commit":"a106e60c89d2d1759bcfef834b3a90cb9f8c356a","subject":"Rename Rocky Linux 8 templates to avoid conflicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a106e60c89d2d1759bcfef834b3a90cb9f8c356a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-07 22:08:04.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-21 16:20:07.000000000","tz":0},"subject":"Support OpenSearch configuration","message":"Support OpenSearch configuration\n\nChange-Id: I6af90d02e619d85b25ad364dc17d411008fd792e\n(cherry picked from commit 12ab39568e029843f800bd0e24bf507a7ea9d364)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/947ea921fd0d8563202980a7e11b81f75db15922"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/947ea921fd0d8563202980a7e11b81f75db15922"}]},"branch":"refs/heads/stable/yoga"},"78f5d4295bedfe8f4a34017c3b98bc1ff5af3868":{"kind":"REWORK","_number":15,"created":"2023-03-22 10:54:03.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/49/877649/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/877649/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/877649/15"}}},"commit":{"parents":[{"commit":"a106e60c89d2d1759bcfef834b3a90cb9f8c356a","subject":"Rename Rocky Linux 8 templates to avoid conflicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a106e60c89d2d1759bcfef834b3a90cb9f8c356a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-07 22:08:04.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-22 10:54:03.000000000","tz":0},"subject":"Support OpenSearch configuration","message":"Support OpenSearch configuration\n\nChange-Id: I6af90d02e619d85b25ad364dc17d411008fd792e\n(cherry picked from commit 12ab39568e029843f800bd0e24bf507a7ea9d364)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/78f5d4295bedfe8f4a34017c3b98bc1ff5af3868"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/78f5d4295bedfe8f4a34017c3b98bc1ff5af3868"}]},"branch":"refs/heads/stable/yoga"},"31959fb2c91ac85bd38f107cabb9234182bdf4f5":{"kind":"REWORK","_number":16,"created":"2023-03-22 12:43:31.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/49/877649/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/877649/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/877649/16"}}},"commit":{"parents":[{"commit":"a106e60c89d2d1759bcfef834b3a90cb9f8c356a","subject":"Rename Rocky Linux 8 templates to avoid conflicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a106e60c89d2d1759bcfef834b3a90cb9f8c356a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-07 22:08:04.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-03-22 12:43:31.000000000","tz":0},"subject":"Support OpenSearch configuration","message":"Support OpenSearch configuration\n\nChange-Id: I6af90d02e619d85b25ad364dc17d411008fd792e\n(cherry picked from commit 12ab39568e029843f800bd0e24bf507a7ea9d364)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/31959fb2c91ac85bd38f107cabb9234182bdf4f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/31959fb2c91ac85bd38f107cabb9234182bdf4f5"}]},"branch":"refs/heads/stable/yoga"},"d0482b658c15cdcdf078deb1c9173157704d26dc":{"kind":"REWORK","_number":17,"created":"2023-03-23 08:57:40.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/49/877649/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/877649/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/877649/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/877649/17"}}},"commit":{"parents":[{"commit":"a106e60c89d2d1759bcfef834b3a90cb9f8c356a","subject":"Rename Rocky Linux 8 templates to avoid conflicts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a106e60c89d2d1759bcfef834b3a90cb9f8c356a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2022-12-07 22:08:04.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-03-23 08:57:40.000000000","tz":0},"subject":"Support OpenSearch configuration","message":"Support OpenSearch configuration\n\nChange-Id: I6af90d02e619d85b25ad364dc17d411008fd792e\n(cherry picked from commit 12ab39568e029843f800bd0e24bf507a7ea9d364)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d0482b658c15cdcdf078deb1c9173157704d26dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d0482b658c15cdcdf078deb1c9173157704d26dc"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
