)]}'
{"id":"openstack%2Fkayobe~878207","triplet_id":"openstack%2Fkayobe~master~Ibffecaa8085bd38ebc8cded9a4bfebe77d59d515","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-05-25 14:13:46.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-05-25 11:43:22.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-05-25 12:31:47.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-05-25 14:13:46.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-05-25 11:54:37.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"Ibffecaa8085bd38ebc8cded9a4bfebe77d59d515","subject":"Bump up Ansible supported versions to 6.x/7.x","status":"MERGED","created":"2023-03-22 11:19:17.000000000","updated":"2023-05-25 14:14:46.000000000","submitted":"2023-05-25 14:13:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878207","meta_rev_id":"635b512d82ef7b3fc7180c770d29cef727ee7271","_number":878207,"virtual_id_number":878207,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-25 14:13:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-05-25 11:43:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-05-25 12:31:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-25 11:54:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-05-25 12:31:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-05-25 11:47:20.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-22 11:21:46.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-03-22 11:21:54.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-03-22 13:30:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-19 09:50:36.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2023-04-19 12:10:31.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"a4d24549d7f62bc6694b71635c042c0bacbb4d53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-22 11:19:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d8d5a1a1c3b7894e51e3a22501d9d71b37a30b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-22 11:22:39.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f5f401efa410a0437d5d9dcc7ecb7564b8fdbbc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-22 11:26:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d86d12c33adf97d549efe2960ea3b024e19c1c43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-22 11:27:34.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"513f33b491ccfcfa8299b92a3fb56905d355f1dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-22 11:31:54.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d1c5ecaf4c4f7c10b1bfdf2766e70e4d96ceab2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-22 13:30:51.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4b2608c381b147b596c049799f195603\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa49fa4640204e668cf9b1c2243a035c : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47f2980d05db44cc8ddac41020e55d2a : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/49cf57bfa5f74d0fa65ad6e1359d78ee : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fc100cb7632c425da99b1afc49c3554a : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea0dcfaa0fd6435e8dc1d84ae62d5be1 : SUCCESS in 7m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04d3e7acf7b5489aa811294eef2ba133 : SUCCESS in 3m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9b8c86b0880e475bbf57767311f3f7a4 : SUCCESS in 4m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/327549ddffbe4cf98520937ea886bee4 : FAILURE in 10m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fcfd1584b6ff4b4b873265a461931205 : SUCCESS in 16m 15s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/90d98793994540eba87514cee2e5c698 : SUCCESS in 59m 54s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/719f5904be5a4df0842772213ac9465c : SUCCESS in 1h 00m 50s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d6f0deb300148209faaa30aaf7a32d6 : SUCCESS in 59m 05s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/cf1cc4c68d334f9fa605323f46923cd7 : SUCCESS in 1h 06m 13s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ddc090aa45994c91ab07a40e4922fea8 : SUCCESS in 19m 53s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/0f4490d0cbe04f62850be6c6076e524d : SUCCESS in 18m 47s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e4ca2a98e9d54fbfbdc5785160ae93be : SUCCESS in 14m 23s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6dbcdc0f5933454c958f1fd607315bd5 : SUCCESS in 1h 51m 43s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0458e09918b4c27af0ed94c570f45ba : FAILURE in 1h 12m 16s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3af4cb5fcad44fa98646c226f4d5d2e6 : SUCCESS in 18m 04s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df848c9f97874f46a957a15790456bc3 : SUCCESS in 16m 24s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/517680cd476448289ab27766c6e2b702 : SUCCESS in 33m 08s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e6de863f369343aa8e7c59e6ad8533b9 : SUCCESS in 27m 21s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b0a7d6cc979a4ef69441b688c8d0f90f : SUCCESS in 27m 46s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4cba0a015dbe4d5796c00bcbd70bb0f6 : SUCCESS in 12m 45s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e42ef3c224a9418b9f869e7860e51b4a : SUCCESS in 10m 43s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a37bfdf8a3bc488cb8d490757e59a35a : SUCCESS in 28m 37s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/68fc4995e1834cf1b76f31a06ac4a1c8 : SUCCESS in 22m 10s","accounts_in_message":[],"_revision_number":5},{"id":"22f80f0cb6ae0cea677fddde887a5359d0a61417","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-13 13:13:11.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2061b5c2f7c271468bd4b87765b0c4511e8258b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-19 07:13:32.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"453dcaedbe8ae789b6282fbd125e024523c9bacf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 08:56:01.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ebfe6b6d7bf46ed9ec2ad3f6846710a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a475064602e04a6fb1940741eb541a02 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af000967cf8146029b6ad35beb2d73d6 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/934dd6d999b849f9b877702688fe0878 : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5fc8d9f5ee774874bf52cd69724eb4be : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aff94319d2af487191bfa0b311f68496 : SUCCESS in 7m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38d6307e76bc4b41a520b0c4921fdf7d : SUCCESS in 3m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/09b4ada9f93346cfb9ccbf4f44e6623b : SUCCESS in 4m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d2088a8b35624ddbb0babc525ad551c5 : SUCCESS in 11m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1d7f7a06e7e64b5cb6f7070b6e3c947c : SUCCESS in 17m 11s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/2e1c2f05aa6b4914bad5f4724f3b3ee6 : SUCCESS in 57m 24s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/feabe4b4c2ca486abe5ed52a428adeab : SUCCESS in 45m 38s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd754ee2740347828005e2856d87f6a3 : SUCCESS in 49m 44s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c1b0f999aade469db7531406606bbe89 : SUCCESS in 1h 11m 58s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/dd72bb157a484a4cb20f7de521404e55 : SUCCESS in 16m 40s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/ed84d3737d4246d4a18476d604b5c8d1 : SUCCESS in 16m 01s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/70aede6bc18b467e9885da43d75ad717 : SUCCESS in 10m 31s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/61087e989ea8466c92a1bc3b3986368b : SUCCESS in 1h 37m 38s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c1ae5da0e7884c27a1dd8c27c8d1394b : SUCCESS in 1h 18m 27s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/175fbf5f815243ddbd0a8d991ae78d8b : SUCCESS in 16m 15s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8416b647a5f442aebaf1b06e17b6d2e5 : SUCCESS in 13m 20s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/a395ee62de994fc4954745355d65589d : SUCCESS in 31m 40s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ad23b3fdb5834aca8c0e7121388503e3 : SUCCESS in 26m 05s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1f25c304d68e4192b102ae49b50b3137 : SUCCESS in 23m 34s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d15c9929159d45ef901c29b7b69be0a1 : SUCCESS in 12m 55s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/224235693675430ca659f08b4fb66460 : SUCCESS in 8m 25s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/48dde1a58098406e8485639d61f34763 : SUCCESS in 32m 33s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea65062242d245368c86f0fc09dbd03d : SUCCESS in 29m 13s","accounts_in_message":[],"_revision_number":6},{"id":"602b9505fc71c652714173d6a75ea15e95b12612","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-04-19 09:50:36.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"f5b1df5aaf88ff8d0da44bc545dea2cf68dbe88f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-19 11:07:38.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"e7b049a70d3866f08080d7b7fcb4eaf040884054","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 11:08:06.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"e32b3e8f33430b93711b317758248eb17ea26fd2","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-19 12:10:31.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"78807d8b366984467d2555495eda21beee0b30f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 13:35:25.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/867adf42b6c847bc8e87e0de107c044a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2ef00838ede497da04522210f717280 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/59459aba66e1445f9fc8e2103c9b905b : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a81201f9ddec4a2ba9bafeb5937f7b1a : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f261ed9bc3746878e358faf4a9d8366 : SUCCESS in 7m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4520afc202e4ae5b21477fb15aa207b : SUCCESS in 3m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cd2e32081e2e420782b7502721336f2f : SUCCESS in 4m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/92a098a456df4c9e80ac56a1a44b9ff3 : SUCCESS in 9m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ad0e15f7a3364ec0862bb08b3a9650d6 : SUCCESS in 14m 06s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/8332db82715540c9a0ef7235a7e0d93f : SUCCESS in 1h 01m 29s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d973daf489564553b2c097619d17d25f : SUCCESS in 47m 54s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/cf09b6a21ce24a81a7385ad8142596d7 : FAILURE in 1h 32m 53s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d1a0ca10b3974cdda4c89e12bcf47af8 : SUCCESS in 15m 54s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a93ce913f6e74345a697ccb48a62bf18 : SUCCESS in 15m 14s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/09af141d51e64a81898a8c39c7436661 : SUCCESS in 1h 32m 29s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce39d84c88bc46cea8f29afacc3b5093 : SUCCESS in 1h 15m 01s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/40514f2ef03742ca95c7ad4b6bc781f5 : SUCCESS in 17m 01s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/18deabaf93d140d4b6d7b52d5852c8b0 : SUCCESS in 17m 46s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0433f08984fd40f99f0b508313df6ed8 : SUCCESS in 25m 00s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5381a02902304b77a412b8a7d307faa4 : SUCCESS in 22m 12s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/21762fe4eca0419fb5a92ec0f7285fd5 : SUCCESS in 12m 44s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0f3ba503f8dc478887b5ea2fc535618a : SUCCESS in 12m 28s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/edc8e9b331b641839f0c4a8049183c02 : SUCCESS in 30m 17s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e67064927bff451f800adf9c2d11fad9 : SUCCESS in 21m 29s","accounts_in_message":[],"_revision_number":6},{"id":"fa86fb36e7c6967515602db94ddddbc863469e48","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-19 13:37:18.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"45c37d36db3a689fa952faee6b67b0b1fe18f6fc","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-04-19 13:40:47.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6eabf9fa3a7c5ea8cd4a8d739f4d075e4d2041cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 15:25:44.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/23db0d4985944f039e25855dd52dcabd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de2227c8785e4cc9a2f6935313bffa72 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02d2e2c96c7342e3bb43625aabf7800b : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/82e6d3ca56844086a3e03d39a263039d : SUCCESS in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/885ceedb6a1440d597806b7f28ffd77a : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d77a413078fa4cc9b92e28e2ba8a107e : SUCCESS in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/524e07dc53874eb6b4ff82f33ea5c4bf : SUCCESS in 4m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/70725e68c1d642908f0338fb453edab4 : SUCCESS in 5m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d834eb31e33647bab5a575edbd0a72e0 : SUCCESS in 9m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2222837c65844d7bb673c380732d036f : SUCCESS in 19m 19s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/1f83709f4a8440e482362d1cc38f3358 : SUCCESS in 1h 02m 55s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/cd959771622f402d9b97b707fdc4f2fd : SUCCESS in 54m 41s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/274bfbfb8d4842d1a5c7fa2fb72ada2f : SUCCESS in 51m 31s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e65a2c2053dc48f69d8b021bd6382d58 : SUCCESS in 1h 11m 16s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0c6522477509469cbd0e53a965d24588 : SUCCESS in 16m 54s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/35442bade7ba426cb35900ae93e57691 : SUCCESS in 14m 24s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d034660fc7c043e59a29715f80c407ba : SUCCESS in 14m 37s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0536a6a5392944f193837346bb00ad20 : SUCCESS in 1h 43m 21s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/731fe341d0214ae095b6920f977b8162 : SUCCESS in 1h 21m 39s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/570f3c431c334af5a14bac1187cdb08d : SUCCESS in 16m 06s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f888e40a7fc54acf933fa9371e815bd5 : SUCCESS in 14m 34s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/56d2a96e336743a8a8019b96b3cf0800 : SUCCESS in 31m 17s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2c192d07c16b4b99ad99c5d26387ce73 : SUCCESS in 27m 48s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a71754914c9b48e88766ceaaf637ceeb : SUCCESS in 23m 29s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4013b4fbd9b245fb8c070e2c4ea7e560 : SUCCESS in 13m 01s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/72d284088b7e4d15949c801dfdad4258 : SUCCESS in 10m 25s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ea2ba6c468dd4d4c92652b2fecc9d28e : SUCCESS in 33m 53s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cf45a999b1d942d9a5ee8e62a4c43220 : FAILURE in 18m 08s","accounts_in_message":[],"_revision_number":6},{"id":"57e169f3614da823700d93af8cb95c5fa357944f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-04-19 20:41:05.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"abfba0cddeba8f16a0d4ca9ecefc89df1ec4600f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 21:42:46.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e0039ad5d54041cfb71bdd86aca3ed3c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c7c7e057e01468aa27ec026c3865e4d : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43fd71da3f6f4676b5b9b0071a667c26 : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/679e940433044162b7bd14d3d73c0cc7 : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7fc58468a8384b1b8ac7f4f41a06a7fb : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9d6201acbbd403787a58466a24a3393 : SUCCESS in 7m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed236afc38d742e7b6f97bf09d859d24 : SUCCESS in 4m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/377f4a1dac3840efad8ef56bcb0d033d : SUCCESS in 4m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e1ade7f9425e429f9ab284ec75850029 : SUCCESS in 11m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f7686f51d6b248fc9189812e88f65d59 : SUCCESS in 12m 47s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/4f9d2e6ceb03499e9d2e574f822a4ed4 : FAILURE in 32m 42s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/1a770dc40ca64ff891c9d84f5c222b9c : SUCCESS in 1h 01m 06s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07d397cc1a1c4600843268cd14fee794 : SUCCESS in 52m 47s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/23dc642405ca432b9952ba44bf3eddfc : FAILURE in 22m 56s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/1b328dacc38a4e5d92804a415dd5f333 : SUCCESS in 17m 28s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/619eb3bf3d0e4df09683e6dfeff09d47 : SUCCESS in 13m 26s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c141bff3a56b40bc86a0370cf07b6aac : SUCCESS in 14m 01s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c53741468ba142309d39cbe2ea0cac38 : FAILURE in 19m 47s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/396e1bb19910421eb5bbbeaf4becebb2 : FAILURE in 13m 11s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ce31e68d96bd4fe4b7fb9f2ad00f3627 : SUCCESS in 14m 32s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f75b29d254941acb7eb60175e9642bd : SUCCESS in 11m 43s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/935ff3a1268241bfa71977d51b7572b4 : SUCCESS in 31m 47s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/117c232bf8af4aadb6dbe3af55d41096 : SUCCESS in 28m 08s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/49ab393cae514dd3a1ada5bed6fb38ac : SUCCESS in 21m 23s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b6593a3db997446299911ee8d3768761 : SUCCESS in 13m 23s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/505922a2f627413e800bf773dcc264f2 : SUCCESS in 11m 22s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ad462651be53487c826153b319754b76 : SUCCESS in 27m 14s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a3ba3f791214491fbd44e7e21816df34 : SUCCESS in 23m 04s","accounts_in_message":[],"_revision_number":6},{"id":"98532055bd349ce19fd53bd03d5c1cbbcdbe48a2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-20 08:44:26.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"15bd4209a1b0558d4c16646f8bb1b9e7313bdf7a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-04-20 09:37:47.000000000","message":"Patch Set 6: -Code-Review Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"75726c0ee9f7076dbed18da771f80d8d561a240c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-04-20 09:38:01.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b76e7489f07e4f6a25a11a7eb924823394ed9d28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 10:28:44.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43aa7b18e0014158b36067995358592c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7eb6c58a24b408c87015f07d5054b90 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6818910707e492891cb22a48955fea0 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29c5dde2a753427cb6d05a89938ab01a : SUCCESS in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b87bb58cc62543438981b5ffdd6679c9 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96f503ba02b14cb492099fdc96359db7 : SUCCESS in 6m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c1ddd9e0457466986797bc05b3f651a : SUCCESS in 3m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e94f255e6e654a3499c8c317c34c7180 : SUCCESS in 4m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7759afaf2f6e431b9609589b45be8a6a : SUCCESS in 9m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/252f2ba3412c4153a9ad669a43d322df : SUCCESS in 11m 13s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/8e4dd4f322d54c58aa629dd32617e51b : SUCCESS in 57m 13s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/2903eb589b7f461980e9fd2f5e5399b4 : SUCCESS in 59m 19s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f8c2340c12604caa9fb9e67ba11920be : SUCCESS in 53m 25s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/2f29bc9e97be4099a679d0bde7a734a6 : SUCCESS in 1h 17m 54s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/dd76c5fd1553499fa2289d1822a7df47 : SUCCESS in 19m 06s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/b64b6ab1fa4e438d93cff9a470d495fb : SUCCESS in 17m 45s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0538b0ac995c4e68a34778b7a97411c3 : SUCCESS in 12m 10s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4b6f1fcb157a4ffeaecddb7bf58ad4e5 : SUCCESS in 1h 39m 25s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ae98f253b5d43a4a4129189f0b09eb9 : SUCCESS in 1h 23m 42s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e2ee0b2de40248ffaf56e3b5da292782 : SUCCESS in 19m 43s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/64367e41c83a442bb2e9de56b1bf5219 : SUCCESS in 17m 07s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/e2f714293c7d4e2a88258331c0fc9507 : SUCCESS in 31m 28s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/cb7e700008a44b49bca490221ca28409 : SUCCESS in 31m 34s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c7bfb64f48724b42be11a558a0d00687 : SUCCESS in 23m 47s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/032a6e282f044801a763b8d9bd40c066 : SUCCESS in 13m 28s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/639dddbbd50e46ad89981a59ca1dffed : SUCCESS in 12m 39s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a07bf3fea03548d2b2bc678f3fc0ae60 : SUCCESS in 32m 24s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8b09879d83cd450dbd8a49a01e34c7ee : SUCCESS in 27m 00s","accounts_in_message":[],"_revision_number":6},{"id":"fb3a389a1b81b818324e9c655ef7155cc4326b61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-04-21 06:46:07.000000000","message":"Patch Set 7: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow+1, Workflow-1\n","accounts_in_message":[],"_revision_number":7},{"id":"a8628808224f723d1f76850e2ade3d536fc50774","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-04-21 06:46:35.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a65b5411a31b87f0450057d038d99bd8d673919a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 08:44:05.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a4fc7c5017a247ae9ccbe10fecf86f5c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99fe3775bb724521ad5cca253efe45c9 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e25bf498a7864e5f912632dd2c1a6d19 : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57133e3adfe548d08c972da4c3c28047 : SUCCESS in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/06815f94291f45159217587cfe561d3f : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c931456e057424abfbe3560d3c7b147 : SUCCESS in 7m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0235fc2828fc49e286fad01c62141e6e : SUCCESS in 3m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ae4387a097d9436da694e6b3f8089cd8 : SUCCESS in 4m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9e527f7cdc6c44b095e79fc5ab7086c6 : FAILURE in 9m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/63b9939d8cfa4fd9a0914c919fb6278e : SUCCESS in 13m 46s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/34f74c8c714d4ef6944711e0f771fdae : SUCCESS in 57m 19s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/ae684adfc9c4402a860d7ea87e01af3e : SUCCESS in 53m 55s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d1f4dbf0bf47424f8ede894b22437784 : SUCCESS in 48m 03s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/675565e09b474b0c8c021b6ba87ed2d1 : SUCCESS in 1h 06m 11s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/89dc25b2f9364c049225dd574802d2b2 : SUCCESS in 18m 58s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/fbda83c0ef5642b7afa7150edf54d4f5 : SUCCESS in 17m 16s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f4326c96902487db97d7af6b1b5ce82 : SUCCESS in 13m 26s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e76463e28d8747a0a3d15fb844b93ae2 : SUCCESS in 1h 53m 13s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/16bdcaa9e2bd43bba8dc95ebd11818be : SUCCESS in 1h 06m 19s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/6b66f89b8e7d41ef823202ee9d027bf0 : SUCCESS in 13m 52s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b9c58dca3e6c4ce98f4120af5d894c93 : SUCCESS in 14m 18s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/792d9484f8874323ac6b2c9644aa3330 : FAILURE in 30m 52s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ddf201dcf14148f0aed5e0eba8fb2fb3 : SUCCESS in 25m 09s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5c686368c474400b962db6932cdf6380 : SUCCESS in 16m 54s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c64ff9e8eee64ccbbb6358c16c348665 : SUCCESS in 14m 09s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8666b6c138e74e63af5c544048c1474f : SUCCESS in 11m 02s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1b9213e4d09c463abb4314244cac046e : SUCCESS in 27m 29s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5c54150619aa435a9fc7eb71f569a068 : SUCCESS in 27m 40s","accounts_in_message":[],"_revision_number":7},{"id":"2512acb7ea3f7db231de932cc3f0c1003af73aca","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-04-21 11:03:32.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"80232d15c0d32cffdf19b81cabc93241c3957bca","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-21 11:05:37.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"15d1bd81b973e5ccb3cad20d3e7835ed13b1008d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-04-21 11:06:07.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"d9b92e97836494a3ed6be584c973fa2d51a9cbdb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-22 06:42:21.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"fd33060e7de18ab4f451f29e62d0eb871b2df852","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 06:42:41.000000000","message":"Patch Set 8: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla-ansible for 881018,23","accounts_in_message":[],"_revision_number":8},{"id":"da5a36c87b00f020c65dd7c16a00929cdd296185","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-22 06:43:56.000000000","message":"Patch Set 9: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"e2d03a6935a47391804133b2a985c455f469f81a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-22 07:44:43.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"db2be4e665a29bf988440d1bd0d777d1f2cc40d8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-05-22 08:57:16.000000000","message":"Patch Set 10: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"3e14ab203a4ef9d19d9410920addc715398d153f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-22 09:12:56.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/990586e7dda843f9afc3fec60fdb2c0d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f386dedd083b4800bc0f866ae6e406aa : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43cb4aadc8ad4e249f7a64f2c4c2868c : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e97c1d839ef246589a3759b696bf44c6 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b41d7e8e4f29499b9b9adb5f6338b664 : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3c78e2f31e74411b8d82bfce723fb3d7 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f21a9cd689744d888d8718caca22f57 : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/85b715ed23c14ad58de3345af66323d7 : SUCCESS in 3m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2b705d16fef8402aa5c0ddb0c8b9068c : SUCCESS in 3m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5c668d6c3abb48a2ba03ac9c58049ce0 : SUCCESS in 9m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/25c7d945de78473cb2afcb5c6fbf13f4 : SUCCESS in 12m 44s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/45ef37b8d9464e7d84830e44f6c2cf3e : FAILURE in 1h 06m 31s (non-voting)\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/dfc9c7682c0e46dda2b0b49eb6dbe4d0 : FAILURE in 1h 05m 46s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd85c36155e24e29b864ecbcfdd3ab2b : SUCCESS in 50m 37s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/0b4ef87e91c94735a3c04e6627fcf365 : FAILURE in 1h 21m 36s (non-voting)\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/82ab5b6e287c4ec88a68b1469e1add92 : SUCCESS in 14m 57s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/b3da5f63b63f4a88a12b6b42fcfd822d : SUCCESS in 17m 58s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7bf43d616cd148eaacd2b1222d39f647 : SUCCESS in 14m 43s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/39ab2449a4b8424d9a09116bd97fdca9 : FAILURE in 1h 09m 29s (non-voting)\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/386ef7156a8a4d32a619c04cf79f429c : FAILURE in 43m 06s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/5321740185a1416995751d1364ba1b08 : SUCCESS in 17m 52s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/22dc201c9dab44f2be7a8ddbe31cbe26 : SUCCESS in 15m 05s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/b6f75b41ee1f4187acdde092f3eae1f4 : FAILURE in 31m 05s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/33d795d6b6ca492299c48b5c543866e5 : SUCCESS in 29m 17s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1dfa123abe4540d79d6994e7a3c7008a : SUCCESS in 25m 25s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/345e23d793b748e9b2417ca5dd0238bf : SUCCESS in 13m 47s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3da3d0b7bb094bef8dbfd2d238da1bd4 : SUCCESS in 12m 13s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/12a640474e054f48807383dfbc53a8dc : SUCCESS in 27m 17s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4a6df1936714aac97b37b326ab6ed28 : SUCCESS in 28m 12s","accounts_in_message":[],"_revision_number":10},{"id":"76bb21ddb6ededd279ae87fbff0e3bd0d5ba7ef6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-25 08:10:17.000000000","message":"Patch Set 11: Published edit on patch set 10.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"7141b300e157b563cb545609fd1c5dec63aee980","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 09:45:31.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f255da48b27245709fef450afd1fad0c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e2d86fbfbc54603a81898e6029b4fe2 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/967ad40271ab46ecba8e5d551b315042 : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9448a2554a09486fbc29701bf0215140 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d569d9970e7d4069a136a5601d994374 : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6cd78887c6cd4bc08bc5968d9d9bbaba : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7361247a281a4033a91b0a8b9b86e0a3 : SUCCESS in 7m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/039488a52a1c4e2fbb5903ebb8972b03 : SUCCESS in 3m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8a94895954b34509a4c60783dab39b2d : SUCCESS in 4m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fe64b85287d64b2ab1916a7a540558c6 : SUCCESS in 9m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ff3073053e7a42c3b8fc4036fe15ad64 : SUCCESS in 12m 54s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/fe525866036d4388908679aa51b8d06c : SUCCESS in 1h 00m 40s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/23495a84fa5b4d64b97c356485c0268b : SUCCESS in 46m 07s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/57879bc8357f4714b5d673a98b7ee2a8 : SUCCESS in 45m 03s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/0dcc206c4cc347e5aa953c3a46e1eb25 : SUCCESS in 1h 03m 50s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b76495ae357d4f869c29caf6c6441110 : SUCCESS in 15m 38s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/3f89632d79aa4445aa74050cdc212e6b : SUCCESS in 17m 21s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/86dc903960674464901127c6ccdecb43 : SUCCESS in 13m 13s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/fd5257269b4844d8bcf5b13d7987cb23 : SUCCESS in 1h 29m 20s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/035851d708974a11aaddb915bcabd428 : SUCCESS in 1h 29m 42s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3b4197ca888f4412b424c3c75ed5e45c : SUCCESS in 15m 57s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4afa16e2dddb468780978dd05e6b9b98 : SUCCESS in 13m 55s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/b50d46b8e3694d7891f08239f0b94200 : FAILURE in 30m 17s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/17d3b09e685d4fcabbf9bb8f39fe01b6 : SUCCESS in 27m 16s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8b4fc85e6a1a4121ab36f33ea1c017ba : SUCCESS in 24m 51s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3d4c95a1ede2448080803318a5197dee : SUCCESS in 12m 04s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c4c7c9b2f2a482782b982b8c9757afe : SUCCESS in 12m 16s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b00499da126e4b9f83e4d9c090b1e138 : SUCCESS in 32m 57s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b6c00bd5b0814cf99ffa74ed07421e97 : SUCCESS in 26m 04s","accounts_in_message":[],"_revision_number":11},{"id":"e5f066fc07d39b8baf5de0be5fc0337be48b52da","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-25 11:43:22.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"724a53078ce4f887051cb96879da82828304ccd5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-25 11:47:20.000000000","message":"Patch Set 11: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"deea3cda06a8e896e7f618033c1d0378d7ded428","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-05-25 11:54:37.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"3b585305560762aa8456000e4f3ffbf82d71eb4c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-05-25 12:31:47.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"90d841ac3a78a2a77db7d48f9ca8c11272b9e3de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 12:32:13.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"4cb82ee82b4fc907b63686329a3f9fbd1739fecb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 14:13:45.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53ad6208b0814519971f1af0e3e435d8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6dd5b7e1d244241b31aaec834f9ebb8 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c8908d0711554a97a272cee0b016a535 : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3930eab2fc9456583edb57969a5308d : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/37073844ebe5441faa4ae28a62392901 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54d54d75fada45839b7d41ada6fc5540 : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a1d6331ed4134781897e5af45c2a92d4 : SUCCESS in 3m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0236d5e7c63043b29e957016faad2e54 : SUCCESS in 4m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a34a222522454a40808f988f7d80f1c4 : SUCCESS in 9m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ad24c9a1aaed417da9da137ef209eb5e : SUCCESS in 11m 24s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ce2079effd3d44a889b6c84c9d9bd29b : SUCCESS in 54m 38s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc2203379acd40c19ac3917fb0c6f8de : SUCCESS in 49m 02s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/d9690486a9e044289d4f5e636028b828 : SUCCESS in 1h 04m 11s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0e47c608abfb431db85aad572796daa0 : SUCCESS in 17m 27s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/51d22da9cc1e4a2c80c2e1a03ca98f52 : SUCCESS in 13m 32s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/92afb1ecff8c4b2591efd7f767e37d53 : SUCCESS in 1h 23m 45s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1ecdfc48219847b280f5194ccfa7b588 : SUCCESS in 1h 40m 26s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a51ed787d2fa4bd4bfd913ba71bd2254 : SUCCESS in 16m 18s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/afb958c8337b48a48bf089e9a7f877d1 : SUCCESS in 14m 00s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/85057dc0a7144909a68a413fe8c212f4 : SUCCESS in 28m 39s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/685af06a88594a1b8f9c8f0018221ae3 : SUCCESS in 24m 30s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d372772746c943748b62738ac5f6cdb0 : SUCCESS in 14m 24s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f5144f12795463facd5d65b9b4d32dc : SUCCESS in 9m 58s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f98962c8ea1f4c29ab24cffac9a97a35 : SUCCESS in 26m 14s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e63cb890ff07435286f780bf27badb09 : SUCCESS in 27m 36s","accounts_in_message":[],"_revision_number":11},{"id":"3de696c63418db36f1eb70d86976fa31b828a033","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 14:13:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"635b512d82ef7b3fc7180c770d29cef727ee7271","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 14:14:46.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d70881df0d9463391d5b8cc70bb6065\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd4fbddda2af4f349157c14cbc323da1 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8db52ca38a3a4706a4450a6934569264 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"773eead34dad528e389a4502f7474c4687a3ddf4","revisions":{"3f05ccc557319f5819964eb47ba630298998555d":{"kind":"REWORK","_number":1,"created":"2023-03-22 11:19:17.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/07/878207/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/07/878207/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/878207/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/878207/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/878207/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/07/878207/1"}}},"commit":{"parents":[{"commit":"137e64b57c86fde852b49f9975ffbca26da79de1","subject":"Fix opensearch container builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/137e64b57c86fde852b49f9975ffbca26da79de1"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-22 11:18:22.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-22 11:19:16.000000000","tz":60},"subject":"Bump up Ansible supported versions to 6.x/7.x","message":"Bump up Ansible supported versions to 6.x/7.x\n\nThis change bumps up the maximum supported Ansible version to 7.x\n(ansible-core 2.43.x) and minimum to 6.x. This synchronises Kayobe with\nKolla Ansible.\n\nChange-Id: Ibffecaa8085bd38ebc8cded9a4bfebe77d59d515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3f05ccc557319f5819964eb47ba630298998555d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3f05ccc557319f5819964eb47ba630298998555d"}]},"branch":"refs/heads/master"},"e67f3bf38866bfa1051f9de3463c0a35f3c9cbfe":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-03-22 11:22:39.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/07/878207/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/07/878207/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/878207/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/878207/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/878207/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/07/878207/2"}}},"commit":{"parents":[{"commit":"137e64b57c86fde852b49f9975ffbca26da79de1","subject":"Fix opensearch container builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/137e64b57c86fde852b49f9975ffbca26da79de1"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-22 11:18:22.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-22 11:22:22.000000000","tz":60},"subject":"Bump up Ansible supported versions to 6.x/7.x","message":"Bump up Ansible supported versions to 6.x/7.x\n\nThis change bumps up the maximum supported Ansible version to 7.x\n(ansible-core 2.43.x) and minimum to 6.x. This synchronises Kayobe with\nKolla Ansible.\n\nDepends-on: https://review.opendev.org/c/openstack/kolla-ansible/+/877697\n\nChange-Id: Ibffecaa8085bd38ebc8cded9a4bfebe77d59d515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e67f3bf38866bfa1051f9de3463c0a35f3c9cbfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e67f3bf38866bfa1051f9de3463c0a35f3c9cbfe"}]},"branch":"refs/heads/master"},"eca0fad7aa42ac0494e9a033d74fc544db75e97f":{"kind":"REWORK","_number":3,"created":"2023-03-22 11:26:45.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/07/878207/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/07/878207/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/878207/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/878207/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/878207/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/07/878207/3"}}},"commit":{"parents":[{"commit":"137e64b57c86fde852b49f9975ffbca26da79de1","subject":"Fix opensearch container builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/137e64b57c86fde852b49f9975ffbca26da79de1"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-22 11:18:22.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-22 11:26:39.000000000","tz":60},"subject":"Bump up Ansible supported versions to 6.x/7.x","message":"Bump up Ansible supported versions to 6.x/7.x\n\nThis change bumps up the maximum supported Ansible version to 7.x\n(ansible-core 2.43.x) and minimum to 6.x. This synchronises Kayobe with\nKolla Ansible.\n\nDepends-on: https://review.opendev.org/c/openstack/kolla-ansible/+/877697\n\nChange-Id: Ibffecaa8085bd38ebc8cded9a4bfebe77d59d515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eca0fad7aa42ac0494e9a033d74fc544db75e97f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eca0fad7aa42ac0494e9a033d74fc544db75e97f"}]},"branch":"refs/heads/master"},"f942ddc672b18e16f6b10a1c513544d384d58662":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2023-03-22 11:27:34.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/07/878207/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/07/878207/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/878207/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/878207/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/878207/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/07/878207/4"}}},"commit":{"parents":[{"commit":"137e64b57c86fde852b49f9975ffbca26da79de1","subject":"Fix opensearch container builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/137e64b57c86fde852b49f9975ffbca26da79de1"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-22 11:18:22.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-22 11:26:48.000000000","tz":60},"subject":"Bump up Ansible supported versions to 6.x/7.x","message":"Bump up Ansible supported versions to 6.x/7.x\n\nThis change bumps up the maximum supported Ansible version to 7.x\n(ansible-core 2.14.x) and minimum to 6.x. This synchronises Kayobe with\nKolla Ansible.\n\nDepends-on: https://review.opendev.org/c/openstack/kolla-ansible/+/877697\n\nChange-Id: Ibffecaa8085bd38ebc8cded9a4bfebe77d59d515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f942ddc672b18e16f6b10a1c513544d384d58662"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f942ddc672b18e16f6b10a1c513544d384d58662"}]},"branch":"refs/heads/master"},"d704e68aba3cea160b069a840b75b0298ad68a09":{"kind":"REWORK","_number":5,"created":"2023-03-22 11:31:54.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/07/878207/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/07/878207/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/878207/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/878207/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/878207/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/07/878207/5"}}},"commit":{"parents":[{"commit":"137e64b57c86fde852b49f9975ffbca26da79de1","subject":"Fix opensearch container builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/137e64b57c86fde852b49f9975ffbca26da79de1"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-22 11:18:22.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-22 11:31:14.000000000","tz":60},"subject":"Bump up Ansible supported versions to 6.x/7.x","message":"Bump up Ansible supported versions to 6.x/7.x\n\nThis change bumps up the maximum supported Ansible version to 7.x\n(ansible-core 2.14.x) and minimum to 6.x (ansible-core 2.13.x).\nThis synchronises Kayobe with Kolla Ansible.\n\nDepends-on: https://review.opendev.org/c/openstack/kolla-ansible/+/877697\n\nChange-Id: Ibffecaa8085bd38ebc8cded9a4bfebe77d59d515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d704e68aba3cea160b069a840b75b0298ad68a09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d704e68aba3cea160b069a840b75b0298ad68a09"}]},"branch":"refs/heads/master"},"f125b7a7644a8f8069ef4bc3a5209bcae131286f":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2023-04-19 07:13:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/878207/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/07/878207/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/878207/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/878207/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/878207/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/07/878207/6"}}},"commit":{"parents":[{"commit":"cb50b0cbc05be450f856b43e1ee27d2f8f119dd4","subject":"Merge \"Stop using kolla-ansible bootstrap-servers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cb50b0cbc05be450f856b43e1ee27d2f8f119dd4"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-22 11:18:22.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-04-19 07:13:32.000000000","tz":0},"subject":"Bump up Ansible supported versions to 6.x/7.x","message":"Bump up Ansible supported versions to 6.x/7.x\n\nThis change bumps up the maximum supported Ansible version to 7.x\n(ansible-core 2.14.x) and minimum to 6.x (ansible-core 2.13.x).\nThis synchronises Kayobe with Kolla Ansible.\n\nDepends-on: https://review.opendev.org/c/openstack/kolla-ansible/+/877697\n\nChange-Id: Ibffecaa8085bd38ebc8cded9a4bfebe77d59d515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f125b7a7644a8f8069ef4bc3a5209bcae131286f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f125b7a7644a8f8069ef4bc3a5209bcae131286f"}]},"branch":"refs/heads/master","description":"Rebase"},"0fbeb60676b2cadaf759e6d823231313fc7f63de":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2023-04-21 06:46:07.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/07/878207/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/07/878207/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/878207/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/878207/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/878207/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/07/878207/7"}}},"commit":{"parents":[{"commit":"cb50b0cbc05be450f856b43e1ee27d2f8f119dd4","subject":"Merge \"Stop using kolla-ansible bootstrap-servers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cb50b0cbc05be450f856b43e1ee27d2f8f119dd4"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-22 11:18:22.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-04-21 06:46:07.000000000","tz":0},"subject":"Bump up Ansible supported versions to 6.x/7.x","message":"Bump up Ansible supported versions to 6.x/7.x\n\nThis change bumps up the maximum supported Ansible version to 7.x\n(ansible-core 2.14.x) and minimum to 6.x (ansible-core 2.13.x).\nThis synchronises Kayobe with Kolla Ansible.\n\nDepends-on: https://review.opendev.org/c/openstack/kolla-ansible/+/881018\n\nChange-Id: Ibffecaa8085bd38ebc8cded9a4bfebe77d59d515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0fbeb60676b2cadaf759e6d823231313fc7f63de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0fbeb60676b2cadaf759e6d823231313fc7f63de"}]},"branch":"refs/heads/master","description":"Edit commit message"},"32aef9f298098bec3576ffc7d13697ac7057c367":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2023-05-22 06:42:21.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/878207/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/07/878207/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/878207/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/878207/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/878207/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/07/878207/8"}}},"commit":{"parents":[{"commit":"805414ef9d29deddd432fd3f591734a5f05da174","subject":"CI: fix molecule job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/805414ef9d29deddd432fd3f591734a5f05da174"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-22 11:18:22.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-22 06:42:21.000000000","tz":0},"subject":"Bump up Ansible supported versions to 6.x/7.x","message":"Bump up Ansible supported versions to 6.x/7.x\n\nThis change bumps up the maximum supported Ansible version to 7.x\n(ansible-core 2.14.x) and minimum to 6.x (ansible-core 2.13.x).\nThis synchronises Kayobe with Kolla Ansible.\n\nDepends-on: https://review.opendev.org/c/openstack/kolla-ansible/+/881018\n\nChange-Id: Ibffecaa8085bd38ebc8cded9a4bfebe77d59d515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/32aef9f298098bec3576ffc7d13697ac7057c367"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/32aef9f298098bec3576ffc7d13697ac7057c367"}]},"branch":"refs/heads/master","description":"Rebase"},"a785c94883cc823ae2c90909ff84108ef200ff75":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2023-05-22 06:43:56.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/878207/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/07/878207/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/878207/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/878207/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/878207/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/07/878207/9"}}},"commit":{"parents":[{"commit":"805414ef9d29deddd432fd3f591734a5f05da174","subject":"CI: fix molecule job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/805414ef9d29deddd432fd3f591734a5f05da174"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-22 11:18:22.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-22 06:43:56.000000000","tz":0},"subject":"Bump up Ansible supported versions to 6.x/7.x","message":"Bump up Ansible supported versions to 6.x/7.x\n\nThis change bumps up the maximum supported Ansible version to 7.x\n(ansible-core 2.14.x) and minimum to 6.x (ansible-core 2.13.x).\nThis synchronises Kayobe with Kolla Ansible.\n\nChange-Id: Ibffecaa8085bd38ebc8cded9a4bfebe77d59d515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a785c94883cc823ae2c90909ff84108ef200ff75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a785c94883cc823ae2c90909ff84108ef200ff75"}]},"branch":"refs/heads/master","description":"Edit commit message"},"bdcd70b5f9efa45b0f8779ec279276005b6b6c9c":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2023-05-22 07:44:43.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/878207/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/07/878207/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/878207/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/878207/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/878207/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/07/878207/10"}}},"commit":{"parents":[{"commit":"d80189611a73aa9fdb5ec9f1f36f253963a11d02","subject":"Add neutron-ovn-agent group to Kolla inventory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d80189611a73aa9fdb5ec9f1f36f253963a11d02"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-22 11:18:22.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-22 07:44:43.000000000","tz":0},"subject":"Bump up Ansible supported versions to 6.x/7.x","message":"Bump up Ansible supported versions to 6.x/7.x\n\nThis change bumps up the maximum supported Ansible version to 7.x\n(ansible-core 2.14.x) and minimum to 6.x (ansible-core 2.13.x).\nThis synchronises Kayobe with Kolla Ansible.\n\nChange-Id: Ibffecaa8085bd38ebc8cded9a4bfebe77d59d515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bdcd70b5f9efa45b0f8779ec279276005b6b6c9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bdcd70b5f9efa45b0f8779ec279276005b6b6c9c"}]},"branch":"refs/heads/master","description":"Rebase"},"773eead34dad528e389a4502f7474c4687a3ddf4":{"kind":"REWORK","_number":11,"created":"2023-05-25 08:10:17.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/07/878207/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/07/878207/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/878207/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/878207/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/878207/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/07/878207/11"}}},"commit":{"parents":[{"commit":"d80189611a73aa9fdb5ec9f1f36f253963a11d02","subject":"Add neutron-ovn-agent group to Kolla inventory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d80189611a73aa9fdb5ec9f1f36f253963a11d02"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-22 11:18:22.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-25 08:10:17.000000000","tz":0},"subject":"Bump up Ansible supported versions to 6.x/7.x","message":"Bump up Ansible supported versions to 6.x/7.x\n\nThis change bumps up the maximum supported Ansible version to 7.x\n(ansible-core 2.14.x) and minimum to 6.x (ansible-core 2.13.x).\nThis synchronises Kayobe with Kolla Ansible.\n\nChange-Id: Ibffecaa8085bd38ebc8cded9a4bfebe77d59d515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/773eead34dad528e389a4502f7474c4687a3ddf4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/773eead34dad528e389a4502f7474c4687a3ddf4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
