)]}'
{"id":"openstack%2Fkayobe~879140","triplet_id":"openstack%2Fkayobe~stable%2Fyoga~I9cf265ffc7acce926ce093d1e4f8d53a12be1f26","project":"openstack/kayobe","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"25600":{"account":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"last_update":"2023-09-12 08:49:28.000000000","reason":"Change was abandoned"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-04-06 13:10:04.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I9cf265ffc7acce926ce093d1e4f8d53a12be1f26","subject":"[DNM] ansible-role-interfaces NetworkManager test","status":"ABANDONED","created":"2023-03-31 08:26:21.000000000","updated":"2023-09-12 08:49:28.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"18dbcb89d36a0ac4b2f5eb4e7841e61b6d4297b3","_number":879140,"virtual_id_number":879140,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-04-06 14:57:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-31 08:28:05.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"state":"CC"},{"updated":"2023-03-31 11:43:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fabf6296857174cae4e477a30651ff880ae17f90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-31 08:26:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2dfded3b7d7959806f2da553b0196c0f9766a409","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-31 08:28:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0a0d51922a45c82fc327881f3dfa7dd052251b57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-31 09:31:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"24e4296200a5aba6fa74c7ae8bcd03e8854aa9ec","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2023-03-31 11:09:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"af8e6ee39e428a6dccf3d4186bab24595889b486","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 11:43:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a1dba1b22b94a4a9599d0a30439940c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58948108f652490d88c7172d5960beeb : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80d9619b87ab499c89bcb6579f585e1c : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3610b30e11e74c078e8e584eb058eee8 : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/51058917075c449e90619fc1900e71e3 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67a7a2c8db9c45a0971a54fe25a9e7c3 : SUCCESS in 11m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3b760d6bc8324801ad5be57032ffce88 : SUCCESS in 6m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/13f9aad3fa414e2481b13fd0beec01fb : SUCCESS in 13m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/762142f940da47eab8738fb57779549f : SUCCESS in 18m 06s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4b6e231da8454c7c9af40b5ec6aa9263 : SUCCESS in 1h 13m 53s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/a5fb38b246a34002bb5b12102e28a4b2 : SUCCESS in 1h 28m 05s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/26e8e24da14247f9937cf8651132221b : SUCCESS in 1h 02m 58s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ff62f65766ec4096bd1967458144a86f : SUCCESS in 53m 22s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05fe7953d90a4557a6687992798353b8 : SUCCESS in 51m 39s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d3d0eaf3c4c41bb85766667618c813e : SUCCESS in 48m 41s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/29f85213d3ea496c94618cc13efd7207 : SUCCESS in 23m 42s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/326df68a17ed46d28a78f5b78b520cd3 : SUCCESS in 24m 56s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/54f3c3ae76684a2aad5cdcc5b0d69673 : SUCCESS in 19m 08s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/de536a029c20411d96096e3445179bf2 : SUCCESS in 19m 21s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f04243f4bf054f72ad0fe63f3959edf8 : SUCCESS in 16m 08s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d083e875aaa64b3faae393a78269804e : SUCCESS in 15m 22s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/8c624ede7b2a4cc993a2348ed2fef35e : SUCCESS in 53m 57s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/616017d382b54e5ab454970a45def39e : SUCCESS in 1h 09m 23s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e4af09a41da54642b8ce743f4380ee45 : SUCCESS in 2h 06m 13s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f8e2e9f393d74dc6a8c6f2e7e50ce140 : SUCCESS in 55m 40s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a9a8e19f55584ad98561416ef932fa80 : SUCCESS in 20m 35s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/32e96dc964d841c6a953c626d25c4980 : SUCCESS in 19m 41s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/2695bfb45f3648f4a7d10704728c0760 : SUCCESS in 18m 34s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7737882b92f48c1baac736015106d8b : SUCCESS in 14m 00s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/85d1b76ed17b465c8980fe8dea200976 : SUCCESS in 16m 29s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/865be5c5dbed435994141cdef76604b9 : SUCCESS in 32m 02s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7de48863a3c49d48701839bf1934181 : SUCCESS in 20m 52s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/483dd4e3d3094ccbaa22fa48be542630 : SUCCESS in 17m 06s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/0f982301b4c949d19073c4487b8754be : SUCCESS in 16m 32s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5c0d5932de6d44d68239b67bdb4795b3 : SUCCESS in 15m 26s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ffb7e5b709e147faa74c4707b7ebc7db : SUCCESS in 11m 21s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5c3cfea66e26419ca15d9b7d56a9357f : SUCCESS in 12m 57s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/5eab47249d214b80bd4e92996f6cc742 : SUCCESS in 23m 25s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/39d0f3785d9b44c59b077f29e74ad1d1 : SUCCESS in 19m 29s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/73e4bb69ccf04874b1d169b4077e37f1 : SUCCESS in 16m 46s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e3c1d61dc9944e7289864194c61768c3 : SUCCESS in 13m 04s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a028e9e07a814bb8beaea284b5061bd0 : SUCCESS in 14m 20s","accounts_in_message":[],"_revision_number":2},{"id":"8d2622e6918988feead34cfe6e668fbfbe0747a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-04-04 15:00:47.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"ce7b032c39ed4b7cc712cc2457f45959509ef179","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-04 17:08:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73ef483386834e0cb21d1e83fd346fde\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/110df43772db401cbe97b9f1e458dd8c : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e05166372d843459daad70f91566c5e : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7790a82acd834efb975077c403bf488d : SUCCESS in 6m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77f4dcdc207e4b3aaad239d87d926729 : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/877d2dd5b1a3466f92dc61b7549b9977 : SUCCESS in 9m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/59659849c20f46c69591004ea25e1f35 : SUCCESS in 5m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5d4e471f2bd7436d9ce9644b5d87dd34 : SUCCESS in 12m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/70f28765d6ee497f8c9ec31e6d3f58ee : SUCCESS in 16m 42s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d51fef4f602f4f77862e583f570bee5f : SUCCESS in 1h 03m 25s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/058293b9182048e98c3966bf4fd48ce4 : SUCCESS in 59m 37s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/cbdfd58a27404dfb87ef29e63e84e857 : SUCCESS in 1h 02m 23s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e94b19968e1b44a0ad5d107e1bf56b9d : SUCCESS in 58m 25s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a05aad9c2e84d9497c7ec67ba1cc7fd : SUCCESS in 53m 26s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5aed0234a07a460ca1dbff5e613d1c27 : SUCCESS in 57m 27s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/9574adec9a154f7da09420ce033a6645 : SUCCESS in 28m 06s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/571c4edb65d847c98cf687d2cf925256 : SUCCESS in 23m 21s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/a57a8ea8193a48ff915d16b68849dc32 : SUCCESS in 25m 03s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f54a612a75544c009ca76197840f2ac8 : SUCCESS in 22m 59s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae268e7341794dc4b3caa81d4902d138 : SUCCESS in 16m 55s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a881a24c258b49ef839b9fd7cc55eb7e : SUCCESS in 15m 40s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/a60045abe60a421b832fd308029f2b83 : SUCCESS in 1h 18m 02s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/91d7c4c989604fd5b932d4ee43438888 : SUCCESS in 1h 03m 32s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2ad8af13a4c143968fb4c781faa24fe2 : SUCCESS in 2h 02m 11s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ae2422d563c450286e354b08b1040cf : SUCCESS in 1h 14m 28s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/4f2f64857bde49c995a114da19fadebf : SUCCESS in 21m 46s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/c8589edf3c8b4336bf380b275d734cbf : SUCCESS in 20m 43s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/33574f2069d349f9ad7156a76ce7febc : SUCCESS in 18m 11s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b04f7393f5d246d781d02aab1bbf1686 : SUCCESS in 17m 57s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cee86a31ced54af99eb211864a5162b7 : SUCCESS in 17m 15s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ec20a770b2ab4fa291adda78d2880435 : SUCCESS in 26m 13s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8a7fefb33924d2ca40dde4015674673 : SUCCESS in 24m 14s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2e7970dd11f64acba82fb53e91a9c056 : SUCCESS in 16m 53s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/99d5e266dda348f2ab58d74234a0a54b : SUCCESS in 15m 45s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6a57b5af33f445bd9aaad071f4f4a2f2 : SUCCESS in 15m 19s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cfeb3c5282d14bfba2fafb56a1eff023 : SUCCESS in 15m 00s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9e53cc6eef1544978f99f1e28a07c40a : SUCCESS in 14m 27s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/7482bc319fcd457d83cde249e4c2afd0 : SUCCESS in 20m 22s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/6a311882e5dd464fa91f1db692a279dc : SUCCESS in 18m 24s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e86ad4faa7534dc6b93193dfc5c0ea95 : SUCCESS in 17m 56s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd16bf1d767c43f89ceebff8c7936641 : SUCCESS in 14m 29s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/850e310526eb454a8d24a63609109070 : SUCCESS in 14m 33s","accounts_in_message":[],"_revision_number":3},{"id":"ec53d75dbb5d0d4fe5535923875f04459258b98b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-04-05 13:36:11.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ad6be1380711eddd7c9722b6031136939069355a","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-04-05 13:36:33.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e52e26e426b16afdf5861abb5d7bd356b1b7a366","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-04-05 14:39:49.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"77dc2d0796e5cf776819ce38449a6adbf91189e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 16:38:05.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc9f5d7fc6be4970934cc38ae8d6ab62\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2f1634977de49de8d39f10bdc90a5b6 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46d167df30ac41af84c499e09e9439d3 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd24ed6fc1a94b0fb670963bdfd56c18 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b6a0e4754044ab3802cf69f02e8969e : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99b6b1d1871f43fa9ab0a225f6c8f905 : SUCCESS in 9m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/085d7e3f293f4c14a383d89e806002a7 : SUCCESS in 5m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e517e1f16d164218949ef64cddd9f0f8 : SUCCESS in 16m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/75f3a2c0ef7449df885905d586607e45 : SUCCESS in 18m 19s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/439a6d47d4754fc9acd4d7f59f82bdd3 : SUCCESS in 1h 12m 59s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/f4da96bb00b04aaa8e10272047d356c9 : SUCCESS in 1h 11m 34s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/1dbe757872d04d4183e88657c1bc4268 : SUCCESS in 1h 34m 04s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/5d0a226a2e4f492fa9d20717caddca1f : SUCCESS in 1h 06m 20s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f03db933c974fca8fa083d338c9c6e2 : SUCCESS in 58m 19s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d1f88288ee49401a98a0e1574254862f : SUCCESS in 54m 29s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b7d64d2c55724f358362cc8dfc60ffa1 : SUCCESS in 25m 29s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/457fcdd24c1147308123ada5a882e401 : SUCCESS in 24m 09s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/19c6061b0a70466a990fa866c0155669 : SUCCESS in 21m 03s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/811836c4ce584825a3cae7fa76cee217 : SUCCESS in 21m 45s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd393c038c374dabbc13347631695354 : SUCCESS in 16m 38s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/193eba9eeb9a408381ea3d66d4af9646 : SUCCESS in 16m 27s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/0b17cc33d7464ffdacf532e38fb893e1 : SUCCESS in 1h 14m 32s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/d00863f198754d26a11cc3291494bbe5 : SUCCESS in 1h 12m 01s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c179721ec2d94ef4a7ccee927be4d318 : SUCCESS in 1h 53m 05s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3acda55127564ad0af79c2351cc05485 : SUCCESS in 1h 17m 48s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7815179ca5154cdab0fb4f25c4cc6b8c : SUCCESS in 22m 36s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/17274e7760554bcbac52e002903def01 : SUCCESS in 16m 14s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e9aef246c53744c0aaf4670bd0b7ae4a : SUCCESS in 20m 17s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c8749f77fa842e2bc95014de482fa08 : SUCCESS in 15m 55s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bcd65d479456449aa1ca6e45c650b840 : SUCCESS in 20m 20s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/785c5a3156074bafa7f4931a9dada2fc : SUCCESS in 29m 02s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e6e4a30213345e696c7a4e985e873e9 : SUCCESS in 26m 04s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2081548fe0994cbab727e7f7003c5a0f : SUCCESS in 16m 37s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/baccaf30f1eb4601897929604b90f692 : SUCCESS in 15m 38s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5c9db14f42094f27b929714d3430747a : SUCCESS in 15m 58s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aede7462f9b241869a8fe16dc469effc : SUCCESS in 11m 03s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d7f9c08a3a24d49b5dca579903c3faf : SUCCESS in 14m 13s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/fd4d0f5174f24367aaae5ced6c6dea31 : SUCCESS in 20m 30s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/fbb85106a06747ac9e6f9339853d6d86 : SUCCESS in 19m 05s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1cc2b1bbf3c14a65a04aeb8c8be9cdcc : SUCCESS in 19m 14s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16fc7a01138b4f879ffd40c337489e51 : SUCCESS in 14m 43s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9511149a583949598901b9c6dc7b885b : SUCCESS in 13m 29s","accounts_in_message":[],"_revision_number":5},{"id":"03769722b5c5fe7199b8e52251b35e2de15b778e","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-04-06 11:22:56.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e739d85a5ee8f5eb26952af9e93dc6ba03dcc17c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 13:04:29.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f94bc975b1f47968eda035f86560da0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9be2950a257a4a1bacae9d83b62ec8f4 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84728c47a72d4e3dae23786ffd8df647 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b30833ee02b24dc295275638bf84652b : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d79e0786fd9a4b088cd1699906e3c800 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/861958e4338247d3b9d3b796ffc1a5e2 : SUCCESS in 9m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/17dfa058641c462ca1ea83f0343035af : SUCCESS in 5m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6d5480c1bea7440580999047a6436beb : SUCCESS in 10m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b4f238a84df74581a2cc544d56427976 : SUCCESS in 18m 22s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6fe6b7dbcaa94da88697e1136be53da9 : SUCCESS in 56m 12s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/57e6817e683a4d5694e91eab5569629a : SUCCESS in 55m 06s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/d2513349f4334cefae915320bcb85312 : SUCCESS in 1h 04m 12s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/2de8068f547a410d974a7e486ab01b71 : SUCCESS in 1h 13m 39s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/539f73404bd14e9bb2906ba545b60929 : SUCCESS in 51m 45s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ecddd726fef40328fdd3270eb863496 : SUCCESS in 52m 53s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/10219271daa2442aa2fc086760dc0800 : SUCCESS in 23m 08s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/7c76e43517f445d1959e7569c88ed1ae : SUCCESS in 22m 17s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/099b467455384767b10e37dbb568570c : SUCCESS in 22m 11s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/131c859d00b64d6fb9880465ce30a5ee : SUCCESS in 20m 39s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52be52a95b9a4518bf6a199fd4593404 : FAILURE in 13m 32s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a5e6f16c7cf44689d100585131574ee : SUCCESS in 17m 03s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/80ae78b433e24bce8163290f0e9de72b : SUCCESS in 1h 11m 17s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/0ca060f103404d69ba3f4d2469c934c1 : SUCCESS in 1h 10m 00s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f0881ad59a9b4e05b9617430ca53dc2e : SUCCESS in 1h 39m 41s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b2170262d1f4e8d99e1ef146e6c35bb : SUCCESS in 1h 13m 49s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/62367646b7ea441085891ae0a983174f : SUCCESS in 22m 00s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/f08dac01434c4c32a6040dcd6eb591c2 : SUCCESS in 21m 44s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/98a80bd0b99847278bac1efd0b3952fd : SUCCESS in 19m 21s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d82961cefd664f61822f41df5310314d : SUCCESS in 16m 45s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7591bf327f4447fcbffe27019b9cf9fe : SUCCESS in 15m 50s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b803799d413b4a9797a83b73a491670b : SUCCESS in 29m 29s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bef63dcaa9634f808d41618fdf9277f1 : SUCCESS in 26m 54s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/562726cb1ebb4c08bde7044fd88193f1 : SUCCESS in 16m 17s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/87a6c68ef2a947e5b23dd0c74915465f : SUCCESS in 15m 55s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/edf61d02063b4292abe5e7139c4f1547 : SUCCESS in 15m 03s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/04e58f33e6dd471dafe5f48a919a637e : SUCCESS in 13m 18s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f972e14e066c405185dc5ffe533fd62b : SUCCESS in 10m 10s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/a8d2605bfb5e45f490afc11da0f9ac9c : SUCCESS in 19m 00s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/57df0b75c13c4719a432075362ecb6a3 : SUCCESS in 17m 59s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7242697920c7449d9e0e058d45ed7b95 : SUCCESS in 18m 26s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/891d919f3368482c9d14ad0c7f1b9f9b : SUCCESS in 14m 20s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc517b49806f4fff8fc5dc88f025670f : SUCCESS in 11m 19s","accounts_in_message":[],"_revision_number":5},{"id":"ca94e02d80b79cdf3d69e0972df4fa83e63f4518","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-04-06 13:10:04.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"63a786f810818ac295c699afceb526d5b744308b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 14:57:16.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4035ffb049aa4934bcdb645b66112135\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/24c138f27f6d4caf8bb259c19e7b9191 : SUCCESS in 8m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28afb9349ea843ff93a1a0f0aba47c98 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7006be9410e84f29bd8673611bed403a : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f8a69b3e80e543daba843ffd3717b782 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae1089de912c42cba14c2603c5654085 : SUCCESS in 11m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2648724ae6d9408aa8567b6c9438f1f8 : SUCCESS in 6m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/846e198023774a73bd5892885e226b58 : SUCCESS in 13m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a139a3330dca4b799a4dad02b8a5d81f : SUCCESS in 16m 21s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/eeba04b1e4884bb2ba414b71afc27b82 : SUCCESS in 1h 10m 55s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/218cf2e297c74ec6b0f3e3eac78b44cd : SUCCESS in 1h 02m 04s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/25d80ca2ad83484c89e92f6a4b092b70 : SUCCESS in 1h 06m 27s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/1ec3f31094254a58bd0b6e41567906ea : SUCCESS in 59m 03s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0cee6f9254cd4c1a8a722600034a6ed6 : SUCCESS in 59m 50s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/42b6dade14594d008a7c46b279713707 : SUCCESS in 51m 47s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/2070ef874cad4d959e31e6b22dfbcfe3 : SUCCESS in 22m 44s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/d29c08aa49074acbb7b65f967ede1e94 : SUCCESS in 23m 02s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/295d0a0661164d9493435d1fe0d906f9 : SUCCESS in 23m 42s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d3ff07575aed4ece9ab82c50908c6f8a : SUCCESS in 21m 02s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c392ab2e57f498083fc9ffa7a8fcf75 : SUCCESS in 17m 53s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8060a3fae32a46a582d0bbc2e227c5d2 : SUCCESS in 13m 37s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/45c8d2299f3440d1b5ff1742231d67d3 : SUCCESS in 1h 16m 30s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/0302aa15e1b64e6191547b623c79d513 : SUCCESS in 57m 07s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/40e022499ef6480098815abe9470ce9f : SUCCESS in 1h 45m 34s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f98542d99aa41788a9345d8817566eb : SUCCESS in 1h 08m 00s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7e72424dd20a4cd6992d19d0bd8bdd86 : SUCCESS in 22m 14s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/16b6697b21964d96a0df51d07f01998e : SUCCESS in 17m 49s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/dd51b7b19feb4cf98e345f6dd3122bc1 : SUCCESS in 16m 00s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4930100c39c146d9b63fce9230b788c7 : SUCCESS in 20m 12s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3eac8271f30d4b8d8802e2c15bb73d82 : SUCCESS in 17m 33s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/df8a45171c204517a9f8f4cefd847a3e : SUCCESS in 27m 01s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/134522e051d04688893164da7992d4bd : SUCCESS in 23m 05s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/053e6231e6d349f5b5cbd50efe311049 : SUCCESS in 17m 57s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/bdbcc995551e4939b8a8560d154a6241 : SUCCESS in 16m 58s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fb6abbf2647f49f084a566506abd17ab : SUCCESS in 16m 32s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d957efc5c77846ddab55780b62d81a96 : SUCCESS in 13m 16s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/04a9387cc656478b9448daea171a19b9 : SUCCESS in 15m 04s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/e828f859f17648a984550add87804339 : SUCCESS in 21m 40s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/c867013e097244c4a47f6d38b35beb18 : SUCCESS in 14m 28s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1bf9209a3bd54e5f944a4b77ed1397fb : SUCCESS in 18m 00s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d96be9087a9748d7970a794aea9c780f : SUCCESS in 18m 20s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cfc3be64935146cb8ae6ca878f5acae3 : SUCCESS in 15m 17s","accounts_in_message":[],"_revision_number":5},{"id":"18dbcb89d36a0ac4b2f5eb4e7841e61b6d4297b3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-09-12 08:49:28.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"de145ec1568118e458d359addf4e222295295d5b","revisions":{"3ed10f7726fc5bfb6742919c57dbe67de542e68a":{"kind":"REWORK","_number":1,"created":"2023-03-31 08:26:21.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/40/879140/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/879140/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/879140/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/879140/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/879140/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/879140/1"}}},"commit":{"parents":[{"commit":"787df61aadfdb7140eb92856f0babb2311cfaf8c","subject":"Merge \"CI: Disable disk image build in seed-upgrade jobs\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/787df61aadfdb7140eb92856f0babb2311cfaf8c"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-01-12 13:39:36.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-31 08:26:20.000000000","tz":120},"subject":"[DNM] ansible-role-interfaces NetworkManager test","message":"[DNM] ansible-role-interfaces NetworkManager test\n\nThe ansible-role-interfaces is keeping backwards-compatible rules\nas strings, but been updated to handle rules as dicts for EL-family.\n\nnmconnections seem to want a table_id rather than name for it\u0027s\nrules, and the dictionary is the preferred way, where the templates\ncan perform a lookup of the id for a given name.\n\nHowever, if using strings for nmconnections the strings must specify\nthe table id, instead of the name.\n\nChange-Id: I9cf265ffc7acce926ce093d1e4f8d53a12be1f26\n(cherry picked from commit 01753259f3b5f7626724de8e634865171293f9f3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3ed10f7726fc5bfb6742919c57dbe67de542e68a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3ed10f7726fc5bfb6742919c57dbe67de542e68a"}]},"branch":"refs/heads/stable/yoga"},"3d7ae6653e12ee348f30e810609f4467cc27a7e8":{"kind":"REWORK","_number":2,"created":"2023-03-31 09:31:54.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/40/879140/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/879140/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/879140/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/879140/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/879140/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/879140/2"}}},"commit":{"parents":[{"commit":"787df61aadfdb7140eb92856f0babb2311cfaf8c","subject":"Merge \"CI: Disable disk image build in seed-upgrade jobs\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/787df61aadfdb7140eb92856f0babb2311cfaf8c"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-01-12 13:39:36.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-31 09:31:47.000000000","tz":120},"subject":"[DNM] ansible-role-interfaces NetworkManager test","message":"[DNM] ansible-role-interfaces NetworkManager test\n\nThe ansible-role-interfaces is keeping backwards-compatible rules\nas strings, but been updated to handle rules as dicts for EL-family.\n\nnmconnections seem to want a table_id rather than name for it\u0027s\nrules, and the dictionary is the preferred way, where the templates\ncan perform a lookup of the id for a given name.\n\nHowever, if using strings for nmconnections the strings must specify\nthe table id, instead of the name.\n\nChange-Id: I9cf265ffc7acce926ce093d1e4f8d53a12be1f26\n(cherry picked from commit 01753259f3b5f7626724de8e634865171293f9f3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3d7ae6653e12ee348f30e810609f4467cc27a7e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3d7ae6653e12ee348f30e810609f4467cc27a7e8"}]},"branch":"refs/heads/stable/yoga"},"b8c131ee779b8344cad2428b96b480edc4faf15c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-04-04 15:00:47.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/40/879140/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/879140/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/879140/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/879140/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/879140/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/879140/3"}}},"commit":{"parents":[{"commit":"20d8f6c3d10f85050a96663abbe64d7f4bb931f2","subject":"CI: Add custom routing table output to logs.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/20d8f6c3d10f85050a96663abbe64d7f4bb931f2"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-01-12 13:39:36.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-04-04 15:00:47.000000000","tz":0},"subject":"[DNM] ansible-role-interfaces NetworkManager test","message":"[DNM] ansible-role-interfaces NetworkManager test\n\nThe ansible-role-interfaces is keeping backwards-compatible rules\nas strings, but been updated to handle rules as dicts for EL-family.\n\nnmconnections seem to want a table_id rather than name for it\u0027s\nrules, and the dictionary is the preferred way, where the templates\ncan perform a lookup of the id for a given name.\n\nHowever, if using strings for nmconnections the strings must specify\nthe table id, instead of the name.\n\nChange-Id: I9cf265ffc7acce926ce093d1e4f8d53a12be1f26\n(cherry picked from commit 01753259f3b5f7626724de8e634865171293f9f3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b8c131ee779b8344cad2428b96b480edc4faf15c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b8c131ee779b8344cad2428b96b480edc4faf15c"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"},"2f6e892e6497c0f46a70f2fb4570212d28f325fa":{"kind":"REWORK","_number":4,"created":"2023-04-05 13:36:11.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/40/879140/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/879140/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/879140/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/879140/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/879140/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/879140/4"}}},"commit":{"parents":[{"commit":"20d8f6c3d10f85050a96663abbe64d7f4bb931f2","subject":"CI: Add custom routing table output to logs.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/20d8f6c3d10f85050a96663abbe64d7f4bb931f2"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-01-12 13:39:36.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-04-05 13:36:06.000000000","tz":120},"subject":"[DNM] ansible-role-interfaces NetworkManager test","message":"[DNM] ansible-role-interfaces NetworkManager test\n\nThe ansible-role-interfaces is keeping backwards-compatible rules\nas strings, but been updated to handle rules as dicts for EL-family.\n\nnmconnections seem to want a table_id rather than name for it\u0027s\nrules, and the dictionary is the preferred way, where the templates\ncan perform a lookup of the id for a given name.\n\nHowever, if using strings for nmconnections the strings must specify\nthe table id, instead of the name.\n\nChange-Id: I9cf265ffc7acce926ce093d1e4f8d53a12be1f26\n(cherry picked from commit 01753259f3b5f7626724de8e634865171293f9f3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2f6e892e6497c0f46a70f2fb4570212d28f325fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2f6e892e6497c0f46a70f2fb4570212d28f325fa"}]},"branch":"refs/heads/stable/yoga"},"de145ec1568118e458d359addf4e222295295d5b":{"kind":"REWORK","_number":5,"created":"2023-04-05 14:39:49.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/40/879140/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/879140/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/879140/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/879140/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/879140/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/879140/5"}}},"commit":{"parents":[{"commit":"20d8f6c3d10f85050a96663abbe64d7f4bb931f2","subject":"CI: Add custom routing table output to logs.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/20d8f6c3d10f85050a96663abbe64d7f4bb931f2"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-01-12 13:39:36.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-04-05 14:39:19.000000000","tz":120},"subject":"[DNM] ansible-role-interfaces NetworkManager test","message":"[DNM] ansible-role-interfaces NetworkManager test\n\nThe ansible-role-interfaces is keeping backwards-compatible rules\nas strings, but been updated to handle rules as dicts for EL-family.\n\nnmconnections seem to want a table_id rather than name for it\u0027s\nrules, and the dictionary is the preferred way, where the templates\ncan perform a lookup of the id for a given name.\n\nHowever, if using strings for nmconnections the strings must specify\nthe table id, instead of the name.\n\nChange-Id: I9cf265ffc7acce926ce093d1e4f8d53a12be1f26\n(cherry picked from commit 01753259f3b5f7626724de8e634865171293f9f3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/de145ec1568118e458d359addf4e222295295d5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/de145ec1568118e458d359addf4e222295295d5b"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
