)]}'
{"id":"openstack%2Fkayobe~879147","triplet_id":"openstack%2Fkayobe~master~Id86acf22cc548b37222a78300c1d5063f4527bf2","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-04-03 14:09:28.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"Id86acf22cc548b37222a78300c1d5063f4527bf2","subject":"CI: Add custom routing table output to logs.","status":"MERGED","created":"2023-03-31 10:08:50.000000000","updated":"2023-04-03 17:20:12.000000000","submitted":"2023-04-03 17:19:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"879147","meta_rev_id":"3c7ab7f8c5fd603cdf07f84044528042b40764af","_number":879147,"virtual_id_number":879147,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-03 17:19:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-04-03 10:32:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-04-03 10:32:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-31 11:57:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-03 10:32:55.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"2f690706c3eb8afddda19c7542687cec60e13337","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-31 10:08:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7e79e6a0253092ad260647b514b4c57b1fb4d0eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 11:57:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a0be977192df4f4cb6365620cdab3f80\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec11e9135f0f48c5a7a825e4d6d71ac1 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e60a2baba444cab96c57496a59bd940 : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/76fc446d4b154ac68661ab376f4fb26f : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b6aa294360464c86908dcb6b391871f0 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6deca0ad9824012bf5561a7d07768d3 : SUCCESS in 8m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6a76e9e5ac4e48a89550a6e22fbb10df : SUCCESS in 3m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3fe03ec984474b1791bb0c91f5313c48 : SUCCESS in 10m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e619cd9c299b4655810bb02ad262d336 : SUCCESS in 14m 35s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/873c779cd4b24bf98f0ad9de28051ef0 : FAILURE in 34m 10s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/ce01c257cab947f491190bab9fd1dcf3 : FAILURE in 37m 57s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3607703115704b799d17eab73e350588 : FAILURE in 32m 48s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ccc0762e36ba4b879b5dae99b49d2d9d : FAILURE in 39m 17s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ba915adbf1c74c70aba141258d15cd34 : SUCCESS in 17m 38s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/f2249c4cbb8b47cda01a761a349d79a6 : SUCCESS in 16m 33s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/301755169e6c47a68895f5878e3ed68f : SUCCESS in 15m 02s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5bc6544fa8b443ec813b251d66c7f530 : FAILURE in 1h 41m 50s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c74d7c8ce1447938425c7bb5d2072df : FAILURE in 1h 09m 59s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/47c9ccd5688f4fe5a4e311e8c1d2fdfe : SUCCESS in 12m 50s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1553894c08b54cc1b48e93a35a436968 : SUCCESS in 14m 20s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/93e3a617ef664fa488efe9c30b4c6ba4 : SUCCESS in 30m 18s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/62f45c37f0e84fafbf60755ef099a251 : SUCCESS in 27m 17s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/484208aca51d4816a3f11bef7e6abc8d : SUCCESS in 22m 43s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/08782ce846054f9895a15dec07be5f6b : SUCCESS in 12m 49s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/810325bdab0c48819c8f688bb7522b92 : SUCCESS in 9m 37s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fb1f3976b1f54e529cb3696ef3eb1e48 : SUCCESS in 16m 42s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7565b4423b1d46f1b3d1bd2c5500a2db : SUCCESS in 13m 08s","accounts_in_message":[],"_revision_number":1},{"id":"cb7904fa7ca4c176b8c384b4c9c3e1e13609b149","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-03-31 13:37:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a2ebffcb12a4cb68153d731f9a2cb1c973a655f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 15:23:34.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/77c1d7a70f0c4869aa37116dc4851386\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59e7fb8c070240e59f59731dfd215ee4 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1c6f61f83604069a1cbd262a0250144 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5f03cc47c87454683797523155b5cd2 : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/344a0896bbc54c4caf25bf69caf56b03 : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42a65e7565b241029811081f1aa73d91 : SUCCESS in 6m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e43ad6000bcd42179c37d17a07fc5452 : SUCCESS in 4m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/138f2af4d28243afb533e1ecc13f3a2a : SUCCESS in 8m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6e1b1a5c0faf4dc0a8f9f72e4f3664f7 : SUCCESS in 13m 49s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/bc2ea9a0b96c4b5cbc5dcae72c75bb1d : FAILURE in 37m 29s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/8c32d397e6c74f3880c5983614670645 : FAILURE in 40m 51s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c3e9a340f0f4406dbcc588c8bf56262c : FAILURE in 32m 13s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/2b95e95d94324f31bebfbd4898c26a07 : FAILURE in 43m 33s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/35823555c2d54d3ab1796b3d34416e2e : SUCCESS in 16m 04s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/b731749677b443c68801fb3942f53d6d : SUCCESS in 17m 23s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/152a1647f3fe42109a3610e295cca566 : SUCCESS in 11m 49s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f485bf8edc1e4e7d95f57b57201009bb : FAILURE in 1h 36m 36s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b69d56ab81744a692c21434bc22cbca : FAILURE in 1h 22m 06s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/2b623ed6875841978f802c55aa3a45a0 : SUCCESS in 16m 11s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cbc9e7e2b60b40cc979df94891e96773 : SUCCESS in 14m 10s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/56e8453f02324510a7c49e6539445da5 : SUCCESS in 32m 25s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d45bd4153dbc4eb2948b67e3f0b4e39f : SUCCESS in 27m 35s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f85107e91cde4e22a8332a4ab2ab6d06 : SUCCESS in 21m 49s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/640a4851b9f14c7a859e7d4943855238 : SUCCESS in 12m 29s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/91913636517b41eb80996d7cb17b05d9 : SUCCESS in 9m 00s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/219139de305d4d5e998d1e70aa46331e : SUCCESS in 17m 12s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de6c1aae130e48a7a0c5961ffd09abbb : SUCCESS in 13m 51s","accounts_in_message":[],"_revision_number":1},{"id":"046efd51e03cc11d63a71225837a2b5a8c72ea21","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-04-03 08:34:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0c6ead4564f7136fc0e252eed2529a816461fa48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 10:19:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e6ba2014d814d25b1ebccd1736f5213\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f8d9775731146a386a8541fb8c9dc2d : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8cba9e9275b4ac5a94218847a657697 : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e5410e4afd1449dcb8d1130469bbe54e : SUCCESS in 4m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba8c6843ed2b405a937284e42a74f594 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd35667e46e646a0b479ab1ca7212170 : SUCCESS in 7m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/86d88912b3474aa0862449c86d8e32dc : SUCCESS in 3m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9ef1c071c09047cf9649f7af75ba28a0 : SUCCESS in 10m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aba2b8109ab14ab3a2b948d75c80ff55 : SUCCESS in 16m 50s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e0e12c079a534234aa80be7cbda7439b : SUCCESS in 44m 46s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/47ce4a05c16247e3876f669aa9ac5104 : SUCCESS in 57m 37s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0113cc210b084ecc8954e1d8ae2cc87f : SUCCESS in 48m 31s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/80aa16fe26784b4f94e6a68a54452c12 : SUCCESS in 1h 00m 50s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e7b6696125d84d6dbece815d03c518e7 : SUCCESS in 15m 51s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/080132b478f8404a8d51cb78811e782c : SUCCESS in 16m 18s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/87f805664f4e42c68da61b41afb80b1d : SUCCESS in 13m 27s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b465f102fc7240e1a2c1ca10f1eea83a : SUCCESS in 1h 38m 51s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8da908826e9345538d5de2a522f9d16b : SUCCESS in 1h 19m 53s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/78b948377d444051a79d914bc9359d23 : SUCCESS in 16m 49s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66116e36cf444a8db6090fbf814802cc : SUCCESS in 10m 52s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/0728d8d935354368957eb10fcaf39dab : SUCCESS in 30m 48s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4af23a7e78154a59bd0be215c0ddae65 : SUCCESS in 28m 58s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fa661784d4ae4f998f90275f0994af6b : SUCCESS in 21m 53s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2fe2cd276ecf4e85b66d54331a79f24e : SUCCESS in 14m 06s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/37fe8c4712e94a9abf5f91fccdbb9a54 : SUCCESS in 20m 27s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fba2a9c275114c56aeb4fcba6da2a4d0 : SUCCESS in 18m 25s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3abbfcc1354b4d1b89d4f07ed97966d6 : SUCCESS in 10m 05s","accounts_in_message":[],"_revision_number":1},{"id":"08dc1556afae3aad0e4c35a320a08301945f1a4c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-03 10:32:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cabb3772322f957f065d2ce4367b05f7b7b85424","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 10:33:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a33aeff9280ce0201a40c576357c1f5a9fb867da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 12:26:46.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d2675d34db914088a9c0a315e968250d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4588246f13544962b6ffd94e5ccbdce2 : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f7fdda1086db4fc7a379570317e4149e : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e740a287514645b68da68458260046b3 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/491245854a6c4b658d957238b77bafa8 : SUCCESS in 6m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5c9917b3a60444ffb7677d9fc9fb249c : SUCCESS in 3m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a731b61ceed24ddbae408b0f26f2728d : SUCCESS in 11m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fe0331e1b26247dbb70bb7fc9d93eb84 : SUCCESS in 15m 33s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/92b1b2ddd2244117bb47c96c126e0abe : SUCCESS in 55m 49s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e0d78d0581e498d8b4a1b81d8ec6317 : SUCCESS in 48m 45s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/8578561fb98c47ad9b55996350275aed : FAILURE in 1h 07m 17s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/39da015d652e4637960da59efec811e3 : SUCCESS in 16m 08s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df4222dbd8234e1ca5ed369dd70f200b : SUCCESS in 15m 44s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/97eddaa297e54bae87b5124c26432dc3 : SUCCESS in 1h 46m 53s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/376bb3e61c844389859859b07303c446 : SUCCESS in 1h 11m 43s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/d4e311ed093040e9999b1776343ce306 : SUCCESS in 16m 19s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67cf0510c9074f4497b6f21f29ef30bd : SUCCESS in 14m 26s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ced4d4f851fd4bc58e960330e0fedd51 : SUCCESS in 25m 47s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7a1dfcbfb46403abf9b31124c62b735 : SUCCESS in 26m 44s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6a08144bdccb4595aaeb9d8c44ec8f81 : SUCCESS in 12m 42s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a73a2ad8ed5244d0abab7b1f24012a63 : SUCCESS in 11m 03s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ce8bd29b90be46fca99cc78f6052c790 : SUCCESS in 15m 52s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a7e41fd87d00433e911051685c95a941 : SUCCESS in 14m 20s","accounts_in_message":[],"_revision_number":1},{"id":"f4b1f8f578c29d734d0d271d30480423df94ba90","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-04-03 14:09:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7c9b3daf01d52bef3c6164652d4eb4f5c470f32a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 16:02:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6bd43cb52ee430b8c40abe4eaa3cf16\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b19d3402a2aa4b6297d6a4451fde1b36 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/beb739efb64348b9aa7740fa332b8dff : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4149cac7472a4ad796a962c725358ca2 : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cc4ec99f2e5c4e53acf77726b2192175 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b2dd95ff60f45fca2a711c7ec7ddd88 : SUCCESS in 7m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d297c782acc7409689acee2853c5d51c : SUCCESS in 4m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/beba937501ea495f973720a1dc3d2d2b : SUCCESS in 11m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/eaa8da7d40034eceb39519830c326487 : SUCCESS in 14m 46s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/fb38fb9f45894c7da43620a6b1fc8d52 : SUCCESS in 58m 50s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/9d058bd382ff4443959cb45aafeea095 : FAILURE in 41m 03s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8450012aec674ad08ab040d5cbc113cc : SUCCESS in 54m 21s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/05b4f60575144606b485a734acd74888 : SUCCESS in 1h 13m 21s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ae48301ac0604a49989da8aab660977a : SUCCESS in 16m 40s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/02b4edf9f20e4d629126a42af9fc2a91 : SUCCESS in 18m 04s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/56855e23ce8449589895ae35cf23babb : SUCCESS in 14m 14s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/bd2b956fb0fc442285ba7f068a1d8091 : SUCCESS in 1h 46m 03s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a33010446724407ab2f211c5b6733e64 : SUCCESS in 1h 14m 04s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/06f9d831566e4dd69f1863dfcb539dc0 : SUCCESS in 19m 42s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/29b1ad6d433d4c4e907e7e062e956c56 : SUCCESS in 15m 01s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/992a811f35704175a7213c858fe186e2 : SUCCESS in 30m 09s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/34600bdbb3da42d7afb7538c4785d871 : SUCCESS in 23m 58s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c27d2114a8e408398ae95ab410eef63 : SUCCESS in 22m 59s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0a850cb00ed744738ba0acc1a557150b : SUCCESS in 12m 27s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/37f2346edf2149439aebe52ee41ee3be : SUCCESS in 12m 33s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/14ce759bcf8e476fa10e91b2c247a6ee : SUCCESS in 18m 42s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d26ac375dfc344bdb29aadde8bea7760 : SUCCESS in 14m 43s","accounts_in_message":[],"_revision_number":1},{"id":"d2d27e3d32ef6e8cd49390b718d76f051e9b262c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 16:03:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"410ccb85675149892ff6e943d8fef0df840fdb5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 17:19:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b3d1b9be3ef49188058ad6c686ba10b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33a973f485d1471e956a06fecb105226 : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3138955b81494c86a2c53d31a00b7e79 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/65d63a61143b4d5d9087741fb2aa512d : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c962739f5fcd45278f055c19f8577d83 : SUCCESS in 7m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1e3627edb01c44a2b0446d259c4b7646 : SUCCESS in 4m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ce5c3684f69e4bb9b4c8d00e6ea91ed7 : SUCCESS in 10m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6e348f4f799a4375b71e7b64f590a977 : SUCCESS in 13m 47s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/475a9b072b224d3080b106f4ee1c4f19 : SUCCESS in 53m 17s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e22061232ea540749fed9acf79e8f495 : SUCCESS in 50m 34s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/f72c9f253635436f99c2c0a838335869 : SUCCESS in 49m 00s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/298f115c812041bd9af143a7f3e584ee : SUCCESS in 14m 03s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a685995b24f24bdfb09baaf40662732a : SUCCESS in 13m 19s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f6714f2a130a4e73bcdf5d38e50634f3 : SUCCESS in 1h 14m 50s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/622689f11331453f8177602d86270566 : SUCCESS in 1h 09m 33s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/850d9674f7c84d51907c95d9b180b364 : SUCCESS in 17m 23s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e41175139e0044158d2acb7181aa8b28 : SUCCESS in 11m 09s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/17993eb16e784c43b3cf9b9178cbeeab : SUCCESS in 23m 20s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8b3e275dc0c74e2892a16b4784615e26 : SUCCESS in 20m 03s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4bc18caff3774e34b449177f528babe6 : SUCCESS in 12m 05s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/71919b477ab74b89ae93341518e92f1f : SUCCESS in 9m 39s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/521edfee97444cc1a80aecaa02f79a40 : SUCCESS in 17m 47s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/554c805676134d8cbd40bc1316453c94 : SUCCESS in 12m 25s","accounts_in_message":[],"_revision_number":1},{"id":"71ffd3e35b6341b147a7d2704d5550906fcd03c5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 17:19:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3c7ab7f8c5fd603cdf07f84044528042b40764af","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-03 17:20:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a20c7d94abdf48da93048b8a05cd5b73\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21bbce53b5dd417ba8e00ab866da6e4c : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"84fcc6ad0bb1cc4b9c1198e8f391396f911dc2ac","revisions":{"84fcc6ad0bb1cc4b9c1198e8f391396f911dc2ac":{"kind":"REWORK","_number":1,"created":"2023-03-31 10:08:50.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/47/879147/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/47/879147/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/879147/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/879147/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/879147/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/47/879147/1"}}},"commit":{"parents":[{"commit":"137e64b57c86fde852b49f9975ffbca26da79de1","subject":"Fix opensearch container builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/137e64b57c86fde852b49f9975ffbca26da79de1"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-31 10:07:50.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-03-31 10:08:49.000000000","tz":120},"subject":"CI: Add custom routing table output to logs.","message":"CI: Add custom routing table output to logs.\n\nChange-Id: Id86acf22cc548b37222a78300c1d5063f4527bf2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/84fcc6ad0bb1cc4b9c1198e8f391396f911dc2ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/84fcc6ad0bb1cc4b9c1198e8f391396f911dc2ac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
