)]}'
{"id":"openstack%2Fkayobe~879190","triplet_id":"openstack%2Fkayobe~master~Ia956cfea7e7bfe47bf1e73c9edcac602caf45579","project":"openstack/kayobe","branch":"master","topic":"add-multipathd-config-support","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-04-19 12:58:48.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-04-19 15:21:03.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-04-17 13:15:34.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}}},"hashtags":[],"change_id":"Ia956cfea7e7bfe47bf1e73c9edcac602caf45579","subject":"Adds support for custom Multipathd configuration.","status":"MERGED","created":"2023-03-31 16:02:11.000000000","updated":"2023-04-19 18:49:45.000000000","submitted":"2023-04-19 18:48:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"879190-add-multipathd-config-support","meta_rev_id":"0eff140f38cf48f78271f4b7234d4cc415e1a1e3","_number":879190,"virtual_id_number":879190,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-19 18:48:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-04-19 12:58:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-04-17 13:15:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-04-19 12:58:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-31 17:45:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-04 20:32:53.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-04-17 20:24:29.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"589813a0ba35f462cf3d64c2cb4535cd186fc419","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-03-31 16:02:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c3f1fd4032a08c825527480cb2b806228e9a00cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-31 17:45:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5bec047b45074280ba9adeb72253b647\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5dad14d0da6949d395a29a710f7aa114 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/941a34c80f6142b19e1b00473fbfc10c : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/773230d8dcb846a0801f38ee6c4889a2 : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4add93c42c5e4f289ac9172d467b46d3 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6464b1582df440a7b83069f8931c3ecd : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f0adec2767047338dc665d5c50962bd : SUCCESS in 3m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4e553f542e26425f8468ce06d50923e4 : SUCCESS in 4m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1daf8a3c3e02439284d66ca97b3d5ef8 : SUCCESS in 10m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ff39241e200445cfbce6d1a29ced365b : FAILURE in 14m 41s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/24c2c6bc47ba46b093a424d27234c913 : SUCCESS in 57m 36s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/e65d9d6f4ef840d9b2ea5db87167fb6a : SUCCESS in 1h 00m 57s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fbaff8a4ed7142bba1d076bad25fac47 : SUCCESS in 51m 42s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e054fabf99e449d59893943e95266bd4 : SUCCESS in 1h 05m 58s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5d112d16eeef4a7e9e3ef9e10c7798dd : SUCCESS in 18m 18s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/d412787b88f64bc2b1d0d2bcde651ab1 : SUCCESS in 19m 12s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/35af8003ad554c31a180e60135058fce : SUCCESS in 24m 03s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ae6d4ea846b74e6eaf57a0a490bcc684 : SUCCESS in 1h 38m 24s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ce6fe9a5b48449f8171a6e038c536c1 : SUCCESS in 1h 18m 31s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/6a90ada257144896a9131fbed5e289a6 : SUCCESS in 17m 11s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aab11658f95441db86b0720295c7726e : SUCCESS in 13m 52s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/7429401fb7734406ac63982e84937ed7 : SUCCESS in 31m 27s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ef6d40f072e14059b3ad812411c51196 : SUCCESS in 24m 56s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/908b9b5b75d74411ae5bc13ab1a86b2d : SUCCESS in 23m 35s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/cfa4d9d4ff6b4535a36521c753e7de0c : SUCCESS in 13m 18s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/419d4bbed97448c2a07ace10b5b6db22 : SUCCESS in 9m 34s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0afc973d4e4f412ba4876d114f1627df : SUCCESS in 14m 39s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a20ae0cfafca4e12a99df9cc7da648ae : SUCCESS in 13m 28s","accounts_in_message":[],"_revision_number":1},{"id":"f252927f06ecd7218ffe017e7d53bf8154c0f8de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-01 12:01:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ea3cfaedd204a457ac895410a564cbb42aa75d9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-01 13:53:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f66dc8d257b4b189e8d7d51dd31b7c3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2aad98ecaf5e4fb0a16a2aacefdbb47b : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d39735d2ff834b2bbbe400d015ff108b : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea7ddd8261a546398efd4a583d217d08 : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0b11460fd14e47c59a19c6c2bb275985 : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5021c20d755c41c6b5ac5576d874b032 : SUCCESS in 6m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc4a0662139740ac9ff6adecfa4c84d3 : SUCCESS in 3m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7e7acce1103b4724b1782b289f2d4486 : SUCCESS in 3m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7624e375d6ef4f338497d0662a9db4fd : SUCCESS in 10m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0f6bc90ea06943959d913a09e13d5d10 : SUCCESS in 17m 03s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/f158577a9206406d8549824115da49cc : SUCCESS in 58m 02s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/ec5a1c62a58c4a09bd01b4f48a0e3c79 : SUCCESS in 44m 19s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f6b9994669d845b9a1c6f0901b62fced : SUCCESS in 36m 53s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/50c126d8633646f891ebfcd681b0f32d : SUCCESS in 1h 03m 00s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d7b95facf0114fc88ebc2a698b1e5d38 : SUCCESS in 16m 02s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/916376ff63ee48b5ac88e3dd89cacee1 : SUCCESS in 13m 14s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9bc6d9e21f534b3e847977791f74f2e6 : SUCCESS in 11m 04s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/82026a7d32f4489a92735b04f609da34 : SUCCESS in 1h 46m 04s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66b7c22d321845e6a3c16e80b5d1b673 : SUCCESS in 51m 41s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/710e3899d3524c2c9ecfb8d5399c1493 : SUCCESS in 16m 54s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/32a3a565eaa848aa9b491f3f8e9c85fc : SUCCESS in 14m 40s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/4b33765ee51c4d8ebab1017ef9779013 : SUCCESS in 32m 12s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3414519c0d31468b9a3a286a3d88ba2a : SUCCESS in 25m 53s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/11777a50932a40b5bf545c3722a1e5f6 : SUCCESS in 17m 19s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2288d357984e4df3bb69ab752daf0092 : SUCCESS in 12m 32s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dad0e153583c496a9d0dcd9da4db0575 : SUCCESS in 8m 46s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/eeee20da21cf4f0eac783e632f4743d0 : SUCCESS in 16m 14s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f47d444b9a9648faa0f50842731ab2d8 : SUCCESS in 11m 02s","accounts_in_message":[],"_revision_number":2},{"id":"440ccd7f7ed493e7ae32f781f95ddd947d227fce","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-13 19:16:16.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a581aec5aefea29b2a99e831487d541809706699","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-13 20:56:14.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"232daf944f6013941e357c408cf7bdd8b5a017bb","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-13 20:57:35.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"ad59116b186201e991e8140542b22a17e8012923","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 22:37:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a0c48a9899a46ba94a92c6d6981ed7d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ed5bef294014564abc7fc303504d278 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cccfb7fbb1a54f5da378086c482532dd : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fad86fa344cb4082b8a8c65ce95ef6b5 : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e143898d10554a6287b918cf5f5384f1 : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df2f3ef641cf414eb18f6460e760c971 : SUCCESS in 8m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3cd131aacfc4c62a8a8cf46f4f26ef6 : SUCCESS in 2m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6741974c09794d5889c0fa83e35e7bdd : SUCCESS in 3m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/703df64a7b384c2b9595b794003c6e5e : SUCCESS in 10m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3afde1caed2a4edd818808f7e9acf24a : SUCCESS in 14m 05s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/17be5a7efb534676b485d9b82e481ac6 : SUCCESS in 57m 50s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/efe623a08b2d4f62b83d536997ec1935 : SUCCESS in 47m 48s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8fa4f48dd3a049b293463be93be57d7f : SUCCESS in 52m 31s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ceeb56944c7b4e4393bfab1a2bbc2126 : SUCCESS in 1h 01m 40s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/20c960904717465c8a25a0856118d033 : SUCCESS in 16m 42s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/a567050dc8fe4a17ab7a0f971c6a1b61 : SUCCESS in 15m 12s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c68da1cd0461442aa9c47c7e12ac4194 : SUCCESS in 13m 53s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b61a7cce8b7b4d0280861a760b874fe1 : SUCCESS in 1h 35m 35s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f87b88b53ee54ef0a28f13d92250ceee : SUCCESS in 1h 12m 12s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/df32bfcc59d541c0961115f0a296f29d : SUCCESS in 17m 39s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/994519127579439481ebf72ae77d530d : SUCCESS in 13m 52s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/4d4465e213d44125b27eea1067d6d1d9 : SUCCESS in 32m 23s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8eaa0e22fb1240b1abc2b38abf6c2189 : SUCCESS in 25m 49s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a39ecf6a27944c3791437f8f37f1445c : SUCCESS in 18m 33s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/68c9afc8cca0423bb73eb0938a56a563 : SUCCESS in 12m 47s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5915e6cfb01f4a7397537f277a451bcf : SUCCESS in 9m 17s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a85ac1e6d9ec4ce09142fccbf776c2ab : SUCCESS in 16m 28s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4f529e24bd1a408fbd14d4dabe98046f : SUCCESS in 13m 23s","accounts_in_message":[],"_revision_number":3},{"id":"4e2369f1fa065e32fcf7422f57509280530d7773","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-17 13:15:34.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"0747301084ebf9ce0f3d29e88b77b72b5b5968c2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-19 12:58:48.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4cc11dccb7d3ebf258fe6306ee633fa58aad1fd2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 12:59:15.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"62d8b4808d6237993675d6b6d6b4a2dc479cd942","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 15:19:29.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ef3d590e29d040a0a3d97e89d82d9495\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8626839beee2433ca5194ff4bd699272 : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6387acd67f024a5aa6122287fc3f63b4 : SUCCESS in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b4e2600099784375bd03e8d3017ae16e : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4109ca0fb624d7fbf2f6bba9d9194d4 : SUCCESS in 7m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e228b9964114d408276d2b64e965336 : SUCCESS in 3m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/46e2f8a969134c29aac7c146ddaca307 : SUCCESS in 4m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9e44d2b2546742308713231eaad885cd : SUCCESS in 11m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d4912fd695cd417799caa11823338c57 : SUCCESS in 20m 28s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/f6f6349a60464d29997520eaf9cef0fa : SUCCESS in 57m 52s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eea82b2d907145b39bf6b374eeaae9ec : SUCCESS in 53m 49s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/2bd0a7fb50cc4a5887a7eb58a6004f30 : FAILURE in 1h 08m 44s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/29a8846f022d424a95b87767847fc89b : SUCCESS in 13m 45s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d8e0869771914861806d63a99bea6066 : SUCCESS in 12m 43s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f78fe595400a4e88a9ea694bb8e9843a : SUCCESS in 1h 41m 16s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c24842d3c494820b1d3e53b891e8188 : SUCCESS in 1h 09m 46s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/946cd71201b648378cda712d6b15e50b : SUCCESS in 16m 41s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/964188db28574c8eb3bbc562cd66cdec : SUCCESS in 15m 11s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f4ad936cb704478bacb66a8c5de6ba96 : SUCCESS in 26m 18s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9f0e1697a32f43f987b9edd97421d185 : SUCCESS in 23m 45s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/911b5a7160aa45e4b9d3bb879ebc0798 : SUCCESS in 11m 38s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/119bdd96c29f4b139915f671a9a1d37a : SUCCESS in 10m 32s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/683ce3c574c940a9bca043170dc33d74 : SUCCESS in 19m 27s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a031ff7ab66b4e3f895fc200b1e5e43f : SUCCESS in 15m 12s","accounts_in_message":[],"_revision_number":3},{"id":"2a0ae6944b6f41b695f32ae79e13e92c973effd1","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-19 15:21:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"86ca37d9f054639f804b06e00beda478ce0a2483","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 17:24:41.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53f1cfb8869645a9b1c07339ff970781\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/731d755d806241299d8ba44d1a71fd5f : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b449347cd3834b51b4924aa7af7a94b3 : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/445f9b237a824ce5837f74959a345d6e : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/916ed800fddb4a299db02b5265494dc9 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d372cb4db8f45a6b3d04c62bc315377 : SUCCESS in 6m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b042ed6b96254410908ce78aea2e9d3a : SUCCESS in 3m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4739ecf2acb74115b8d3f014b60f9d32 : SUCCESS in 4m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5fb573560bb441c7b1bceb19cc698de9 : SUCCESS in 13m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/86d3cdcee708431dbe4e75fd70916aec : SUCCESS in 17m 22s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/6e5a2e53007945b88b612900319fc8de : SUCCESS in 1h 06m 15s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/06ed0bdae1d141d8bcf04ee51ed3c4b3 : SUCCESS in 53m 37s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae835e8302f74f1d9db79786bbefd547 : SUCCESS in 50m 17s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/052fea3c0b9a4cb6829feff7303a0d6b : SUCCESS in 1h 05m 23s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a507a61cde974e8e86de8d30db7101c5 : SUCCESS in 13m 53s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/b9778484a1a54f7fb267d3d5fd5bc6d2 : SUCCESS in 17m 39s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/27e6fa719653481e9990ce14eccd3bcd : SUCCESS in 17m 33s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e8e301aaffde4fad8656ee16ded3b8d3 : SUCCESS in 1h 54m 33s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/44ef26abeb104118a595b49e86837cc9 : SUCCESS in 1h 24m 51s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/62645bd6358f43b3938854d1e0bf1d28 : SUCCESS in 17m 11s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39f924e80a4e43678e81f5292c29c296 : SUCCESS in 15m 10s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/2a543cb2ad56444e885239dca34bed11 : SUCCESS in 30m 58s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/312e03fc65e14dddbd112e4717c680a3 : SUCCESS in 28m 46s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e1a12dccbb94e2a95997c9463ce7044 : SUCCESS in 22m 29s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8a88518dc65f4436b1542253c69e698c : SUCCESS in 12m 38s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0a87db8d1ed74260abaec9aaa0027937 : SUCCESS in 17m 03s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9c1463f750d94603aaea1a6101eb69db : SUCCESS in 17m 05s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e87b394911f74f31a4f0c4e23e818153 : SUCCESS in 13m 31s","accounts_in_message":[],"_revision_number":3},{"id":"e5c2022d344f82cce2e5f2920752ccdfed3ed41c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 17:25:06.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5bc558eb91cff24780f6552462a6d3860c31e4b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 18:48:44.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/820dd461f0f64450a43f0df0a7e569b3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c7be30dc44244209cbbb411cfd48142 : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc59d2d1441443d7a7f03043ce3e933f : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dc7859910a0b45c6bb6046d70898b6f7 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54fa00a7d5634d878b8f13eca7d1f248 : SUCCESS in 13m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/204d3b314c6345579044c3ffe2ace462 : SUCCESS in 2m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fb5b7db90c83447cb9a1e3dff08fc1bd : SUCCESS in 4m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d52c9b0e109e4b4f88ee108c22261d42 : SUCCESS in 9m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a499702bb7ff49bebfaf284af593d182 : SUCCESS in 15m 36s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/0825e0a4f3bf46f38a8e8fdee96411df : SUCCESS in 58m 18s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c08fe965ced4474eab32c982f6573d4c : SUCCESS in 47m 18s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/9e3283b97f6848fc9e3ac7ac11766a77 : SUCCESS in 48m 21s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/84606dfa611f4dd3be9fa605a3ccfe50 : SUCCESS in 16m 07s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/105870c191de4fd993138ac3cf6feeae : SUCCESS in 13m 52s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/83608123f2f0458089fdddd49a945bed : SUCCESS in 1h 17m 34s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e3ae0c5929bb47089df6740d9ef45887 : SUCCESS in 1h 12m 19s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/233eaa4f71dd431b9c28faebf627bc5e : SUCCESS in 13m 16s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fa6729c8fb244cc3b2274040ed39b41a : SUCCESS in 13m 32s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d0bd98c4ca28457aa0909f37b3f5cb7a : SUCCESS in 26m 47s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eca0f68f1c5f474e87c47bc99829c241 : SUCCESS in 20m 53s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2c7964a79535430c9ed9f9b1d4a414bd : SUCCESS in 13m 01s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/682ff82512d944fc802a3e7c36fbfda5 : SUCCESS in 10m 01s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/48befefb03484dc280707d21b74d571f : SUCCESS in 13m 06s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2325110ae1984e8b829c0ca5e9d2e5cb : SUCCESS in 12m 38s","accounts_in_message":[],"_revision_number":3},{"id":"36181952d0327f19619fef0e2b4ecbdf965760a0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 18:48:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"0eff140f38cf48f78271f4b7234d4cc415e1a1e3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 18:49:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc37160ed1564e229670be7a12d8e00a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/483add6058f9406b9915850989a69638 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1764f18acf21468d95bb51d8fc178334 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"25fe761e62f647ed828dc5441d651fb4433c8bf0","revisions":{"bf2ee8d29436dfa50fbbd3313a51bf966cd8327d":{"kind":"REWORK","_number":1,"created":"2023-03-31 16:02:11.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/90/879190/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/90/879190/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/879190/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/879190/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/879190/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/90/879190/1"}}},"commit":{"parents":[{"commit":"137e64b57c86fde852b49f9975ffbca26da79de1","subject":"Fix opensearch container builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/137e64b57c86fde852b49f9975ffbca26da79de1"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-03-31 15:39:56.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-03-31 15:58:29.000000000","tz":180},"subject":"Adds support for custom Multipathd configuration.","message":"Adds support for custom Multipathd configuration.\n\nChange-Id: Ia956cfea7e7bfe47bf1e73c9edcac602caf45579\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bf2ee8d29436dfa50fbbd3313a51bf966cd8327d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bf2ee8d29436dfa50fbbd3313a51bf966cd8327d"}]},"branch":"refs/heads/master"},"e894952c06e73c2d594d13618dd307fa0d507252":{"kind":"REWORK","_number":2,"created":"2023-04-01 12:01:21.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/90/879190/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/90/879190/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/879190/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/879190/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/879190/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/90/879190/2"}}},"commit":{"parents":[{"commit":"137e64b57c86fde852b49f9975ffbca26da79de1","subject":"Fix opensearch container builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/137e64b57c86fde852b49f9975ffbca26da79de1"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-03-31 15:39:56.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-01 12:01:09.000000000","tz":180},"subject":"Adds support for custom Multipathd configuration.","message":"Adds support for custom Multipathd configuration.\n\nChange-Id: Ia956cfea7e7bfe47bf1e73c9edcac602caf45579\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e894952c06e73c2d594d13618dd307fa0d507252"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e894952c06e73c2d594d13618dd307fa0d507252"}]},"branch":"refs/heads/master"},"25fe761e62f647ed828dc5441d651fb4433c8bf0":{"kind":"REWORK","_number":3,"created":"2023-04-13 20:56:14.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/90/879190/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/90/879190/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/879190/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/879190/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/879190/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/90/879190/3"}}},"commit":{"parents":[{"commit":"137e64b57c86fde852b49f9975ffbca26da79de1","subject":"Fix opensearch container builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/137e64b57c86fde852b49f9975ffbca26da79de1"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-03-31 15:39:56.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-13 20:56:01.000000000","tz":180},"subject":"Adds support for custom Multipathd configuration.","message":"Adds support for custom Multipathd configuration.\n\nChange-Id: Ia956cfea7e7bfe47bf1e73c9edcac602caf45579\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/25fe761e62f647ed828dc5441d651fb4433c8bf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/25fe761e62f647ed828dc5441d651fb4433c8bf0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
