)]}'
{"id":"openstack%2Fkayobe~879539","triplet_id":"openstack%2Fkayobe~master~Ica5cd17ddd784aee8ebdf0726b8c7098fb4a897f","project":"openstack/kayobe","branch":"master","topic":"fix-absent-path-for-passwords-yml","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-04-12 14:14:11.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-04-13 16:41:50.000000000","reason":"Change was abandoned"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-04-13 15:33:53.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}}},"hashtags":[],"change_id":"Ica5cd17ddd784aee8ebdf0726b8c7098fb4a897f","subject":"Create directory for passwords.yml if absent","status":"ABANDONED","created":"2023-04-05 10:10:58.000000000","updated":"2023-04-13 16:41:50.000000000","total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f347a3d2a6cbe9b5af7fcd0fab0392fc76e9a4bf","_number":879539,"virtual_id_number":879539,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-04-10 14:14:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":-1,"date":"2023-04-13 15:33:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-05 11:38:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-12 14:12:33.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-04-12 14:14:11.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"8008993bb5358a671b6f7e11668e80b5ba859bb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-05 10:10:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"343ae691c637c9f0821991dc2bf6f4921dabacd8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-05 10:15:02.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9df0b9ef47778feda5388914b5d9cb9e59f506e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 11:38:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b501f0d42b4940d1a4375d65db0fd360\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/69c48236e6b847248135ed40624cd55a : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad6bba82989e43588312adc4dc4066a7 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19c8e3899eac4fc9b594d67fe8f9e648 : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5f7fd5ad35024527be383de294960d41 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3df633d83e5406ebf79e0689de2d827 : SUCCESS in 7m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7b6268ecf760464381ba8e09604c3266 : SUCCESS in 4m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0222999c25084e5dad31b0d55d900808 : SUCCESS in 10m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1232c3c25fcb4d93adb10293cc189870 : SUCCESS in 30m 58s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/415382efe8c74f49a206d0776e92167b : SUCCESS in 57m 18s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/6e028adb8f0c4f8c9bb837e643cccf84 : SUCCESS in 48m 32s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca8d7b1be89143d7a34fd7d5940a2a4e : SUCCESS in 48m 04s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/12ec6c01c1bc463c937d616bfe06ec8f : SUCCESS in 57m 29s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/fca9ec8a81b04422b40cc1b35bc65fcf : SUCCESS in 14m 08s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/fa33dc3eea054ac0bd4e72683d0b6560 : SUCCESS in 16m 46s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da3c467a90724c4f82a2f041b8c6c76d : SUCCESS in 14m 00s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b1fe21410659467fa019d8ee64aacbd4 : FAILURE in 1h 23m 02s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/890b301b29df49b48bd54b41c3669132 : FAILURE in 1h 00m 40s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/1513882b5c7844c2805a311d98537566 : SUCCESS in 18m 05s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c6d279047d05402d904cfd0b3f95afe8 : SUCCESS in 13m 12s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/b01299fd9a02487b9812d524c77d2a69 : FAILURE in 14m 52s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0ba7e8b8ac0f4b7c9dcdd0c68245dc1a : SUCCESS in 28m 15s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b9fc991826a8417eb6f4f434bd41b5e5 : SUCCESS in 22m 11s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/70f67f389e324ff5a6867204872fb4c5 : SUCCESS in 10m 39s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c5d717150c8a4620bfcceadcb94c79c6 : SUCCESS in 10m 15s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/df84f2ceabfd4d308eb50fc6180f07a2 : SUCCESS in 15m 42s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/397f36c0ab564ab4b12826dbb1e51c9d : SUCCESS in 13m 33s","accounts_in_message":[],"_revision_number":2},{"id":"08c6cc7ff84b5ec5e61fb6af0ddf1ede28efabc1","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-05 13:34:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4a0f0cf0886d608d934604afda5a5a772aef1bbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 15:05:54.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b0fc7a2855074c48b35a59420feaea53\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec64cb282fa548e2a6221f7ea7a1878b : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f60b39722f10488894bd50bd3efdf542 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d5ea081e8a04056bb629ed619c08c8f : SUCCESS in 5m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7d42d5d6ed984d1487b138cf290632db : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84048598a2a04fbcb7b5b3d2bb4cfc57 : SUCCESS in 6m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/75a001380fc7429e9bbcb9dfc1707645 : SUCCESS in 4m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/92810f791a0f47eaac4ac3b5b9588ca7 : SUCCESS in 8m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0041a99c9d754f3e8ece64f2f559724e : SUCCESS in 16m 59s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/170df6e48f864c84bb0805bc7f22dde2 : SUCCESS in 57m 48s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/64bef04efdd54ef9a59198cd1d85b3de : SUCCESS in 51m 51s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/193eb6f231d44ff29eab09d8171dc494 : SUCCESS in 51m 42s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/673ec4524a6648d6a9345065ec5eafc4 : FAILURE in 1h 07m 55s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5e220d1ed02e46bd82f1e21853efaa26 : SUCCESS in 13m 35s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/72cef2d58b7445e9a6eedcfda3453d57 : SUCCESS in 17m 29s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c563d1cbc1e1455bb4cce55ddd5f9114 : SUCCESS in 12m 15s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/aaf42154ccff4cf798b661323f73b2a6 : FAILURE in 1h 27m 43s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9d0ee881fb3449b2af3d449ed3b92dc8 : FAILURE in 58m 22s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f19813d17b504013bb493741ddba92e2 : SUCCESS in 17m 04s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d9215f4db174589974500392e9bcdd5 : SUCCESS in 14m 12s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/f951b4a18f4347859a846dcab2be2717 : FAILURE in 18m 00s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7361e62691994f3183878d407750583a : SUCCESS in 26m 09s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/45fed940cd4844a194ce8ed5b1e2110e : SUCCESS in 19m 21s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1607a9996373499e9ac48f39f5df4050 : SUCCESS in 13m 04s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1573ee2e88bb4abfaedd45c5ab9eb504 : SUCCESS in 10m 51s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/96c5769350024855836d262644c7fe26 : SUCCESS in 16m 29s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33fca27705a14eea8e6085656ab27117 : SUCCESS in 11m 01s","accounts_in_message":[],"_revision_number":2},{"id":"6fb35d2027cf77e44db864bddc27de23420ab429","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-05 15:10:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2729ecb1b66d2c8c2d4f3ddb03f5a400137776a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-05 17:02:16.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/46a27ea4ce4042269cd6e72d10b23488\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab5288c9f184497e9dab5044613c327c : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66f8fa2c9a9c45f48f7069718f5b4e38 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13a7a74ae62245d6b15c06ef66356e4b : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9931e08e3b5c4e299646687dc84a499e : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fef30c14da19457887995b4301570d5c : SUCCESS in 6m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6da307d1cf7a4bcaa4cc82710d1b2fa8 : SUCCESS in 3m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ce1527b84cf4427c85977ed72b735128 : SUCCESS in 9m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/be96a9afdeb542718ca23672bf551b60 : SUCCESS in 14m 34s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/8424956a3d97417db9b47480b98464f1 : SUCCESS in 1h 04m 11s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/32fb740ffbc14499841385ab2da08933 : SUCCESS in 52m 45s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a09def11ac1247d69edbc46c7c1f228f : SUCCESS in 44m 59s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/063d28c4261948519607fd11340e3891 : SUCCESS in 1h 03m 06s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9c6fa881f6ce459b84fc19232bf5782e : SUCCESS in 15m 45s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/3759526a826d4331be5985a0008e91d3 : SUCCESS in 15m 37s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a7924562b800442c9a1a5bf35efd0084 : SUCCESS in 13m 27s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/55df722888d94442b4590ef5445ad8e6 : FAILURE in 1h 44m 18s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ecf4bae864445adb78997b937dc1891 : FAILURE in 1h 22m 11s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/d241ca39192c492793349f4bfa7b3635 : SUCCESS in 17m 09s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7999ecc9f3d84a45956b529d80aa57d4 : SUCCESS in 14m 33s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/d90b1c5460be4d0fbe3b565fe69f8f9c : FAILURE in 21m 59s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c2d6653b4da84efb9d46292c82b1590e : SUCCESS in 25m 29s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/82246f7eb56c4d03be2887c7b7fb70f4 : SUCCESS in 26m 07s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3dce23826da5442181684f4ae82b3b6e : SUCCESS in 13m 59s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/358873f3f46a4e09bf2444d6cac73a31 : SUCCESS in 9m 37s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3301f4a606c946dc8161466315368391 : SUCCESS in 17m 46s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c20e891503848f88c54925004812404 : SUCCESS in 13m 01s","accounts_in_message":[],"_revision_number":2},{"id":"97a71c052dda3d4466ee5ac26c0f6467c2d77712","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-06 14:11:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b1cfc0bf03faeb966839c32f8ad249ab00755e40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-06 16:10:55.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a4770cc36a7f438ba2d822d967b2b361\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b86ab656ed8f427b9e527389420328b2 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8c532c11b1846fca7ca0ae4cb468c3d : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4fb5f8f6ff39402fac46a0789f66414c : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d943d3cbc7d44fdf8047055efd7aa122 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7afe63a1b0fb4ccaa8917717eb6b9d14 : SUCCESS in 7m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3a47b7b1e4084205887ebdf66332223b : SUCCESS in 5m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8b60251cc2f3420c83114c2b956374e2 : SUCCESS in 10m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4cdeef96d25e48e7bc7f166d5e9a8b81 : SUCCESS in 19m 33s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/362a074ae30943fcbbc56425f2a09356 : SUCCESS in 1h 06m 19s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/dfaef0126ac44fcf80935ac9b040e1c5 : SUCCESS in 59m 24s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/20ea7004bdda4318980f053b355b1683 : SUCCESS in 58m 33s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/2d47c52e79de452ba61c4cca81e09f22 : SUCCESS in 1h 04m 47s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/aa95ff501c204f90afa1c2918207ed46 : SUCCESS in 16m 59s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/02c36da316db4f009d7c6de5cd8c00f4 : SUCCESS in 19m 13s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6351524026fb45e097769ec6cfeb73c1 : SUCCESS in 15m 58s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9c2276e5f26d46a5bf248927e9e95ebd : FAILURE in 1h 52m 21s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ab96cd4ae77f42f68540719699f26d57 : FAILURE in 1h 11m 57s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3c82d22af3484212ae1054caa42a14e5 : SUCCESS in 16m 27s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/244c1e98d4424f5797a06f1f64285949 : SUCCESS in 11m 59s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/a5eed3a979844f1c89fefae7153ac307 : SUCCESS in 33m 35s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2fdf8ab5017649a39493a6c12797e0fe : SUCCESS in 28m 26s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/62a3e9468abe45f2b639251951d024fc : SUCCESS in 18m 55s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e3a0898b312940c2b98d268493cde9b3 : SUCCESS in 12m 50s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c4d4759c6dae40dda3da4bf5459c75d0 : SUCCESS in 9m 51s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a88adefaa6e14f0abb954a189307a287 : SUCCESS in 12m 53s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1d1c8818a62f482a825fde81de355ea8 : SUCCESS in 12m 55s","accounts_in_message":[],"_revision_number":2},{"id":"a1f6466dfa302214aa76deab9adc8bfb6a16241c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-07 15:30:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"843f62e8e810cff3726ce92867996fb6730de8b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-07 17:10:06.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5388ab28daed49429b5cf1a8f4060dd1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/439aae99067049e69073d588c336a75b : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8afb9e00edfa4c27823fa7a0b820ef78 : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6355f790d5eb4c509a4bb3304f5838d0 : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/181c3a0441354c8ca1ff18beccf95c1d : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d67988455d1141b3b137f7caa9e94fd0 : SUCCESS in 6m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f94adc8ffd944b27b3b540485812ae05 : SUCCESS in 4m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9d00ac7cd1fc44d586d66011a254d78b : SUCCESS in 10m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8d3c736b935a4c58a65202b9cb77bb66 : SUCCESS in 15m 32s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/27013d7c97b9437a80871b392c3e441d : SUCCESS in 1h 00m 17s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/2e1bf98f52c34237a3f740e5b6aabc3b : SUCCESS in 53m 48s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6361b64a01164972a3139154ea29582b : SUCCESS in 48m 52s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/de7890d43c8d436fa3e3f7fecd7cd5e5 : SUCCESS in 1h 08m 37s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3c362f98dac343c1bf13ab36de081e44 : SUCCESS in 13m 46s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/be8ac18e58d540e2a531acf91c2c07cf : SUCCESS in 18m 18s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6d28f179236d4c8090f84960eba7347a : SUCCESS in 15m 01s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6ff39390fcba4f0c90c4afe5e9f3e76d : FAILURE in 1h 33m 38s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f11c283af81c4a76bbed5fa6e82532a9 : SUCCESS in 1h 29m 38s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/df6f7eb2170d4635aa7c1bdbdb47e0f6 : SUCCESS in 17m 33s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a7eef8a5bbb24c12a7a8fad6382c53f0 : SUCCESS in 14m 31s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/15a9c43da59049738b1f93af9caa18a5 : SUCCESS in 31m 05s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/77e00b3016064137bc23d367dc2b81e9 : SUCCESS in 29m 30s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4a9632af6e6417db8c265719afd3d27 : SUCCESS in 23m 45s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4d4caba73fb442f4b9cea0beee5aa086 : SUCCESS in 14m 49s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/979d8ec163664a47b1560d8663755fee : SUCCESS in 9m 18s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/febcf1c4852c4760af9cadb3d5c03358 : SUCCESS in 12m 36s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97abe33080634fc890c8b198fb40b79b : SUCCESS in 14m 42s","accounts_in_message":[],"_revision_number":2},{"id":"2a1406ce37a9d3ad05329f23ace73a18b30618c0","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-10 12:32:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b981727a00f45e688b7162d2048b6bd0ddebc095","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-10 14:14:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1fab0d019f0449b88c85b9f043c81de\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b32f893e7fee492ea11e277eb8bf40ef : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c811298a02444d0784a03694977e96a8 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a976892434d5485086eee013e0612a91 : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2b4320415d994fcf9f1a553e7e6e60b2 : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95f6a2873885403da358a09174c68002 : SUCCESS in 7m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3ffa6da08b8147da90ed9ac2d4107d13 : SUCCESS in 3m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a4834d73b8c044498ee091c057c8c808 : SUCCESS in 11m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f14e3b26eecf4d6fa3570887fe1caf14 : SUCCESS in 14m 55s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/bf588252ff764f65a9f5ee177f842c3f : SUCCESS in 1h 02m 36s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/cc570c27030346e6bb1db2cc7b68ae09 : SUCCESS in 53m 35s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6114e3c8f2c2416c9eedea9a458f2d01 : SUCCESS in 56m 53s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/d354c0561dce4f4e875e2edc1b1b7f33 : SUCCESS in 1h 03m 26s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f3f1c9a903904dce89c5d8cfcda1bf84 : SUCCESS in 16m 47s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/4c8f19f7be154aebb24eb73d73990d3e : SUCCESS in 14m 52s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97b308b5a3d3488aab941f8152a70cff : SUCCESS in 15m 06s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6ec749facc79412097fe135643eb2831 : SUCCESS in 1h 37m 14s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ecf0403c3c2b4d0190b18b61f6cce5c5 : SUCCESS in 1h 02m 34s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/30135935ef33416ca6746e67c2cf50ae : SUCCESS in 16m 41s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/427cc1035000459987a6f8a7e5cd2221 : SUCCESS in 13m 08s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/7c010235b1e3447383f445e2d434a8be : SUCCESS in 31m 59s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/528186fc2ad1442a9b824f752b4312c4 : SUCCESS in 27m 09s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2c01013acdb34754840b93aa6b30db8e : SUCCESS in 19m 02s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bbd95fc8e44b4ecd830d54341499568b : SUCCESS in 12m 19s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6017bde8516b4c8f929dcc32f5689b01 : SUCCESS in 11m 19s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d066853564ec4fad98358d0d8bd00ef5 : SUCCESS in 16m 03s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a3a6068da8d44a0891a30ced9cb3d79f : SUCCESS in 12m 43s","accounts_in_message":[],"_revision_number":2},{"id":"8347dec0fe6e5d0d8b3eeb8baa8daad37829c91c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-13 15:33:53.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f347a3d2a6cbe9b5af7fcd0fab0392fc76e9a4bf","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-13 16:41:50.000000000","message":"Abandoned\n\nhttps://review.opendev.org/c/openstack/kayobe/+/854236","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0c8d9e838aac81eabe0164f45c642f779759ed9a","revisions":{"3f308ed274055b503a27f62a9d6fcfa8c02bc9e5":{"kind":"REWORK","_number":1,"created":"2023-04-05 10:10:58.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/39/879539/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/39/879539/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/879539/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/879539/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/879539/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/39/879539/1"}}},"commit":{"parents":[{"commit":"e6432ce671dc65b3380f7f298e5c87bed937df0d","subject":"Merge \"CI: Add custom routing table output to logs.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e6432ce671dc65b3380f7f298e5c87bed937df0d"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-05 10:06:50.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-05 10:10:15.000000000","tz":180},"subject":"Create direcopry for passwords.yml if absent","message":"Create direcopry for passwords.yml if absent\n\nThis change creates direcory for Kolla-Ansible passwords.yml in case\nwhen Kayobe enviroments are used and destination directory not exists.\n\nCloses-Bug: #2015093\nChange-Id: Ica5cd17ddd784aee8ebdf0726b8c7098fb4a897f\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3f308ed274055b503a27f62a9d6fcfa8c02bc9e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3f308ed274055b503a27f62a9d6fcfa8c02bc9e5"}]},"branch":"refs/heads/master"},"0c8d9e838aac81eabe0164f45c642f779759ed9a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2023-04-05 10:15:02.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/39/879539/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/39/879539/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/879539/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/879539/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/39/879539/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/39/879539/2"}}},"commit":{"parents":[{"commit":"e6432ce671dc65b3380f7f298e5c87bed937df0d","subject":"Merge \"CI: Add custom routing table output to logs.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e6432ce671dc65b3380f7f298e5c87bed937df0d"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-05 10:06:50.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-05 10:14:39.000000000","tz":180},"subject":"Create directory for passwords.yml if absent","message":"Create directory for passwords.yml if absent\n\nThis change creates directory for Kolla-Ansible passwords.yml in case\nwhen Kayobe environments are used and destination directory not exists.\n\nCloses-Bug: #2015093\nChange-Id: Ica5cd17ddd784aee8ebdf0726b8c7098fb4a897f\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0c8d9e838aac81eabe0164f45c642f779759ed9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0c8d9e838aac81eabe0164f45c642f779759ed9a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
