)]}'
{"id":"openstack%2Fkayobe~880367","triplet_id":"openstack%2Fkayobe~stable%2Fxena~I7b9ced54ad886b3f92f7c757adc2efafa4cbfa4e","project":"openstack/kayobe","branch":"stable/xena","topic":"fix-globals-description","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-04-17 17:34:39.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-04-17 13:16:09.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}}},"hashtags":[],"change_id":"I7b9ced54ad886b3f92f7c757adc2efafa4cbfa4e","subject":"Fix kayobe_config_path description in globals.yml","status":"MERGED","created":"2023-04-13 21:01:05.000000000","updated":"2023-04-17 21:05:04.000000000","submitted":"2023-04-17 21:04:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880367-fix-globals-description","meta_rev_id":"149da42246367dae26e6859508a6f5c634755f51","_number":880367,"virtual_id_number":880367,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-17 21:04:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2023-04-17 13:16:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2023-04-17 13:16:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-13 21:01:53.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-04-13 23:02:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b0900bc780032ea65697fd671264026191bcd8ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-13 21:01:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0992ebfdc876ea858a70fcec41f1e6277f4f790a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-13 23:02:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec6acf73c0674e2b99ce19d9af8c2b4e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/664c742d970a40fea1680b972b228c30 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25363e423ed34a9790dc7502a16760d6 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/88a28a92ffbe4b1f85c2074919dc9a4f : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b43bb6ec43646b69ad8d29671bf4c16 : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e40b59a82ae4b46b8c57de8d66ddf2a : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d640708001f4718b4c2970f94ad7eb7 : SUCCESS in 9m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fa6abc2c40d5413d9d16039eac5d7249 : SUCCESS in 5m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a195139aff824d3398f3c4cb0e3e1a34 : SUCCESS in 12m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/78afd488b66548699498774d3223badc : SUCCESS in 20m 07s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7feedabb154d499aaf54b957cd0bfe75 : SUCCESS in 1h 10m 04s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/80ffb770b795491f96cd383956356f65 : SUCCESS in 1h 05m 58s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61982d3834f34c3b86b64d64d75fb8fc : SUCCESS in 51m 54s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5067b723d0a6420faf0059586b1bdf0a : SUCCESS in 1h 14m 51s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0d1fb56c7259495f93f850eeca6a4fa4 : SUCCESS in 29m 21s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/28a994fc311846ac8066c538c1a2bcb4 : SUCCESS in 24m 53s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02f187f9d4b74fe89b4ec42ac625034e : SUCCESS in 14m 41s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e995455a1480479fb89d73fa0692f11e : SUCCESS in 1h 54m 52s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/abcb14d2fe0e442b83b603c295571269 : SUCCESS in 1h 19m 10s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ac84076df1004b30bd5825d238517567 : SUCCESS in 19m 23s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/48f9f1bbf25c4160a05f24da8c7aa504 : SUCCESS in 18m 29s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec073759b97b49019059f668967dfe77 : SUCCESS in 14m 35s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/3dbbb797bb2642d0a859a1bff03d555e : SUCCESS in 44m 03s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b3394517b9b44c2daddbfff1d3ea9d1a : SUCCESS in 31m 18s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e2c7185619a4a2889bc46d1c9e67ac0 : SUCCESS in 22m 01s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f04c2dd86406455388259b78725a7682 : SUCCESS in 16m 34s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/6e96b66eaf034b7681019994b6efb438 : SUCCESS in 16m 05s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0df8cc5fea7546e48f931bf66e3fec65 : SUCCESS in 11m 11s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/d6123513553946d2bad8d177cf790934 : SUCCESS in 18m 16s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/791af3b3180c4527b3ea7f0918d9a190 : SUCCESS in 18m 26s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af24b1f925bb4dfbaa109f62aa2c7231 : SUCCESS in 12m 13s","accounts_in_message":[],"_revision_number":1},{"id":"b43f99e96fb2a0171fefb95eb7ffed3d20d85874","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-17 13:16:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"140a4863ed28ebcd94ace8a0f40b3e7d86bec6e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 13:16:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5b5ea9a9a53d8779bd8595909ae87c88d6f8504d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 15:26:06.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1634452f2433468f99b6c1d17da2ebe3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8efaa8451e4248f19710aa40278bea3c : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c54103f81f8741adaf224a55d405569d : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7a1fc440c374c5ca6fd9712a7ad328d : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/756a5f62cce14905bddb44144b0aeef3 : SUCCESS in 8m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b2b7a5b10c8e48b69f52b166ffc256c6 : SUCCESS in 6m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/eb8b32ee00b845529023ff6d24358980 : SUCCESS in 13m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c22deab9c9334e95bcd91c7408640006 : SUCCESS in 21m 21s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/85e0a211e2d5440d9c0ec4c4a9c5dda1 : SUCCESS in 1h 34m 35s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/70d929ffef3947259a7e59ef9b6e2cf2 : SUCCESS in 1h 21m 18s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d07a44a4cdbb4331aad13adabe822382 : SUCCESS in 1h 08m 23s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c8cb9772f61649b382cfb1ba48015ffb : SUCCESS in 1h 31m 50s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/eb41501ba6384d1bb5bc68ea9aade739 : SUCCESS in 26m 07s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/4b49f08b06824eb098830a6badb2ca4c : SUCCESS in 24m 55s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b1574b7b6644a3b88f95f6c4a5e05da : SUCCESS in 18m 08s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/eb8eece9accf4679bff48f898abd73c7 : SUCCESS in 2h 00m 33s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f466efede90a4ffbb855804153e12817 : FAILURE in 47m 39s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/9d8fff28d3d6479999a8e8a8e17799d9 : SUCCESS in 24m 18s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/73e902549f50411a8c737496727792dd : SUCCESS in 22m 15s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d6323e92c70f40f28338c33136d0b938 : SUCCESS in 17m 00s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c13fa67fa2e946ba96b158f2ccdceeca : SUCCESS in 36m 50s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee5efe81bfe44758884502b862f5133c : SUCCESS in 24m 22s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/8f25decea332435ca4e11896e3d1b202 : SUCCESS in 19m 14s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/cd2a0f7446a944109fc20a2dd9f3ce8b : SUCCESS in 17m 36s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6c2cdf92f1a4908add4ed9c87b6eb57 : SUCCESS in 11m 55s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/9ec4b2fc2ee84d119a5c9c9f06b5708e : SUCCESS in 22m 19s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/a5ee45ddf3d94f5ba433273d057ba67a : SUCCESS in 19m 18s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0051edf4132140fe8eac362fd0233331 : SUCCESS in 14m 17s","accounts_in_message":[],"_revision_number":1},{"id":"8c36b5ab769a6cbdec6bbbbfe66508c6a6939029","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-17 17:34:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1a6dde08922157a3a148e75a134b7c32c68d773a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 19:11:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86b55c7b201d4d22b9d280d050f12618\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4cde1d4ee84043149f86c7568fe75d5c : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bb6c4b1378941b2ad41311fef65ad31 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74a473a83b44433a81c56f234881ad30 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c801e1007aab4c2592489aab92539513 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e758ee31dc842e98bfe44b96af0dddc : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a54eaabc14c34bc4926be7045870501e : SUCCESS in 7m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/323146b893a34596bb1c9a29c3388189 : SUCCESS in 5m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cdb008569a974215ba24bf1ae19660e7 : SUCCESS in 11m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/330b5a713f8d4224af72c40f87f8f091 : SUCCESS in 19m 26s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/83895884ce7d4634adcb58db875075cb : SUCCESS in 1h 11m 47s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/d6d1e8edec3c499ea92a48445822c33e : SUCCESS in 1h 16m 49s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c1cc90c0767a422d91f30b68f1aa5a6c : SUCCESS in 57m 41s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e80406ca128f40b089678a9e70a68f55 : SUCCESS in 58m 46s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/24a37b6db5e94331a8b39081c49c788a : SUCCESS in 26m 08s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/63af2ef7abc34be3a6d4a100780d85e8 : SUCCESS in 23m 17s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5bd863b748ca49b6ae6db7b6fe8d4f5b : SUCCESS in 12m 34s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/040e8f14997445c09760a0ff831013e0 : SUCCESS in 1h 35m 34s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8ba0eeb61084e9c8c9937d505238fb7 : SUCCESS in 1h 01m 51s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/77220c58de5c4c4eb8d4b8e83fdc1391 : SUCCESS in 22m 23s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/291d6e2c3326404d9afbb28c9f885544 : SUCCESS in 20m 19s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb38d196e7334eeb8cff2bbe2768777d : SUCCESS in 14m 54s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/dbdec84f96404ee089564163f1616b60 : SUCCESS in 38m 14s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/edaa7bbf627e4ea78fee586c3833dddf : SUCCESS in 32m 57s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0cbd955e68c54e57bc5718213d881c31 : SUCCESS in 20m 13s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/3aade00021db417abd30d275357a5f38 : SUCCESS in 18m 13s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/accfe625cb12482f997be7996e0fb5b4 : SUCCESS in 18m 30s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f0f16de224d474996b21036e62402aa : SUCCESS in 8m 55s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/9a99be74dced4202bc1658fc8cb40d0d : SUCCESS in 21m 45s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/0451743d91814053a86a2875ea6ea30f : SUCCESS in 19m 23s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16cb07700f844dcea04e27b2e0583ba1 : SUCCESS in 11m 13s","accounts_in_message":[],"_revision_number":1},{"id":"d975cfff2349da1240046b2c798184bcc4c9d722","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 19:12:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9d7ec13d6895c2ec296837b6c2266c09103cc398","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 21:04:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0da02256bff41be9c9f26ec7f9b09b7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67f6895c08dd43c096fce240d5f716e6 : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b34c55b5f6be4b7d9d6bc256681ee364 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a599c3c357664d759444b1c4ced8dc55 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e54755a617843f4ab424f55b7c0ec9c : SUCCESS in 8m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/52c0e61c6cca45c89495a1dc8d86dc50 : SUCCESS in 5m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3c3d498e8a064258b2d102079f7be2ba : SUCCESS in 12m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e673c7d4283e42d4bdccf6b2dff498fb : SUCCESS in 18m 07s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3ec75ba4423c42d18169011262688222 : SUCCESS in 1h 18m 18s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/a6be222037754034be342d32866e135f : SUCCESS in 1h 05m 16s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c00cd35f3c024780ad05bd358a198b0e : SUCCESS in 56m 11s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5b311902187a4dd6b1aef83cf21e64db : SUCCESS in 1h 19m 55s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c6d8cf969c014944a6e698913e53cc46 : SUCCESS in 24m 48s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/30dfe4ac2e294a63baf778414a8aad71 : SUCCESS in 23m 54s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24badbaf68c54d44a23b4c72a46535c4 : SUCCESS in 16m 09s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ef38f7f60f06433fb8f3d9cfbbb2d04c : SUCCESS in 1h 47m 41s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c4cce9330754455b802e472292ce9d3 : SUCCESS in 1h 17m 22s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/5098f6d473a444f796a156b0c1eed93d : SUCCESS in 21m 36s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/06ea08a1504f4eb99209250387513d5c : SUCCESS in 19m 51s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2360279a47d940d68ef1d746b13cbd95 : SUCCESS in 15m 26s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e718c3d4eab040699f72f8c5693566b3 : SUCCESS in 31m 43s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c06c295beb084058b451de064892beb3 : SUCCESS in 21m 21s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ffa6da1913264bedafbe5a9e5c752d66 : SUCCESS in 17m 19s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/0f5caa16d8844f018687e05e96a7dec1 : SUCCESS in 15m 58s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8cb1db66c574557a5ec34b887a92f4f : SUCCESS in 11m 58s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/087a3721281e43578ba3e1f516602902 : SUCCESS in 21m 09s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/61c26dfefed9406880e5b84eb2ff0fda : SUCCESS in 17m 44s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af759879fed546d2ac400ce2b790afa7 : SUCCESS in 13m 13s","accounts_in_message":[],"_revision_number":1},{"id":"cf898db208327c28f5bf789872657a8488a8a1af","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 21:04:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"149da42246367dae26e6859508a6f5c634755f51","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 21:05:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0939fdf32b3a423b9649929295c72533\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/409b7ae7e15941a29f956a9d82dd1f07 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aec7807b2d3c2c0b0c88d783c860561b81bd3377","revisions":{"aec7807b2d3c2c0b0c88d783c860561b81bd3377":{"kind":"REWORK","_number":1,"created":"2023-04-13 21:01:05.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/67/880367/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/880367/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/880367/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/880367/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/880367/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/880367/1"}}},"commit":{"parents":[{"commit":"76fbde18334bbbf420bd6269d5ebf1799490ecf0","subject":"Merge \"Support configuring VLANs with systemd-networkd syntax\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/76fbde18334bbbf420bd6269d5ebf1799490ecf0"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-01 22:32:38.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-13 21:00:49.000000000","tz":180},"subject":"Fix kayobe_config_path description in globals.yml","message":"Fix kayobe_config_path description in globals.yml\n\nThis change fixes the description accidently copied from the\n\u0027kayobe_env_config_path\u0027 description.\n\nTrivialFix\n\nChange-Id: I7b9ced54ad886b3f92f7c757adc2efafa4cbfa4e\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 2ab498f955c87f0f92651600be45d7dda9264af4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aec7807b2d3c2c0b0c88d783c860561b81bd3377"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aec7807b2d3c2c0b0c88d783c860561b81bd3377"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
