)]}'
{"id":"openstack%2Fkayobe~880388","triplet_id":"openstack%2Fkayobe~stable%2Fyoga~I9dce73a8a205c0c0ad02eba3a10e02b82f5b191d","project":"openstack/kayobe","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-04-17 15:02:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9dce73a8a205c0c0ad02eba3a10e02b82f5b191d","subject":"Fix passwords.yml generation when parent directory doesn\u0027t exist","status":"MERGED","created":"2023-04-14 07:39:05.000000000","updated":"2023-04-17 15:03:48.000000000","submitted":"2023-04-17 15:02:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880388","cherry_pick_of_change":854236,"cherry_pick_of_patch_set":3,"meta_rev_id":"e437a8d427f885478d704b25ed253bb5fe55832d","_number":880388,"virtual_id_number":880388,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-17 15:02:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2023-04-14 08:29:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-04-17 13:01:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-17 07:58:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-04-17 13:01:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-14 07:39:05.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2023-04-14 08:29:09.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-04-14 09:28:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-17 07:58:34.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2023-04-17 13:01:33.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"01397e5cf4086b7e8a0e279b3dbbefd7a5907e4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-14 07:39:05.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e3893bd3d22358cfbe4cbbe59c620453a71b3500","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-14 08:29:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"447af78a67c125deb8d18777198e8e1ed6f7b482","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 09:28:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc126dbbbf884672a6459cce59ea0c9a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f82f23b67ee84748a2af7394b453b211 : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39f49f5fce0c456f9d3233233aea2d0e : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69f1046d2294433f830359294e225e82 : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba119ee984274f85ab438d5dac3f1337 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e33d7fa4fad941e08725d3e5421486d5 : SUCCESS in 9m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7e1f26448734b64b4870f2fbcda445f : SUCCESS in 2m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b89ee6ba9b1a406cad01cda6025c91f0 : SUCCESS in 5m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/16c53f9671c9486f88a9de0534d23f93 : SUCCESS in 14m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cca4afea2e864f99b8a530189147a983 : SUCCESS in 16m 54s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d6bb14d1385444439c95a4e937fb43ee : SUCCESS in 1h 02m 52s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/0f7413317a474d0383e3c005ca55ca66 : SUCCESS in 1h 07m 07s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/dfa482552a8b487eab74e79e4a641854 : SUCCESS in 1h 02m 07s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a0811b86a4e14e5382f81e1eef450d01 : SUCCESS in 58m 01s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/673c54cda6334134be231a0eda91cc30 : SUCCESS in 56m 31s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0aeaea7cd7304183a1a4c2c68b81d9a8 : SUCCESS in 49m 47s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7c0e4957d0054b298435c22e89c2dc9c : SUCCESS in 23m 42s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/9df2fd29a7a24133bc9709dcdc1bd65d : SUCCESS in 22m 06s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/22887167de624546a2ad3e142cab3e4d : SUCCESS in 20m 14s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/b1d42fc9f19e4dcebe18f4d564ce9b0d : SUCCESS in 20m 57s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f4a16d667a14e96998f657c42ae3b81 : SUCCESS in 17m 18s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6126385f501c438087b93b3a1000b23f : SUCCESS in 15m 31s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c5f54f8906a24d20b7161958801b2aa7 : SUCCESS in 1h 08m 35s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/fae81bb5065547f290a08a9d3f527025 : SUCCESS in 1h 07m 57s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/807b438d3ea043f4944bba3ec629cefd : SUCCESS in 1h 45m 09s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7529d80ba0d84a4cba4e8712ca28019e : SUCCESS in 1h 15m 04s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/26948040b805405789c9e248614971c3 : SUCCESS in 23m 11s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/46ceff72f8634b49a1054f2c5dc5a485 : SUCCESS in 21m 04s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/5c82494bfcad4b2db209431ff6492528 : SUCCESS in 18m 16s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46e38ead81154ba2b22bb2bc75e9fcf5 : SUCCESS in 17m 56s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bac0bbef23754a8696b34560f950cc3b : SUCCESS in 16m 59s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/56651e47c2d84720bc6069489682afb2 : SUCCESS in 27m 38s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/407f1c5d689c41adb6c152c129014c4b : SUCCESS in 27m 15s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/4a2fd6d652b94dc7a42d94a9f491ac23 : SUCCESS in 16m 21s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/329a89d3127142a193e19d44140e4d57 : SUCCESS in 17m 16s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2705a9a7cee04fdda32db842517e9745 : SUCCESS in 16m 20s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f1558b86a59402dabb97aacf681665f : SUCCESS in 12m 16s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c937b4a6d3dc4319a694a9a36e510ced : SUCCESS in 13m 40s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/49872720b16c497994a4f0d2ecb1d486 : SUCCESS in 22m 14s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/e2174d03580b486597a02ef04b8acd4d : SUCCESS in 17m 49s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ef627c33b25a4569bb69fc75150959e5 : SUCCESS in 18m 22s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0c9081208384dd59c7bf9a479bee155 : SUCCESS in 14m 23s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c2593dd6c7846eb91b9cc41594f5171 : SUCCESS in 13m 58s","accounts_in_message":[],"_revision_number":1},{"id":"755916c2a4bb4b076d457ba6d3588927408e1d71","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-04-17 07:58:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"127ba4498f43db1b67787d9cfe2a9ba88ff46965","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-17 13:01:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"620204957413c8ff6df4f9f1d1dc9216be3c08b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 13:02:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b5e3cdc9d234d8f47f5f91e07e09a9fce4a25d6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 15:02:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a91ce90a1ca44d1985d5d5893ed1b526\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6fb5e0593c04c1391127dff665e7adf : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a116cbcb9239481faa79403c5810bf9a : SUCCESS in 6m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e30cb4e313d493481210e5c09869d85 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eefe858d8c62434d829ec854e0d24006 : SUCCESS in 9m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d278903ea7849fdb0bdecbf01339a32 : SUCCESS in 2m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f4071f88d34745be8803ce345f7b63bc : SUCCESS in 6m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b2a742599616418a8cea1019a20d87c5 : SUCCESS in 13m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0d30865d1e904ee6bc9d18a4242ebd81 : SUCCESS in 19m 41s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4cc637a4b04644c8b5b88f657c870264 : SUCCESS in 1h 10m 08s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/966e362b4492426a8f90bf401e9afcdf : SUCCESS in 1h 11m 37s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/2d3491a129e243219ee11ce2222fc776 : SUCCESS in 55m 11s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d66e68be05ef478480f4e7daecb401c8 : SUCCESS in 59m 18s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6cd8d4706f36462e93fd412074c5e33c : SUCCESS in 56m 33s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d169071394114a41a63fd54a69528fb7 : SUCCESS in 28m 33s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/f2728f428c2f4e0eb4035382caa77aca : SUCCESS in 23m 03s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/8c76fa2b0bea4cbbaed70bf4e60f2bde : SUCCESS in 23m 59s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cf0d42d5e551438fa665de9be601d888 : SUCCESS in 14m 02s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9acff947cb34e679aa3bc12728f6386 : SUCCESS in 17m 39s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/61cc1e6c66d642dc89ae30f3cdcc14ee : SUCCESS in 1h 08m 45s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/93e143bd447f47e8bd1b62dc71b65add : SUCCESS in 1h 14m 49s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e186ccc6b7b34b35b063bfc5f444dfe8 : SUCCESS in 1h 49m 15s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/72e0a3e42d0a4ddaa66dd2f1dc298874 : SUCCESS in 1h 22m 23s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0e468a885ad14521a7e01cb724009eff : SUCCESS in 20m 23s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/670dbd7896eb490483a140a05c32f92d : SUCCESS in 22m 54s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/420353e6ad374e60b79ea08bff737361 : SUCCESS in 17m 40s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/13cdc72ec31548c0a5012720a26045df : SUCCESS in 18m 50s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bcc40c688b4f4b5891426c0668cc2a53 : SUCCESS in 15m 49s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/78beb6c8a5344213bb96d447e0a7aa0a : SUCCESS in 33m 47s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3457e9cf86d74c5dba86076b9dde2fa2 : SUCCESS in 22m 50s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/4fe0169fd1b144c385a23643219f651a : SUCCESS in 20m 25s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/989e388b49764cc3af2b2f9bd1e2a956 : SUCCESS in 15m 30s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c9eae75da67048459ae46bf9bdb2f7b6 : SUCCESS in 17m 11s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/622e7bafc9d748ebb64ac549c2c10a28 : SUCCESS in 12m 01s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a74d2f540c3546b5b8ed673138b8dda5 : SUCCESS in 15m 19s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/faa9229b27ce4b9e8ee31f0e5e4bb059 : SUCCESS in 19m 32s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/d28bfc8fcdc048ae9b627c5ce662cb8c : SUCCESS in 21m 29s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7e1ab9d7e1b74a41be7fc84d1458a8d8 : SUCCESS in 16m 07s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/369489c6635b48c99c0cba746d504a94 : SUCCESS in 13m 10s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55565e0986fe4291b1a352f1bd193815 : SUCCESS in 17m 12s","accounts_in_message":[],"_revision_number":1},{"id":"872ddb8fd80cf6ec332f170282381064055a3ef9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 15:02:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e437a8d427f885478d704b25ed253bb5fe55832d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 15:03:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3dada349ff824fa9afd17d9d87913703\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e18b7461fdf40f6ab0062882e3af4f4 : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9bf99edabd074f49b4b502edd9dd6d59 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bbcd462dffebab377363f0f8b5ed00651a94517b","revisions":{"bbcd462dffebab377363f0f8b5ed00651a94517b":{"kind":"REWORK","_number":1,"created":"2023-04-14 07:39:05.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/88/880388/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/88/880388/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/880388/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/880388/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/880388/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/88/880388/1"}}},"commit":{"parents":[{"commit":"6bcadd984f0169c4c616f1d0920a8304d1811746","subject":"Merge \"Ubuntu: Fix networkd config for multiple VLANs without MTU\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6bcadd984f0169c4c616f1d0920a8304d1811746"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-08-23 15:46:29.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-04-14 07:39:05.000000000","tz":0},"subject":"Fix passwords.yml generation when parent directory doesn\u0027t exist","message":"Fix passwords.yml generation when parent directory doesn\u0027t exist\n\nFixes an issue where generation of passwords.yml for Kolla Ansible\ncould fail if the directory containing the file does not exist. This is\ntypical in a multiple environment setup, when creating a new\nenvironment.\n\nStory: 2010293\nTask: 46275\nCloses-Bug: #2015093\nChange-Id: I9dce73a8a205c0c0ad02eba3a10e02b82f5b191d\n(cherry picked from commit 5287083116da2512b887529e67af4c4919e1e296)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bbcd462dffebab377363f0f8b5ed00651a94517b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bbcd462dffebab377363f0f8b5ed00651a94517b"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
