)]}'
{"id":"openstack%2Fkayobe~880464","triplet_id":"openstack%2Fkayobe~stable%2Fxena~Ie2d9a0501fe29bfd854eb31258f282b197855948","project":"openstack/kayobe","branch":"stable/xena","topic":"fix-merge-action-plugins","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-04-17 16:05:46.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-04-17 13:09:46.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}}},"hashtags":[],"change_id":"Ie2d9a0501fe29bfd854eb31258f282b197855948","subject":"Fix merge action plugins verbose output","status":"MERGED","created":"2023-04-14 08:44:49.000000000","updated":"2023-04-17 16:06:46.000000000","submitted":"2023-04-17 16:05:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880464-fix-merge-action-plugins","meta_rev_id":"6ee5cbaee390a30cb426c10b1fa1be6111346507","_number":880464,"virtual_id_number":880464,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-17 16:05:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-04-17 13:21:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-04-17 13:09:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-04-17 13:21:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-14 08:46:15.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-04-14 10:36:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-17 13:21:08.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"002e32da5940c160e6ac3d84abaad1b1d722fafe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-14 08:44:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4972130b27874d11bbc16dc7070f516098217c34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-14 10:36:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd64d7ad33234a36909a5bf3c9d102ac\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4623b8e72ea3442e8fbf6950cac83727 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6efd17f7f904de2b529288cd61b17ee : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b889f9c5f7f744149ebf3057047a458f : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0554d367d4a546a4900589aac614b006 : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec3384168869479d91543b890d41a2cc : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb6e1cad1e704a0eb950373712734a64 : SUCCESS in 7m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c75f4527656b4afb95b57204bc55f859 : SUCCESS in 5m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7fa3d88f82534926a0fc364935c1df08 : SUCCESS in 11m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f7cf7a2e5413447d9a9f04554b14b45d : SUCCESS in 19m 21s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/aba9385e677a45639e0a63f701b602e2 : SUCCESS in 1h 09m 29s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/d86ed967b0bb41edb4d4a50e2ddf7ef4 : SUCCESS in 1h 05m 24s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/327c34e8d5c4461a8d360a933ecbcc90 : SUCCESS in 52m 32s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b00fd65e5ad34f5782878e8a72831a86 : SUCCESS in 1h 19m 09s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/4c13ee9dfb154250ab4aee18efa46b0a : SUCCESS in 22m 34s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/c3f2868223974d51960208eac0c96a49 : SUCCESS in 23m 32s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2cfac605ffec4f728844f3d0869b5705 : SUCCESS in 12m 28s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/52806fbf2a5945ac99277c583341aa68 : SUCCESS in 1h 47m 28s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/085bce762c0b4ab298fc86fdcddb2834 : SUCCESS in 1h 09m 05s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ab477c44225c4f6280e5897e075a4ade : SUCCESS in 24m 17s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/de0c41e0631f4599824236216f98f2d8 : SUCCESS in 17m 38s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e814a9a565d84c2c85f3b60eab4f3e94 : SUCCESS in 15m 39s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/3493c8d212674ca391aed9e5a6252787 : SUCCESS in 43m 54s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6e1214e1b7fd48668444e9d34a1344bb : SUCCESS in 27m 19s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56bee74b02ff40f28ea44d86acfbc3f9 : SUCCESS in 21m 34s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/51e0b27388f342bfa98b318ee3bc25ab : SUCCESS in 19m 15s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/ee4878db14474142b1499fc98567f32d : SUCCESS in 15m 33s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01b684a18b7d4b03987abc0b5b6b7562 : SUCCESS in 10m 59s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/cf674b32e3f849e7be05dab113b78a19 : SUCCESS in 18m 21s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/57066dea03704a4c9214818f187863cd : SUCCESS in 17m 37s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0f3947cb15d43f484df760fc46a76f7 : SUCCESS in 12m 22s","accounts_in_message":[],"_revision_number":1},{"id":"9f4f175ecf44a8f4026aadc1fa425eb0342a0e0b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-17 13:09:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3c7d24779f3ee82b35528f81898f67594f92aaa0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-17 13:21:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d7b445604e0dbbc5e614c7f8ca997a6023601ef5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 13:21:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"030bf7fc68bd6a836041378a5a01e49554b831a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 16:05:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d7795f852804c268f7ffbc91fdf3a8b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/463833ae19a54cb29a784c4d7fd7c047 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4440045c34aa41338ce801d327624dd1 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f38bc3bf10974f8680df0ba1b410d8fe : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14dafe6f651540ae9f683c2e4753238e : SUCCESS in 6m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b517f6dc90a5438eb07683b6a778542f : SUCCESS in 5m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/756cc800726f4203aeb438179adc688b : SUCCESS in 12m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dfea2286d3c649f59f7ab1dd25f80d83 : SUCCESS in 15m 06s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/81d7cf79f7134c879015d3d5c6ed7663 : SUCCESS in 1h 02m 26s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/30582e3c81894c0aa35631852f3622c7 : SUCCESS in 1h 08m 44s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2836eaa184d49ac92b790037938c59e : SUCCESS in 52m 55s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/311f5b19f30c4a4a86afb7eaabb6a62f : SUCCESS in 1h 10m 08s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e105cd53662249d6879bf6622ead7887 : SUCCESS in 25m 01s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/5e7fc75104c54611a301cbbb70e979e6 : SUCCESS in 24m 24s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98a1e4941ae248ff81d4801dce0ddd67 : SUCCESS in 16m 11s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9d74988dfb2a48458e0383f0821296a1 : SUCCESS in 1h 47m 51s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/452a959a041b407e894d95254bb400b5 : SUCCESS in 1h 18m 35s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/bdb5c19e6d2846e19dfab29a36ca78c9 : SUCCESS in 20m 23s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/1b9c47c165934cf786e95f80c93ad14a : SUCCESS in 22m 24s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/293e46e75a7145748297b65b0e9838f9 : SUCCESS in 14m 01s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/93d3b8fbc149412f95278088bd91e68c : SUCCESS in 30m 59s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b506db306d7452fbc1fbc66f3e402d5 : SUCCESS in 26m 00s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/91ca61d6e33e4223b652c0980c99dfd4 : SUCCESS in 16m 52s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/18a06fc4762841bb8819c8c3cb74c324 : SUCCESS in 17m 08s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e79a749c4c094846af27f8a194a920b2 : SUCCESS in 10m 58s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/967471dc45d64e179aebc3936d36e052 : SUCCESS in 22m 24s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/b73f2a8ef40d461b858f38eee862d026 : SUCCESS in 18m 28s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/290490d54e5d45bda61c4ce876e01cfa : SUCCESS in 14m 27s","accounts_in_message":[],"_revision_number":1},{"id":"ffc7fba68320ca2b970b667ff6e1633f28b52e7d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 16:05:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6ee5cbaee390a30cb426c10b1fa1be6111346507","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 16:06:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/571a05b4ab844ce589041b18f19a1fdb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15171030d5f548c8b8cc0567b70b9c63 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e8b9b8c0edb8e1482f3e4a0cde23dd735668d5d1","revisions":{"e8b9b8c0edb8e1482f3e4a0cde23dd735668d5d1":{"kind":"REWORK","_number":1,"created":"2023-04-14 08:44:49.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/64/880464/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/64/880464/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/880464/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/880464/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/880464/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/64/880464/1"}}},"commit":{"parents":[{"commit":"76fbde18334bbbf420bd6269d5ebf1799490ecf0","subject":"Merge \"Support configuring VLANs with systemd-networkd syntax\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/76fbde18334bbbf420bd6269d5ebf1799490ecf0"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-01 21:39:02.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-14 08:44:38.000000000","tz":180},"subject":"Fix merge action plugins verbose output","message":"Fix merge action plugins verbose output\n\nThis change fixes the output \u0027module_args\u0027 information of the plugins\n\u0027merge_configs\u0027 and \u0027merge_yaml\u0027 when Ansible is executed in maximum\nverbose mode. Now all the plugin options are displayed instead of\nstandard \u0027copy\u0027 plugin options only.\n\nChange-Id: Ie2d9a0501fe29bfd854eb31258f282b197855948\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 46000d4e05b6b7fc133114e8100da6fc64afb70e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e8b9b8c0edb8e1482f3e4a0cde23dd735668d5d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e8b9b8c0edb8e1482f3e4a0cde23dd735668d5d1"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
