)]}'
{"id":"openstack%2Fkayobe~880876","triplet_id":"openstack%2Fkayobe~stable%2Fyoga~I339042d9ce405f59aba936dd98df7d89a88bb41e","project":"openstack/kayobe","branch":"stable/yoga","topic":"dib-yaml-width","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-04-20 10:42:44.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-04-20 10:44:50.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_14200\u003e by clicking the attention icon","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-04-20 08:26:06.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}}},"hashtags":[],"change_id":"I339042d9ce405f59aba936dd98df7d89a88bb41e","subject":"Fix maximum width of the DIB Multiline-YAML","status":"MERGED","created":"2023-04-19 18:15:07.000000000","updated":"2023-04-20 12:21:18.000000000","submitted":"2023-04-20 12:20:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880876-dib-yaml-width","meta_rev_id":"65f444052445569cea4b0c82298fe63db9f9a51a","_number":880876,"virtual_id_number":880876,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-20 12:20:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-04-20 10:42:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-04-20 08:26:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-04-20 10:42:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-19 20:24:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-19 20:26:38.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-04-19 20:26:38.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"5b26e03fe811b759996d06192a004523067018b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-19 18:15:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a7b6e32305d387ede8df2b3eb95f1c3dfb2ea3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 20:24:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2823a2304a4047a29884d3c6653399b1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/add16077298340288f9cd8169c2e6262 : SUCCESS in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6437bdef02a44c1fb9e028e8d0371c17 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6dbb86ff6a834a988fedb661ffb415d3 : SUCCESS in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7b42bbc935347caa1035a59f08898e4 : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d81c8fb97a24459e8bdbd75288dc2582 : SUCCESS in 8m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cbffbe0a86014a22a4d3052d169d2bf6 : SUCCESS in 5m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2abc8ef0d9014cc68692740a93ec5099 : SUCCESS in 12m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c2eae90c6e3c48098c9dbc182054c726 : SUCCESS in 17m 32s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/fd1de42d4c3a43d380fce48352eb1445 : SUCCESS in 1h 14m 45s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/c2033a87f91344d489924a015656dd0a : SUCCESS in 54m 48s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/60344e660aca4a8f909fa6d2694861a2 : SUCCESS in 1h 10m 18s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/9af24bfc527840a69a7742abf816c3ca : SUCCESS in 56m 39s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/117b92714f504643b61bae179bfa9e17 : SUCCESS in 54m 11s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce7dbbb98abd478cadeb16695d17913f : SUCCESS in 48m 15s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0d5da203018547539fd3e092e67ef336 : SUCCESS in 24m 51s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/a785a82693ef4d31842754c4e9b39a7b : SUCCESS in 20m 38s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/2118c63d50ba4726a23ec46db5d25d77 : SUCCESS in 20m 48s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ad55d888e2bf4c9fbb05b190d3068d50 : SUCCESS in 19m 35s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f53f95a9fb948558b6ac1003141cc72 : SUCCESS in 13m 04s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d898c436132345cd9b229e7b6f37bf05 : SUCCESS in 16m 29s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/4ced6a837c5042ba825fb2adf78bce78 : SUCCESS in 1h 07m 08s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c0bc99bcebc341128ec8ad9491f6b87d : SUCCESS in 1h 07m 32s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5fc47e8a6576448596eb63e7b7e8ba6c : SUCCESS in 2h 01m 51s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b9f5d9a154f47218cd85fb788560bff : SUCCESS in 1h 12m 30s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a0981074b7354bd6ae75da6ceb29d150 : SUCCESS in 21m 43s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/3fc1014c42ea4574a77f1d85c2ae036e : SUCCESS in 21m 43s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/46e4f5371fd645dc92ef8df8793088ab : SUCCESS in 19m 42s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a3e7a8eadd54aba8991322cb804de60 : SUCCESS in 18m 04s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f36e11fecdb849dfb6a3ee3afd7e54df : SUCCESS in 18m 32s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/13bac7eb1e464a818f989d0215ac0c70 : SUCCESS in 26m 09s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce0b191865d7405bb8fa8eac8f937db3 : SUCCESS in 24m 23s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/7c52bc57bb52411c9ee3b47bf07735b3 : SUCCESS in 18m 01s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/86893c451b2b44b4a0ac8b4aae08b1c9 : SUCCESS in 16m 43s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d82fdaf83cd345fe9c2e1dd0cd7fbc74 : SUCCESS in 15m 31s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0502fda4b534ba98605cbb1aa28cc88 : SUCCESS in 10m 16s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fde6fec7fa2a4e219c87403dd1ce0265 : SUCCESS in 12m 48s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/1196ff5b31f14a278444c7005e568c8b : SUCCESS in 18m 33s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/1115808c60ab4bb1a0f54548d91468ce : SUCCESS in 21m 10s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0d54a3df51344a59838beb7a5c1a1f43 : SUCCESS in 16m 16s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6428d6a95acc48b9895e14a266904623 : SUCCESS in 14m 57s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a82181cc11814305812228e48935f92e : SUCCESS in 14m 26s","accounts_in_message":[],"_revision_number":1},{"id":"d59d46076e942f557611375c99312cf5f66068ae","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-20 08:26:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"63bc0983e3951df0e7f4f03ff27473da575fe466","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-20 10:42:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3bff6204d38b508931371c7f7e5fd3a7643ceb9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 10:43:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3f964d2b1151a685318547414898c04e32e5ddbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 12:20:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9faf45a080014213898b31fe28b20a5f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08f6024e902848e7bbf60a50cd31d099 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60a031877fc247be8ea5962283996ccf : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5ad7ddb4d3044c19af67cff9a1986e2 : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1368873cb94541b293865388269fb798 : SUCCESS in 9m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9c8775aed5054dd1b5e570d6e144d088 : SUCCESS in 6m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/baaf314f58c94764901f1e5b74f4d586 : SUCCESS in 12m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3a2715b52aae4f6397a9463d56d762be : SUCCESS in 13m 52s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d4acf5c0e2dd48a6a96599269d69ca83 : SUCCESS in 1h 03m 25s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/8221b7725ccf49cead59d06eeecb467d : SUCCESS in 1h 07m 08s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/2fffdfaeeb5c442aa70acee9e2e16799 : SUCCESS in 43m 36s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dcb1bb2ed50042f6b93d334431f4e53d : SUCCESS in 48m 43s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52fc963ca68b4bfda81b7159c83430df : SUCCESS in 57m 18s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/54f609b766084136bb29fb93a7155eb9 : SUCCESS in 24m 08s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/72a0e6411ff04856bf569d8a30359f21 : SUCCESS in 24m 40s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ba778f7ed9904cc898bf8b47926807c3 : SUCCESS in 20m 28s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a2406bc8e89548fdbc171eaeed80ff22 : SUCCESS in 15m 18s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b15d7f677bc0405aa1f369350fe1c895 : SUCCESS in 14m 21s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d9c5ed9f100248d2bc6b949a36c5decc : SUCCESS in 1h 02m 19s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/cdbddff557a84fa8ac40227079d668b0 : SUCCESS in 1h 01m 26s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/021df2e5935e47c994049cc84da82ae2 : SUCCESS in 1h 32m 39s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f592ff2c309043cbbdbba4a71d999dc6 : SUCCESS in 1h 10m 16s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d9d8138df3ec4c54a8131edb2f9336fd : SUCCESS in 24m 53s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/70f66c763098402daf90033479efa146 : SUCCESS in 15m 23s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/16d04e90ee1e4ed39ba628a2ffff5e95 : SUCCESS in 19m 04s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99a65ba62be2461f9fba173f3f42c89e : SUCCESS in 16m 47s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4539ae8ae587431bb50418544f04db42 : SUCCESS in 15m 50s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1d10ea29fadc48cb9b88ed409947bbe8 : SUCCESS in 25m 56s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/15c3d239916542e4a63ed4c85ee2bba1 : SUCCESS in 24m 22s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/36f3e366672141d4ba295124cd0439dd : SUCCESS in 19m 17s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/911fc90073f94b9d962eaacb8ea88397 : SUCCESS in 18m 15s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/50073c0815e94b7082c2bce773bb4c5f : SUCCESS in 12m 37s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ceacbda8f004f8e8df07d316daefbf8 : SUCCESS in 13m 19s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ac972c70ec442c6b167d8b5dabd07bd : SUCCESS in 15m 01s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/3371654b9d0a4070a65124fd1b25397f : SUCCESS in 17m 34s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/bd56386d82414900843b1620f5535038 : SUCCESS in 22m 38s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fb6bfabc295d4a0da78ae333d1119c8c : SUCCESS in 16m 41s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1c957998d514a398b2e04b28cf2a7d5 : SUCCESS in 12m 13s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ba1069ce19df4030ad7cfd11b0ef7c12 : SUCCESS in 15m 12s","accounts_in_message":[],"_revision_number":1},{"id":"760466961bcf690af7adaf6778344d8c54749c84","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 12:20:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"65f444052445569cea4b0c82298fe63db9f9a51a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 12:21:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82c51faf11824c54bc179637cb90e5f9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3caef6d9f4d04e6f8b4793e04ef97015 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b394ff396365991b8d16aee179f5c40d0db59507","revisions":{"b394ff396365991b8d16aee179f5c40d0db59507":{"kind":"REWORK","_number":1,"created":"2023-04-19 18:15:07.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/76/880876/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/76/880876/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/880876/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/880876/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/880876/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/76/880876/1"}}},"commit":{"parents":[{"commit":"2d4d609ee444c70965631579196be841f65a3103","subject":"Merge \"Fix merge action plugins verbose output\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2d4d609ee444c70965631579196be841f65a3103"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-02-28 22:12:17.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-19 18:14:56.000000000","tz":180},"subject":"Fix maximum width of the DIB Multiline-YAML","message":"Fix maximum width of the DIB Multiline-YAML\n\nThe dib_env_vars variable in the Bifrost\u0027s dib.yml file can contain\nthe DIB_BLOCK_DEVICE_CONFIG environment variable which is always the\nMultiline-YAML data. By default, the format of the data is not\npreserved while the configuration is merged and saved for the\nbifrost-deploy container.\n\nThis is because Ansible uses the PyYAML library which has a default\n80 symbol string length limit. The official Ansible documentation [1]\nrecommends using to_yaml or to_nice_yaml filters with width parameter.\nThis change adds the same ability to the merge_yaml Ansible plugin.\n\n1. https://docs.ansible.com/ansible/latest/user_guide/playbooks_filters.html#formatting-data-yaml-and-json\n\nThe related change for the diskimage-builder to solve the issue with\nincorrect data provided by Kolla-Ansible is also provided:\nI3b74ede69eb064ad813a9108ec68a228e549e8bb\n\nThe related change for Kolla-Ansible is also provided:\nId79445c0311916ac6c1beb3986e14f652ee5a63c\n\nCloses-Bug: #2014981\nRelated-Bug: #2014980\nChange-Id: I339042d9ce405f59aba936dd98df7d89a88bb41e\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 5cf750c3135a8cbbcbe8b003269cb79ef6f0e8ab)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b394ff396365991b8d16aee179f5c40d0db59507"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b394ff396365991b8d16aee179f5c40d0db59507"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
