)]}'
{"id":"openstack%2Fkayobe~880877","triplet_id":"openstack%2Fkayobe~stable%2Fxena~I339042d9ce405f59aba936dd98df7d89a88bb41e","project":"openstack/kayobe","branch":"stable/xena","topic":"dib-yaml-width","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-04-20 10:42:52.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-04-20 10:44:52.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_14200\u003e by clicking the attention icon","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-04-20 08:26:16.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}}},"hashtags":[],"change_id":"I339042d9ce405f59aba936dd98df7d89a88bb41e","subject":"Fix maximum width of the DIB Multiline-YAML","status":"MERGED","created":"2023-04-19 18:15:27.000000000","updated":"2023-04-20 12:39:28.000000000","submitted":"2023-04-20 12:38:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880877-dib-yaml-width","meta_rev_id":"ef4110c4fec54cbb771c975e566722c8757c1e75","_number":880877,"virtual_id_number":880877,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-20 12:38:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-04-20 10:42:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-04-20 08:26:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-04-20 10:42:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-19 20:25:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-19 20:27:15.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-04-19 20:27:15.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"e89b2cadb19ff4e980607613bc0d6b22af82f7c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-19 18:15:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"677edc8da6ff5ccf2d4a9ece0961e0647c0dd38d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 20:25:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e78573dc658d4b509b995612478ea435\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f3e4a3b9a264ba2801217619d9abe11 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb11e3ce42d2423d932f8c2a7a964c26 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48a73cc17d9f45dd8f4e61934b921f73 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f65b1a166d049c48971fa49ba55bf9f : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/42d08f74d2a845b0b518c038d9b27c24 : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cafcea7d0724aa08c556b35842de911 : SUCCESS in 8m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0f38d947d3f042a7ae64db5502277c84 : SUCCESS in 5m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/624c772ad5e8452b833c5581dd344d58 : SUCCESS in 10m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/464939d36f384ba39d6bf0364cee3194 : SUCCESS in 18m 58s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d1d22e65caad47838a2c3f61fa31a8db : SUCCESS in 1h 08m 37s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/827a2bd48bfd4b66811cd91102d65348 : SUCCESS in 1h 12m 35s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bff67de96b364ea383c285dcde773061 : SUCCESS in 45m 13s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d968f231d9b34df1aa2d5a3303192ce5 : SUCCESS in 1h 15m 10s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e7581b07b1c04f9685de77ea1b7d68c7 : SUCCESS in 21m 33s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/632f7eb8f6dd47e78023ac509de4fe55 : SUCCESS in 25m 42s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/226ed4ab781b4d0796211b7ff4d2d550 : SUCCESS in 15m 16s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/74d1e53561fc477cb68b0f7e02c4df2b : SUCCESS in 2h 04m 32s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5852f8c84fa64cceaafede80b40f6cf5 : SUCCESS in 1h 10m 30s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/02891a84ba5c4392bd09f9d3b6923f8a : SUCCESS in 20m 37s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/577e0dd88fd942ecbf15e63a37334a7b : SUCCESS in 20m 33s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7cc1b97e050c485a827bc8574df4c38c : SUCCESS in 14m 42s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/ae08c3a88e7f4221b335b66e9dfce47c : SUCCESS in 46m 31s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c81a27b899524ac6bfd8ce885055cc9b : SUCCESS in 27m 16s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e8363b478e740f28f332f8a91ebfb8e : SUCCESS in 19m 56s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/72689ac07e8e4c64931cc7a64ddb3d20 : SUCCESS in 18m 53s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/75825da2966d43c6a6f2829464bc86d6 : SUCCESS in 16m 36s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b31ae39ead4a46d892b65a6f5dfa6efc : SUCCESS in 11m 45s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/fc66736fee1e4377a5acd776f41c23c2 : SUCCESS in 19m 22s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/5cf1a29e325241c78977d8e697ab9a73 : SUCCESS in 17m 08s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b476a8eec85f41318a397c73c477c783 : SUCCESS in 13m 28s","accounts_in_message":[],"_revision_number":1},{"id":"67c7c53712f3159621290196ca354b83a7a85e78","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-20 08:26:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c504cd3c6f9363911263add1be8a9596d13a4fa6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-20 10:42:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"056ce2db9f1fe3d2b1a2f030a8958225d710c048","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 10:43:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"72eb57c0f3b8f54150b747eab6a1b2b25f084a98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 12:38:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/591777fe154c414899c8b6c9c9e92855\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aebb4343fbd844708e3aaf11a061d0bd : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2a0d66591b44b23a0ff8ed256aa6c63 : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/341db36492f7456690db67d73e0dc4ab : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07198062d2654032a0ce046870257c31 : SUCCESS in 8m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d944f29694fa41cd9944fe2040676b0c : SUCCESS in 5m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4b21adf442f348b58e602e087e0e96fc : SUCCESS in 13m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ac0c7d29fb4742d4a6979790da5112a3 : SUCCESS in 15m 53s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4d28a4d1909140afa1d59d3d22078ae3 : SUCCESS in 1h 09m 52s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/9fd86a7e65174301b291251a0a3387fc : SUCCESS in 1h 10m 17s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ad7e73a647d4a319a54f66c44a8816c : SUCCESS in 56m 17s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/aa125ba861524176b1f521d3df6809b1 : SUCCESS in 1h 09m 33s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/130f34ee0a7042819719978b44a5ed72 : SUCCESS in 25m 39s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/f4afec3edece41ab994eb1d11e140c5e : SUCCESS in 23m 57s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f31db5618894437da4db5a78dfc75c15 : SUCCESS in 13m 43s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/91bdb8fbb5cb4c06b048da59b858695c : SUCCESS in 1h 52m 52s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31f3042112364a21a723234ee0fad476 : SUCCESS in 1h 15m 08s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/86e1ccee03fa4a5f837ff2e4d6a1ca6d : SUCCESS in 22m 00s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/2c57a8cc337f44058bc4b011dc2eb527 : SUCCESS in 15m 34s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3fcbe71382234b5ab082ae9a1b47e0ec : SUCCESS in 17m 38s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6185f54e4cb24002984e88c874e45772 : SUCCESS in 25m 46s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff3cec2fd7974edfa9ec9eb73fb3096e : SUCCESS in 26m 52s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b3ef2869dddf40e5bd481d0d4076a0e6 : SUCCESS in 14m 20s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/5a4e62baea5547e885ed878cf379fd9b : SUCCESS in 16m 23s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42ce944ae6cb4eb0bd420f12a3eb0780 : SUCCESS in 13m 29s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/9ca177b7054647198ce9567e5cf5486a : SUCCESS in 19m 52s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/f4b50af02f9949ce927a69ee85fb01b1 : SUCCESS in 19m 39s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e0e215cae4842898537d7bec0b17fe4 : SUCCESS in 13m 38s","accounts_in_message":[],"_revision_number":1},{"id":"73c8f573592609110ee90a8b8bc209a54d277388","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 12:38:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ef4110c4fec54cbb771c975e566722c8757c1e75","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 12:39:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/098e9a4574b84b46a2efb62c4e502898\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/061634af31ca4054975a726042b44115 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2e31c7f32e094f862d0e1c5b84070cb35af61e66","revisions":{"2e31c7f32e094f862d0e1c5b84070cb35af61e66":{"kind":"REWORK","_number":1,"created":"2023-04-19 18:15:27.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/77/880877/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/77/880877/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/880877/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/880877/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/77/880877/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/77/880877/1"}}},"commit":{"parents":[{"commit":"466651ab0242c67a4bdedfbb7cd78d38f7f974c9","subject":"Merge \"Fix kayobe_config_path description in globals.yml\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/466651ab0242c67a4bdedfbb7cd78d38f7f974c9"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-02-28 22:12:17.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-19 18:15:16.000000000","tz":180},"subject":"Fix maximum width of the DIB Multiline-YAML","message":"Fix maximum width of the DIB Multiline-YAML\n\nThe dib_env_vars variable in the Bifrost\u0027s dib.yml file can contain\nthe DIB_BLOCK_DEVICE_CONFIG environment variable which is always the\nMultiline-YAML data. By default, the format of the data is not\npreserved while the configuration is merged and saved for the\nbifrost-deploy container.\n\nThis is because Ansible uses the PyYAML library which has a default\n80 symbol string length limit. The official Ansible documentation [1]\nrecommends using to_yaml or to_nice_yaml filters with width parameter.\nThis change adds the same ability to the merge_yaml Ansible plugin.\n\n1. https://docs.ansible.com/ansible/latest/user_guide/playbooks_filters.html#formatting-data-yaml-and-json\n\nThe related change for the diskimage-builder to solve the issue with\nincorrect data provided by Kolla-Ansible is also provided:\nI3b74ede69eb064ad813a9108ec68a228e549e8bb\n\nThe related change for Kolla-Ansible is also provided:\nId79445c0311916ac6c1beb3986e14f652ee5a63c\n\nCloses-Bug: #2014981\nRelated-Bug: #2014980\nChange-Id: I339042d9ce405f59aba936dd98df7d89a88bb41e\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 5cf750c3135a8cbbcbe8b003269cb79ef6f0e8ab)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2e31c7f32e094f862d0e1c5b84070cb35af61e66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2e31c7f32e094f862d0e1c5b84070cb35af61e66"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
