)]}'
{"id":"openstack%2Fkayobe~880879","triplet_id":"openstack%2Fkayobe~stable%2Fyoga~I03a826e98a18b158774ba100cfa2987299eb6c25","project":"openstack/kayobe","branch":"stable/yoga","topic":"fix-acl-forgotten","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-04-24 07:07:26.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-04-24 09:21:48.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-04-20 08:42:14.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}}},"hashtags":[],"change_id":"I03a826e98a18b158774ba100cfa2987299eb6c25","subject":"Fix an issue when \u0027acl\u0027 package can be forgotten","status":"MERGED","created":"2023-04-19 18:22:06.000000000","updated":"2023-04-24 09:22:50.000000000","submitted":"2023-04-24 09:21:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880879-fix-acl-forgotten","meta_rev_id":"6d874a9e3e38df663348ec47ca7fc7147355e5be","_number":880879,"virtual_id_number":880879,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-24 09:21:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-04-24 07:07:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-04-20 08:42:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-04-24 07:07:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-19 20:32:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-19 20:42:32.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-04-19 20:42:32.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"8b1543a79de39ae5aeb3bed6ecf8a724d8de7d1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-19 18:22:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"98c18f34fc27f97445862e5abc0b2d88952cf74c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 20:32:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a134c36ee9f4ad69840164d46c871e1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b505170d8f3b46c38c4d5ad68dd4e9c8 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d44d8ce884a4fb083918df990a57c9d : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22e64d147dd843efbaec6c8cbfae1899 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b1a41d9524b420d8191ccd654b039fa : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a361cf6a62de485bba020d717fb7edbe : SUCCESS in 9m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/64007998e2e141f294354a05fea8a2bb : SUCCESS in 5m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/88f6b32a89f5432abf4ae0bb032f45c3 : SUCCESS in 12m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6bb2037918f4415c8f2301c9df447966 : SUCCESS in 16m 09s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5efe6dda4d144f74a110cf8b79a679bb : SUCCESS in 1h 03m 33s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/3b15be0bf84f4bc3bb7b55c7bdd33ea3 : SUCCESS in 57m 39s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/740b6bc3e0c447c99f001ad03e31402e : SUCCESS in 57m 29s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/2a7d12d2c4d74ee7b200c75d3664211a : SUCCESS in 59m 39s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e183331022954a34a5f4bd5428d6d614 : SUCCESS in 54m 28s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e0c830d82c64ffca4d0534455c12386 : SUCCESS in 48m 47s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c87d6667e9e1400fa455f1562c967b03 : SUCCESS in 22m 51s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/4ee02eca7e1a45cebc0a59e87743a537 : SUCCESS in 21m 26s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/02bb592efe9e43cd9001316612d3a623 : SUCCESS in 23m 13s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f7ff502b09ad4fefb1e23887e780969b : SUCCESS in 19m 32s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/442a57b1223443db951d772f3df311a5 : SUCCESS in 15m 34s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/63e0cb8ed5ec4cf0901104d42ae6edd2 : SUCCESS in 15m 55s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/57956a4f857844678904e3c057ec99a3 : SUCCESS in 1h 08m 04s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/da24d9da588f4a6ab44f5ae224e698dd : SUCCESS in 1h 01m 28s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d296dee9d8bb476e8e0d8c1bd674a7cd : SUCCESS in 1h 52m 49s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3da6fe1c06b94aee8ecae5cb8b2a7375 : SUCCESS in 1h 10m 33s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/21b271dfd4d14401978db1921558868a : SUCCESS in 21m 50s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/68de61411b0447a8837b101298964af8 : SUCCESS in 18m 26s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/fb9904ea53b948d488bc270a66b89e27 : SUCCESS in 18m 52s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1eccbe6efcd64661840e500417b9a0e6 : SUCCESS in 19m 29s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e3997ea958d4ff2a770555bbf59127b : SUCCESS in 16m 45s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/56d4e262069e40c988a353124f3ec627 : SUCCESS in 28m 59s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3412f3bc999f4c7aa21da15998faa97a : SUCCESS in 23m 01s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c3211c3bd62a42f3838e4b7d71f14d62 : SUCCESS in 17m 54s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/fd13c811025b484dbb21ae995d96d3d3 : SUCCESS in 17m 13s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/476c18752145436593d85175d8a60f41 : SUCCESS in 16m 41s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d920d675187645f3b99fcdf80796d803 : SUCCESS in 13m 21s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b6c15cdea9994355b2883a3f28998dae : SUCCESS in 14m 01s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/663a9b7a16c944d1bd42ea2031a60f4f : SUCCESS in 20m 48s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/e2b8a4f5c73048c68d47b0dc53ae5a7b : SUCCESS in 18m 46s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/317c394a73a74de1a45e1461134a74f0 : SUCCESS in 17m 16s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e17c558a6196496c8b4326ff7eb2067d : SUCCESS in 14m 23s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f2813aa983df4b349078294ead97201f : SUCCESS in 15m 28s","accounts_in_message":[],"_revision_number":1},{"id":"f17c3538d10d891343bb008671f410f0f5c7f8ca","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-20 08:42:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cf4d838f7ba6d0ec7fee1039b8fd8ad0f7cf63f3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-24 07:07:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"164c7e65e162a7aefebb51ed465181232707f64f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 07:07:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d44da56d1ddde6f21ecd9cac7b26e6db1a74738c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 09:21:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a078d7a338244e39afa1b984156eb5f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/829eeb0b321e4756821ff4ec7b9ae93c : SUCCESS in 7m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c285d25e25c433a907da32ee634a876 : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7068f3e10b474e5eb7e91ff7d44d26f3 : SUCCESS in 7m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32d18f0976de4a87b9cee9ed41a54429 : SUCCESS in 8m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/04f443d8d01b448fbea9a80409aea328 : SUCCESS in 4m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6daee74681244776a1e701e5001164e7 : SUCCESS in 14m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/08dd9c3ae31745af80b69af76e23e7c7 : SUCCESS in 15m 34s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8a8f6a00944f49c3a4e2c5d390631995 : SUCCESS in 1h 00m 29s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/dba2efbf5e0e403598c69390a08da09c : SUCCESS in 1h 08m 15s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/f3806ea7a34a458f9a3494b8409a1353 : SUCCESS in 1h 04m 02s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a18a7b4b22ea4f2d92945baf6d566871 : SUCCESS in 50m 07s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3749ee168f7c491fab210512ddcc2ec7 : SUCCESS in 52m 24s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/896a948d699a4cb78042262fbe439716 : SUCCESS in 19m 54s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/da43768f9e5a4d6cb4a75ec43c11d60b : SUCCESS in 20m 52s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a98b703515954fe787a91e7e3f08cf45 : SUCCESS in 21m 28s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a86052e63a444568c0acd8833bb780f : SUCCESS in 12m 42s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de207090670441ff94efba1695455acc : SUCCESS in 13m 19s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5396b0b5548f447a921769f0e5fe39e6 : SUCCESS in 1h 14m 31s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/848d9b1aef57416ea1d21c6c71db6a3a : SUCCESS in 1h 03m 45s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5ae6bf42278f4a9584747225f5273e0e : SUCCESS in 1h 59m 32s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a3f0df59a61743b797343608af0f4c50 : SUCCESS in 1h 10m 51s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/9958b7bbeefa4672a3616071d89b376f : SUCCESS in 19m 30s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/c469738c28644b6db16b15ff57523faa : SUCCESS in 21m 01s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/18fca2a7bd1f458188b554cec6bc74d8 : SUCCESS in 20m 12s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18b46ce106b64bffa3483651dfbb5744 : SUCCESS in 17m 45s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4f5f303305cc4c449083e76bf9237af1 : SUCCESS in 15m 16s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a1e3f0cec2ea4fbcadd679bbce9ea04e : SUCCESS in 25m 36s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b45143bd6e9145a39a1e1d26e19c2e5c : SUCCESS in 23m 25s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/6bc4c882d2b74bb8b69d096623d11450 : SUCCESS in 14m 02s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/9d20206217a74d32a34c0637577f12ce : SUCCESS in 12m 30s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2dc7961d519f4b9ab9faa39d8dc8e377 : SUCCESS in 15m 46s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/04728c10261d41c48cea157d42613514 : SUCCESS in 12m 40s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bd4824bb07f44a2eb4d8a5580916481d : SUCCESS in 13m 46s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/ecc7febcd1ba42b0949a4101b2fa7931 : SUCCESS in 20m 32s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/017022c3c271434daa5853bdbf207151 : SUCCESS in 14m 48s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1584132b601c47caae9193c64cf2ae24 : SUCCESS in 16m 14s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3477fa412811494b98517a547f1dca41 : SUCCESS in 14m 22s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/236c3f6b4bec4858ae66a946ad4baeb5 : SUCCESS in 14m 04s","accounts_in_message":[],"_revision_number":1},{"id":"6c280b5721a77501cc0a57c1cc6dee1e6ca109f3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 09:21:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6d874a9e3e38df663348ec47ca7fc7147355e5be","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-24 09:22:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a10f84f3ba304af4b57b61017fa88ebd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb6f30e6108748f387bbda4eda3cd4e2 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"81e8f16205eb61b42673163bf0e214e2fb184402","revisions":{"81e8f16205eb61b42673163bf0e214e2fb184402":{"kind":"REWORK","_number":1,"created":"2023-04-19 18:22:06.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/79/880879/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/79/880879/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/880879/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/880879/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/880879/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/79/880879/1"}}},"commit":{"parents":[{"commit":"2d4d609ee444c70965631579196be841f65a3103","subject":"Merge \"Fix merge action plugins verbose output\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2d4d609ee444c70965631579196be841f65a3103"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-09-29 12:31:54.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-19 18:21:54.000000000","tz":180},"subject":"Fix an issue when \u0027acl\u0027 package can be forgotten","message":"Fix an issue when \u0027acl\u0027 package can be forgotten\n\nThis is folllowup on I69bf810632d09eddaa3983ae56e833debe9fd03b to\navoid user accidentely forgot the \u0027acl\u0027 package when override the\ndev_tools_packages_default in their custom configuration. Also this\nadds an ability to customise list of packages installed in addition\nto the default list.\n\nChange-Id: I03a826e98a18b158774ba100cfa2987299eb6c25\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 5e8a8233da46b321f03d9323a746e72422ba7670)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/81e8f16205eb61b42673163bf0e214e2fb184402"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/81e8f16205eb61b42673163bf0e214e2fb184402"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
