)]}'
{"id":"openstack%2Fkayobe~880883","triplet_id":"openstack%2Fkayobe~stable%2Fxena~Ia956cfea7e7bfe47bf1e73c9edcac602caf45579","project":"openstack/kayobe","branch":"stable/xena","topic":"add-multipathd-config-support","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-04-24 10:08:27.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-04-25 15:10:39.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-04-25 11:15:18.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}}},"hashtags":[],"change_id":"Ia956cfea7e7bfe47bf1e73c9edcac602caf45579","subject":"Adds support for custom Multipathd configuration.","status":"MERGED","created":"2023-04-19 19:13:14.000000000","updated":"2023-04-25 15:11:40.000000000","submitted":"2023-04-25 15:10:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880883-add-multipathd-config-support","meta_rev_id":"a778828d673f6ef5d1c6b2cee590df6c9589b4b5","_number":880883,"virtual_id_number":880883,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-25 15:10:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-04-24 10:08:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2023-04-25 11:15:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-04-25 11:15:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-19 21:27:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-19 21:28:46.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-04-19 21:28:46.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"5e890c6e355d6f352c90a46d06ca2b42fead0331","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-19 19:13:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bd57ba0b26eaf075af2dd0e972d941bf60f4d34c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 21:27:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/880b7ea16e4640a78c83d5a648ec96f9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84dece34f18444938a3041ba783b4ecd : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b27b86059e24e16ac52aaeec032143b : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41cd579d4cca4fe38335135c61650a17 : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf1ded52dae14ee2bb50ed0db8b7c0ba : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb7838c2eca9425bb90f46ac6a2b3d28 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57196bcfc07a4ec0b590365bd96fd053 : SUCCESS in 8m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/514c568349c54374a88e081376736c8f : SUCCESS in 3m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/31ee9308eff4445daaeac25a8671a22c : SUCCESS in 4m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ed3626be1efe4a9d9e890311b16d5b31 : SUCCESS in 12m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bd278a8058564b909a02d3a92b515343 : SUCCESS in 17m 42s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/cb0a5f8b868f40f9b48d3ce27f459ccb : SUCCESS in 1h 08m 41s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/e66cea0e2c794114a2437120904abeb4 : SUCCESS in 1h 01m 13s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7296549d95b442ff903867d67b970fdc : SUCCESS in 54m 55s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9330c50f70344dd2a2bfbbdf80992679 : SUCCESS in 1h 09m 54s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0dc28ac80cd4498d91190269183551e9 : SUCCESS in 27m 17s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/b5edf311f5ee4850a12faa0fe0f6195b : SUCCESS in 21m 59s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8cb4e20ae705435981b0f3750bdcf7b9 : SUCCESS in 15m 11s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c861539a25354b1ab0f22be4b0d7a1c0 : SUCCESS in 2h 09m 13s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53d560a47e8f4779b4a0960a6f398028 : SUCCESS in 1h 22m 45s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ccd0e440cb534af6838d07dd542cd0cc : SUCCESS in 21m 24s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/5fc67bbd8b4349c3bd91a0fce8cd286f : SUCCESS in 20m 47s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7586619c9e44578a178ea34a7548685 : SUCCESS in 16m 56s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/a074db5e977d4d0b884535cd536186d4 : SUCCESS in 51m 27s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3f3e35925c4e47df8aa6dcbd5588b114 : SUCCESS in 30m 15s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef6d3259bac94614bec927355ee28b60 : SUCCESS in 21m 00s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/62099959ddd04e8b847b6ced5fd50215 : SUCCESS in 17m 54s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/a575c3d9ebda43c38fc0164cea4a8328 : SUCCESS in 16m 35s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8af2f1e067d04d0b99fa7ad374c4f3af : SUCCESS in 12m 18s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/2c08ac67c779445b97cb6ebaffa33363 : SUCCESS in 18m 17s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/c037a397f65842d6a6dd43f5a7881093 : SUCCESS in 20m 13s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e2329fb34444422984e918247681d299 : SUCCESS in 13m 45s","accounts_in_message":[],"_revision_number":1},{"id":"c4e43f5b0f2778d5bc3428fa04ab0870b037cb37","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-24 10:08:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ab9f3bff9f9fc902462481d5d018f52698a62e26","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-25 11:15:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3ed3461b67357cc6c85b9342a81dbd167485bb8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 11:15:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aa3182358cd1c63b926104bfc6be6eadef4ae3a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 15:10:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b9f330389ba4b38ab4dbc194ebd9000\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77b5a84d7d0544849313dc0a15790291 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b64af31189644c6b8c9dd09a32bcdc6b : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e6b1f89d8084d4ca2fcd9bdec39471e : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/953c0b285923401b8c95b3568a80da0a : SUCCESS in 8m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/728a56a1b09c46199a819af7e314fb56 : SUCCESS in 4m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c21f7825fb1b42338119f1f0d62c8ed9 : SUCCESS in 5m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5e46b520c7d54aad9d9c37e9cafd2fb1 : SUCCESS in 11m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0c87eba8ff5845f3a3a08a5b95bd39e3 : SUCCESS in 18m 52s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/2fc8966188714196b4cda3e1d5feb87a : SUCCESS in 1h 09m 04s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/4b535cb458d74128a31bd8d16c57110f : SUCCESS in 1h 06m 23s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8a3e7165d6246adb351923097f2ade4 : SUCCESS in 51m 05s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/1cdbe2a31d14411f859bf82eb5c97258 : SUCCESS in 1h 10m 22s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7b590de3fdff4781aea5a270c20775d9 : SUCCESS in 26m 58s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/fde116573a87424ebc56e980e69d8eb2 : SUCCESS in 23m 51s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0841718d147e4538aa47b665bb427668 : SUCCESS in 15m 34s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2a10468e9b864d86b3ba3ba604905ce1 : SUCCESS in 1h 37m 40s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e33a5d98ac204b18ae77f6d3cf659536 : SUCCESS in 1h 23m 42s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/1d3cd7791b404adfbd8c16879997a47f : SUCCESS in 21m 51s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/3e4d18be67f5473d981955cc92bf6dfc : SUCCESS in 21m 41s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b83953f7b8014c6990542cc002f47eb6 : SUCCESS in 12m 59s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ab8cba70e2074106a840da1cf570c76c : SUCCESS in 28m 36s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1867db1bf4fb48229958795b74789238 : SUCCESS in 26m 16s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/367393775e764a7b908ee0346254e927 : SUCCESS in 17m 20s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/92f537ad579647e1bebaefedd0b60a12 : SUCCESS in 19m 29s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1e5f6a60465426a8e53cbeb891c41df : SUCCESS in 11m 05s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/68eee0cb454f4b12b8d1e10720138104 : SUCCESS in 17m 14s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/3d76bba7574d4fdc84fbfb8456df8544 : SUCCESS in 16m 08s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3fb1cbc867f14a3eb48289561edeecc3 : SUCCESS in 11m 39s","accounts_in_message":[],"_revision_number":1},{"id":"02132209b33d3bb6f25dbc8b23dd502514bb6258","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 15:10:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a778828d673f6ef5d1c6b2cee590df6c9589b4b5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-25 15:11:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33b13c9af82e4e5aaecef730dd9767f2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0981f32195f749908502146049aa4eb0 : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/964253be89fa473f88046d3268a47f6b : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8249a39ea72adb3297221538b6d0085ede1a7840","revisions":{"8249a39ea72adb3297221538b6d0085ede1a7840":{"kind":"REWORK","_number":1,"created":"2023-04-19 19:13:14.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/83/880883/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/83/880883/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/880883/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/880883/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/880883/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/83/880883/1"}}},"commit":{"parents":[{"commit":"466651ab0242c67a4bdedfbb7cd78d38f7f974c9","subject":"Merge \"Fix kayobe_config_path description in globals.yml\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/466651ab0242c67a4bdedfbb7cd78d38f7f974c9"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-03-31 15:39:56.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-19 19:12:53.000000000","tz":180},"subject":"Adds support for custom Multipathd configuration.","message":"Adds support for custom Multipathd configuration.\n\nChange-Id: Ia956cfea7e7bfe47bf1e73c9edcac602caf45579\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 25fe761e62f647ed828dc5441d651fb4433c8bf0)\n(cherry picked from commit a7a65d3646b11a7744a4402b55574fe2c118433d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8249a39ea72adb3297221538b6d0085ede1a7840"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8249a39ea72adb3297221538b6d0085ede1a7840"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
