)]}'
{"id":"openstack%2Fkayobe~880923","triplet_id":"openstack%2Fkayobe~stable%2Fzed~I0d7ab0d8ff5b16ac5de8e50e63400bd455996555","project":"openstack/kayobe","branch":"stable/zed","topic":"user-provided-repos-yaml","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-06-07 11:32:02.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-06-07 11:32:02.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-06-05 19:01:47.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-06-07 09:57:24.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I0d7ab0d8ff5b16ac5de8e50e63400bd455996555","subject":"Allow to use own repos.y[a]ml file to build containers","status":"MERGED","created":"2023-04-20 08:53:35.000000000","updated":"2023-06-07 11:33:12.000000000","submitted":"2023-06-07 11:32:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880923-user-provided-repos-yaml","meta_rev_id":"11ba612f2a99c1ecaf1de31cf585147f8530d5a7","_number":880923,"virtual_id_number":880923,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-07 11:32:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-06-05 19:01:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2023-06-07 09:57:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2023-06-07 09:57:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-20 10:33:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-20 12:13:41.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-04-20 12:13:41.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-06-06 10:23:18.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"e4e4e73812d3cfadbe55e108baa82d43bdfd1615","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-20 08:53:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ca0419b6d698bbe839fc08cb0203869851d54238","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 10:33:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7a432e4374146b881f9d589a15285c8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb22b90443604b749d30011926f26f74 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af3583d68bd342bb8384e3d90a6940ac : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd652e031574495883f6af32934cb5f8 : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/528ce31de0a9419ab370c24f1bd3fbda : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/88521e8c1be34f3bb28f9a1be4e825a7 : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34538b717d4a45989d75ad4504dd67c8 : SUCCESS in 7m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9cfecc2d5e964b2da15ffbaac5eb06ef : SUCCESS in 3m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1727637dae8344b2b52816c31dc79669 : SUCCESS in 5m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/555394d0877d46e7a5251e1e948cf58a : SUCCESS in 9m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3d04da78bf3c4e33a80555740b750a11 : SUCCESS in 13m 31s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/8d4b9665d23e445ab4a96fd1ba915aa5 : SUCCESS in 55m 26s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/e1e11d840c6c42afac07ea38af4a5e59 : SUCCESS in 51m 29s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/86670c1833d64447b72cd4e800e22f55 : SUCCESS in 57m 08s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/494d1fe0616a41a39de2fcee2efb39c7 : SUCCESS in 1h 01m 23s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/8500a74aea344ec1b4c97b0702331262 : SUCCESS in 17m 55s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/8fa66296714143718906149154f49777 : SUCCESS in 15m 30s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/842ebb453baa4532b8e29ddc2417b8ae : SUCCESS in 26m 21s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0614aa2c56f4f89b171f8ae5e08c2d0 : SUCCESS in 1h 30m 37s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ca64cca49e77417489314a81b4b1e19e : SUCCESS in 20m 17s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b6313d62c4ca4dd8b129a5061fcb4145 : SUCCESS in 26m 44s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/618b1675504f49c6be2e210c85177c49 : FAILURE in 39m 57s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a837d818f8da4187b0e99f534c44b5c4 : SUCCESS in 33m 13s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/cf23d4f6905f4e9d9062ee1e15f1f1fe : SUCCESS in 19m 15s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/37308ce330794662b8819563a0540cff : SUCCESS in 14m 41s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7f0657f3859a469aa09cc9b0a7b2daec : SUCCESS in 18m 12s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/636b42d587044025b17a23c939860f6e : SUCCESS in 11m 17s","accounts_in_message":[],"_revision_number":1},{"id":"cbdcb473dc81ce81aa5a7bd7d4cdc2dff0217ff4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-24 07:08:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c2b45e40f628595796cb0442a44013d9c1d5cf3f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-25 11:13:15.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2e504271401e68388d939b26b4a1558623990080","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-25 11:14:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dd3a8bd42ff8024c0dd81c0ab9887d5fe5f46406","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-25 11:16:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0dbe49f31e385b2f87a723de5274a2434817fa78","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-25 11:17:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1b563052134a61e2d9c09e8cce54d2be8e969b8e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-25 11:22:01.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"75d4164421b7060c5119ce17cdfe74138e0f2d85","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-30 16:06:33.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"af8576dc772a23776001e02983f08bcad57d6ff1","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-01 20:34:00.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"ddda6c3ff93f9b89e8b6726a0022207f47602e00","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-01 20:47:26.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"a8966475abeb13857a5fbf0d5033072577c85b1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 22:11:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a2966a381cb4edd873adfe96bf9ed38\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49d91667beed488ba109ebad2e32cdce : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7debbcef67b40879f60d18072e1797f : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd7df8cb93244f44a4d3108efffa68a6 : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5556dce740d4e44ba140da8d6d8bbc4 : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/43a4929372284eceacbe567ccf056186 : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9280b17541c44ce09d07a22acb14c1b1 : SUCCESS in 7m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1fe7316527554d7c9b7880489f1bd8aa : SUCCESS in 3m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/24ae51b7f2794dc0b1c37ee3c11fd537 : SUCCESS in 3m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2c51d7d086e54019a060ef5dbf4c2ff4 : SUCCESS in 9m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f3b71112c5e04ef6b9b2bbeb06a2f400 : SUCCESS in 12m 54s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/218b9c0b537642e2bdcb55a4e2642450 : SUCCESS in 38m 47s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/2ce197aae3dc4487b6ada362dcbf9990 : SUCCESS in 39m 53s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dcd3a3a117b34bb7aae8997f1353fb3b : SUCCESS in 38m 30s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/07395695b17a4c558b303d87f4863056 : SUCCESS in 43m 17s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f5754daa56c34cd4a6d71b9f97a90f6f : SUCCESS in 15m 04s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/7972dcfb8b0e4809876bdf273372b917 : SUCCESS in 17m 38s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8082a55711a741909fd3804405bbd45b : SUCCESS in 15m 55s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f01f05b9ec74d2881a2564c2da190ca : SUCCESS in 1h 19m 54s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/964e5f663cf14be9b534f3b528b064e7 : SUCCESS in 17m 41s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3904aa89cdf948249085831ea0657577 : SUCCESS in 13m 17s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/f9f72ad73e8949f5aae49a9038811401 : SUCCESS in 28m 24s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b91aa32b58543f98448bd33d1619ddd : SUCCESS in 21m 18s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a7c8298ef68c4b75bcf25cc4830a3242 : SUCCESS in 13m 49s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/32283f4c8f004c0e85080f852c043779 : SUCCESS in 8m 56s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d7483aa5f22b4e4b9ddce29b961fe8d0 : SUCCESS in 16m 19s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c635af7c6b7453495bbca8d9973a414 : SUCCESS in 12m 15s","accounts_in_message":[],"_revision_number":2},{"id":"094517e44f1c1fb7861ba2cdc8453990206504e2","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-05 13:25:29.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"1aacb93b977df8d3ee1865cf2c02b9cf9ba9c429","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-06-05 19:01:47.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f3717a88bfc427096fd1f8dfd97fd2329456621f","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-06-07 09:57:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"13e448a4ff940232d16d7e0d0582909be7c306c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 09:57:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"82b94e933e5703760ff2662309fe584180b7a30f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 11:32:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8a06486e829489ab3c0ca0c6f9b297f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8732dc93c3684768975a33fbfa501625 : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/732adcee5303450ab97dbc15a4606111 : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c0aa6ece4064db5bfa2f5b73419a680 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fa70c3cf151442da83322adf5c415ac : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f995023137e94ae79f30bb35b9de13fc : SUCCESS in 3m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/20b2cf8181484a0289d9d528984fa24c : SUCCESS in 4m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2a688f2522354b9a93329934475b7784 : SUCCESS in 10m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cdec5267483f4fed97991017ac85b96f : SUCCESS in 16m 37s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/3532e4465215458e92d315a3514d4a9a : SUCCESS in 45m 45s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/efa3899603a94e67990d1b25e65a7309 : SUCCESS in 53m 54s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/795a4e968d1543709ce1cd78d1443413 : SUCCESS in 48m 59s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/449535afe43649138b6abb5160eaef6b : SUCCESS in 20m 21s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8392d67bbb244c48ae64a5c0d5dd62a6 : SUCCESS in 17m 23s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d11ff64d4b746098ed2e3cba36f958d : SUCCESS in 1h 19m 53s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b587e4becb8842ac83432743a5b2ffd4 : SUCCESS in 13m 31s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b7c7da4699541c695d717f48b527c8c : SUCCESS in 14m 44s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3e6eed195674b37bb8e9e83f11e7019 : SUCCESS in 25m 46s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/be4f6a5a08db4828955e401e38fa64b5 : SUCCESS in 14m 14s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/58cceb49ed8e47a3840d2c67fac1d286 : SUCCESS in 11m 05s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e40d45904b754c8aa3c4805a1e7410b6 : SUCCESS in 15m 50s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d00b2225f37a4c639f415fee005d353d : SUCCESS in 14m 03s","accounts_in_message":[],"_revision_number":2},{"id":"d88d7f741022cdbe929c23cb8e0aa0bcbfd31fa8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 11:32:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"11ba612f2a99c1ecaf1de31cf585147f8530d5a7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 11:33:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47c6a03303ab4df48b53cc546b04495c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/991db60b38504fc2b0ab3537f7cc5fa5 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45fc5373f49f4db59478e3672a52670c : SUCCESS in 35s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"46b1974271fbcf55698227fb729797f8db12d3c4","revisions":{"cc0fb0d13b4cd6c04250a9dbe249b09d9ec3d6ae":{"kind":"REWORK","_number":1,"created":"2023-04-20 08:53:35.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/23/880923/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/23/880923/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/880923/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/880923/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/880923/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/23/880923/1"}}},"commit":{"parents":[{"commit":"5108aac96802620c072adbd3b23fc3b5cef1af02","subject":"Merge \"Fix merge action plugins verbose output\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5108aac96802620c072adbd3b23fc3b5cef1af02"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-06-09 19:54:07.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-20 08:53:05.000000000","tz":180},"subject":"Allow to use own repos.y[a]ml file to build containers","message":"Allow to use own repos.y[a]ml file to build containers\n\nFollowup the I0b07da22fea27e0ff4e90aaad19e50d84ff9a121 from Kolla.\nThis change allow to use own repos.y[a]ml file to build containers.\nMultiple Environments supported.\n\nChange-Id: I0d7ab0d8ff5b16ac5de8e50e63400bd455996555\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 070621a45df11fd2487ddc38f715c1075524ee93)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cc0fb0d13b4cd6c04250a9dbe249b09d9ec3d6ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cc0fb0d13b4cd6c04250a9dbe249b09d9ec3d6ae"}]},"branch":"refs/heads/stable/zed"},"46b1974271fbcf55698227fb729797f8db12d3c4":{"kind":"REWORK","_number":2,"created":"2023-06-01 20:47:26.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/23/880923/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/23/880923/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/880923/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/880923/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/880923/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/23/880923/2"}}},"commit":{"parents":[{"commit":"586d23550224d1f74a97d6e204fb23c84421f826","subject":"Show an error when *_bridge_ports is not a list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/586d23550224d1f74a97d6e204fb23c84421f826"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-06-09 19:54:07.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-01 20:45:38.000000000","tz":180},"subject":"Allow to use own repos.y[a]ml file to build containers","message":"Allow to use own repos.y[a]ml file to build containers\n\nFollowup the I0b07da22fea27e0ff4e90aaad19e50d84ff9a121 from Kolla.\nThis change allow to use own repos.y[a]ml file to build containers.\nMultiple Environments supported.\n\nAlso, this change contains I209459ecf5b1a6e5f541cbc72d74aa7defb459d4\n\nChange-Id: I0d7ab0d8ff5b16ac5de8e50e63400bd455996555\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 070621a45df11fd2487ddc38f715c1075524ee93)\n(cherry picked from commit 7088ff2c2f10ba9897d1cd5424fea6d9268020a9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46b1974271fbcf55698227fb729797f8db12d3c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46b1974271fbcf55698227fb729797f8db12d3c4"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
