)]}'
{"id":"openstack%2Fkayobe~880924","triplet_id":"openstack%2Fkayobe~stable%2Fyoga~I0d7ab0d8ff5b16ac5de8e50e63400bd455996555","project":"openstack/kayobe","branch":"stable/yoga","topic":"user-provided-repos-yaml","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-06-07 12:03:47.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-06-07 12:03:47.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-06-05 19:03:52.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-06-07 09:57:31.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I0d7ab0d8ff5b16ac5de8e50e63400bd455996555","subject":"Allow to use own repos.y[a]ml file to build containers","status":"MERGED","created":"2023-04-20 08:53:51.000000000","updated":"2023-06-07 12:04:46.000000000","submitted":"2023-06-07 12:03:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880924-user-provided-repos-yaml","meta_rev_id":"19b98b1b41c9c4d6115a4efb14f724421537d19d","_number":880924,"virtual_id_number":880924,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-07 12:03:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-06-05 19:03:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-06-07 09:57:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-06-07 09:57:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-20 10:58:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-20 12:14:04.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-04-20 12:14:04.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-06-06 10:23:28.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"965ad5d014562ca4e89ac0210c7de4163c24bbe6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-20 08:53:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"41ea42c6afc0cd30041991687a552511abed254d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 10:58:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55a88b303e8f4d1287fa835bc1d3d54a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f51faf2f656b43de9b1cd5ebefcf0b11 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16cc854b24484137b121c6d8ab2f9f59 : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b3056d4349444ffaf273ccafbbfa0d2 : SUCCESS in 7m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c2a7190674634001912ec07623b564d6 : SUCCESS in 7m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d733d652c7f846108880d3d08d548597 : SUCCESS in 11m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05885bfaabaf4d068e0331511b428dcc : SUCCESS in 3m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f3fa9cb8c2894986a33a96d31f04333f : SUCCESS in 7m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0a8d7f8398d04fc7b0a6d867b7c0fe39 : SUCCESS in 13m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3e2ee08ebd5f45719e9cf8864f954a0f : SUCCESS in 25m 09s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3135846bca674b21b71386ab0b718b21 : SUCCESS in 1h 09m 05s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/fb1c242c30ed4eb1a3e858daed41625e : SUCCESS in 59m 49s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/9d492c71958c443fa5aebe01187319b7 : SUCCESS in 53m 13s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/c69c1f08a06b49f9963fbbe9ec9d3246 : SUCCESS in 1h 25m 22s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c1e56db1d1ef4a678a88b194abf6238d : SUCCESS in 44m 49s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f1bef42406d040d3902439bd68fe81ab : SUCCESS in 1h 07m 17s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/bb8bd5511ed341c7b2f0f127bf0bee87 : SUCCESS in 30m 31s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/0218defa92bc4ca4a46738c58e64eb02 : SUCCESS in 21m 09s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/688fe57856364be6a0437874df5d0933 : SUCCESS in 25m 27s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f7f0ad1b50c14c5a815fb9cb19db4034 : SUCCESS in 23m 20s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42f87aad7d114a30ba55f41c238d9b1d : SUCCESS in 18m 36s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a629c45f081548d39c7266faa0e3d306 : SUCCESS in 16m 48s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/37b82f2924ed4f23accee01d74a50690 : SUCCESS in 1h 21m 33s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/96f7e0330b6241e7a104e2bac61d2529 : SUCCESS in 1h 08m 04s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f149ef640fbe4cf7b008c7b31fc8741b : SUCCESS in 1h 56m 12s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/864cd1f1d727420d89cb2f557127b0b7 : SUCCESS in 1h 15m 50s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/16b4cd3e7b404edd91da3feaa2b4996b : SUCCESS in 24m 52s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/fa555a5986da4fa4b79b66ae42f659df : SUCCESS in 22m 10s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/9856b2e9719443afb39ae64adcb96b51 : SUCCESS in 21m 08s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ca980278ea24a0d8b197d1e665629be : SUCCESS in 17m 19s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/786f373566c64bf3acc939b105431215 : SUCCESS in 16m 49s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/92a73c7809aa4df2b03e1991b36b60de : SUCCESS in 30m 21s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7992fc1b3c841a682b3afcb409f019d : SUCCESS in 26m 43s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/0f45e8346e8c44edba312a446766ff74 : SUCCESS in 19m 18s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/77c538529809444899d588dbd901a585 : SUCCESS in 17m 40s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/84c04880bc724aa8ba3c0d863e226ded : SUCCESS in 16m 09s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/813bbb4348604071996a49bccd706025 : SUCCESS in 14m 24s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8308df09bc694842b7775ad86e7dff65 : SUCCESS in 13m 38s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/6b60920be73c472db8ab0ae3f5c2d988 : SUCCESS in 21m 08s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/10e0200d59364768b77af7f909412f75 : SUCCESS in 20m 06s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ac88dd73d84d4efe89a28449dc2cb4d0 : SUCCESS in 18m 06s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9585c99910d4659ac6c19b9865a5742 : SUCCESS in 14m 33s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e74a486c41ef4a4991bb90258d39e93d : SUCCESS in 16m 45s","accounts_in_message":[],"_revision_number":1},{"id":"6283265e9c1227eaae83c72d5cd0bb09eabf83d3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-24 07:08:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"71215e1bf84bfc49b9eda50613af432c2c237d43","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-25 11:22:09.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"1ded90e790a5f5bc7eab96e9398ec823fb3951f6","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-30 16:06:17.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"7ad85f4c4297246424231e90f4a61c9891a327e1","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-01 20:35:02.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"fa8ee0bd36d6f4125dadd2127bf9389db8bd51b2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-01 20:53:02.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"30e08acc6994216b9172adc537b6f494dbb926a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 22:35:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f979af82c1534f85993cd3de7b4aa16c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/421b5024c7574593bc79f78fccbb0a06 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/671cd2ed44e343b6a21bfa82ebe6cec3 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c40aed2591fe49d0bba7dd70c08667e7 : SUCCESS in 6m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/86040e126b1147e2957484e03270c26a : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3938b600deb45829ac8cdbd326d4cbf : SUCCESS in 9m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/154fbc19cbdc4b9183b267cdc2148af8 : SUCCESS in 3m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9678d07078c74fc5817135f5f9f0fa54 : SUCCESS in 6m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d0907025646e49888ba8ccae466a0f80 : SUCCESS in 13m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4e25fc13ac6b4d4697c3a36bccbc677e : SUCCESS in 16m 45s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b8f1be19b8cc40bfb64476657d9c8e81 : SUCCESS in 1h 01m 06s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/b0d0b59c6e1e458aa079b747966740f5 : SUCCESS in 40m 21s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/8515966895494fd49008585935df3892 : SUCCESS in 1h 03m 54s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/225e0cede25344be83d40016ead6150a : SUCCESS in 41m 18s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e2822d923fd47c1b65b0a9fd0ae0fdd : SUCCESS in 47m 07s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e6669c540f7847c7ba69d154eba482d8 : SUCCESS in 52m 23s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/1466629060ce422188fb40a1a43562cc : SUCCESS in 24m 20s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/0f4917aaa69347afb574810fa28983fe : SUCCESS in 19m 37s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/6b114e34782e487494b0b15bc780f010 : SUCCESS in 21m 51s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c65b435114f34d849d122f0827fe9fa3 : SUCCESS in 19m 19s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d1fd77f5fe747fcb412c9abc6ba7c53 : SUCCESS in 15m 04s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/77f99d0097074111bf2566667314dbc0 : SUCCESS in 14m 01s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e45d74798bcd44eb8f43662ca7223e79 : SUCCESS in 1h 04m 30s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a3ad3009ec6748d6abe4b12e17c98d37 : SUCCESS in 49m 01s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f32296b289854c38ac741f9d4dfc478e : SUCCESS in 1h 40m 14s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1559d829bde84a70b19891e3277b4e67 : SUCCESS in 1h 07m 29s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/55b894b1c5944252962497708350dded : SUCCESS in 23m 10s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/070d45a738304e44886b1bd45cfc9a75 : SUCCESS in 22m 03s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4c4dc221313940ce9a8b1fe424dede81 : SUCCESS in 18m 03s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c18b44a07f504baaa629a1c905ef69f3 : SUCCESS in 17m 49s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/daad22bedc7b421c87a85404dbac68bc : SUCCESS in 15m 08s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bd90e8e7923247908117c740bcfe3169 : SUCCESS in 21m 23s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e293d99eb659416eb92cd6f2833d45d1 : SUCCESS in 22m 14s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5dde710a9460436ab0ea6975ade45d71 : SUCCESS in 18m 58s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/c6c2c5562d6e416e86f21a0f5523e77c : SUCCESS in 18m 23s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b5a4f27675854de7bbf9fc3fc1bbfe27 : SUCCESS in 16m 53s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd714d6d7942451995d9251c9a84ea9e : SUCCESS in 13m 08s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb45faeae03848aea037b60d71057d11 : SUCCESS in 13m 42s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/4754cb88297d46c4aec69c6936f48edc : SUCCESS in 21m 54s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/ac8f485b90234fb19b598f545ce8459b : SUCCESS in 20m 02s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/536e817bb12c4a2ab1f6f36ca8b72083 : SUCCESS in 19m 23s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/475baa04b63f4501a5f43b3bf0db90ee : SUCCESS in 17m 20s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d3378226bfd4ec2a1a35429433f4aed : SUCCESS in 16m 00s","accounts_in_message":[],"_revision_number":2},{"id":"8eb96a5e0a1d4eb809b234732ff4ad2a669b11dd","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-05 13:25:27.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"1a0a8e558505da623ba9ecc96bba3d1a6d05f2dc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-06-05 19:03:52.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"938b782a030499e0e6a89bc14868fe64e3fa8a08","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-06-07 09:57:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a70b5f64a8ae1f0076069a8c0be9121ed536ff5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 09:58:02.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"78bb21e9eea6c23f32027d8552a0c0a3435fb89b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 12:03:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d5e82dd2f6d49c482a97249ad8d6010\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/981a6deb77b247ad91a12747b645004f : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/73e4cc81239c414a80077faf8855854f : SUCCESS in 6m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fcdfbe078ed44ede9950d0e12acc6864 : SUCCESS in 7m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afa410ff9a1c40d288f5de29e119fac3 : SUCCESS in 9m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30e7314497c34634ae86f3f5095fac95 : SUCCESS in 3m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5d18fc8abaae426d8ad83da285c1db91 : SUCCESS in 7m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f2592c65c3894273aa7dbfd68dc4cf56 : SUCCESS in 13m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fea566d008c94cfb9cc8bf31125dfaf1 : SUCCESS in 20m 10s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/83b55ec0286b4563aebe8b7c74baa9dd : SUCCESS in 1h 13m 39s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/5ac05eed1db04a1ebba7e627f13e61f2 : SUCCESS in 1h 07m 34s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/62bb0d28ebf548f493d7328f011f96bb : SUCCESS in 53m 57s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e02a6d955a3c4e6896a5fd062dc57b56 : SUCCESS in 55m 30s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e27907d79bf4cfca7de7340e315724c : SUCCESS in 55m 56s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/6694263eca2c4caf980cd050d13056c5 : SUCCESS in 23m 30s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/78fcfac91c0e4f5e93225fcf42620e7a : SUCCESS in 21m 53s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f44a4accc01946178c38eb033a070ad1 : SUCCESS in 20m 11s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/64bd890bfc5543e790856e95d21a6930 : SUCCESS in 13m 13s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cbf4c2c748fc4a9eb28949c5f0ffa775 : SUCCESS in 15m 42s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/86f5b3221f6142d79ec42fd13af16fb3 : SUCCESS in 1h 25m 19s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/92d15949c5734d098f6b73fbe9422deb : SUCCESS in 56m 48s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bc465128836c4b4da5a19b8377888f07 : SUCCESS in 1h 56m 10s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/975b12e8fcee496cbda2b989dd17783e : SUCCESS in 1h 35m 28s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/da590ca95c0448dd958baf4249554d5c : SUCCESS in 24m 07s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/95e2550e09e848fcada27c3c32eb0024 : SUCCESS in 22m 06s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ce91542a3aba4bf5ad16240ff9203bea : SUCCESS in 20m 45s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5c57b303f8049f593de14bdb42dd920 : SUCCESS in 18m 01s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a105ee3c5d034dceb397498c56f39edb : SUCCESS in 18m 54s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/be25415b943240538faad750130a2bef : SUCCESS in 32m 22s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/82ea69e17c924693a7394a2516fd8fa0 : SUCCESS in 28m 38s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/6426385335344df18d211489da662cb2 : SUCCESS in 18m 38s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/a0d10049546f4af3833bf3fd83a96c7e : SUCCESS in 23m 22s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fc193e64371a474b92f572b44be1c5f8 : SUCCESS in 13m 59s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e719ca8ca6c74fc8a463010c94ad8565 : SUCCESS in 13m 26s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2450cc86f3e246e8bcee9d884ad56722 : SUCCESS in 16m 33s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/d6aaff89de85422ab7e6c5de02397333 : SUCCESS in 22m 49s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/6f8d1cb349ef476d8e7af752e2f574bd : SUCCESS in 16m 55s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/78c1d0f8411f471d9cf7654fb9e32013 : SUCCESS in 18m 09s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1206ebd790cd41ba9a2386a987aff256 : SUCCESS in 15m 01s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/75faf359c2f94d56862f17fb62c7c99d : SUCCESS in 15m 59s","accounts_in_message":[],"_revision_number":2},{"id":"5b0fd30837184bdaa3e0646ed61872d28699baac","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 12:03:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"19b98b1b41c9c4d6115a4efb14f724421537d19d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 12:04:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4f50ff9921249549ffdb80ae6d1a1f4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09caa97a49144f7784e474158ec931ec : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/645113ba9f5d48b2ae37bd5ac7fe8d88 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8e26267e1d46271c5e7d42cb775641d27aa02d59","revisions":{"cd15405b168a005f633f30401fe073c59f5d68f2":{"kind":"REWORK","_number":1,"created":"2023-04-20 08:53:51.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/24/880924/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/24/880924/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/24/880924/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/24/880924/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/24/880924/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/24/880924/1"}}},"commit":{"parents":[{"commit":"2d4d609ee444c70965631579196be841f65a3103","subject":"Merge \"Fix merge action plugins verbose output\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2d4d609ee444c70965631579196be841f65a3103"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-06-09 19:54:07.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-20 08:53:41.000000000","tz":180},"subject":"Allow to use own repos.y[a]ml file to build containers","message":"Allow to use own repos.y[a]ml file to build containers\n\nFollowup the I0b07da22fea27e0ff4e90aaad19e50d84ff9a121 from Kolla.\nThis change allow to use own repos.y[a]ml file to build containers.\nMultiple Environments supported.\n\nChange-Id: I0d7ab0d8ff5b16ac5de8e50e63400bd455996555\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 070621a45df11fd2487ddc38f715c1075524ee93)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cd15405b168a005f633f30401fe073c59f5d68f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cd15405b168a005f633f30401fe073c59f5d68f2"}]},"branch":"refs/heads/stable/yoga"},"8e26267e1d46271c5e7d42cb775641d27aa02d59":{"kind":"REWORK","_number":2,"created":"2023-06-01 20:53:02.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/24/880924/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/24/880924/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/24/880924/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/24/880924/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/24/880924/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/24/880924/2"}}},"commit":{"parents":[{"commit":"74911c52daf74347cdc479c87bcd4a065a5bc535","subject":"Show an error when *_bridge_ports is not a list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/74911c52daf74347cdc479c87bcd4a065a5bc535"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-06-09 19:54:07.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-01 20:51:58.000000000","tz":180},"subject":"Allow to use own repos.y[a]ml file to build containers","message":"Allow to use own repos.y[a]ml file to build containers\n\nFollowup the I0b07da22fea27e0ff4e90aaad19e50d84ff9a121 from Kolla.\nThis change allow to use own repos.y[a]ml file to build containers.\nMultiple Environments supported.\n\nAlso, this change contains I209459ecf5b1a6e5f541cbc72d74aa7defb459d4\n\nChange-Id: I0d7ab0d8ff5b16ac5de8e50e63400bd455996555\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 070621a45df11fd2487ddc38f715c1075524ee93)\n(cherry picked from commit 7088ff2c2f10ba9897d1cd5424fea6d9268020a9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8e26267e1d46271c5e7d42cb775641d27aa02d59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8e26267e1d46271c5e7d42cb775641d27aa02d59"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
