)]}'
{"id":"openstack%2Fkayobe~880925","triplet_id":"openstack%2Fkayobe~stable%2Fxena~I0d7ab0d8ff5b16ac5de8e50e63400bd455996555","project":"openstack/kayobe","branch":"stable/xena","topic":"user-provided-repos-yaml","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-06-07 12:08:35.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-06-07 12:08:35.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-06-05 19:04:18.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-06-07 09:57:38.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I0d7ab0d8ff5b16ac5de8e50e63400bd455996555","subject":"Allow to use own repos.y[a]ml file to build containers","status":"MERGED","created":"2023-04-20 08:54:17.000000000","updated":"2023-06-07 12:09:34.000000000","submitted":"2023-06-07 12:08:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"880925-user-provided-repos-yaml","meta_rev_id":"3dd8ea69a9c26912452ac981952be60998eaf3ae","_number":880925,"virtual_id_number":880925,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-07 12:08:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-06-05 19:04:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2023-06-07 09:57:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2023-06-07 09:57:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-20 10:54:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-20 12:13:52.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-04-20 12:13:52.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-06-06 10:23:38.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"f384e37b9f4dc88a88e4a2c0bd1f537f16671f44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-20 08:54:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d58c53a11a48356ec32af5c5b6a41f098b94c759","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 10:54:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd933cba63884beaae9b17ff532bfd6c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8315c243d4d4375a7c5a52cc7ac567b : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ca6a0ee682145369d523f74b402a2ad : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/430438fa61a8473c882a26880958f85e : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e9b2ce2de8047ff8b233960463c0a94 : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ebd4aaaf7741461cb6527583f02218d6 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23230e9e665d4f90960938fc6d22ebf5 : SUCCESS in 8m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83c134d853c249f5ae402027dc49f8b7 : SUCCESS in 2m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/edec1c72d13547b5ab1deefea134696f : SUCCESS in 5m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/85df80569a7e413a9780ff6a9d86fe30 : SUCCESS in 12m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8282306bbe3746ad8a3e131d2df99c2a : SUCCESS in 17m 22s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/c96f018d76ea466587016fc8263f7666 : SUCCESS in 1h 04m 20s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/6dbd8694daba4f9697f0f9137bf06029 : SUCCESS in 1h 04m 13s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1c310a28065e4851bb4322f236e41f55 : SUCCESS in 48m 41s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/a7f16b83ebcd471aad37fb251c543c5c : SUCCESS in 1h 03m 29s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/5e4a97e517594563a47e5d04ec52e96a : SUCCESS in 25m 09s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/110e80d399664ff9a09cbcd55853f563 : SUCCESS in 22m 53s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f374eae6332c42cf96907c742c3af68a : SUCCESS in 17m 50s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ff3002d709fd4d9b8d14137ccbc15a5a : SUCCESS in 1h 48m 25s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a265f2bbe9a4c5c9d79734f7b90b03b : SUCCESS in 1h 05m 12s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/bf943d119bae4d4e8d0f62191e350151 : SUCCESS in 19m 55s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/03c2908f0618461ab2621972be5a5ba2 : SUCCESS in 18m 22s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee427b24f3374b12b7738828a77a3842 : SUCCESS in 14m 14s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/c821d8814f2b417aa2ee219d43f4d6c3 : FAILURE in 42m 10s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/83de177347c34a80818fdb14160e5b75 : SUCCESS in 26m 09s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e5d72b594f549b6bf0978ffa2873331 : SUCCESS in 22m 57s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5cae9cf2c0eb4a8e8597a1f07fd6a78c : SUCCESS in 17m 58s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/8164bc21d25e4d64a9aa232602b456f3 : SUCCESS in 15m 03s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3dd558bbefac482d8833c149bb9284bd : SUCCESS in 11m 17s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/37d90ef8b88b4626a0d490c9a6b47b0a : SUCCESS in 16m 25s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/f22be0a5b87847b29a63321d00c72a98 : SUCCESS in 19m 24s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c759837049de4d15b06e323f48069347 : SUCCESS in 18m 14s","accounts_in_message":[],"_revision_number":1},{"id":"e610770a0935c14df1827df15e8ce5b1798daf03","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-24 07:08:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b28a5ed826db4909c1447af1b5a12783aaeb79b9","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-25 11:22:19.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"b14f5edc41d96d3d99d0fa9fe73787dbb041db31","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-30 16:06:01.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"a881aee823d47c19a1306f0d0b90e1beac5d9254","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-01 20:36:24.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"ea5228bd932cabf08b3f9deb0ec701ff5c9f9f61","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-01 20:58:27.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"1a16ed3fb52af7f7a88ec9174bcc8295b5bca8d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 22:59:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b447e2db87b6473d81407fbd8a971e5e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55ba5eaba27c435c8afd8d3f80fd2460 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c8c3d123ee74b81ac6c4687f95a6c66 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2be6f625ac294a74887803e1be755b47 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3b5e6caf209413fa6a6bd9fe4b45895 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cfc17e6182824ca99af424670b5ac9f4 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64c0b3eef1cc4b7fa167360aa6d99bad : SUCCESS in 7m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/523b8b17425d4d7d950aa5028f85d244 : SUCCESS in 3m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d1510407ec2a43a7afde8d03c0141992 : SUCCESS in 4m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3e5cce7103fa4b91ba9f14ecfe4bb3ec : SUCCESS in 12m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9fd3151a79b146feaf22022ca25ad57f : SUCCESS in 15m 30s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d09f9e91caa54a0096ad660fbdad014b : SUCCESS in 1h 03m 25s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/03effd4a1ab84960b7dfccc12d8a33e7 : SUCCESS in 1h 09m 43s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af67387e90134aaebe14950b94b30fa8 : SUCCESS in 52m 05s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/2c18092b0b464ba0aad8693de199f048 : SUCCESS in 1h 02m 40s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e38772960a43435797d84b3c05577abe : SUCCESS in 25m 48s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/a1b225a651034ee9b2f2b5edffb1657c : SUCCESS in 21m 26s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f10fc84f6a30442ca7dd05c63e8e9c3e : SUCCESS in 13m 38s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fbae0e0039d14b808c6c70cb531c8288 : SUCCESS in 1h 56m 12s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b3f260a6f6b468e88fd33495cd36097 : SUCCESS in 1h 16m 49s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/46ea40b9a3854dad9503a0acbae40f08 : SUCCESS in 20m 55s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/49ebdea7859e4f5c90525b6b2d0683e3 : SUCCESS in 20m 26s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7dc8a6d60a894d8994f94410974fe590 : SUCCESS in 15m 39s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/483d51442efd4bb2ae8726ce16d55245 : SUCCESS in 38m 54s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/56f1ecdb92594a71b755c6e531667564 : SUCCESS in 26m 07s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b6a39d9960f485aa93c6494a44ecda3 : SUCCESS in 24m 31s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/6f898758e62d474ea80742efe6043c21 : SUCCESS in 16m 18s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/a19dcefea09e4b5a80e532860c045cc4 : SUCCESS in 16m 12s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba176dbfc9c14ecca03552f25d57f614 : SUCCESS in 11m 35s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/a0f68d89668545e487373f6f3b419868 : SUCCESS in 19m 49s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/ceaa50a331d54a46a6edc06861363e8a : SUCCESS in 19m 58s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/025edc2802cd40cbad9739cadd676c05 : SUCCESS in 13m 46s","accounts_in_message":[],"_revision_number":2},{"id":"2059129d969781d2c8884e1e71a3999e0d5237a4","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-05 13:25:25.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"56c8485fb86121ad3acb5df8e593ef7e1dee194c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-06-05 19:04:18.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3da7e3cf6be2295cd706117b3914ce67303f8416","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-06-07 09:57:38.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"88f3527a6e7c6f992810973d729da278e747b855","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 09:58:14.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ae7ebda9e0cff128898517f376bdd8b2d575cc51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 12:08:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f349a3866f04eac8d579fc5495b6181\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d11b2c669fc4df2a4cfdd21ea375cd6 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59699dce0a4743c8a65b39ed14b28853 : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4bb727ab77dd4aad9e5a8f0870bfe37f : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca495f0193064429b880fa9edac35178 : SUCCESS in 10m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a42cac3725bd4de0be516fdd3d16cf84 : SUCCESS in 3m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2f516158b3dc447a8368b181e9e31f86 : SUCCESS in 6m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9ae0bdb4ce7a430e9bd1ee5f993632fc : SUCCESS in 13m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2f605a6aae0b49c09054b096c320cdf4 : SUCCESS in 18m 34s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8c111a0fe7cb4fc284fe32f307ee29fb : SUCCESS in 1h 14m 04s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/7e395f386c2b481a84df58eb5aef453c : SUCCESS in 1h 11m 50s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/83df88fe6ceb49b888ff8c7eb2353ecb : SUCCESS in 58m 47s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/379333af6f3747c8a382e54b6c843797 : SUCCESS in 1h 13m 20s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/2fcfd699132545e19d31d39e8ed7451e : SUCCESS in 27m 26s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/b99b21d6f19d48c2b05f7b61019aafe9 : SUCCESS in 24m 04s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/19cdf9ea225f4b25a58a658c04670ba8 : SUCCESS in 15m 50s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1493cf5fb3c149c48aa1dcd8562b6927 : SUCCESS in 2h 01m 04s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/663c20728bec409cbc8aeb727caf2d3a : SUCCESS in 1h 20m 44s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/1d74317148894cb5a62171e17e6011c1 : SUCCESS in 22m 08s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/ea71b5549eb647cd989181bc75a5ac97 : SUCCESS in 21m 00s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c723e182b3b4f55875af8c510ef2fa3 : SUCCESS in 17m 07s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/02862ad3f76a4b4496cfc1d605dfd54e : SUCCESS in 32m 32s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/10fd5e098a0247478794fdbbfb5df141 : SUCCESS in 25m 13s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/33190ef25bc9420a997d14e2c20b179a : SUCCESS in 18m 00s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/0cfdc8e234d244178346a90725382b0f : SUCCESS in 17m 46s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91dbea4f705a4837a8e33e97729c0085 : SUCCESS in 11m 48s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/c1efa8c6d3b34cafb0a388a64d9db350 : SUCCESS in 20m 53s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/120fe789cbe24d339d54eb466672c4e0 : SUCCESS in 18m 16s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b4b379728c64a76bea9787f5aa2a024 : SUCCESS in 14m 30s","accounts_in_message":[],"_revision_number":2},{"id":"19998c0d5aae65df0d7435830b21aa45da9a179e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 12:08:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"3dd8ea69a9c26912452ac981952be60998eaf3ae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 12:09:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69ba33e3268a405dad321119367011f2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8366e39803dd49f5bf30a94dcc3842a3 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82831301b3894a83b1121b30f905e031 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3d92659a5006d5d5bf7e758280120f71f99adaf3","revisions":{"b4373aa29bdb9829fcf27ecddb00766a0f9aaf58":{"kind":"REWORK","_number":1,"created":"2023-04-20 08:54:17.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/25/880925/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/25/880925/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/880925/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/880925/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/880925/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/25/880925/1"}}},"commit":{"parents":[{"commit":"466651ab0242c67a4bdedfbb7cd78d38f7f974c9","subject":"Merge \"Fix kayobe_config_path description in globals.yml\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/466651ab0242c67a4bdedfbb7cd78d38f7f974c9"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-06-09 19:54:07.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-20 08:54:01.000000000","tz":180},"subject":"Allow to use own repos.y[a]ml file to build containers","message":"Allow to use own repos.y[a]ml file to build containers\n\nFollowup the I0b07da22fea27e0ff4e90aaad19e50d84ff9a121 from Kolla.\nThis change allow to use own repos.y[a]ml file to build containers.\nMultiple Environments supported.\n\nChange-Id: I0d7ab0d8ff5b16ac5de8e50e63400bd455996555\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 070621a45df11fd2487ddc38f715c1075524ee93)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b4373aa29bdb9829fcf27ecddb00766a0f9aaf58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b4373aa29bdb9829fcf27ecddb00766a0f9aaf58"}]},"branch":"refs/heads/stable/xena"},"3d92659a5006d5d5bf7e758280120f71f99adaf3":{"kind":"REWORK","_number":2,"created":"2023-06-01 20:58:27.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/25/880925/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/25/880925/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/880925/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/880925/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/880925/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/25/880925/2"}}},"commit":{"parents":[{"commit":"72950d2c6e737ae44950dd1e758a7cca212e9d83","subject":"Merge \"Bump mrlesmithjr.lvm to skip swap resizes\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/72950d2c6e737ae44950dd1e758a7cca212e9d83"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2022-06-09 19:54:07.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-01 20:56:41.000000000","tz":180},"subject":"Allow to use own repos.y[a]ml file to build containers","message":"Allow to use own repos.y[a]ml file to build containers\n\nFollowup the I0b07da22fea27e0ff4e90aaad19e50d84ff9a121 from Kolla.\nThis change allow to use own repos.y[a]ml file to build containers.\nMultiple Environments supported.\n\nAlso, this change contains I209459ecf5b1a6e5f541cbc72d74aa7defb459d4\n\nChange-Id: I0d7ab0d8ff5b16ac5de8e50e63400bd455996555\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 070621a45df11fd2487ddc38f715c1075524ee93)\n(cherry picked from commit 7088ff2c2f10ba9897d1cd5424fea6d9268020a9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3d92659a5006d5d5bf7e758280120f71f99adaf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3d92659a5006d5d5bf7e758280120f71f99adaf3"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
