)]}'
{"id":"openstack%2Fkayobe~881016","triplet_id":"openstack%2Fkayobe~stable%2Fyoga~I23584fe76554680251307d75078dec34e3a9d384","project":"openstack/kayobe","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-04-21 08:37:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I23584fe76554680251307d75078dec34e3a9d384","subject":"Revert back to MichaelRigart.interfaces","status":"MERGED","created":"2023-04-20 15:12:35.000000000","updated":"2023-04-21 08:38:20.000000000","submitted":"2023-04-21 08:37:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"881016","cherry_pick_of_change":880742,"cherry_pick_of_patch_set":2,"meta_rev_id":"50ec3bd668eecd06e37269e3a40f4b5fdafab3d4","_number":881016,"virtual_id_number":881016,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-04-21 08:37:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-04-20 17:09:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-04-20 15:30:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-04-21 06:37:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-04-21 06:37:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-20 15:12:35.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2023-04-20 15:30:50.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-04-20 17:09:59.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-04-20 17:13:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-21 06:37:24.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"46d4b38db1f3ddc402b73c3af9b3aaf6bf2d733d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-20 15:12:35.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"cd89525717066d632568a911513055d94aa24956","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-20 15:30:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"262b921afa6ee4e9a3b78f520394affbf9f25321","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-20 17:09:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a5696d5e5b1ac8888971b12c991f43f16b985a55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 17:13:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/661495b6a39d43eb822b76417e0fd829\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f1a1c613886478dbf63f276b308f3d2 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b89188333a5489b980ed968bc45221a : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e85d6ce3548a46c1afc55ad83e8974c2 : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cbf95a876ed4419eb9b50846f0cb74ab : SUCCESS in 7m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ec5ccf9f46846779a56000e995461d0 : SUCCESS in 11m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59484545152f49ce8841a75646222216 : SUCCESS in 3m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c5a36741d9d3437b8275e8476775ecd4 : SUCCESS in 7m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e48e7caefc1f442ca55a33cc3a8b4110 : SUCCESS in 14m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0465ffc11f2144a9bd06f928441ef70d : SUCCESS in 18m 07s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8a89b2757f8242ac9739b31874f366a3 : SUCCESS in 1h 29m 46s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/12b18b1745b84e4ba5e322fb945bfd7b : SUCCESS in 1h 02m 11s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/2a7366ec49a84741bc57fad4b2569c99 : SUCCESS in 1h 09m 24s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/5add6960396d4e61b1555e16f8f1a10c : SUCCESS in 56m 49s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00b26e3d4e9c4a62a317b16759810865 : SUCCESS in 53m 26s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e049f13b13344478a8c8878491e313f3 : SUCCESS in 55m 05s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/dffc974ccf2e4a6bb1be7c4e4371a722 : SUCCESS in 25m 23s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/cc42ba7c24bc475aa3190716201338d7 : SUCCESS in 23m 48s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/b168ad3aa775408fabd939e066d66879 : SUCCESS in 22m 33s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4913a8c53c8145648341e2bb32ccc2a1 : SUCCESS in 22m 50s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f0d069c23d44588a332f282469dff03 : SUCCESS in 17m 26s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c1fe5d80471d432b9d04a0c3fa679e40 : SUCCESS in 18m 34s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/af1bfe3356104c958869db3bac023341 : SUCCESS in 1h 24m 19s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c0254cfdfa644057adf3c433f47ed6e4 : SUCCESS in 1h 10m 15s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/66deb47bdc824b6695d03b919b028c5a : SUCCESS in 1h 52m 19s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4197ef2e8b96418e9f41b02107175394 : SUCCESS in 1h 19m 23s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c1f73a0844924ea0a3534aa1fcc52b64 : SUCCESS in 24m 58s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/538a26ccf7334dfcaf40220da994a104 : SUCCESS in 21m 33s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/2ff1ad452a2642c9a69eb1e2d26a8d74 : SUCCESS in 21m 49s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/990176948a3d45ff9bf97cd2164a3967 : SUCCESS in 19m 24s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/01e1b04766434a02a4a5e64a1204de0a : SUCCESS in 18m 23s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4fcf4e36954945dbb79adba846908285 : SUCCESS in 29m 32s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41768ba434b04a8491544b3f1f7fe4eb : SUCCESS in 24m 54s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2d6c306123f94f0fa5bfdad7b68b9d6d : SUCCESS in 17m 53s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/49eef658ad0f4e45ac8339f8edeb1d3a : SUCCESS in 16m 20s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d91b4b8493bc422998126ebf92ec8551 : SUCCESS in 15m 19s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b48134135e674b87a94e83f6d0839181 : SUCCESS in 12m 57s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fb159c452bd44811b67038453cd19752 : SUCCESS in 13m 14s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/5ac3ce35ddc0422682bdcbcd43070b1c : SUCCESS in 22m 04s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/5f0b4323ad524bb3958f22f6368983e8 : SUCCESS in 19m 30s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5f0a4a7337534ca0b46838feff434062 : SUCCESS in 17m 06s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c3c2b8ffaf31420d8ab139d34ea95d72 : SUCCESS in 14m 13s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6ea83132cf1b4f7da54f8f8a07fa2591 : SUCCESS in 14m 02s","accounts_in_message":[],"_revision_number":1},{"id":"457ffb50eb318f476a4e328d1e9ab7542b997001","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-04-21 06:37:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f0f1fdcac5a13da89769610ccc53358ec8af351c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 06:37:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"76fa0b97c71f595a2c6b62153fb2f4946ccd308d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 08:37:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d61abcdc327e43fcb65136599f12b56d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed5a0a963a364bbd9144c2fad5ff4434 : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5fff8cc9e0124f3db87c4f3523cf61a6 : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f72ca9ba6f0744d8b2dca13d1e60bee5 : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b9957f59a5e43bbab06b028ded92bd2 : SUCCESS in 9m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c9bb25f370249b8a722307676306387 : SUCCESS in 3m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fdde536ac84648c89d77d34de10018ad : SUCCESS in 5m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7fc7ce11bf3f46a2bd1e4a4dc1d32972 : SUCCESS in 12m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f5860affbcc44a48b31ec153d34096ef : SUCCESS in 12m 38s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/54a669d9d4ae4898878a45f336586ce3 : SUCCESS in 1h 08m 51s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/4dc3a47bcfcd4f17b2c4e51390ae69ae : SUCCESS in 1h 08m 54s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/af49d8b64cd84abaac1bd92a8e9f7a16 : SUCCESS in 48m 11s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e1b7f358a7b4a75b8aaad7a9a5ca13f : SUCCESS in 49m 29s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/303262372f004e7db9b2da8cc224ebb4 : SUCCESS in 37m 06s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/5346617c7e3a47628403f61adcc083b9 : SUCCESS in 25m 22s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/0f0c6abe1a2149ae81606afb0d2985f5 : SUCCESS in 18m 29s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ae608f27a06b406a920657481e0ad6d9 : SUCCESS in 15m 11s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23f981c4ba3e4088ae9f01656b084992 : SUCCESS in 16m 06s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2a00c84c3c304cc58cb0c5372011cd0e : SUCCESS in 15m 31s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b004a50f946e463b91ecf2abf68894bd : SUCCESS in 1h 10m 03s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/3d75417190034e0cb1e7255df9fd9a3f : SUCCESS in 1h 06m 30s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a3a6d92800154b369f450a67b0f5b016 : SUCCESS in 1h 54m 04s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd16a7dff7414952a1ca9d41cf9a02f6 : SUCCESS in 1h 18m 19s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d14007d1df734b7fb50d3485e1ea4588 : SUCCESS in 21m 30s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/89c1878bc10f42f29e7e815a378a7865 : SUCCESS in 20m 08s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/d1ebdfd7822342fbbf2848af9a8d85a1 : SUCCESS in 18m 15s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2748cf98e4594cd1a89e8c4e4fc7d7a6 : SUCCESS in 17m 01s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d0fda6165ade4dd09ffb6b8a92657e47 : SUCCESS in 15m 48s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7855e9c98bed4a06a9564dbe05c2f30f : SUCCESS in 25m 36s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94e9b490d9844ddca7176153dafaeea7 : SUCCESS in 26m 46s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/86239000e2f94e41a8ceed7530cb9be4 : SUCCESS in 17m 39s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/b7cbceecc86b42ee8b4789097cf8cc80 : SUCCESS in 15m 52s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/705e0062862a45e9b1be1ca2e5dd23ed : SUCCESS in 16m 52s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9862f40dd6674b55b56bec8c68e34218 : SUCCESS in 12m 51s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5a32c5439b6d4820a44ba5cf780aaf34 : SUCCESS in 13m 45s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/d26db82a76eb46ce98a0e14f23321a0e : SUCCESS in 20m 09s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/c54b871397bd4fdd80c672903cc46a83 : SUCCESS in 19m 34s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/268195ddae2c45e189dd25f9efff1f66 : SUCCESS in 13m 51s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4dbb7a70d9f04fd2a4e05a24d3553e27 : SUCCESS in 14m 52s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/076abf7c41b843e58a211abd04bc708a : SUCCESS in 14m 52s","accounts_in_message":[],"_revision_number":1},{"id":"4bc320cff280c0d4324cd4c93addbd6f57e14c6c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 08:37:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"50ec3bd668eecd06e37269e3a40f4b5fdafab3d4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-21 08:38:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b80c66427a749c69854aff1cc8c647c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e0c0aabc9ed4777a41ba616fbf5a181 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e695b6fc0474f638f908db8d4f978f2 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fba1f964e43c039290261b97b9edc7594f3288b4","revisions":{"fba1f964e43c039290261b97b9edc7594f3288b4":{"kind":"REWORK","_number":1,"created":"2023-04-20 15:12:35.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/16/881016/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/881016/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/881016/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/881016/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/881016/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/881016/1"}}},"commit":{"parents":[{"commit":"b394ff396365991b8d16aee179f5c40d0db59507","subject":"Fix maximum width of the DIB Multiline-YAML","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b394ff396365991b8d16aee179f5c40d0db59507"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2023-04-18 13:27:32.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-04-20 15:12:35.000000000","tz":0},"subject":"Revert back to MichaelRigart.interfaces","message":"Revert back to MichaelRigart.interfaces\n\nNetworkManager support has now landed in upstream, see:\n\nhttps://github.com/michaelrigart/ansible-role-interfaces/releases/tag/v1.14.0\n\nThis also fixes a couple of bugs in the fork we were using (see linked\nbug reports).\n\nCloses-Bug: #2016970\nCloses-Bug: #2016971\n\nChange-Id: I23584fe76554680251307d75078dec34e3a9d384\n(cherry picked from commit 3f4010d35ad7086d37fad04e224d781c77ef7453)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fba1f964e43c039290261b97b9edc7594f3288b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fba1f964e43c039290261b97b9edc7594f3288b4"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
