)]}'
{"id":"openstack%2Fkayobe~881679","triplet_id":"openstack%2Fkayobe~master~I209459ecf5b1a6e5f541cbc72d74aa7defb459d4","project":"openstack/kayobe","branch":"master","topic":"kolla-build-repos-yaml","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-06-05 13:24:12.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-06-05 13:24:12.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-06-01 10:11:17.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-06-05 11:49:31.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-06-05 13:24:12.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2023-06-01 11:21:13.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"I209459ecf5b1a6e5f541cbc72d74aa7defb459d4","subject":"CI: Fix kayobe-seed-images-* jobs","status":"MERGED","created":"2023-04-27 00:58:04.000000000","updated":"2023-06-05 13:25:10.000000000","submitted":"2023-06-05 13:24:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":23,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"881679-kolla-build-repos-yaml","meta_rev_id":"396e20265e1136ac99ab93c8551e2f564c636401","_number":881679,"virtual_id_number":881679,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-05 13:24:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-06-01 10:11:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2023-06-05 11:49:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-06-01 11:21:13.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2023-06-05 11:49:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-05-31 22:16:16.000000000","_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-27 02:50:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-04-28 00:42:00.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-04-28 11:20:41.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2023-04-28 11:51:07.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-05-17 13:44:54.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2023-06-01 12:28:18.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-06-01 21:13:58.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"61f00baa26990ce8092bda41cb8653ad2c3e2514","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-27 00:58:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"50010eceeba9b786a5d88fa6b2ed55454a050f53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 02:50:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d18aa601d3444c2a79fd1b5c71cfd94\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b39d3eb51304c87a184d9469bca5186 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02dc732732234024bca811eb386d98ff : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a4dcc2d78170481998429c6e31510676 : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1a3408f82e3747fe9ad03ec462e397dd : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/579b101dfb894208b1eedec3b95d9ac3 : SUCCESS in 5m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5d2548693ae74660b10d387daab69f9b : SUCCESS in 4m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e3a3a8934efa48e49264c1c83c568b23 : SUCCESS in 10m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7d1b1047079d42cd9417ded642ad6e22 : SUCCESS in 12m 34s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/0f74dfa7de394974957f0acb1dcb436d : SUCCESS in 1h 03m 52s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/118762efa30f4f94a5fb19b074fafa23 : SUCCESS in 58m 22s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/889bb4abf38042b8aa0c6ffdc20dad38 : SUCCESS in 54m 51s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/8abd71169c0e4f05be2ab7422e703de1 : SUCCESS in 1h 02m 39s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a8a6c1b1276b4639bebcb87172fb67a1 : SUCCESS in 15m 56s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/ebcd6961769c450e9872d7ea4dd8a0a2 : SUCCESS in 15m 40s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/af957bd3c1fc45c389dee7d47e20c681 : SUCCESS in 12m 16s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e639e146af374077992bdd5fd280cdf2 : SUCCESS in 1h 46m 51s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c9874360c974736ac18740bf58270ca : SUCCESS in 1h 12m 10s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/2b9e486b42bf4e2ab1a6321b522e1d75 : SUCCESS in 16m 48s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d8ba766435534616a63b2b8f729fba63 : SUCCESS in 13m 28s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/502facbde46541c4b949eface39568b4 : SUCCESS in 30m 08s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/cc20673fc9c845778a221e95bfeb6364 : SUCCESS in 21m 04s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/658f42928fa548408c831033c43d176c : SUCCESS in 22m 13s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7496d7a7c403491b95fe00b45377c01c : SUCCESS in 12m 34s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a5f842da9b984b44824afde8d8eb1ad1 : SUCCESS in 8m 23s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/cd2c1caa75bc47e1b525253f3b7ec5aa : SUCCESS in 14m 22s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c80fe56323394fa7ae64b40af337be57 : SUCCESS in 14m 05s","accounts_in_message":[],"_revision_number":1},{"id":"61c009e9edbdbe275d12a2f4b11dae0692746ffc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-27 07:44:32.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"0e735dfbb13f209427751fc0abb8fe96131a1aae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-27 07:49:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"57343e1225041065cb68f92a57f99681e0c0ab8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 07:50:15.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kayobe on branch master.  The error was:\n\n  expected a list for dictionary value @ data[\u0027templates\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      queue: kayobe\n      templates:\n        # - openstack-cover-jobs\n        # - openstack-python3-jobs\n        # - publish-openstack-docs-pti\n        # - release-notes-jobs-python3\n      check:\n        jobs:\n          # - kayobe-tox-ansible-syntax\n          # - kayobe-tox-ansible\n          # - kayobe-tox-molecule\n          # - kayobe-overcloud-rocky9\n          # - kayobe-overcloud-centos9s\n          # - kayobe-overcloud-ubuntu-jammy\n          # - kayobe-overcloud-tls-rocky9\n          # - kayobe-overcloud-host-configure-rocky9\n          # - kayobe-overcloud-host-configure-centos9s\n          # - kayobe-overcloud-host-configure-ubuntu-jammy\n          # - kayobe-overcloud-upgrade-rocky9\n          # - kayobe-overcloud-upgrade-ubuntu-jammy\n          # - kayobe-seed-rocky9\n          # - kayobe-seed-ubuntu-jammy\n          - kayobe-seed-images-rocky9\n          # - kayobe-seed-upgrade-rocky9\n          # - kayobe-seed-upgrade-ubuntu-jammy\n          # - kayobe-seed-vm-rocky9\n          # - kayobe-seed-vm-ubuntu-jammy\n          # - kayobe-infra-vm-rocky9\n          # - kayobe-infra-vm-ubuntu-jammy\n  \n      gate:\n        jobs:\n          - kayobe-tox-ansible-syntax\n          - kayobe-tox-ansible\n          - kayobe-tox-molecule\n          - kayobe-overcloud-rocky9\n          - kayobe-overcloud-ubuntu-jammy\n          - kayobe-overcloud-tls-rocky9\n          - kayobe-overcloud-host-configure-rocky9\n          - kayobe-overcloud-host-configure-ubuntu-jammy\n          - kayobe-overcloud-upgrade-rocky9\n          - kayobe-overcloud-upgrade-ubuntu-jammy\n          - kayobe-seed-rocky9\n          - kayobe-seed-ubuntu-jammy\n          - kayobe-seed-upgrade-rocky9\n          - kayobe-seed-upgrade-ubuntu-jammy\n          - kayobe-seed-vm-rocky9\n          - kayobe-seed-vm-ubuntu-jammy\n          - kayobe-infra-vm-rocky9\n          - kayobe-infra-vm-ubuntu-jammy\n  \n      experimental:\n        jobs:\n          - kayobe-seed-images-centos9s\n          - kayobe-seed-images-ubuntu-jammy\n\n  in \"openstack/kayobe/zuul.d/project.yaml@master\", line 2, column 3","accounts_in_message":[],"_revision_number":3},{"id":"246acf2b0f54c5161681af30fe7e62b3bdeea8d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-27 07:51:27.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"7446f9e972014168ccff7b2b7e8ea5d96478a381","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-27 07:55:09.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7ae4f6ac807dc46040c7da5ff5b6efcad33901bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 07:55:38.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kayobe on branch master.  The error was:\n\n  expected a list for dictionary value @ data[\u0027experimental\u0027][\u0027jobs\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      queue: kayobe\n      #templates:\n      #  # - openstack-cover-jobs\n      #  # - openstack-python3-jobs\n      #  # - publish-openstack-docs-pti\n      #  # - release-notes-jobs-python3\n      check:\n        jobs:\n          # - kayobe-tox-ansible-syntax\n          # - kayobe-tox-ansible\n          # - kayobe-tox-molecule\n          # - kayobe-overcloud-rocky9\n          # - kayobe-overcloud-centos9s\n          # - kayobe-overcloud-ubuntu-jammy\n          # - kayobe-overcloud-tls-rocky9\n          # - kayobe-overcloud-host-configure-rocky9\n          # - kayobe-overcloud-host-configure-centos9s\n          # - kayobe-overcloud-host-configure-ubuntu-jammy\n          # - kayobe-overcloud-upgrade-rocky9\n          # - kayobe-overcloud-upgrade-ubuntu-jammy\n          # - kayobe-seed-rocky9\n          # - kayobe-seed-ubuntu-jammy\n          - kayobe-seed-images-rocky9\n          - kayobe-seed-images-centos9s\n          - kayobe-seed-images-ubuntu-jammy\n          # - kayobe-seed-upgrade-rocky9\n          # - kayobe-seed-upgrade-ubuntu-jammy\n          # - kayobe-seed-vm-rocky9\n          # - kayobe-seed-vm-ubuntu-jammy\n          # - kayobe-infra-vm-rocky9\n          # - kayobe-infra-vm-ubuntu-jammy\n  \n      gate:\n        jobs:\n          - kayobe-tox-ansible-syntax\n          - kayobe-tox-ansible\n          - kayobe-tox-molecule\n          - kayobe-overcloud-rocky9\n          - kayobe-overcloud-ubuntu-jammy\n          - kayobe-overcloud-tls-rocky9\n          - kayobe-overcloud-host-configure-rocky9\n          - kayobe-overcloud-host-configure-ubuntu-jammy\n          - kayobe-overcloud-upgrade-rocky9\n          - kayobe-overcloud-upgrade-ubuntu-jammy\n          - kayobe-seed-rocky9\n          - kayobe-seed-ubuntu-jammy\n          - kayobe-seed-upgrade-rocky9\n          - kayobe-seed-upgrade-ubuntu-jammy\n          - kayobe-seed-vm-rocky9\n          - kayobe-seed-vm-ubuntu-jammy\n          - kayobe-infra-vm-rocky9\n          - kayobe-infra-vm-ubuntu-jammy\n  \n      experimental:\n        jobs:\n\n  in \"openstack/kayobe/zuul.d/project.yaml@master\", line 2, column 3","accounts_in_message":[],"_revision_number":5},{"id":"1e3c30fdb863ead30ed2ae202a409b9598ceea0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-27 08:02:20.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"dda7f1bf1c7cf1fd70cf957db61e14219bd1b5d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 09:42:26.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ced5bc4f36894f10a5ec08a6283c2007\n\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/6189f12d20644391af5b65d18b1a91a9 : SUCCESS in 30m 54s\n- kayobe-seed-images-centos9s https://zuul.opendev.org/t/openstack/build/cb56dbd12a0b4ba4bf36af9a13919677 : SUCCESS in 29m 22s\n- kayobe-seed-images-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/411723a19c8541079030639256d21dbc : TIMED_OUT in 1h 34m 45s","accounts_in_message":[],"_revision_number":6},{"id":"ff1125d85f92e84a1f420cc419c0e94255f8d940","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-27 11:34:16.000000000","message":"Uploaded patch set 7: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"df435ea3917eb30452bd40d95dfad166b680a903","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 12:13:46.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/944801eb12e94495858fe8cdb0c5b5e4\n\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/8de1dd6db0004f23a8230614d860b3d6 : FAILURE in 33m 05s\n- kayobe-seed-images-centos9s https://zuul.opendev.org/t/openstack/build/0915306fa7374f82bccfc53b64eb1147 : FAILURE in 29m 37s\n- kayobe-seed-images-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97c36aa3a6354b2083d019b8f5c2e445 : SUCCESS in 29m 40s","accounts_in_message":[],"_revision_number":7},{"id":"9ed04bb500c4e2066b2ced7f4f8b83b4ff0d8e78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-27 12:33:56.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"6a8ede6809b34305936a68fe4820a16701c41190","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 13:09:53.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bfb6aa7246d84d768cdaf0ee611d6e9b\n\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/6f5f2cb3fb2b48258fb2b84b650491b9 : FAILURE in 26m 45s\n- kayobe-seed-images-centos9s https://zuul.opendev.org/t/openstack/build/68d88d95c7f348f39cbfb21be7e3369c : FAILURE in 34m 10s\n- kayobe-seed-images-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ef01dc16aac24b63abf99eb5bdbfd169 : FAILURE in 23m 30s","accounts_in_message":[],"_revision_number":8},{"id":"43c0b828ccd276958338698527dc204f356dacc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-27 13:42:52.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"fb79ccd2cf3308df2f117b7b9cf74e7bc321c181","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 14:14:29.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d3fe3da12fe942119306b917ac038d2f\n\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/8de40b4ccdb348dcada3d61a899f26bc : FAILURE in 27m 37s\n- kayobe-seed-images-centos9s https://zuul.opendev.org/t/openstack/build/65097ef062ac4273b332bccbc394c4ae : FAILURE in 27m 33s\n- kayobe-seed-images-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4abff6b1de0648fda7067b42080538d7 : SUCCESS in 30m 57s","accounts_in_message":[],"_revision_number":9},{"id":"dc0609279b2cfdf4ae50b28aa4dc8983281bd6bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-27 14:33:10.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"fe470b3416132762c0495352e5cdab28518fa38c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-27 15:35:56.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"045e9fae448af27660f9a8329b86510d6da4e999","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 17:14:27.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cf18d6f05e50403d91675990ad33a52c\n\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/ee36e04f708f44a89307383fe2609cfd : SUCCESS in 27m 06s\n- kayobe-seed-images-centos9s https://zuul.opendev.org/t/openstack/build/4c7fd3ab05134df6802fa451feb30c33 : SUCCESS in 32m 20s\n- kayobe-seed-images-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca53b8c6214a48ab8df0f4ef749fe9a4 : TIMED_OUT in 1h 36m 52s","accounts_in_message":[],"_revision_number":11},{"id":"693bda1f4abb9ce1e08213399c21ca3f8b714120","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-27 23:08:13.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"75793e2f64b67eeda7ef352507bce91775b7abea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-27 23:10:20.000000000","message":"Patch Set 13: Commit message was updated.","accounts_in_message":[],"_revision_number":13},{"id":"d5b9fbedd56d4f8596ca1b69eb52a199c910100f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 00:58:12.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36f10177a3594c25a59f76ea6fa2010d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a360d03d76884fafb872b114ecbece89 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08b0b15cd6564798890560da7c54e920 : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5bce2be4b1dd4cfb8f0f9247179f2569 : SUCCESS in 4m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a0c4069992b946f5ab4145f06ef74ea6 : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e59e63728b6e429cbd289cc5dcdbc65e : SUCCESS in 7m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/50406402be364798a622b792d17b21fd : SUCCESS in 5m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9a4ad1633b1443da92da58594a2d4859 : SUCCESS in 11m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4030fd389e244a7ebeca7b43fad62566 : SUCCESS in 13m 39s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/57aa803fdfe74ba28bbb641595e9f30b : SUCCESS in 47m 27s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/372042311aa34bf287709cad343e88b8 : SUCCESS in 1h 00m 53s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5492b0f257a342dfbbdb49bc448eaf1e : SUCCESS in 56m 54s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/5b243d80cbbc4dbaa3909c40c90dc752 : SUCCESS in 1h 07m 10s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/8f39b55217b044a7bd5d38bce5333c8a : SUCCESS in 16m 35s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/3c6bc121cc13451cb6636d984b732228 : SUCCESS in 19m 43s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c6f18f6e2ec940529afdc2cb83acec10 : SUCCESS in 13m 39s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9dac60bd5b5547099c86ca7bb63a912d : SUCCESS in 1h 33m 33s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d4eb53db8d654b36bffb3520ff40d196 : SUCCESS in 1h 29m 32s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/05348fd43e864a98aafe20449473b817 : SUCCESS in 18m 55s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/57120824493245d8a77e89e9b074abef : SUCCESS in 12m 15s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/0a22ed541e68458e96949550df98f6ab : SUCCESS in 29m 17s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/101b9042b126474faea9df17545f49b3 : SUCCESS in 30m 25s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b95efe89e8ee42e1b49eb18b3a075434 : SUCCESS in 22m 56s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3428d36a6d2e4c02bc223fadf0897aee : SUCCESS in 12m 40s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/966f7103b83945d08b1dac6adc148674 : SUCCESS in 14m 55s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a106217b29d546e09d67020828df40d7 : SUCCESS in 17m 59s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc03fb974d9e4e5fa9d4e687bd23f91e : SUCCESS in 13m 55s","accounts_in_message":[],"_revision_number":13},{"id":"129f2828671a6b40fdd3b66aa4fa2ce17b740d64","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-04-28 11:20:41.000000000","message":"Patch Set 13: Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"ce774a184314fd83196c4af0e82901ccec3657dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-10 14:17:52.000000000","message":"Patch Set 14: Patch Set 13 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"147f80adaa740762b93420e88e5500331f31cbce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 16:03:03.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/049a479728c64bbbaa0e4669c6ef0ac7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8215d9a7cf4455095a9da2a5e23df7c : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8a8f62bd8eb420bb26f6ca5460dfef7 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/045de0e10af24354b7307134a15b59c9 : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/089e9cdd799d4aabb018c306e2abc207 : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/789ea548142145e0a114374e6f8248a9 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97a3d2ed99a64e599ea867f8b9414d49 : SUCCESS in 7m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6c9191c35c63443db12ab1ee96355e0b : SUCCESS in 4m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/174bfaff6c59422198f31bbc0c22f6c0 : SUCCESS in 8m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/29cb6b1bbb8543348c2a443104665bdb : SUCCESS in 12m 23s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/bdae127a6511460f8e21b11f9eae46e2 : FAILURE in 51m 58s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/bbe3bc38b88e4d1abbcec08bd3f98267 : SUCCESS in 1h 04m 17s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cdae1d8bcc384aaa829f1e6cdae636d6 : SUCCESS in 53m 08s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/86148b5730d844a3afcecc4c4008fc68 : SUCCESS in 1h 01m 54s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/1c4269cd02bb40a7b3c6d14683501137 : SUCCESS in 16m 43s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/be6643244ac44d9aa100e66c9ef1ada9 : SUCCESS in 18m 55s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b108ddc26bc94b66935ef89711b48f06 : SUCCESS in 15m 14s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/60e79bc645aa42e4a07874452f950a88 : FAILURE in 49m 10s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa146e12b7724239a126e11f65019217 : SUCCESS in 1h 14m 08s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/696fcce373a044e48c6798dbe4ee43d6 : SUCCESS in 14m 56s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b76f8352fb4d41758d663a9f5db35318 : SUCCESS in 15m 24s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/b66a9bfe11c0450f9423c15868d184c9 : FAILURE in 34m 41s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0384594172f749cb861df3eac8469d88 : SUCCESS in 27m 59s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8bde48534c4344d5ad9d7721850a115c : SUCCESS in 21m 07s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5f4f85b693184b30be14e4603c4fbbbe : SUCCESS in 12m 30s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f42fb52700c42239f075b2fbe94d020 : SUCCESS in 10m 11s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/58989801c1c242058010d637e2e83a26 : SUCCESS in 16m 06s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fae59af928d84876a747ecebe2f325fa : SUCCESS in 13m 26s","accounts_in_message":[],"_revision_number":14},{"id":"e8f9ff7419b69ed6341b57e7cb7f0ad904ba44a8","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-10 17:59:41.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"d54da1a09a6592f4563fa420fa3e3e2d932345c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-10 19:40:02.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dba04bf229904123a23c5ce692ca06db\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f33d727ebf2b47a8a7b111d194aa3dec : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5060dcb9aa244f0c842456d6b6f43fcb : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e83944441cac492e99170f25375acd1a : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c470d0b18e7e4bb8a4471934f59c48a7 : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2f34cb550d154b9c912f6c9065e55168 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/271dfb8e249745329849e24f2991d08e : SUCCESS in 6m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/958dd2b5408342c8974d39c103e8d39a : SUCCESS in 4m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/13f975a574d54998a0753998e1399ae3 : SUCCESS in 10m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fb15758cc6b94629bdcff869acff3002 : SUCCESS in 14m 00s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ffbe968860ef4666b9b91b1e23eee87d : SUCCESS in 59m 10s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/4efba44740104d5ea9916b7a28857d0e : SUCCESS in 1h 01m 22s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5228d064f35c44c5971523a3ca40edf7 : SUCCESS in 52m 59s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/5f5e9c0bbbb9463fa06c9c4be76610f4 : SUCCESS in 1h 02m 54s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/04d01d04b04c481a90599809bf421fab : SUCCESS in 12m 57s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/a4258dbe50b64c20a3bc5b50d24950fc : SUCCESS in 16m 35s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ab29638ca5e04d68b4bbc6750abea75f : SUCCESS in 11m 56s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d542efba99c44c6f94f3be2b24ed5bbb : SUCCESS in 1h 20m 39s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c78d76d04ea940a0be2b634741f0aec6 : SUCCESS in 56m 13s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/6901c69dbe394c80b8abcda29397f31d : SUCCESS in 16m 18s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/686239325b7d4de899ebcd3f9b295245 : SUCCESS in 11m 39s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/6e44459132794cb4b871158df84a83ed : SUCCESS in 25m 20s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/045cafcd949148e3b6e3fbd6db227acd : SUCCESS in 20m 59s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/509910052f224be2b7821cd13a3bec52 : SUCCESS in 22m 13s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1296c4354daa4feab3b6ed213c117357 : SUCCESS in 9m 58s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f2bf2107f94243c0bac33faa72c4e641 : SUCCESS in 10m 15s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/789b169cc04c470a9d3baf44e28d4b4b : SUCCESS in 17m 43s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/791dbd243f344e1fb2f7168a996fb08c : SUCCESS in 13m 23s","accounts_in_message":[],"_revision_number":14},{"id":"f5ff16136faee485a9487bff14e23968ef8b283d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-17 11:32:29.000000000","message":"Patch Set 14: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"3cafe6c23da4390ed8ec0780b0d0252104111ca3","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-17 11:55:02.000000000","message":"Patch Set 14: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"f93122538a1be582ec1f6076213f68db1f52eafb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-05-17 13:44:54.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"ae5e2c8ba6ebb5bcfa339c45e86777ba5e8f8d27","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-17 13:48:32.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"cada3729879c1dfe5f33de70ba51252262d4684a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-05-25 15:11:46.000000000","message":"Patch Set 14: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":14},{"id":"98b4b17893da761f0a4d0d719c60ff5270e0af0e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-05-25 15:12:06.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"3150826c69d5f28060dfea123e8a2347177fc3b9","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-25 21:47:46.000000000","message":"Patch Set 14: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":14},{"id":"3db0b0c779d1b8c6632eea0ea7837387091fd376","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-31 13:35:36.000000000","message":"Patch Set 14: -Code-Review\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"9eeff3d2966b34b3f21e50b27a6dee761cc2921c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-31 13:47:53.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"f9cd528b0e253c9e89d3ea5464f8642045d37671","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-31 14:20:35.000000000","message":"Patch Set 14: -Code-Review","accounts_in_message":[],"_revision_number":14},{"id":"f839c09ea7ae10d5c4ac3f3678b7319ae002c363","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-31 14:20:46.000000000","message":"Patch Set 15: Published edit on patch set 14.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"8cf8a6dbe3cfcc1a5c77c60700804c4e5c09242b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-31 14:22:46.000000000","message":"Patch Set 15: Backport-Candidate+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":15},{"id":"9bef9cfbab29f554c88388a72983e56c06f0557a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 16:05:26.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7b303e0f73c64ff996f2e92c6cf3cae2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cdc614f745fb40bdb4b7875dccf6d005 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5947fc3e31140d0bcfa082fac1e9410 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d47c563dc13b4a219958eab93c6e56df : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a46fef53c3894fa39cbed1f15f223c70 : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/132397dd75604418964e466df699a08c : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/706f2f125e06451ab4f741ccc3f53771 : SUCCESS in 7m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/aed279d5e770483f935243e1749ed162 : SUCCESS in 4m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2a11850ebb9f41379cf9a52bac643385 : SUCCESS in 11m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/640408fa097e4a35baa5d7f1b51f662d : SUCCESS in 13m 58s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/81af49f5a84b4f2babfe55bab9c9d791 : SUCCESS in 53m 10s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/6e572f79badc4abeb5e71fa7f50354c5 : SUCCESS in 58m 59s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/57d8e717d65447fea9e27641c23e9d6a : SUCCESS in 57m 35s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/036c8f8403d44f3d922a09199bc723b9 : SUCCESS in 1h 13m 33s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0b27bbde1ae64c8f817cdd47e8fc0cfa : SUCCESS in 20m 43s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/7a141cb29125458d8310ab6f47c17aba : SUCCESS in 19m 01s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/44c3adf532054ac1927603031c8f69bc : SUCCESS in 33m 57s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/593f54f1094040dc9d85a49f22ec72af : SUCCESS in 1h 35m 48s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e44aa174091b485fa282637bbef4372b : FAILURE in 47m 54s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/9a43f0255054493692e3525c7a7a35ee : SUCCESS in 17m 02s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d44aa251b974440c99611a39e177358a : SUCCESS in 16m 18s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/2118f8c0d40b498893aa84de1fe6c95c : SUCCESS in 30m 52s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f77782c61aa34302ab27dbeb4e2bafb3 : SUCCESS in 28m 14s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e7f80cb0e8b4a60877da6006bff83e9 : SUCCESS in 24m 51s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/cca73020033d4c75a238695675d5163a : SUCCESS in 14m 54s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f066334685af41fea50e24fad158325d : SUCCESS in 12m 12s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b6400d199bad4c1a9979ad3d530763bd : SUCCESS in 32m 16s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/79c28599873e42eba0499d69872722e0 : SUCCESS in 26m 03s","accounts_in_message":[],"_revision_number":15},{"id":"a2492869b6a854da55f6f6062d932151ca852aa6","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-31 16:14:16.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"72b1b3c49fbde7e3c6557eaa7797a7f0e35f423b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 18:18:33.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1195892aa10b4e96991589ca41674a85\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e4f30ab1e1040c29ce2f9d2b354881a : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4ad829cfb874d359ab84176158bb498 : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89514f054aa748e88daa7be5fc4bf951 : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e89b7be36c184d7bb62c23e4d6d091bc : SUCCESS in 4m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dfe6ac74df3a4389ba6a4d3f1d92f159 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/973478939c204993a5704b1471ed49b5 : SUCCESS in 8m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/124d78542b8f41e1ad64cfd6429d06b8 : SUCCESS in 4m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1f46ede2f737448087f34ded18236cf6 : SUCCESS in 10m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9e36421ffe67474ba140138fde01fa30 : SUCCESS in 14m 55s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/77146993b40e441c84436dc227c8d59a : SUCCESS in 56m 30s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/c16d00b3b6f1407baced4b562ea89559 : SUCCESS in 52m 33s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/89fe8362b2d6469cb2d4b48efc8a6344 : SUCCESS in 55m 48s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/078255f8df374b34a13da68bc04ed1f2 : SUCCESS in 58m 23s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/064016c7e1f6494ab16ef475322929e6 : SUCCESS in 17m 21s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/1797edbff6824e52bcccbf420ee2516f : SUCCESS in 17m 58s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/680f7a3f9d624a2e8de99b52b4945675 : SUCCESS in 14m 50s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/63f7076e490144128e091838a6216c7a : SUCCESS in 1h 31m 20s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5a808024df8846d29efb26fc4493dfb5 : SUCCESS in 1h 54m 10s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/cb42237f8a774a61bd56fdc9a0a00e2f : SUCCESS in 18m 14s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf19c4d2ccc14cba9cdcb92d0ee787f9 : SUCCESS in 15m 37s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/e86f893b28534ba3a5aa5b9cc1276115 : SUCCESS in 33m 10s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1e681a51fb714892bf280812b0f1674c : SUCCESS in 44m 07s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a3d26c5a94a347f4b360457c6ac2d7a9 : SUCCESS in 22m 47s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3bad4ad7c57944ab8016263b866dcc88 : SUCCESS in 15m 11s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/68e715f334ff48fcb97f9319039b4dd3 : SUCCESS in 9m 36s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/94cc2bdee8874e3a96ee8baaa84308f0 : SUCCESS in 31m 57s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff52e4ceff434581bf4b9ed74bdf0b33 : SUCCESS in 26m 44s","accounts_in_message":[],"_revision_number":15},{"id":"c519a7df32020b370c8a76d2f7ac0b8ef4fdc806","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-31 19:22:38.000000000","message":"Patch Set 16: Published edit on patch set 15.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"ec2f61c31605315a9a118a9327f6e8455875b2a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-31 19:24:15.000000000","message":"Patch Set 17: Commit message was updated.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n","accounts_in_message":[],"_revision_number":17},{"id":"d076a01bced53d40790b3fea5c6bb7e44a94dc72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-31 19:28:33.000000000","message":"Patch Set 18: Patch Set 17 was rebased\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n","accounts_in_message":[],"_revision_number":18},{"id":"54950aa241991b02f884746dba35a30669bf2830","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 21:02:22.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7936b5f2031c49749948bb63f0f81787\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8958483163724ea2ad28dc1f55174142 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4d76ad73b21415f975ef5186055dcd4 : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a8ae2a1d9424615ac3ea3e66e734ad0 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f60a9e6059bf44d094a8647030d0fdc0 : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a0786c342b2240a2adb3e27470bb0230 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9411f87ccef48d4959bfa2fe9fdad30 : SUCCESS in 7m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/64b7e9b45d454078a6441a1920579ac6 : SUCCESS in 5m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0a2cef70752e4e69a7250f795789ae8b : SUCCESS in 12m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/caff709b408548eeb04c9681c26d5819 : SUCCESS in 13m 34s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e458b511ed5e426e923efbabdc457ae1 : SUCCESS in 48m 27s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/fcb747eec4f84bbbb6b43e61e0390710 : SUCCESS in 51m 09s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7acec635f60c40f2afa02b46f4a5f6c1 : FAILURE in 16m 30s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/9d2eda21fff74ce38797afc6ab471a69 : SUCCESS in 1h 02m 05s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5201f610efc94256b46b09233e2b4a37 : SUCCESS in 16m 22s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/91a682dab909427485b352c7e61f6a29 : SUCCESS in 17m 00s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4fa86e9df08b4e1aab1e20b2e7aa904f : SUCCESS in 13m 54s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9807efa1cbae44778108d1eb45b1687d : SUCCESS in 1h 30m 17s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9be9d99813b34e509d9ae2d3b5fc5cd7 : FAILURE in 1h 28m 36s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/640019bf0fbf4ce89411c382dc52fa32 : SUCCESS in 15m 33s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f1f882a958bd40c8aff58a0b77105147 : SUCCESS in 18m 09s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/9905f6390825449ca3ac3cb32cd74f9a : SUCCESS in 34m 00s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/26b9356358a84376b6f6ba3402ebe55c : SUCCESS in 29m 31s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97e27095339644379ad3f99aed9d3f7e : SUCCESS in 20m 30s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f01225c77a164463b3ee326d00584fdf : SUCCESS in 13m 21s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7facf35b049a41e993b2f7bc3b9f7681 : SUCCESS in 12m 21s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/81008a6e6ff84e9ab24d8deb2adbdaec : SUCCESS in 34m 36s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2cdab11049114717b3634eae803291b3 : SUCCESS in 21m 00s","accounts_in_message":[],"_revision_number":18},{"id":"a6f97f5f6a6c17106252ae480d826816c40ff4e8","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-31 22:16:16.000000000","message":"Patch Set 18: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"a6720e1fc0a0288bed7b86d07557aadd03b7dca1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-31 23:53:03.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd58a4eb33b84c7a8be8bcc1607d949d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7196d1806eec474a9418e0cba70f1be5 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d51950b2db144e0cb83026abe7c03766 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ae069f0f23e4cffb3b0220be15d99c8 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff1d3681b79a46f0a8b65e71c7f34800 : SUCCESS in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4b06e94645134634a6bc7b52d576ecc1 : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd0f775adbf34bda85ed76c7d89608e6 : SUCCESS in 9m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fc76d73b15214bc2846e55851eeeb46c : SUCCESS in 3m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a8af3ed2b0d248db83527ce8ae27f9dc : SUCCESS in 9m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e3eee0989ee54bb3b31726cfa5c25fa9 : SUCCESS in 11m 13s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/bfa1bf613e634ad193a6ce6d7d1172e4 : SUCCESS in 39m 01s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/fcbf29ad2aa342958b83a598fb5e85a5 : SUCCESS in 42m 31s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39dcc49217ed4ce8b061b9c0aeb87ea6 : SUCCESS in 41m 06s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/9819e1ff9e9c4f5ca21dba1f32788025 : SUCCESS in 54m 27s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9798f1d6ed514496b979171fa42b2b9a : SUCCESS in 18m 50s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/f670798741564a78aac086aafda2d6e9 : SUCCESS in 14m 53s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aef55b42a22e42c783d6239c031ba319 : SUCCESS in 12m 56s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3af72d8b275140a3bfdde7a7b5f4a838 : SUCCESS in 1h 36m 16s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cb2a7194c3a44f06985db282a00d8c42 : SUCCESS in 1h 15m 31s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4d8d998b668d44559eab53d85e77ea42 : SUCCESS in 13m 52s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/769b34177e4143889c4fb44e7702fc84 : SUCCESS in 11m 45s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/afb4b63c3b254c73a5c5acf176d5d96e : SUCCESS in 27m 37s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2c5904cd18e547db8e767946c6e26066 : SUCCESS in 28m 32s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e0e15bee9ff1498a84cfdba5585ddfea : SUCCESS in 18m 31s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/00bc7f96db98412484b533fc73d15ee6 : SUCCESS in 17m 13s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce38a63d71924fa5938819e179ce64cd : SUCCESS in 9m 32s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a06b939b07ca43708d771c1c63398bb0 : SUCCESS in 33m 55s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e3f8c86ffca34f2495b800858c6fe6bc : SUCCESS in 33m 07s","accounts_in_message":[],"_revision_number":18},{"id":"e64304b3a56ddde40dcb2e8601d52bc4e21d235b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-06-01 10:11:17.000000000","message":"Patch Set 18: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":18},{"id":"e27188410fb7fb81786c51c3e493bed4296ca872","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-06-01 11:21:13.000000000","message":"Patch Set 18: Code-Review+1","accounts_in_message":[],"_revision_number":18},{"id":"91d61bc266617ae8c1db12cc7fcc5b03776da3e6","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-06-05 11:49:31.000000000","message":"Patch Set 18: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":18},{"id":"b1711848d86b454cba73c433fa1af125e8dbb75f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 11:50:00.000000000","message":"Patch Set 18: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":18},{"id":"69934edbf275f8ca4f7423ae0a402eda4b1d356e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 13:24:11.000000000","message":"Patch Set 18: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1807bedc310436ead1a2b73e42104fd\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8a4f4e8103849519009a0dc0242d771 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d532785fee7242a08b4292cbd25cfc6d : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9458a90f63164daca65335411eae4e1e : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a5a0139145c4478f96cfec8bf060846d : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a8fc16791fb46aaad0d6085043e8c9a : SUCCESS in 6m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7817bf208e5e48f5b78436e9d3b9b22b : SUCCESS in 4m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cd36239d8712414ea52193f06e8f6487 : SUCCESS in 9m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/513d3db590714d9e9081d342a93ce6ac : SUCCESS in 12m 12s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ac922e32b6cc46de8063c1a483760890 : SUCCESS in 1h 03m 14s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/509362cf5b1c4f1f8ff9f6617e20bcc0 : SUCCESS in 50m 28s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/874ba2a3a4a54567b2eb717697a95658 : SUCCESS in 50m 52s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/1d65daecf8a14b29ac2c09f844d1db54 : SUCCESS in 13m 20s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0d929598390b44239a5c20549df36ed6 : SUCCESS in 14m 32s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/1bdf953a390948cda53e79e41155d858 : SUCCESS in 1h 19m 29s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d04b8d58ac2e4ee2a57600d24d7e025a : SUCCESS in 1h 22m 02s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/7e63df784eed4909b2992b8d243ecb7c : SUCCESS in 13m 51s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3bd24f1074cd4df1a4c5ddbe14d772e9 : SUCCESS in 14m 22s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a255cbb3496e42538a0aef5fc406667c : SUCCESS in 21m 49s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c989cdef71e4357b2c505fbd44ba577 : SUCCESS in 20m 44s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/97af0911e0174035a24e75575b114c95 : SUCCESS in 11m 40s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9893e4a29d394bd0a689c45674703bbb : SUCCESS in 11m 34s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/08c859935a374172a3df7d2d528c3ca6 : SUCCESS in 23m 57s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/287e5e3a788449db969ad4153fca5b97 : SUCCESS in 30m 45s","accounts_in_message":[],"_revision_number":18},{"id":"8f14d1e911e6d19b35d6363885508478ca323341","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 13:24:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":18},{"id":"396e20265e1136ac99ab93c8551e2f564c636401","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 13:25:10.000000000","message":"Patch Set 18:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3291aed487a9437088db8285ff539509\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3e3aeeae1d743c5be6f238f209092f0 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":18}],"current_revision_number":18,"current_revision":"7088ff2c2f10ba9897d1cd5424fea6d9268020a9","revisions":{"135df18139e33014f015fc3520a3be77e6f7fc28":{"kind":"REWORK","_number":1,"created":"2023-04-27 00:58:04.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/79/881679/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/79/881679/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/79/881679/1"}}},"commit":{"parents":[{"commit":"9ba7cfc9e72ddb2dd3c0ef1a44c2a04ba8709aac","subject":"Merge \"Revert back to MichaelRigart.interfaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ba7cfc9e72ddb2dd3c0ef1a44c2a04ba8709aac"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 00:57:00.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 00:57:00.000000000","tz":180},"subject":"[DNM] test kayobe-seed-images-rocky9","message":"[DNM] test kayobe-seed-images-rocky9\n\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\nChange-Id: I209459ecf5b1a6e5f541cbc72d74aa7defb459d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/135df18139e33014f015fc3520a3be77e6f7fc28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/135df18139e33014f015fc3520a3be77e6f7fc28"}]},"branch":"refs/heads/master"},"d854d7987c9bdee84f35bd5d8293dc7395a376db":{"kind":"REWORK","_number":2,"created":"2023-04-27 07:44:32.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/79/881679/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/79/881679/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/79/881679/2"}}},"commit":{"parents":[{"commit":"9ba7cfc9e72ddb2dd3c0ef1a44c2a04ba8709aac","subject":"Merge \"Revert back to MichaelRigart.interfaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ba7cfc9e72ddb2dd3c0ef1a44c2a04ba8709aac"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 00:57:00.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 07:44:14.000000000","tz":180},"subject":"[DNM] test kayobe-seed-images-rocky9","message":"[DNM] test kayobe-seed-images-rocky9\n\nChange-Id: I209459ecf5b1a6e5f541cbc72d74aa7defb459d4\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d854d7987c9bdee84f35bd5d8293dc7395a376db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d854d7987c9bdee84f35bd5d8293dc7395a376db"}]},"branch":"refs/heads/master"},"8f4c03756d29820f27da138287f96a1cad879d06":{"kind":"REWORK","_number":3,"created":"2023-04-27 07:49:45.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/79/881679/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/79/881679/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/79/881679/3"}}},"commit":{"parents":[{"commit":"9ba7cfc9e72ddb2dd3c0ef1a44c2a04ba8709aac","subject":"Merge \"Revert back to MichaelRigart.interfaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ba7cfc9e72ddb2dd3c0ef1a44c2a04ba8709aac"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 00:57:00.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 07:49:35.000000000","tz":180},"subject":"[DNM] test kayobe-seed-images-rocky9","message":"[DNM] test kayobe-seed-images-rocky9\n\nChange-Id: I209459ecf5b1a6e5f541cbc72d74aa7defb459d4\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8f4c03756d29820f27da138287f96a1cad879d06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8f4c03756d29820f27da138287f96a1cad879d06"}]},"branch":"refs/heads/master"},"f61ddcd9b37687182a7d62057bc66ed00f57f63d":{"kind":"REWORK","_number":4,"created":"2023-04-27 07:51:27.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/79/881679/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/79/881679/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/79/881679/4"}}},"commit":{"parents":[{"commit":"9ba7cfc9e72ddb2dd3c0ef1a44c2a04ba8709aac","subject":"Merge \"Revert back to MichaelRigart.interfaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ba7cfc9e72ddb2dd3c0ef1a44c2a04ba8709aac"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 00:57:00.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 07:51:18.000000000","tz":180},"subject":"[DNM] test kayobe-seed-images-rocky9","message":"[DNM] test kayobe-seed-images-rocky9\n\nChange-Id: I209459ecf5b1a6e5f541cbc72d74aa7defb459d4\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f61ddcd9b37687182a7d62057bc66ed00f57f63d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f61ddcd9b37687182a7d62057bc66ed00f57f63d"}]},"branch":"refs/heads/master"},"4066127f2d37da33ffbaaf1e5bd80c7c86a6515a":{"kind":"REWORK","_number":5,"created":"2023-04-27 07:55:09.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/79/881679/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/79/881679/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/79/881679/5"}}},"commit":{"parents":[{"commit":"9ba7cfc9e72ddb2dd3c0ef1a44c2a04ba8709aac","subject":"Merge \"Revert back to MichaelRigart.interfaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ba7cfc9e72ddb2dd3c0ef1a44c2a04ba8709aac"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 00:57:00.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 07:54:54.000000000","tz":180},"subject":"[DNM] test kayobe-seed-images-rocky9","message":"[DNM] test kayobe-seed-images-rocky9\n\nChange-Id: I209459ecf5b1a6e5f541cbc72d74aa7defb459d4\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4066127f2d37da33ffbaaf1e5bd80c7c86a6515a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4066127f2d37da33ffbaaf1e5bd80c7c86a6515a"}]},"branch":"refs/heads/master"},"0f40c2159bba72176016a180aecb8fae92786074":{"kind":"REWORK","_number":6,"created":"2023-04-27 08:02:20.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/79/881679/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/79/881679/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/79/881679/6"}}},"commit":{"parents":[{"commit":"9ba7cfc9e72ddb2dd3c0ef1a44c2a04ba8709aac","subject":"Merge \"Revert back to MichaelRigart.interfaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ba7cfc9e72ddb2dd3c0ef1a44c2a04ba8709aac"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 00:57:00.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 08:01:06.000000000","tz":180},"subject":"[DNM] test kayobe-seed-images-rocky9","message":"[DNM] test kayobe-seed-images-rocky9\n\nChange-Id: I209459ecf5b1a6e5f541cbc72d74aa7defb459d4\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0f40c2159bba72176016a180aecb8fae92786074"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0f40c2159bba72176016a180aecb8fae92786074"}]},"branch":"refs/heads/master"},"207f7ef1171e61e66b85878a42ec5e86703fd42f":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2023-04-27 11:34:16.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/79/881679/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/79/881679/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/79/881679/7"}}},"commit":{"parents":[{"commit":"9ba7cfc9e72ddb2dd3c0ef1a44c2a04ba8709aac","subject":"Merge \"Revert back to MichaelRigart.interfaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ba7cfc9e72ddb2dd3c0ef1a44c2a04ba8709aac"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 00:57:00.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 11:32:47.000000000","tz":180},"subject":"[DNM] test kayobe-seed-images","message":"[DNM] test kayobe-seed-images\n\nDepends-On: I2a31e374a6d59c365e1132882236d6abb0c5b62b\nChange-Id: I209459ecf5b1a6e5f541cbc72d74aa7defb459d4\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/207f7ef1171e61e66b85878a42ec5e86703fd42f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/207f7ef1171e61e66b85878a42ec5e86703fd42f"}]},"branch":"refs/heads/master"},"d27f3f1dc4d3d99116b2e568c6acf9e0e693c1df":{"kind":"REWORK","_number":8,"created":"2023-04-27 12:33:56.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/79/881679/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/79/881679/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/79/881679/8"}}},"commit":{"parents":[{"commit":"9ba7cfc9e72ddb2dd3c0ef1a44c2a04ba8709aac","subject":"Merge \"Revert back to MichaelRigart.interfaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ba7cfc9e72ddb2dd3c0ef1a44c2a04ba8709aac"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 00:57:00.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 12:33:43.000000000","tz":180},"subject":"[DNM] test kayobe-seed-images","message":"[DNM] test kayobe-seed-images\n\nDepends-On: I2a31e374a6d59c365e1132882236d6abb0c5b62b\nChange-Id: I209459ecf5b1a6e5f541cbc72d74aa7defb459d4\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d27f3f1dc4d3d99116b2e568c6acf9e0e693c1df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d27f3f1dc4d3d99116b2e568c6acf9e0e693c1df"}]},"branch":"refs/heads/master"},"f893dc8c0d1560066edb8d533656e3e92f5c07ef":{"kind":"REWORK","_number":9,"created":"2023-04-27 13:42:52.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/79/881679/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/79/881679/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/79/881679/9"}}},"commit":{"parents":[{"commit":"9ba7cfc9e72ddb2dd3c0ef1a44c2a04ba8709aac","subject":"Merge \"Revert back to MichaelRigart.interfaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ba7cfc9e72ddb2dd3c0ef1a44c2a04ba8709aac"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 00:57:00.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 13:42:05.000000000","tz":180},"subject":"[DNM] test kayobe-seed-images","message":"[DNM] test kayobe-seed-images\n\nDepends-On: I2a31e374a6d59c365e1132882236d6abb0c5b62b\nChange-Id: I209459ecf5b1a6e5f541cbc72d74aa7defb459d4\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f893dc8c0d1560066edb8d533656e3e92f5c07ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f893dc8c0d1560066edb8d533656e3e92f5c07ef"}]},"branch":"refs/heads/master"},"289a5be87cf17dd37add40d2253fcf34b0180148":{"kind":"REWORK","_number":10,"created":"2023-04-27 14:33:10.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/79/881679/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/79/881679/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/79/881679/10"}}},"commit":{"parents":[{"commit":"9ba7cfc9e72ddb2dd3c0ef1a44c2a04ba8709aac","subject":"Merge \"Revert back to MichaelRigart.interfaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ba7cfc9e72ddb2dd3c0ef1a44c2a04ba8709aac"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 00:57:00.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 14:32:57.000000000","tz":180},"subject":"[DNM] test kayobe-seed-images","message":"[DNM] test kayobe-seed-images\n\nDepends-On: I2a31e374a6d59c365e1132882236d6abb0c5b62b\nChange-Id: I209459ecf5b1a6e5f541cbc72d74aa7defb459d4\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/289a5be87cf17dd37add40d2253fcf34b0180148"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/289a5be87cf17dd37add40d2253fcf34b0180148"}]},"branch":"refs/heads/master"},"ffe6d684d8cc60277fd72ede3608fda5c08fcdf6":{"kind":"REWORK","_number":11,"created":"2023-04-27 15:35:56.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/79/881679/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/79/881679/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/79/881679/11"}}},"commit":{"parents":[{"commit":"9ba7cfc9e72ddb2dd3c0ef1a44c2a04ba8709aac","subject":"Merge \"Revert back to MichaelRigart.interfaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ba7cfc9e72ddb2dd3c0ef1a44c2a04ba8709aac"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 00:57:00.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 15:35:38.000000000","tz":180},"subject":"[DNM] test kayobe-seed-images","message":"[DNM] test kayobe-seed-images\n\nDepends-On: I2a31e374a6d59c365e1132882236d6abb0c5b62b\nChange-Id: I209459ecf5b1a6e5f541cbc72d74aa7defb459d4\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ffe6d684d8cc60277fd72ede3608fda5c08fcdf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ffe6d684d8cc60277fd72ede3608fda5c08fcdf6"}]},"branch":"refs/heads/master"},"58ecf018a62a1049938699c8c69f3064ee3b6cae":{"kind":"REWORK","_number":12,"created":"2023-04-27 23:08:13.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/79/881679/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/79/881679/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/79/881679/12"}}},"commit":{"parents":[{"commit":"9ba7cfc9e72ddb2dd3c0ef1a44c2a04ba8709aac","subject":"Merge \"Revert back to MichaelRigart.interfaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ba7cfc9e72ddb2dd3c0ef1a44c2a04ba8709aac"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 00:57:00.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 23:04:12.000000000","tz":180},"subject":"CI: Fix kayobe-seed-images-* jobs","message":"CI: Fix kayobe-seed-images-* jobs\n\nFollowup I0d7ab0d8ff5b16ac5de8e50e63400bd455996555.\nFix issue with \u0027permission denied\u0027 while doing stat files.\nAdds more verbose build log and collet kolla-build configs for CI.\n\nChange-Id: I209459ecf5b1a6e5f541cbc72d74aa7defb459d4\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/58ecf018a62a1049938699c8c69f3064ee3b6cae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/58ecf018a62a1049938699c8c69f3064ee3b6cae"}]},"branch":"refs/heads/master"},"ef5f6a9cce6cd9c8af64aa0815d67d28f2ab9015":{"kind":"NO_CODE_CHANGE","_number":13,"created":"2023-04-27 23:10:20.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/79/881679/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/79/881679/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/79/881679/13"}}},"commit":{"parents":[{"commit":"9ba7cfc9e72ddb2dd3c0ef1a44c2a04ba8709aac","subject":"Merge \"Revert back to MichaelRigart.interfaces\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ba7cfc9e72ddb2dd3c0ef1a44c2a04ba8709aac"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 00:57:00.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 23:10:20.000000000","tz":0},"subject":"CI: Fix kayobe-seed-images-* jobs","message":"CI: Fix kayobe-seed-images-* jobs\n\nFollowup on I0d7ab0d8ff5b16ac5de8e50e63400bd455996555.\nFix issue with \u0027permission denied\u0027 while doing stat files.\nAdds more verbose build log and collect kolla-build configs for CI.\n\nChange-Id: I209459ecf5b1a6e5f541cbc72d74aa7defb459d4\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ef5f6a9cce6cd9c8af64aa0815d67d28f2ab9015"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ef5f6a9cce6cd9c8af64aa0815d67d28f2ab9015"}]},"branch":"refs/heads/master","description":"Edit commit message"},"61a808bd43430aca4c9b09eb84ee73f0673f5a4f":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2023-05-10 14:17:52.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/79/881679/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/79/881679/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/79/881679/14"}}},"commit":{"parents":[{"commit":"805414ef9d29deddd432fd3f591734a5f05da174","subject":"CI: fix molecule job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/805414ef9d29deddd432fd3f591734a5f05da174"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 00:57:00.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-05-10 14:17:52.000000000","tz":0},"subject":"CI: Fix kayobe-seed-images-* jobs","message":"CI: Fix kayobe-seed-images-* jobs\n\nFollowup on I0d7ab0d8ff5b16ac5de8e50e63400bd455996555.\nFix issue with \u0027permission denied\u0027 while doing stat files.\nAdds more verbose build log and collect kolla-build configs for CI.\n\nChange-Id: I209459ecf5b1a6e5f541cbc72d74aa7defb459d4\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/61a808bd43430aca4c9b09eb84ee73f0673f5a4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/61a808bd43430aca4c9b09eb84ee73f0673f5a4f"}]},"branch":"refs/heads/master","description":"Rebase"},"99ede9f18504ada966d881dc69749e9ffa08896d":{"kind":"REWORK","_number":15,"created":"2023-05-31 14:20:46.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/79/881679/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/79/881679/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/79/881679/15"}}},"commit":{"parents":[{"commit":"805414ef9d29deddd432fd3f591734a5f05da174","subject":"CI: fix molecule job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/805414ef9d29deddd432fd3f591734a5f05da174"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 00:57:00.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-05-31 14:20:16.000000000","tz":0},"subject":"CI: Fix kayobe-seed-images-* jobs","message":"CI: Fix kayobe-seed-images-* jobs\n\nFollowup on I0d7ab0d8ff5b16ac5de8e50e63400bd455996555.\nFix issue with \u0027permission denied\u0027 while doing stat files.\nAdds more verbose build log and collect kolla-build configs for CI.\n\nChange-Id: I209459ecf5b1a6e5f541cbc72d74aa7defb459d4\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/99ede9f18504ada966d881dc69749e9ffa08896d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/99ede9f18504ada966d881dc69749e9ffa08896d"}]},"branch":"refs/heads/master"},"1ca35c2cf917c620c7d23b6fb99ff93db745e302":{"kind":"REWORK","_number":16,"created":"2023-05-31 19:22:38.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/79/881679/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/79/881679/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/79/881679/16"}}},"commit":{"parents":[{"commit":"805414ef9d29deddd432fd3f591734a5f05da174","subject":"CI: fix molecule job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/805414ef9d29deddd432fd3f591734a5f05da174"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 00:57:00.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-05-31 19:22:38.000000000","tz":0},"subject":"CI: Fix kayobe-seed-images-* jobs","message":"CI: Fix kayobe-seed-images-* jobs\n\nFollowup on I0d7ab0d8ff5b16ac5de8e50e63400bd455996555.\nFix issue with \u0027permission denied\u0027 while doing stat files.\nAdds more verbose build log and collect kolla-build configs for CI.\n\nChange-Id: I209459ecf5b1a6e5f541cbc72d74aa7defb459d4\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1ca35c2cf917c620c7d23b6fb99ff93db745e302"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1ca35c2cf917c620c7d23b6fb99ff93db745e302"}]},"branch":"refs/heads/master"},"4245750515db71b4420421b32d1a3293b76a7fb5":{"kind":"NO_CODE_CHANGE","_number":17,"created":"2023-05-31 19:24:15.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/79/881679/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/79/881679/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/79/881679/17"}}},"commit":{"parents":[{"commit":"805414ef9d29deddd432fd3f591734a5f05da174","subject":"CI: fix molecule job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/805414ef9d29deddd432fd3f591734a5f05da174"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 00:57:00.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-05-31 19:24:13.000000000","tz":0},"subject":"CI: Fix kayobe-seed-images-* jobs","message":"CI: Fix kayobe-seed-images-* jobs\n\nFollowup on I0d7ab0d8ff5b16ac5de8e50e63400bd455996555.\nFix issue with \u0027permission denied\u0027 while doing stat files.\nAlso we collect kolla-build configs for CI.\n\nChange-Id: I209459ecf5b1a6e5f541cbc72d74aa7defb459d4\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4245750515db71b4420421b32d1a3293b76a7fb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4245750515db71b4420421b32d1a3293b76a7fb5"}]},"branch":"refs/heads/master","description":"Edit commit message"},"7088ff2c2f10ba9897d1cd5424fea6d9268020a9":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2023-05-31 19:28:33.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/79/881679/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/79/881679/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/881679/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/79/881679/18"}}},"commit":{"parents":[{"commit":"fdc941565110b1d27343255e4b23f540b28ee151","subject":"Update master for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fdc941565110b1d27343255e4b23f540b28ee151"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 00:57:00.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-05-31 19:28:33.000000000","tz":0},"subject":"CI: Fix kayobe-seed-images-* jobs","message":"CI: Fix kayobe-seed-images-* jobs\n\nFollowup on I0d7ab0d8ff5b16ac5de8e50e63400bd455996555.\nFix issue with \u0027permission denied\u0027 while doing stat files.\nAlso we collect kolla-build configs for CI.\n\nChange-Id: I209459ecf5b1a6e5f541cbc72d74aa7defb459d4\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7088ff2c2f10ba9897d1cd5424fea6d9268020a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7088ff2c2f10ba9897d1cd5424fea6d9268020a9"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
