)]}'
{"id":"openstack%2Fkayobe~883263","triplet_id":"openstack%2Fkayobe~master~I201faf14da5d143670250052ab15fc285f24868c","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-08-31 12:33:28.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-08-31 12:33:28.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-08-31 08:47:43.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I201faf14da5d143670250052ab15fc285f24868c","subject":"Fix firewalld configuration for monitoring hosts","status":"MERGED","created":"2023-05-16 13:58:09.000000000","updated":"2023-08-31 12:34:32.000000000","submitted":"2023-08-31 12:33:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"883263","meta_rev_id":"0fcd219e15c3fb644c971f16fa12ac8aff05f610","_number":883263,"virtual_id_number":883263,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-31 12:33:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-08-24 12:42:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-08-24 12:42:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-08-28 08:26:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":1,"date":"2023-08-24 12:42:14.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-08-28 08:26:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-08-24 12:42:14.000000000","_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-16 14:02:41.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-05-16 14:07:04.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-05-16 14:09:55.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2023-05-16 15:23:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-26 13:12:50.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"3e7b5ae7fca5b4e8830619d60d0cc9d1349d490d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-05-16 13:58:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7d9d4badbc8fac3d6679225caccd5ce7217b4aeb","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-16 14:02:41.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0df66eb4b2eb5ca7310e9603b27e4578d3ebb676","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-16 14:07:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"678175482e3a5f8e18202208ee3df9d5249f1fcd","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-05-16 14:09:55.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bbb4185dfd793de5032d082c3661df635fa97dc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-16 15:23:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f8c5aa04bb2142c68d930be21b269409\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3b9b6bd98824d59a95308bd828ee78b : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9540b47b661b400a97308c47c5d8fa9c : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04cd0c80bc4a4b87a43218a0c1bdf1f1 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bddfa7ad4b7f4fadaac1943bec8cbea2 : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1b5f997df2714b5384963596d0307cb1 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab11e77ad72c4c76b5b4cb1eb3a9d3a5 : SUCCESS in 7m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3f3ad4681c6e497cbc7ec3acbc542005 : SUCCESS in 4m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3c8fee96717247b183ec629396d6158d : SUCCESS in 10m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1550e15a6f2c4471b09b56c6c4991a7a : SUCCESS in 16m 29s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/b41e2f4eed6d4d3d9668b72ca4d9f0ea : FAILURE in 37m 59s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/b41a94b14f0242bfb96815b9c0e21719 : SUCCESS in 1h 07m 30s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/269aa2e42ed74c1ebf9eef7d20b0bdc6 : SUCCESS in 51m 04s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ae0d22524ba14da2afb8865534b8bd44 : SUCCESS in 1h 06m 30s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ec834f80f5c74b4e9c0dd6a96fa8db30 : SUCCESS in 18m 23s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/92f5ba6fbd8142fe8ee5e22667f404d0 : SUCCESS in 16m 04s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/91701b2e51a84c6aae7f993bc7a0cd7c : SUCCESS in 13m 55s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/fedc1e4a363748e0af03a3fd9c37fa4f : FAILURE in 45m 41s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/84d151af49f247b29faf6023b2a378e1 : SUCCESS in 1h 17m 42s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/ffb8b3eb15ca404eb0e017567eb20e41 : SUCCESS in 16m 11s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/113ec6346f7b4d56b5198598ffdb1a6d : SUCCESS in 14m 30s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/5617be7dbe214da8bc271a2c4deee842 : FAILURE in 28m 02s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2493a5b91abf411aa69467abecb0cb24 : SUCCESS in 26m 55s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a51dcc1f68984f81b9ba17c33531a90b : SUCCESS in 20m 32s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f5c1c688a70d4b8f94d5e2dcab670c76 : SUCCESS in 13m 38s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d4644b390d2f4ca5af4a9c9a134576f9 : SUCCESS in 11m 40s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/cadbf10916bb4015a18fd4b23b74deb6 : SUCCESS in 15m 43s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b212c89d9cd142cbb75b62692d5d77d3 : SUCCESS in 11m 53s","accounts_in_message":[],"_revision_number":1},{"id":"e1cc95660f03854d2487bca7d2b6a2cf14875f04","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-07-26 13:12:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"55d2c222e8242b9ea4677804e38e1a817794ba29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-07-26 13:17:20.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c24f9ce2d6be88eb3d9670f3058b1887314d2fe8","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-07-26 13:25:54.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"24a49998fcab59126968ed3180fe58c96bac0fdb","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-07-26 13:28:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b7207fead9740512cdfe8229e564f5f5f400f51a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-26 15:03:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/db194da207f74f8783dacaab073df08e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2375e0eef044a8f9e247bb03e2c3a5f : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d373c3f596a4f518b9ab2915a38d529 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a7343094bfc4a52b4692514a131031f : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/17a84753180a4c808494ef0b1b15a5e2 : SUCCESS in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/95d0e61f5bc14ae19fbce8ab72f4f1cb : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c37cab487fa4aeb88daa3bc50fa53f5 : SUCCESS in 7m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1519324e78664c5f98a24b0017d4443e : SUCCESS in 7m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fc3f3325646a414b95ad3b5770d79913 : SUCCESS in 13m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/29fd4f7399894411b50471b596c3c033 : SUCCESS in 15m 48s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/47c378bafa0541309cc5f2b27dfcff95 : SUCCESS in 45m 33s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/59cc862ef5cc4c989aca064d8067d11e : SUCCESS in 52m 49s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1733db7837a54a718eaa15c702d01c4c : SUCCESS in 53m 37s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/7a4edabf30844c08adea76c394ca4e6b : SUCCESS in 50m 19s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/7342c90e1cd84e7b991ec726c5f275e2 : SUCCESS in 19m 18s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/fe003237e2854134a23e767cf02c9fa2 : SUCCESS in 20m 24s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f460925e5b6b4754a2d13af8a418b34c : SUCCESS in 16m 44s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/718d31e5c0f84eb3867dd77ae88f6026 : FAILURE in 1h 37m 47s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1cf1fb76355b439e8dd7e1138a6d1932 : FAILURE in 1h 31m 23s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a675c1a4669841dd8979ea9a9cc8832a : SUCCESS in 18m 33s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/00ba307458a54882995637b5d6701609 : SUCCESS in 15m 20s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/4101babeaf164da5b12d5baadf45a217 : SUCCESS in 32m 03s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/075b8a353fb6469fa2e3c484a3bff7f5 : SUCCESS in 27m 28s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6983a05e61bf4832a16a5b132b6c4daf : SUCCESS in 22m 15s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f260c49630e44649a0ffb91221b4b1ea : SUCCESS in 13m 28s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b47ac3a82f5848ba8ede87286f9f7989 : SUCCESS in 11m 31s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/23d869a8d3b34744983c39e00e582cef : SUCCESS in 25m 28s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2a0b0ef490214867a0737dbcde9dea17 : SUCCESS in 21m 49s","accounts_in_message":[],"_revision_number":2},{"id":"f153f9d0e0befc87e33b92be680336bdcbac140d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-08-24 08:44:31.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bfae784678565d165befd14618edbda43ef7a468","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 10:48:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c743140429a84003a9777ea29d471c6d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/70a0ab1665094ef9b0bd41031792ff98 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/089ffe1092624e2fa7e7b58c935ba973 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3eed2d01b00f47f69e124cf305b0f5e2 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ca64dc58c32d4a9abc58fa06ed09023e : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3a34528e02e74767834288eff97e18b2 : SUCCESS in 4m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f5e3bd948a4e449dbe236430e6715c3b : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca004ae5fe3b4db39e2d47e74900d666 : SUCCESS in 7m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b891df2ef9b743559d0431d83c16c07f : SUCCESS in 4m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ac06c23bfda74581993726919e346d80 : SUCCESS in 9m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bcc0206479db4dd1a8e1f7373afee653 : SUCCESS in 12m 11s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a3d9f2ddfb194c468dc96b6b63a3f451 : SUCCESS in 1h 06m 21s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/83df6f2f58f4484d871f50ab13e9be34 : SUCCESS in 52m 24s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c33f6123c934bd386a61a14085bb289 : SUCCESS in 46m 14s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/0c6bbdf762ae453fbdb5f55cd7938914 : SUCCESS in 53m 06s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a679a58884a9416dbbbb1e97c821bc74 : SUCCESS in 17m 05s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/a165b47ef8ec4442995b099a30b858ae : SUCCESS in 16m 29s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/04a4e39139564d90ad29881c5855009c : SUCCESS in 11m 17s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c57fcbcb461644389a2d0cb905076d08 : SUCCESS in 1h 58m 18s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5cb5894eb4c1415abc205adc6be2a4a6 : SUCCESS in 1h 42m 52s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/2f3b5f813b814daa8aea5c8b7c13a6e2 : SUCCESS in 16m 18s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be27a4a612f34ed89ed5e965bfdab2fc : SUCCESS in 14m 05s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/efa8adf227344611875dd0ec30f749a5 : SUCCESS in 30m 03s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e934ce081be94d70a6f9d7bf5278ae35 : SUCCESS in 26m 26s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/76447ea8ebb64dffb6fe7978d5bf3bdf : SUCCESS in 25m 55s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/91391091a75543c3a3ec566a56828570 : SUCCESS in 10m 24s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1f58d08fed4c479ab1f46b9f9f5c58be : SUCCESS in 10m 15s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a553c7701c67433983e932a33f057925 : SUCCESS in 33m 20s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7ee2436d33a243bda42cc0ca5e02e5ed : SUCCESS in 30m 41s","accounts_in_message":[],"_revision_number":2},{"id":"d0bc3f5ac961e12efd56f7fc4eed1c3b2afe4a15","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-08-24 10:56:10.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5bbcb51386afdb15190bdf6d5c2ee0e018d2a005","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 10:56:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f551c94e0ae59d83190d54b3b95635b66b925a32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 12:41:15.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/712218c320de4bfdae9769fc69c3fee0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad289cd96a9445d5805eb47060e63a27 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/384363da044141baae1f119f93709019 : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/24dcf875b755472c9aa8cb13b58f28c7 : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0cae627db0de4c60b102a4c977982a1d : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bb4896a5d024813b7abd3e8ae2b71f4 : SUCCESS in 7m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/125810793de44b14847cdd9bcb5289da : SUCCESS in 4m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ec02b4dd4e104e88a45cd4bf094743d7 : SUCCESS in 11m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/312495cf7a2844579bb2c5e1ca8f8c07 : FAILURE in 36m 44s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e493add82d5c4614bf46d48d310b6df2 : SUCCESS in 56m 11s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/350c967876d54ea19d4916da5f06df4f : SUCCESS in 1h 01m 23s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e6c8a7e40ff247c28ea14e7c0d65ec0a : SUCCESS in 1h 02m 55s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/13aea9c98b0c44cd8db5692618a86b72 : SUCCESS in 17m 26s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39fd3adc8af1493ab583c43adc4fb535 : SUCCESS in 13m 54s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e210a5ff5ed74e2daa97abd77de1ead6 : SUCCESS in 1h 42m 00s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2ecf2a7ff9ed42a986ab3c3688a56fb4 : FAILURE in 47m 58s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/0baa24c15a3b4b99bc49abac1fa137c0 : SUCCESS in 17m 40s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/96ab728bb0344c36b965d9986d79b8a2 : SUCCESS in 15m 44s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8b416dbedbce4759aabb748a2afc6bd3 : SUCCESS in 26m 14s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e3ceab3f1b04b6da47b65ece2bb03f9 : SUCCESS in 24m 32s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a06ff9983ab54b3fa3a6e9df3b6feea8 : SUCCESS in 12m 20s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2bbd97a9927246b7aced1d84eb91f6b8 : SUCCESS in 11m 00s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/fad891cbcf624f319dd7d4603c01c44e : SUCCESS in 31m 51s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3be1754d3f9e4c25bfbb3d6221831b79 : SUCCESS in 27m 00s","accounts_in_message":[],"_revision_number":2},{"id":"1b8edd5ea4191da5836072f0b75e8e058ed255b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-08-24 12:42:14.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"ac43d6da6e5b69a74cd9cc8a799e4b2f565890f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-24 14:40:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b5a3c4dcb974fa6992b6825a6b7873d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe423d3220e34d098e2be84cd7814958 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6133435bfb6b4e9f99113db8cde1fb94 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/782ca04cda544c58b9e0dc69f52e4a6b : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fcd169833bfd4ba9bcd55c612937a283 : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/60055c1f98694ae98bb1b39f770b6821 : SUCCESS in 4m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8be1b7dcffea4ed78d1d37813a5a1835 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e4826326f094bacbe9d7cff2cf8cbd3 : SUCCESS in 8m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8b7018fa99fd4fe2971b4fc592b481d1 : SUCCESS in 4m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/12e95c5f3658451d95aa0a8c041942a5 : SUCCESS in 11m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a6f965e978d44b01bfd8f8c8f31d0f53 : SUCCESS in 11m 11s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e6c0d553f41341a6941a032507dab90f : SUCCESS in 53m 34s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/17ac7040c6fe41a3938618632e034c6d : SUCCESS in 55m 58s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/84712933e7ac40249e28365c60060618 : SUCCESS in 59m 47s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/b0df34e3110340eebcefef8dd1a94fc9 : SUCCESS in 1h 03m 25s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/fb56c885f784412dbdfca30557b651cc : SUCCESS in 14m 27s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/945f0851da254896954429b2d7ba7add : SUCCESS in 18m 30s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38fb694ed57d4b9e93091f8088735f70 : SUCCESS in 12m 38s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e7280e4193434b6e878ee5d6e02df85e : SUCCESS in 1h 53m 31s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b0a09fbd40364c75abe73d6286c36eb7 : SUCCESS in 1h 48m 07s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/607399f90c694dfc941e4bdb97e7c052 : SUCCESS in 17m 09s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/09c6aeae10a64072a160745c7c47c1fb : SUCCESS in 13m 06s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/32ef4ac59f194fa0b2c0a9c39dcc9d3c : SUCCESS in 27m 58s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5b6f221cbe674b1c8ccb82c0ad269a2d : SUCCESS in 30m 53s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5b5503c3cb724bfcb6bc936175221dfd : SUCCESS in 26m 20s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b3cc349ea5334d21b906d21bff71e5fe : SUCCESS in 13m 18s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4dfdcd755d5f4713923a33726aafd40e : SUCCESS in 11m 07s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/44b13adcc44742bdb791405d26da2ca2 : SUCCESS in 30m 09s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/26b9b2ce0a104acfa9e2290a78ee3e55 : SUCCESS in 31m 25s","accounts_in_message":[],"_revision_number":3},{"id":"494fbe63587d4628deab945ea7032b49bd88eef7","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-08-28 08:26:47.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"aad3caf97cde7a9561081497656b01c4ac2d5128","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 08:27:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"95af015bbb25aafceeba5cb64f4323d125ac26be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-28 09:44:05.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a4b2bb58ddec4836b406f554df0a3fad\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36ed4b8f5d8146ab9d8e0d9e7dc708d2 : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44ac4e7ea16e48afa679073391b68fe6 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/24e7bc1b3339493cbad501baa0af0b97 : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5089ee5003274b5a92e52850feb0985b : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/765f84a7b3a14806b23ac828d80ff34f : SUCCESS in 8m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ee289e7a60ce488aaf942358250af365 : SUCCESS in 4m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5d259cece50d459cb500c513f912fc75 : SUCCESS in 11m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6d3af79eb89f4c14b4dbcef02b664508 : SUCCESS in 14m 33s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/69cd65405a8d49308c61157138f8c8e9 : SUCCESS in 46m 55s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/14c110906f4a4fa8aff150447f558f96 : SUCCESS in 51m 21s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/b6538493b7e94774b48c193a4b743d51 : SUCCESS in 1h 09m 15s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/2d5a355d16854265a1440f7cd9cb215a : SUCCESS in 19m 35s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9586269e3b54203b2be0d4cc1f10a65 : SUCCESS in 19m 26s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/93329f0a92fa4e58b26ca0f88755b10e : FAILURE in 59m 38s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc9d8c3c22ba445b871c57b364d540c4 : FAILURE in 1h 12m 01s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/7c7edb71f5004e0fb2102458f7258aa7 : SUCCESS in 17m 55s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8629cbca0f7445d1b1e845425ac3eeda : SUCCESS in 11m 57s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/022bb31cb1e24a8f8ad1e60357447abf : SUCCESS in 27m 49s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/080d8d24dbff4c35994e931f2a08d2e6 : SUCCESS in 27m 11s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4fda7f91ab084807a2f3929673fa7f61 : SUCCESS in 16m 25s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/414212c4eb8c46808f751f335a0e4431 : SUCCESS in 11m 29s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/dc3edc009eb24b4eb44bdfe339689e64 : SUCCESS in 33m 46s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/53da762b060f47b2a51c721a52e3c8c6 : SUCCESS in 28m 09s","accounts_in_message":[],"_revision_number":3},{"id":"352f7bcce6403bc5e28676bab636e98be370ffb6","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-08-31 08:47:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"18dac086a809243dca73ef3c923799b55ebbdc95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 10:39:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a07f01b1bcfa45bda6689398e85c7ef5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e7f9be0064b410691e19c31d0c84fc1 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb276dcd612446ba8572a257bb23f760 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5e91697b3e440f1aaa4f2d5a2d900e1 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e70d1e431a443bab3e49a0cbf60b1dc : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d849edc2cad745dab4549b3bd336a92e : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ba896f75c814449997640399f1e5d84e : SUCCESS in 3m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62cd0a426d054a5da4b2659976164086 : SUCCESS in 7m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d5f6284d78ad4a489c295d6f4ab98f81 : SUCCESS in 4m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d23f10b0f3ae4bd4b969d90a2bc1aa2c : SUCCESS in 12m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8ca20ed76bd54186ab62aaa150ae725d : SUCCESS in 12m 17s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/bdb0a6d80d06480ca4b911cdf2d00321 : SUCCESS in 1h 04m 09s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/2a9f3412b6f44799b3180c1a62cc37a1 : SUCCESS in 56m 19s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ffede3280b7441e2bd76bdceb4eeb7d4 : SUCCESS in 57m 31s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/44ed59edf31249aeb543ab325efb150b : SUCCESS in 1h 08m 48s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/15fd36d3ae4a46398eec2486559aa5a4 : SUCCESS in 17m 15s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/73b0a3e0db274fb0bcda98cc8e551cd4 : SUCCESS in 16m 01s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/61a19d879b8b429395d5650a38c97703 : SUCCESS in 15m 06s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f5ffdd8b17e341bdb9fa9b4f0b0d6862 : SUCCESS in 1h 46m 35s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/916a43681de64cdea443bd58fc91f07a : SUCCESS in 1h 50m 58s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/20e187cd1b1a40d1806a0b29def99435 : SUCCESS in 17m 04s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/84ba5cdff29e4de983d47ab91ca75611 : SUCCESS in 16m 25s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/f523d3b0240b4904b95ce3c30aeda16d : SUCCESS in 30m 04s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e3751138a82e44358287c440c9ffa040 : SUCCESS in 30m 32s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0df075b5d3324104938df5df8f61efc5 : SUCCESS in 26m 53s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a46491c6d83f482aa373ec3e6d3eea0b : SUCCESS in 14m 40s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/22be4d98befa4317bfa35f1fee98e4af : SUCCESS in 12m 29s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2d1bf5cc62ab42e8959c99e6a6438d1e : SUCCESS in 36m 47s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07c07bc7b8374db8b5fe5e78a263f588 : SUCCESS in 29m 50s","accounts_in_message":[],"_revision_number":3},{"id":"1e9a2a0bda64db3eb10b6342552194ab8da54fcc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 10:39:48.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"715aad8dba53030a713c9f3391117df4f6e76cb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 12:33:27.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/243c37d83af14cc1b3903e53f1ec2c19\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e525c8a4c6af4a758487858c2e1ae8ba : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c3a1b9b6cf04c91a941fa13a2a40920 : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fad32d3710274728a9b3f8f49ff3a754 : SUCCESS in 5m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5cf3abbf965643568166c353101bf4e6 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d91e6af5ffdc498c95843d527b0e6b71 : SUCCESS in 8m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/01785356146f47958cb668ba714849e8 : SUCCESS in 4m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/92af7177f4564b768293f95e85108a75 : SUCCESS in 13m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bcbcf065ef2046668a5ffd54bdcc70b9 : SUCCESS in 13m 42s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/086f7f39a48047929eea0905cfc21443 : SUCCESS in 54m 07s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52c02937a2e8463d9ca7639ba99d803f : SUCCESS in 47m 04s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/c7e3058e94a149e79dddb3030b759aa0 : SUCCESS in 1h 05m 56s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/66ebf07ae4f042cb92f5ac6412094127 : SUCCESS in 17m 29s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ee7107f0583a4af2be2d1c6f0c3f9719 : SUCCESS in 14m 36s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c0f8153198c8430cb37aceb4195f02c8 : SUCCESS in 1h 49m 30s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1d1c54201380438e9b0bd57c624551ca : SUCCESS in 1h 43m 44s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/57e6b3efec4d4a0c8dc36963166a4b36 : SUCCESS in 17m 24s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a2e462e67c614c50b803178600ff9419 : SUCCESS in 16m 08s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/385406b22ba34f02a3be35a9da622114 : SUCCESS in 30m 06s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d7035cab27ab4d80b5691a75e63b03ae : SUCCESS in 26m 42s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9c3d496519194eb48f4a9b9cf9cf83b1 : SUCCESS in 14m 43s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/05816b2a555b47a4885142cf04d0a2e6 : SUCCESS in 12m 07s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3e57ad5c0a814d3d8164470465365af0 : SUCCESS in 34m 12s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bbe6cbcb0cb441049cfb89682110b93b : SUCCESS in 29m 25s","accounts_in_message":[],"_revision_number":3},{"id":"4a1d7a4ef85e03da97caf25e9896d82b4f650121","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 12:33:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"0fcd219e15c3fb644c971f16fa12ac8aff05f610","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-31 12:34:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a805555fe5649d188ecec07f697767b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b2f860306a04e94970f05c6deff5d70 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b7e95c83d3ea22bdccb5a5440e677c8a36f4cf61","revisions":{"dca2079ac89567939efe644493219806821d8406":{"kind":"REWORK","_number":1,"created":"2023-05-16 13:58:09.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/63/883263/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/63/883263/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/883263/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/883263/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/883263/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/63/883263/1"}}},"commit":{"parents":[{"commit":"2dc17631cb0c5af802ea04e8baf18094148f16c6","subject":"Merge \"Allow to use own repos.y[a]ml file to build containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2dc17631cb0c5af802ea04e8baf18094148f16c6"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-05-16 13:55:52.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-05-16 13:58:03.000000000","tz":60},"subject":"Fix firewalld configuration for monitoring hosts","message":"Fix firewalld configuration for monitoring hosts\n\nThere was a stray double quote in the firewalld config for monitoring\nhosts, resulting in the following error when firewalld is enabled:\n\n  ValueError: No closing quotation\n\nCloses-Bug: 2019867\nChange-Id: I201faf14da5d143670250052ab15fc285f24868c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dca2079ac89567939efe644493219806821d8406"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dca2079ac89567939efe644493219806821d8406"}]},"branch":"refs/heads/master"},"5b6105a7b18e6dbf1ab266276cf53d0ddf6b7cc6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-07-26 13:17:20.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/63/883263/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/63/883263/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/883263/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/883263/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/883263/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/63/883263/2"}}},"commit":{"parents":[{"commit":"c7b0ffd266c4d567ecee8c5b596f252e3e3c531e","subject":"docs: Add a warning on creating network loops","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c7b0ffd266c4d567ecee8c5b596f252e3e3c531e"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-05-16 13:55:52.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-07-26 13:17:20.000000000","tz":0},"subject":"Fix firewalld configuration for monitoring hosts","message":"Fix firewalld configuration for monitoring hosts\n\nThere was a stray double quote in the firewalld config for monitoring\nhosts, resulting in the following error when firewalld is enabled:\n\n  ValueError: No closing quotation\n\nCloses-Bug: 2019867\nChange-Id: I201faf14da5d143670250052ab15fc285f24868c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5b6105a7b18e6dbf1ab266276cf53d0ddf6b7cc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5b6105a7b18e6dbf1ab266276cf53d0ddf6b7cc6"}]},"branch":"refs/heads/master","description":"Rebase"},"b7e95c83d3ea22bdccb5a5440e677c8a36f4cf61":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2023-08-24 12:42:14.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/63/883263/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/63/883263/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/883263/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/883263/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/883263/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/63/883263/3"}}},"commit":{"parents":[{"commit":"0258141d378f77a70601df28fb6cbfb5277be1fb","subject":"Merge \"Fix for seed-containers being unable to use password protected registry\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0258141d378f77a70601df28fb6cbfb5277be1fb"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-05-16 13:55:52.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-08-24 12:42:14.000000000","tz":0},"subject":"Fix firewalld configuration for monitoring hosts","message":"Fix firewalld configuration for monitoring hosts\n\nThere was a stray double quote in the firewalld config for monitoring\nhosts, resulting in the following error when firewalld is enabled:\n\n  ValueError: No closing quotation\n\nCloses-Bug: 2019867\nChange-Id: I201faf14da5d143670250052ab15fc285f24868c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b7e95c83d3ea22bdccb5a5440e677c8a36f4cf61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b7e95c83d3ea22bdccb5a5440e677c8a36f4cf61"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
