)]}'
{"id":"openstack%2Fkayobe~884244","triplet_id":"openstack%2Fkayobe~stable%2Fxena~I000ae473bb759e19b48ecefdbf5d5dc19c833088","project":"openstack/kayobe","branch":"stable/xena","topic":"fix-bad-bridge-ports","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-05-30 19:12:24.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-05-30 19:12:24.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-05-30 17:19:03.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I000ae473bb759e19b48ecefdbf5d5dc19c833088","subject":"Show an error when *_bridge_ports is not a list","status":"MERGED","created":"2023-05-24 16:03:03.000000000","updated":"2023-05-30 19:13:27.000000000","submitted":"2023-05-30 19:12:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"884244-fix-bad-bridge-ports","meta_rev_id":"cf4925feed87ab4b6deedfb869c67051d15d729a","_number":884244,"virtual_id_number":884244,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-30 19:12:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2023-05-25 15:36:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2023-05-30 17:19:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2023-05-30 17:19:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-24 17:59:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-25 15:36:19.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-05-30 15:58:07.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-05-30 15:58:07.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"8de34947cae31a0ff91e727456a35cc0809621b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-24 16:03:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"079e2552aaaf90fa057692939fc1bcee9c7a1fab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-24 17:59:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5947aa88a724e1091d106ab6bd8d5a1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4bf3a4e870b3482a8434d1502c3df6ae : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bd33780bbe142b994d0531ad4e5e0db : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f7fa5910c06c454fa36f7a5e367cb80f : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ba46551739349de88d6237fd2615571 : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c9d3eae035e4f31bb4cbe431ccb4971 : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2ffd838d2ce452c993a521685e5e82d : SUCCESS in 8m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7c0f2c8cafa44a02be55dbf2a00df77c : SUCCESS in 5m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/633c638dd3e24c8eadc8160a8de1e888 : SUCCESS in 13m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/04b4333826e04bba8569f233e0c2e3e7 : SUCCESS in 17m 09s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e9aafa80bda641a0acb1196a21ef8cdd : SUCCESS in 1h 07m 03s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/4a632c30158d4edcaf656edc0956e622 : SUCCESS in 1h 20m 55s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01c134ba3e194045867ee186d99eda9c : SUCCESS in 56m 44s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/aaa8f138faf548c593d50b35865a00fd : SUCCESS in 1h 07m 45s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0e2c104a776d4f55b91b513bb78555c0 : SUCCESS in 28m 54s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/3d833d16044446bf8f0f7a4c9f3e9550 : SUCCESS in 23m 29s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac15482f218c444cba52ddf481cfcc39 : SUCCESS in 15m 48s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0fcd9c6aa2564f4c8eb8a10083fb6d5c : SUCCESS in 1h 51m 52s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0aafc8915aa54609a439a41258db1799 : SUCCESS in 1h 20m 00s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a5bea0e7a18149189bb215caeff57a7f : SUCCESS in 19m 37s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/f9e465c093c9482aaa30bee1d4070d07 : SUCCESS in 21m 25s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/abeeb6cd604849ee9b05238c800e9c72 : SUCCESS in 15m 32s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/1337b1e233d54a508d9b5e3f2d2fbe54 : SUCCESS in 1h 06m 08s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2d181f7dc50f4224b124a380acb9268a : SUCCESS in 33m 23s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/64460227b02b40e2be3972b6d1b2b449 : SUCCESS in 23m 35s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/452a85b2d6bb410eb15f2876d96ef16d : SUCCESS in 16m 51s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/3347feb19be04db788aae5fb2bb444f3 : SUCCESS in 15m 53s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e85465957434b08918944794bb6af1b : SUCCESS in 11m 46s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/414ebf539608418586e3bc9da4d0178b : SUCCESS in 18m 38s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/04c7e0827ed541f8a774d73fac242e1f : SUCCESS in 18m 41s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3460d29176824c428540fa882c6e43e9 : SUCCESS in 12m 02s","accounts_in_message":[],"_revision_number":1},{"id":"c3cdf44c232be0d148b6d2493319c0e9800a6b6e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-05-25 15:36:19.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4c6890ad2da720192aa47b62722d4febb7fc1ea7","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-05-30 17:19:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d8c6f1ffe239280c8d22a7911597141fc3cecafe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 17:19:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bc12b45f7cc3289f84a8724644efedface1f889e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 19:12:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ab1a7192f4847b3a81b9840c0fa1fad\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac9c3e5d1ee94b31adc0d3418978f9c1 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d7747f18a2f465cb463f6cc3d02e975 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51006413644b417391524f9c2860e58e : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21dac3709c1645e9b21907387e603175 : SUCCESS in 9m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/05e7d8dd451248e1a0aadee18fc16353 : SUCCESS in 5m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/69e0f48abd244d4b935fedd92f4224ee : SUCCESS in 15m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/edc8183c28e04d7295e0f77c276d7db0 : SUCCESS in 17m 58s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/bfd8431b28bf4c44aa8e045f4c818fa5 : SUCCESS in 1h 17m 23s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/2c4950896a464c6796545dc05dd91069 : SUCCESS in 1h 06m 55s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/73d07caedcf14dd6b549372ebc7e337e : SUCCESS in 56m 58s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/448d8d38d026415d8ef879ae145fed7e : SUCCESS in 1h 08m 58s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a57f2f4776b14a02b036ac49799e6e04 : SUCCESS in 27m 18s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/b11252a3483246b8a623166413d97242 : SUCCESS in 24m 13s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6c42533b0ed4dbaade6e508519d5006 : SUCCESS in 15m 10s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/32decb2ef90f44a5bea247b43e35083d : SUCCESS in 1h 46m 48s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9380e3387c44119a600f96c48ea73c0 : SUCCESS in 1h 24m 39s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7cab8ea572a04d4286986446c64429a3 : SUCCESS in 25m 23s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/59cfd65841e14d44b5ddf0828725dcb8 : SUCCESS in 22m 15s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f228df2b4a744ee5b4f525ab47cf8f94 : SUCCESS in 17m 24s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0b0bfc08385640faacebabdaef28b2c5 : SUCCESS in 30m 44s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7aae10e8baef4b258151c9971bea0fc7 : SUCCESS in 24m 06s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e7e0cad8e3fb4d5fb40aa12574626458 : SUCCESS in 21m 04s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/3845b0db58c64ebaac02db70e1548d3c : SUCCESS in 16m 56s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e4a4970b7c5420cad9506a52b332e89 : SUCCESS in 11m 47s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/782d3b7057a840c9aac43f3a0b2ef68e : SUCCESS in 20m 58s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/c30dc670b4c845a3bdce432039063a98 : SUCCESS in 16m 54s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e07504c62634cf98ea79af7d9716654 : SUCCESS in 12m 27s","accounts_in_message":[],"_revision_number":1},{"id":"eeb5818f87016bcab4a632b82f0406cb65ff497b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 19:12:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cf4925feed87ab4b6deedfb869c67051d15d729a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 19:13:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e28eed3b32a4037ac7ff01046e92d69\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a016061b054b4b69873f9dcfbf5b64bf : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0bf45b190c62785514991866a38ca70344b006c6","revisions":{"0bf45b190c62785514991866a38ca70344b006c6":{"kind":"REWORK","_number":1,"created":"2023-05-24 16:03:03.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/44/884244/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/884244/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/884244/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/884244/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/884244/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/884244/1"}}},"commit":{"parents":[{"commit":"190611fbadb68a8631893625dcc5423590b0b54d","subject":"Merge \"Adds support for custom Multipathd configuration.\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/190611fbadb68a8631893625dcc5423590b0b54d"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-05-22 18:44:25.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-05-24 16:02:49.000000000","tz":180},"subject":"Show an error when *_bridge_ports is not a list","message":"Show an error when *_bridge_ports is not a list\n\nCloses-Bug: #2020378\nChange-Id: I000ae473bb759e19b48ecefdbf5d5dc19c833088\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 9b58320f6b40de3c33bee179d97dd0bf5bf61986)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0bf45b190c62785514991866a38ca70344b006c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0bf45b190c62785514991866a38ca70344b006c6"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
