)]}'
{"id":"openstack%2Fkayobe~884472","triplet_id":"openstack%2Fkayobe~master~Id5e946fd7b0526faea99536f4a055f0f5c2b8176","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-05-30 08:43:20.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-05-30 07:14:58.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2023-05-30 12:05:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id5e946fd7b0526faea99536f4a055f0f5c2b8176","subject":"Use OpenStack 2023.1 release","status":"MERGED","created":"2023-05-26 09:51:00.000000000","updated":"2023-06-19 12:09:32.000000000","submitted":"2023-05-30 12:05:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"884472","meta_rev_id":"44f25441dea6b4d2eee5dab970a6fa70a74fe884","_number":884472,"virtual_id_number":884472,"owner":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-05-30 12:05:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-05-30 06:54:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-05-29 11:21:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-05-30 07:14:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":2,"date":"2023-05-30 07:18:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-05-30 07:14:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-26 10:57:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-26 15:02:52.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2023-05-29 10:56:16.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2023-05-29 10:57:47.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-05-29 11:21:46.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-05-30 06:54:52.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-05-30 07:18:06.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"2557296767bd06bea2166b047222909676dab180","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-05-26 09:51:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a5df2269f59c5c8fb9012cac20595fca975b8fa2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-26 10:57:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aeeddc2439df4b94a1bb174b5d5313fd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0df925264814f328ff8d13a70005671 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8487544175974cf89dfcf27c1d94ba99 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ec8ffa91dd547d285791f3e8c4a31e0 : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a8d0e2be0c4c4f88a43340ddba24d2de : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d8383aade562455daa0b38c0f441a1cd : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b75953535dce46edaebc78ded7c1cf0d : SUCCESS in 6m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db57f8019b1b475497e3b66cc53e064b : SUCCESS in 3m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/10c1079357354ccd8d18169bb6f2dd1e : SUCCESS in 4m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/17b977c367a546eda2d6071872aef831 : SUCCESS in 9m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2b8057e56e964ecd9d69322ead36b529 : SUCCESS in 13m 27s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/31dd19dfb69d4832b56e05c75057e6be : FAILURE in 14m 35s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/b3594508638b4c75b02a5a7820761057 : FAILURE in 14m 01s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d3e8d4d85f9a4664a913a3a771689e3d : FAILURE in 12m 45s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/9529b9bb688842be9cf82099e1f0f7e4 : FAILURE in 15m 18s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/11868a70ed85448395ac552f5aa633c9 : SUCCESS in 14m 48s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/8fa6254f34fa4289b8629075a5a0b45b : SUCCESS in 18m 45s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a734c25d5bba48bfbd6e1b976fdc34ad : SUCCESS in 11m 27s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/6db32198dcc54099b6fa939e37bdea55 : FAILURE in 1h 01m 39s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/991e9d84243f4f048ef265753b200a6e : FAILURE in 52m 06s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e2fe91e49e78409d9074bfc517330eee : FAILURE in 14m 03s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/296238dcfab94d4596ec4a4b41513d47 : FAILURE in 9m 26s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/00bb013a17e64e73ace3008a0319e6e6 : FAILURE in 13m 45s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a765822bd96b4f6ba954553ba923dcfc : FAILURE in 24m 52s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/271c5eab3f97415081b386be70b336a7 : FAILURE in 20m 08s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/431e14f6af9840e18d49bdb6dd091c43 : SUCCESS in 13m 59s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c27afba49c664da8ae938b96ae2211c5 : SUCCESS in 10m 50s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e8770b328f854d558d310dafb7e4cc6e : SUCCESS in 33m 02s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f87b3d8539649a1b6aa363f14424208 : SUCCESS in 27m 32s","accounts_in_message":[],"_revision_number":1},{"id":"07947169e084dc2eb54dfba3b480d5a00c1ea342","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-05-26 15:02:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8470e423edadd72c25ca26dfe3055a5f4ef51a68","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-05-26 15:03:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4b998cf5147f4bbf667e2a0d35ee4b52e257c9d0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-05-26 15:05:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3f732031bf0734a46a35df2ece20e49cb41a6752","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-29 10:56:08.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"766d8bb0468dc91179059b47fb1eff2e0a6108af","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-29 10:56:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"00e8fa59a8ab7a699e8beb172b1d5724edaa2400","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-29 11:21:46.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"40727f198ccd934a7e83c256df8b8cd62878fa86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-29 12:39:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/285dc1c484c8401e954196fad649c614\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38c219b3dee54e17a56ef0de755ea74d : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ec234a87d4842b1b8a9721527c93cfd : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/347963fedc6749ba87137b458aa8b50f : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7d9c9710f01d48fa9c89c180d8a93b81 : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a986e55603d8418f98f3d1c415c52f13 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6de90c635cf4106b621d8ec10940a3a : SUCCESS in 7m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6cbf038d9e484c4e9623e1322cf7c16e : SUCCESS in 3m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f0f5549fc8ad4d9ba49f5136604a3b0b : SUCCESS in 4m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/93700b384e6a4b7583c36cf55fb83ce7 : SUCCESS in 9m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/77a382f6291741c9bc122c36d02848ea : SUCCESS in 13m 27s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/2248ee7ef0714ec88db844bb5b050d3e : SUCCESS in 56m 02s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/6ad39ed8d5c74296aea728bf98d6ce27 : FAILURE in 20m 44s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/37319c64a13f4ebe81fe5ee3787a4ed7 : SUCCESS in 57m 05s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/35a95ebd77fb4155b46160ae63063984 : SUCCESS in 1h 01m 05s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/50297d65c7fc453485b5cc501a38e77a : SUCCESS in 16m 41s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/4044caa48ddb4809839495d179d75db8 : SUCCESS in 18m 46s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30d87ea61bf64255877a45ee2ce2e281 : SUCCESS in 14m 01s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5687db2447914d52b1d2344be8d5dd29 : SUCCESS in 1h 36m 42s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e551d75e4272421d9684c0cb80e89ce3 : FAILURE in 16m 50s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/05b096bb56ca4a258537358beb1a8a07 : FAILURE in 14m 50s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dacd9114761544ec9b06b935852a30e8 : SUCCESS in 16m 36s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/98fc5be2678543d7bbdb36fde9a0ddde : FAILURE in 11m 52s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4d5ffed672824f538a75becb7587bf32 : FAILURE in 22m 44s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3c84c5c5ac9644deb0d5afc51f4fcf4b : SUCCESS in 24m 18s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5659c8105f28471387f88fe3968f7b4e : SUCCESS in 13m 25s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/16fbd8e0c8804f9db44b12c99136f0ed : SUCCESS in 10m 27s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e01570d033cb4921a817891ec5eb2c0c : SUCCESS in 29m 06s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/125e856208df475cb33efc1d71ddbf7b : SUCCESS in 21m 25s","accounts_in_message":[],"_revision_number":2},{"id":"7ec7153d519008bf70f877b298ef12bc742d1497","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-29 12:42:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"44c80ea949cac37b41a079a4d669e5b4c516aea8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-30 04:02:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bcbaae268146ca09aa521f88eeb99769e00540b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 05:54:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/458aab3eccfa4f0f9a05c42a4b24a6fa\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8fc2ecfbede4a2191ff95be970cc74f : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c46fb8655e964909b30f4c13765ea390 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4f717f9986c4504a849a0e01aec06f6 : SUCCESS in 12m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/20e1b7b864b04728bacd8422cec41219 : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6b00a494c719472aa3bb6c9dafa724c1 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8a13d5e3a994879b7bf9009d6a57a02 : SUCCESS in 7m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4019b389fe2e4c34ae26385e294fe666 : SUCCESS in 3m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/006dd6e027c2416b914494d2950743a1 : SUCCESS in 4m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cf3eb3b374594a0eb4b1257fd0839341 : SUCCESS in 20m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f84b562c7f3349f5a2d33e26402dc864 : SUCCESS in 16m 58s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d722f3d155b64286a00b06229f26d58f : SUCCESS in 55m 02s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/f33aadfc3c2741e8be62276cbac3d6c0 : SUCCESS in 1h 05m 49s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/892789bdf5c84ff788355c822879f7e3 : SUCCESS in 47m 33s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/7063dfd0da78454c9ce672fd7357255f : SUCCESS in 1h 12m 38s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a4f31071f92c4aaaa5fb2e5d748df921 : SUCCESS in 16m 37s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/51c971516dbd40ffb7df2e73262b3b48 : SUCCESS in 16m 26s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc1bf49b28b54bd2a7cafd6c3403f754 : SUCCESS in 13m 47s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/fe743dc8e43343799800b36a69a028eb : SUCCESS in 1h 25m 03s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/444fd7ca9d93499a89d7387cb1819c96 : SUCCESS in 1h 44m 35s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/03c8796ce5e8494b8d5f37253e3410b5 : SUCCESS in 14m 09s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5fc82f0cb0b448b286cd2305b2e49ad6 : SUCCESS in 11m 33s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/4669295d08c34d858ca640cf86f93cf6 : FAILURE in 31m 16s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/405e9921cff5406cad85e30dcf5fe540 : SUCCESS in 32m 47s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd3f5a02c5014f5dbbcce8fdb08ead77 : SUCCESS in 18m 56s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2f396c179fb041cfb7ca6d4fcce143e0 : SUCCESS in 13m 20s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/635dcc8a9fa54cfd862b04d8c7319f72 : SUCCESS in 9m 37s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/26fe1fa4f88e452daa3fb435014d42c8 : SUCCESS in 25m 30s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ed9c3beac7fd470181e298efe1aeb77d : SUCCESS in 20m 23s","accounts_in_message":[],"_revision_number":2},{"id":"8423ffccec64c529fbb36904d85c78596adf9b29","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-30 06:54:52.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"62c7a0b84b4facc6021cb61204c0b147f965690d","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-05-30 07:14:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"041da9ce21cd82757eb7a4c14b4dea17cf8be4b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 07:15:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d08a9d679ec38dc655f9db0398ddc3d89982ea30","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-05-30 07:18:06.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"35396866827a83b33b7a113afe21792a844275af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 08:37:39.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5749452b322d4cfaaff3ef9d5f09b894\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18b5f7a31b844015b50dee692f6080ec : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf9e27da46dc41f8945186edff0b95cc : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f87c07eda28245978c4940f7538313e4 : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d2e005f602904a74b90c1f6beaa58b14 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58bac3e8c5b743f7bd96d15adb305f72 : SUCCESS in 6m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2bf3b16c7a0247ec9baa1561acaee6d0 : SUCCESS in 3m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3e4d47918de44745b6352ce82e6e17b4 : SUCCESS in 4m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d036d6d2964e487189e6c20a0eb6b95c : SUCCESS in 11m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4b39f4b821c943378c30042f0f1f94ac : SUCCESS in 15m 28s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d8a04c14c5724be5ac3b77402a0698d2 : SUCCESS in 59m 42s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/28afaa9192d6430c8a9ba4823e047ddc : SUCCESS in 47m 24s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/93c127b3db4f4bfd829578590ff68e84 : SUCCESS in 56m 38s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f70db308c13649e1aa290a6459de33ab : SUCCESS in 15m 11s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc95b8caff2645a1ad2088bd94c5bdc7 : SUCCESS in 12m 50s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8612e4078cc14e92931499760105918f : SUCCESS in 1h 17m 28s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df6cc17ff95c4282af2bf6f249beb204 : FAILURE in 41m 03s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/6d98f018efe84b64b482c25437c7013d : SUCCESS in 16m 52s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9103bbec80064aeb8a1a0b9198031ccf : SUCCESS in 14m 00s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/253137ad2ae24bdabbcf664e2257a9cb : SUCCESS in 26m 39s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f0872fc1015b449cb12650e9195f0849 : SUCCESS in 22m 58s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/24375365b0554c64b4235342295241c2 : SUCCESS in 14m 00s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/129f1d9c998147e6a0eece8bf3f453a8 : SUCCESS in 9m 32s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e5e8fddbf38d472ab15ac8e985e1ff7c : SUCCESS in 27m 21s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/18b89b60a7c945639f3280696d51a217 : SUCCESS in 21m 19s","accounts_in_message":[],"_revision_number":2},{"id":"25b316057ae1fe58849eb10503bdfc22335b0ed1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-05-30 08:43:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c511ef28ea61ec1293d676dbb79aa15c85dc2c02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 10:28:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81e2da14881741f08e27e88b053e7fa5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c65497c2f584d189c8d38c3bba69151 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47695dfb500f4a568a89bef198271a5e : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1719b0794ab743a4b06c5cbeddc7ecf3 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/580a25c428ab43e5ae65e4705145c59d : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f3be5b8e5d3648e7b48776e6edb3f316 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed290e50165744909ba09f1268b96067 : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/34faeb93c81f42fc97be39dba892a920 : SUCCESS in 3m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/aa7533c5b80148728613818eab25800b : SUCCESS in 4m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1503aa1ab30f4d6a82d486d31495f1c3 : SUCCESS in 10m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b1640e35ed97475a8e2c1436ad455074 : SUCCESS in 14m 00s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/19b5c5f7c59f47d1955a8b66be6297d0 : SUCCESS in 1h 00m 54s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/86eed8d12173423eab071dccac3e3693 : SUCCESS in 50m 19s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6028d65825194583a38f5a4f02bb41ff : SUCCESS in 45m 46s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e21b8488e798459394e961c3a9e527bc : SUCCESS in 56m 12s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d249beb9c1a741df9964847f39cc3de4 : SUCCESS in 14m 15s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/4c0d747e168444479afc8bcc78ad570d : SUCCESS in 16m 10s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b3481c136ee04af59d2b805297fa33d3 : SUCCESS in 11m 19s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/28315362dda54bb6a2f640644c9dc7ca : SUCCESS in 1h 18m 35s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fccf57c8a6764cea99eec5c87f78b66f : SUCCESS in 1h 28m 50s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/5484c3449eff460a83924d8f0b1afb03 : SUCCESS in 14m 37s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b5ed2a17f6034589a6721fe3a1e8d13e : SUCCESS in 13m 48s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/730a797a930d4a8093b493347b5a5873 : FAILURE in 35m 22s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9e3e93cd028a40fda91623f8bffcc4d2 : SUCCESS in 30m 39s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/46d1db1db19e4b31bd6da7b2f46ae6c9 : SUCCESS in 25m 37s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/563408e7ea2c454b89d27e4a375cb77e : SUCCESS in 14m 42s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dbac77644aa545ef8398c423e475dcd0 : SUCCESS in 10m 45s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/77fe2464c8304d6bb34004b7d9d39b52 : SUCCESS in 31m 10s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/571e16bdb42f4d5b9a247ff7e69490ac : SUCCESS in 32m 03s","accounts_in_message":[],"_revision_number":2},{"id":"ebc4c7678f89d5d6e234cc00ce2b2e5d8e5e0e1d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 10:28:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d8018b4c33ff70c757a94b7a19ad6d2a31fc835d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 12:05:06.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13e6151b8aea4944acf73a249ae85140\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98b1d7c6fcc5413697b5fa75888621e5 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6ba17aa42c74242a04d894cf741d781 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eed054ed2d1b40f08ccb401f0e33c19f : SUCCESS in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/556743bc9b5141d6bc9ac5335056de69 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a073800f4dfb4ab398f69bb38c6594b9 : SUCCESS in 8m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9aed5259bbca4983ad617f03d77fe22a : SUCCESS in 3m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/eade8a4b82b045a3ad957f19c2eaedba : SUCCESS in 3m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4b62b7b16dc843d7963f1126ad3d4309 : SUCCESS in 11m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a66eedb3ceb84830813db2221471504e : SUCCESS in 11m 10s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/bf4592ae048b466980b35ac95d07fd13 : SUCCESS in 54m 48s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/41fbd01a560549c4925fd62e7b25e48d : SUCCESS in 46m 47s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/5aed2060066d4e1996d18c54f72ea26d : SUCCESS in 1h 00m 07s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/689f59dcf2834ccc84cc926b1902404e : SUCCESS in 14m 05s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d8c10f442f72435995be74c683c76223 : SUCCESS in 15m 57s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5fdaffbb3f4c42169d3226b244c0bc31 : SUCCESS in 1h 22m 21s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0afddd7bf4984975a68bbcfeaebe56e2 : SUCCESS in 1h 17m 15s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/f65bd2bd8fd8412db12392b6b446787c : SUCCESS in 16m 35s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/95c1c906d9724f8a88e725bb123d6795 : SUCCESS in 12m 28s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/74eb70bca8dc40839a16c1371afe5dec : SUCCESS in 32m 30s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b619998244de4775a3d0a3d6c9dea67b : SUCCESS in 23m 58s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5c54ae5eb80043319493adc037a87e22 : SUCCESS in 10m 59s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/953e1809d2ce40dd9737852743c19c7f : SUCCESS in 12m 23s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f303674ebb5f474a9990029ba036cffe : SUCCESS in 26m 46s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6350b72fadaf4045a30b7b9c6a6e5040 : SUCCESS in 24m 28s","accounts_in_message":[],"_revision_number":2},{"id":"9b418b65a9ccb6170ab28644a0f5d6dada65cd8d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 12:05:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1664c7286b3a52d2814288a42096174b8fdae7f7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 12:06:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad1d2d887b1f45ef89a7df77cd614eb9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd2daae5858f451686412c2c8f248835 : SUCCESS in 35s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78bba52828404ccb830946a671992c0f : SUCCESS in 36s","accounts_in_message":[],"_revision_number":2},{"id":"44f25441dea6b4d2eee5dab970a6fa70a74fe884","tag":"autogenerated:gerrit:revert","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-06-19 12:09:32.000000000","message":"Created a revert of this change as Ief6b68144dd2d161ca6e98888934888cdfee49a4","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"16a61da41a9516445459c1ac75464e67b5f37ee2","revisions":{"6f180cfeffc13d62c12e935910eae8bb7f34a464":{"kind":"REWORK","_number":1,"created":"2023-05-26 09:51:00.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/72/884472/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/884472/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/884472/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/884472/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/884472/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/884472/1"}}},"commit":{"parents":[{"commit":"7f4ba451913f6a571aee56ce4a5016cd3012cc2d","subject":"Merge \"Limit interface names to 15 characters\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7f4ba451913f6a571aee56ce4a5016cd3012cc2d"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-05-26 09:50:20.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-05-26 09:50:20.000000000","tz":60},"subject":"Use OpenStack 2023.1 release","message":"Use OpenStack 2023.1 release\n\nSwitch all dependencies to use the OpenStack 2023.1 release.\n\nThis commit should be reverted on the master branch once the\nKayobe stable/2023.1 branch has been cut and RC1 released.\n\nChange-Id: Id5e946fd7b0526faea99536f4a055f0f5c2b8176\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6f180cfeffc13d62c12e935910eae8bb7f34a464"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6f180cfeffc13d62c12e935910eae8bb7f34a464"}]},"branch":"refs/heads/master"},"16a61da41a9516445459c1ac75464e67b5f37ee2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-05-29 10:56:08.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/72/884472/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/884472/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/884472/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/884472/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/884472/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/884472/2"}}},"commit":{"parents":[{"commit":"0e984c4d479d256983719c65ebcbb514d924fc52","subject":"2023.1 RC1 - Sync KA feature flags \u0026 inventories","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0e984c4d479d256983719c65ebcbb514d924fc52"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-05-26 09:50:20.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-05-29 10:56:08.000000000","tz":0},"subject":"Use OpenStack 2023.1 release","message":"Use OpenStack 2023.1 release\n\nSwitch all dependencies to use the OpenStack 2023.1 release.\n\nThis commit should be reverted on the master branch once the\nKayobe stable/2023.1 branch has been cut and RC1 released.\n\nChange-Id: Id5e946fd7b0526faea99536f4a055f0f5c2b8176\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/16a61da41a9516445459c1ac75464e67b5f37ee2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/16a61da41a9516445459c1ac75464e67b5f37ee2"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
