)]}'
{"id":"openstack%2Fkayobe~884540","triplet_id":"openstack%2Fkayobe~stable%2Fzed~I5b890e24195d033897a597a0a93a1cacfb2030d2","project":"openstack/kayobe","branch":"stable/zed","topic":"truncate_veth-stable/zed","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-09-21 11:06:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5b890e24195d033897a597a0a93a1cacfb2030d2","subject":"Limit interface names to 15 characters","status":"MERGED","created":"2023-05-30 12:39:39.000000000","updated":"2023-09-21 11:07:23.000000000","submitted":"2023-09-21 11:06:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"884540-truncate_veth-stable/zed","cherry_pick_of_change":869744,"cherry_pick_of_patch_set":3,"meta_rev_id":"dbb5cd3645459c509ec36e1ca725b558a892bd5d","_number":884540,"virtual_id_number":884540,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-21 11:06:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"all":[{"value":1,"date":"2023-05-30 13:25:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":2,"date":"2023-09-21 10:00:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-05-30 13:19:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-06-01 11:08:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2023-09-21 10:00:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-05-30 13:43:02.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":1,"date":"2023-09-21 10:00:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2023-09-21 10:00:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-30 12:39:39.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"state":"CC"},{"updated":"2023-05-30 13:19:04.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-05-30 13:25:35.000000000","updated_by":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"reviewer":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"state":"REVIEWER"},{"updated":"2023-05-30 13:43:02.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2023-05-30 14:07:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-30 14:39:03.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2023-09-21 10:00:34.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-09-21 10:00:36.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"c7eb56949e86603faeb152b30f812a5e45f84dfe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-05-30 12:39:39.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"494531418075384b5f52d833bc21b0168b5a732e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-30 13:19:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e9c6fe16228de666b7acd3d8e4e0b9a3c3fe7d58","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2023-05-30 13:25:35.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d7680b7f59a55158d7aa16b986ef3822c6f09829","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-05-30 13:43:02.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ad88f4dd21018bd81ea344891e13c2137ce42c35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 14:07:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/09f9997a3ee049c18eb6d18ef8691205\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/70d4746ab8ae4e978e8ad4005849095b : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfe92d6870624cac85b81fadf9deefa8 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a53cdfbd70b4c8e9ee978454aa7e241 : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce59fe4f111e40859766ed53ce0652fc : SUCCESS in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e9ef787e134e439da52a643254fd60e2 : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31825719f11a4c4180e58f7aafce2bea : SUCCESS in 7m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0dc4f347361a4d26bd72930d58c5d5c5 : SUCCESS in 3m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/07ee74dce082450fb77f508eec17a7ce : SUCCESS in 13m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3b0d29c88a3a4f2cb0b2e374d9ef8bc8 : SUCCESS in 13m 56s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d9d7b98b88fc47faa46eaafa4e18cb0c : FAILURE in 1h 06m 53s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/fddc20c57ad744e08a43ef0428d9828f : FAILURE in 1h 04m 09s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4245404b65724a73b83e0655d9ceef8d : SUCCESS in 56m 26s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/487ae2fb334640f796a2185dfdffdb7a : FAILURE in 1h 21m 11s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/613ab77d0f074e63833c88717211c06e : SUCCESS in 16m 31s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/71dd44f49a394e37a49df3eadd9558fc : SUCCESS in 18m 09s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2c4a495b059143199319be769fb474ad : SUCCESS in 14m 21s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/def11c5b3ea249eaa4029edef1a71581 : SUCCESS in 1h 09m 13s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/8a87139aecd14166bb30d5795536a478 : SUCCESS in 17m 04s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c03b4f2a7df4e2fb8956576dd13ce0c : SUCCESS in 15m 25s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/2503c5c3cec64fc78577159f9b6330eb : SUCCESS in 30m 35s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d92338dd11c345a58a2256a0e608bedc : SUCCESS in 25m 10s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c4144fc181b04944890de6bbeec03a43 : SUCCESS in 13m 16s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe521bf6eaeb4f01904ada79ef8d266e : SUCCESS in 10m 46s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6edf06e8c07548dcb9232a12f23ef155 : SUCCESS in 16m 54s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b23e56f57254bd28f50b6677ae0c952 : SUCCESS in 13m 24s","accounts_in_message":[],"_revision_number":1},{"id":"fd447e719a0358e5fad8685222d908a5e22a6b11","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-05-30 14:39:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f9682756b9321466d5b9ecc6f7d25b300bf86836","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-06-01 11:08:35.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"99a65a86c268b25deb4ad3f5a71a8d58f8c76871","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-09-12 08:49:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ae1fc90f397f5417b03b130392e1cb10ffb1a8ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 09:48:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e32e351cba24a94bde818012ab4905a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59375bc913554a71912257634d419afd : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfaff93257314fdbaa41861105829a08 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/264d7db79c17458282220c8397b19724 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f7dd56fc199b4fef8b02e85e61eb0012 : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a406e9aa1266455daa8b79c529edaf5c : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/498136f00b1241e989721b2f955d2ba1 : SUCCESS in 7m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ff5eba17f9a2427e8157ac48f96fe68c : SUCCESS in 5m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/627583055f0f4720a37fa497a8a29f98 : SUCCESS in 9m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/827bad63534c43a084b924c326cea10b : SUCCESS in 12m 51s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/999a016fea1947079e017907fce10c2d : SUCCESS in 48m 10s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/d372f619ca55405ba42805c1ca8cf028 : SUCCESS in 35m 30s (non-voting)\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/f0c6803599d64824a8e6db8b81fb583b : SUCCESS in 52m 12s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/655fa70c181a49ef9eb9f8a71658b5d7 : SUCCESS in 16m 54s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/9859d8dcb8e746ab8753647644e2f04d : SUCCESS in 15m 00s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a09c90180d64969b60f58516464fe8e : SUCCESS in 13m 47s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e0236d6a596443c58e604ee9482407fc : SUCCESS in 15m 45s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5affe79672654aa986118a3ef09ed50b : SUCCESS in 14m 42s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/ee689d32fb4f452b99c0bd1268be9a84 : SUCCESS in 32m 28s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea3ba51d52164203883538be84ce33b3 : SUCCESS in 26m 58s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3843192a8db84409a99e148f3689f1ee : SUCCESS in 12m 56s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5dadbc8ef7f144b8a1cd23ed414d282c : SUCCESS in 12m 10s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8dc58bde5baf4532bb9c8eef42b85a7d : SUCCESS in 17m 40s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/236a9373bc30431d963c29c430feea1c : SUCCESS in 14m 58s","accounts_in_message":[],"_revision_number":1},{"id":"22a66fd86d6354db3b5feca3d65524dd3311604b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-21 10:00:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a8439cc2112f56e5776762403cc14a2e975844c7","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-09-21 10:00:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"59c93e013b05bea0be7a45febc11fb0965193302","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 10:01:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c41a0c01e0e728e8d5c4d8980e35676380576cd2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 11:06:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12c51c63e94a425ca722bf62186ea62d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/030872c23d1e4c23911f105b7463d658 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5b978798ce849baafb32523cece4cc5 : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1e237cde76e411eb5468cf92deb350f : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/684cd4c5415f4d5a83fdbd660ae7ed2d : SUCCESS in 8m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cbc7779e3ede4fbe9becaeca49e516dc : SUCCESS in 3m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/60bbf73de3cd4c13812dce4410c7f42e : SUCCESS in 9m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/eb620509addd4d30a6430ac1505cf955 : SUCCESS in 12m 30s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/0ac9572d6bbc4981b5ea93ec791e3739 : SUCCESS in 38m 03s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/53b768c5c83f40d8a21a7c004b681ae5 : SUCCESS in 44m 42s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4c2a37b222154371b8598346d2336974 : SUCCESS in 18m 54s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/913ac857ef184a9eaaac3c5e68ef26be : SUCCESS in 15m 51s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c39f07e5aa3a40ff91bf043d6a578a5c : SUCCESS in 13m 47s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c97f5c83d9b04bbab06f5b551dca34d5 : SUCCESS in 13m 47s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a75fb41ea6524d42b548f5f13620fd59 : SUCCESS in 26m 25s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3039c5cabed74af7a0e6c1133c77b9b4 : SUCCESS in 10m 06s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d787ae52d25840939d22052c95f7f659 : SUCCESS in 10m 54s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/57f7debab1af4a9a9f315da1f2f6bf95 : SUCCESS in 18m 07s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/884d7d812b02415f9b973a2955560351 : SUCCESS in 12m 36s","accounts_in_message":[],"_revision_number":1},{"id":"d834ebc1c9c55c5fcedbd2912a1b4269413f5591","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 11:06:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dbb5cd3645459c509ec36e1ca725b558a892bd5d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 11:07:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b15edb5bd3a49fbb9dab169cd3e72a8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d324203b92c4da39a5292b5796dd008 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"abff761c364b7488683e34731e03fe3b646d2ebe","revisions":{"abff761c364b7488683e34731e03fe3b646d2ebe":{"kind":"REWORK","_number":1,"created":"2023-05-30 12:39:39.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/40/884540/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/884540/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/884540/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/884540/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/884540/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/884540/1"}}},"commit":{"parents":[{"commit":"12f80e9e7d531039a5f7f1498dd8db4a85f8742c","subject":"Merge \"Adds support for custom Multipathd configuration.\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/12f80e9e7d531039a5f7f1498dd8db4a85f8742c"}]}],"author":{"name":"Erik Berg","email":"openstack@slipsprogrammor.no","date":"2023-01-10 16:33:53.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-05-30 12:39:39.000000000","tz":0},"subject":"Limit interface names to 15 characters","message":"Limit interface names to 15 characters\n\nIf you give your bridge a long enough name, eg. `br-external`. The\nextra characters added to the veth pairs can make the interface name\ngo beyond the 15 character limit. We can solve this by truncating the\nname of the bridge used in the veth names.\n\nChange-Id: I5b890e24195d033897a597a0a93a1cacfb2030d2\n(cherry picked from commit b9b90a5d5ff38e5a64b8dbdd8d3dea32d097b3b1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/abff761c364b7488683e34731e03fe3b646d2ebe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/abff761c364b7488683e34731e03fe3b646d2ebe"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
