)]}'
{"id":"openstack%2Fkayobe~884541","triplet_id":"openstack%2Fkayobe~stable%2Fyoga~I5b890e24195d033897a597a0a93a1cacfb2030d2","project":"openstack/kayobe","branch":"stable/yoga","topic":"truncate_veth-stable/yoga","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-09-21 15:19:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5b890e24195d033897a597a0a93a1cacfb2030d2","subject":"Limit interface names to 15 characters","status":"MERGED","created":"2023-05-30 12:39:46.000000000","updated":"2023-09-21 15:20:54.000000000","submitted":"2023-09-21 15:19:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"884541-truncate_veth-stable/yoga","cherry_pick_of_change":869744,"cherry_pick_of_patch_set":3,"meta_rev_id":"383c8650e603a66797ba10d79f34b1cc5de3c94d","_number":884541,"virtual_id_number":884541,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-21 15:19:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"all":[{"value":1,"date":"2023-05-30 13:25:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":2,"date":"2023-09-21 10:00:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-05-30 13:18:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-06-01 11:08:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2023-09-21 10:00:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-05-30 13:43:08.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":1,"date":"2023-09-21 10:00:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2023-09-21 10:00:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-30 12:39:46.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"state":"CC"},{"updated":"2023-05-30 13:18:27.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-05-30 13:25:47.000000000","updated_by":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"reviewer":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"state":"REVIEWER"},{"updated":"2023-05-30 13:43:08.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2023-05-30 14:40:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-21 10:00:27.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-09-21 10:00:43.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"c944cd5ec53ab4efb7ada0c85b28ef887a9b0096","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-05-30 12:39:46.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5ee38bb009183c6355bfec827fb3e6a01e987401","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-30 13:18:27.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f8dad46be01f192b3c99c3d2d1f762d66b3b844c","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2023-05-30 13:25:47.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2bb5a5cb86f98ec7af1864da14f7636fccd88293","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-05-30 13:43:08.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fab2323432464628942a63213e0ca15382095307","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 14:40:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8687591ea109467d9bbfe76d2912bf40\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25b083aa99364fdeb9936badb6928429 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2e75c1205e3427395b59151390531cf : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96f5f943ef15406086bb3c86b9f04601 : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/749fa4b754a24ff99cde2df015da84bd : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a2d9719154e4a1195b9564060199a77 : SUCCESS in 9m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bcba730ef04b4eddb4786a1f2700a33b : SUCCESS in 6m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d32a62f7914f4d088b49a03cba2a315b : SUCCESS in 14m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/920fdbe2b82b4f4b901fd6e377e3ef40 : SUCCESS in 17m 30s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/965b7ccff11044d1b7bf2b108b6c9de5 : SUCCESS in 1h 04m 32s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/175846c399a646ec91b5b9d9eb617378 : FAILURE in 1h 01m 17s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/4f26e9b88d9242a69df83f140e9df3c2 : SUCCESS in 1h 06m 53s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e4afb3ad507c4db6b7451e761e8de47a : FAILURE in 1h 03m 11s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56d25f099e86406b89cb4628d754f008 : SUCCESS in 51m 06s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce1342fdad1346aba5f47a164a5c2c06 : SUCCESS in 50m 32s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/77eff5c37b2b49ed9056d926f09eb333 : SUCCESS in 23m 36s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/3e2e69014f7a42a29ebd9b648cac5064 : SUCCESS in 22m 37s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/a65141e685f4489ab0010ff9121958dd : SUCCESS in 21m 07s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/399a79073a674d4ab86574d08c03cb5c : SUCCESS in 23m 01s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b656c44662c47b5a83c72f3a606479b : SUCCESS in 16m 11s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b6d89539a8a1456e9653c2f9ac8edfec : SUCCESS in 19m 24s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/1cfb06bd1b5a4d60950311e4e096e3bb : SUCCESS in 1h 17m 43s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/3d3f9507c98c4b96b8075001e375e5c7 : FAILURE in 1h 10m 12s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/85640f8c3bbe4c17b59809f79267c3f8 : SUCCESS in 1h 58m 40s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54e0b0233ccf45c681a4d80c5a5606d0 : SUCCESS in 1h 18m 45s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e2c8a68d9115458aa4b5c40dd8f5adf7 : SUCCESS in 19m 57s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/725bf0babdbf46e4b3386adc2c40981c : SUCCESS in 20m 40s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/71c4350ca2574902ac3406a2ed132e4d : SUCCESS in 21m 03s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2aa3301babcd4373948851c04514cf70 : SUCCESS in 16m 51s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7fb785de408c455599a6e96ea8459fe1 : SUCCESS in 12m 57s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f8adf3befc1c47ac9b95fb8a8db17c18 : SUCCESS in 30m 06s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa5956287aee48db9e4662a895d28c69 : SUCCESS in 24m 09s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e5294ffca0054c9a9c64f3764c033967 : SUCCESS in 17m 58s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/e98366d25faf4059ac308a7cd1acee01 : SUCCESS in 13m 41s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e05f09289b4443e1804c8a02b247c25e : SUCCESS in 16m 24s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7d616bff03a4a96be39c4c06b0d2b0d : SUCCESS in 11m 54s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ad8a05fad6c462589d4a8df639fd29d : SUCCESS in 13m 41s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/52a1fba2664b4203808a8095b76a8591 : SUCCESS in 21m 33s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/96eed995880541819f0125df120655fe : SUCCESS in 21m 16s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/324a8155e4074b1e85ef7e590e2b3bd1 : SUCCESS in 15m 00s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c9812a656e247b99326385ecf5db982 : SUCCESS in 14m 08s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f818debf03a34e54a71a605baea428e9 : SUCCESS in 15m 24s","accounts_in_message":[],"_revision_number":1},{"id":"ddbbe8d0d057786a6fd852ce51ad4d1484b21f45","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-06-01 11:08:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3811c1c559b7960646830ae14d0e41d6e849242f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-09-12 08:50:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c742c3983a69b8ae4f65180fb991a1cf5fda9783","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-12 10:50:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c2fcb86595543c086aec44308f7d704\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/078e60f3905f4e2d8f2b9d7ca3baf89a : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d5dbfc136914b979196fedd6e5a7d53 : SUCCESS in 7m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ad7add3e6f74840aafc41b3126ef061 : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9f2a1ab9a594649ad67df1cd79294e3 : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d3839f829424d639bff8fa1d91120db : SUCCESS in 7m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7a44cc2e496f4aed9c8a3f3fbe98011f : SUCCESS in 7m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5ed686cb832e4e49b74e6d11abd6bb2b : SUCCESS in 16m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0df17d47e9f0408086faff46a87e9e56 : SUCCESS in 15m 42s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/abb1ef0355f3421b88bb59ce11bd3640 : SUCCESS in 1h 06m 24s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/6725b29da1ea42d9b20d242c8fd6d94b : SUCCESS in 51m 20s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/6ed6a7a305304c2c89bcbee6c42c3392 : SUCCESS in 1h 03m 00s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/751bf2b9d56447e89ea9890c97fe04e2 : SUCCESS in 47m 36s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f31ce76700644e528f13a2179417d4c8 : SUCCESS in 25m 33s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/a87c2fa4a6ef49d0ba8c5f8360e493c3 : SUCCESS in 20m 15s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/558cf331887941f782bec27de9825f14 : SUCCESS in 22m 15s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9a5dbd7727ec4b4097b80b937234c3c3 : SUCCESS in 23m 38s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34c976faf91943b19d4c3c103dc1a77b : SUCCESS in 18m 06s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b5c0ed1a3d204778acd45dce97bb2002 : SUCCESS in 17m 39s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d750da26873444cd87b6b1a687355335 : SUCCESS in 1h 14m 53s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/74cd30fd6da4414e8ff68de8478bb339 : SUCCESS in 53m 46s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f4c3265f77674160b05bc748b8c9f6ab : SUCCESS in 1h 55m 44s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/80be0d345ec9423c94241591c86a7f42 : SUCCESS in 23m 26s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/018e561dc1e042da93fb3995b4eeaef1 : SUCCESS in 19m 26s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/96bf5ae9eb084df1b7993413d32543b2 : SUCCESS in 21m 13s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b44e4702ff21478eabe2401672fb9f62 : SUCCESS in 17m 25s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e6710e146f314cd48486e37c3bddd7e6 : SUCCESS in 16m 52s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/42ff147330b54761b4a4f9ec13eea720 : SUCCESS in 28m 56s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fdaa940f54394a51a5d0dcddd3b019bf : SUCCESS in 28m 00s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b993b85426984641a2110b882180d0a9 : SUCCESS in 19m 52s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/1adf6a6387e3402c8e16d9813866ed60 : SUCCESS in 17m 54s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8bab79edd43c4c418a6369b1d9818d13 : SUCCESS in 17m 29s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2cb615ba4cfa4d8fa7cccf74de986a04 : SUCCESS in 12m 58s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f5a76c10265f41889809ea196298f017 : SUCCESS in 11m 18s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/48d100ef3f874a0d8fd12aa3aa2fc745 : SUCCESS in 22m 50s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/333a52904f0e45cd952535b521837c65 : SUCCESS in 21m 34s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c82673ff748549608912aa5dfbe246c4 : SUCCESS in 19m 29s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/edd7504339b94e95abfff795919a9a3b : SUCCESS in 15m 03s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/94918672ee7c4c408d96345935b53b72 : SUCCESS in 13m 55s","accounts_in_message":[],"_revision_number":1},{"id":"d3386af21e4d3bca0c1c94a90257b65886bb79bd","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-21 10:00:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d52e1a9d4002968eca252b7c3c6db5595698b442","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-09-21 10:00:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"21ab6bef1bc211b648af49b28cbd2550d49d44bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 10:00:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a17b5f11e19908095a13f796489d8d6c99c80703","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 11:41:44.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b7a346e33ff84aa1afc5b6364b0880e8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d3a4c383dc04b67aa138b24363bf3ca : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5be23eaf516744e7938b364125afb45a : SUCCESS in 6m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fff1678d134a4b32a411ca6a15352975 : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00ba71cd1e784b7db2e7b19ef3187814 : SUCCESS in 12m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/618cd374d31a4debaac5ac89c97059b5 : SUCCESS in 7m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2abfa6b0fd324f77a9a389af992db728 : SUCCESS in 15m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7c398fa0ee574815aef294556418680d : SUCCESS in 17m 15s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/729b7e2905374e12a60ee49e7882d03c : SUCCESS in 1h 16m 24s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/f6566c6f1daf4874b07aa1d947292bbf : FAILURE in 16m 13s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/14dc5975aec8469194473485c4d09502 : SUCCESS in 58m 41s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e93bec14537044049e96801f030c832c : SUCCESS in 24m 54s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/2aa575e2d16748d88d3c5a4150057056 : SUCCESS in 20m 26s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ccfaad710f8d4dc7a95ea304f96b83b3 : SUCCESS in 20m 03s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c48ce5a980646c999a0ee9eb31b12f5 : SUCCESS in 15m 39s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/62de7ecf51514c4e8a9596415905df54 : SUCCESS in 16m 10s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b0bb84276f3744fd8eec24491c18a26b : SUCCESS in 1h 07m 19s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/889225a85a884c96828211d718dc0ca9 : SUCCESS in 43m 41s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/347ae8d0a0394c18b04cf274cc19c81f : SUCCESS in 1h 40m 44s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/1e75439dcada402d9c397756de65e722 : SUCCESS in 19m 20s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/99c19e6bb9124603a9515fbb7493764b : SUCCESS in 17m 52s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/deaa00f144f14fecb5f6f75098e5212a : SUCCESS in 22m 34s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f85887b103bf49128b0fb2f75a5e768e : SUCCESS in 14m 57s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/25abbc1fa047479d9716f78a8e5803a3 : SUCCESS in 19m 09s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/837454331dbd4494a545bc176b535e34 : SUCCESS in 32m 28s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ab1be9419f2442e8d4418f49cc34546 : SUCCESS in 22m 41s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/4cb002dc5de94fb982e4eb307556ebaa : SUCCESS in 18m 54s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/30c832b4cafe476a8908abe14bd1caa3 : SUCCESS in 17m 53s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ef9057e507b94adca6c8c3caa38174f8 : SUCCESS in 18m 44s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/189090e370104ccabd9f03d497d97678 : SUCCESS in 13m 27s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d36a9d8469164216ba19039b34287e00 : SUCCESS in 15m 07s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/bb870c4262414ccc9a99a517eaf67ef8 : SUCCESS in 22m 27s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/683d64acc722462a85e638e2279796e1 : SUCCESS in 19m 18s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/13a68a275f964ee182c0cb8fe3696128 : SUCCESS in 17m 57s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62680f8c352446fb8f28feecf32910f3 : SUCCESS in 13m 13s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c8b0c8c2f6a4bcd883af1e91e4e4b2c : SUCCESS in 14m 57s","accounts_in_message":[],"_revision_number":1},{"id":"4392e6892e99ae56f55725df228d0681fe42cfe6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-09-21 11:44:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"952ed1bfaa0f1dd2b6d8b92eb69f93f66a7332f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 13:15:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5650a56e75bf47b5a465b9b2071e4ff4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e1510248c3b410c9311ae6af81dbaee : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49f2d9c6f0304ac9bec21bb892feb638 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f92f5fec32ab402c839208ece30f66c3 : SUCCESS in 7m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c9998fe0df84ec0874f874d8402ad95 : SUCCESS in 6m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f00c1f1fb7248e1aa81dbba02eb4d03 : SUCCESS in 10m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ab52316ce4f54e8caf4736896531dccd : SUCCESS in 6m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/43e2171a61164762a7ad46bb243540f5 : SUCCESS in 13m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a4702c117afe4c48a95eec8729eddfcf : SUCCESS in 17m 18s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6bced3bb9a6a43d09ddf14e09e943f47 : SUCCESS in 1h 00m 33s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/10f7eb1b099e45df82921c3203def20c : SUCCESS in 54m 12s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/b025a63cb4ec4814b6a50ec542181d0d : SUCCESS in 1h 05m 08s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/603c7724c971432aa0664591db2c88e5 : SUCCESS in 51m 21s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/00e44686b0d24a3ea0ba33b53c5c9546 : SUCCESS in 27m 07s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/99a5cbe329834dbd9efe0170617c52eb : SUCCESS in 19m 17s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/38031ba542f84689916e8a1a6b275352 : SUCCESS in 23m 23s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/18c04251354c4f1eb20446593934fc0c : SUCCESS in 21m 16s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6651e1082ba24f999033c54556bb5fb9 : SUCCESS in 16m 49s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/98b861e5d8104919bdd7e927df1db788 : SUCCESS in 13m 28s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5d0d64962ef24ed28938ee255a930cb0 : SUCCESS in 1h 12m 44s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/2118b648d3e243a2931c30d3d347a34c : SUCCESS in 1h 02m 21s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6a100598d51c4aa69106a70cbd8b5526 : SUCCESS in 1h 29m 20s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c8bb4cb7e8fc40309eb627f74021891e : SUCCESS in 22m 51s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/f4b45a88016640e3b8ac6fdb5014814d : SUCCESS in 22m 32s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/59fd0d807a2d4d13afabc0caeb9c2946 : SUCCESS in 17m 38s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84cd6bb528c64bf98bdc00cf424e81ae : SUCCESS in 18m 42s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f126a06a666549419d55591744c8b368 : SUCCESS in 15m 13s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/853a950a0c4a4425ae8f20b7f2f40c7b : SUCCESS in 28m 53s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/64fc414c6376445f9fd9fd663a6edc34 : SUCCESS in 25m 19s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/d1295cfcbb9b408d910206ad33bc34f8 : SUCCESS in 19m 40s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/c350828134e9467d958a7763034246b5 : SUCCESS in 16m 41s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/87cb27a549a1407687dc0ef976cdcc82 : SUCCESS in 15m 59s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d2267d766eb348a18c26f3763f997e83 : SUCCESS in 14m 02s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d8cce71aa214b9d9402d3c2ee494e02 : SUCCESS in 13m 42s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/70668c0e89794d58929b24c7956b1bf8 : SUCCESS in 22m 02s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/2c065efaf8dc4fb2ae9a5f04c29a1f6d : SUCCESS in 22m 32s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ee9d51176834408dbbe5bf85880e3d2f : SUCCESS in 16m 45s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a400505b4f104eea8b356c3a89b1ac5a : SUCCESS in 13m 33s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc01ed95c93945cbb5d4f858cb6ddb16 : SUCCESS in 15m 54s","accounts_in_message":[],"_revision_number":1},{"id":"f003b5723411f73bd55f1e866ba4feb825f598fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 13:16:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b4832149d680c44dfca421a926b7e121e311429f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 15:19:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae72c38289b9401baedc42f0219e626e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc0165e25d334650b63965f3a6747f59 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d5928953acf4589a64426b1829ed1d0 : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16f329b3d1ba424084a8933952e38084 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f02bfac728054639a7c9f29436cfe55b : SUCCESS in 8m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fbdbd1a28c83481cb0bea2894948b61a : SUCCESS in 6m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/09311f17e5194531b0664b39817b2ca0 : SUCCESS in 16m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d34216097e954682a33372220328190d : SUCCESS in 15m 37s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d2c1c803367341f7800cc10ea85dc661 : SUCCESS in 1h 13m 22s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/b6addee747a6442681bb082505834801 : SUCCESS in 1h 20m 03s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/7cfa0fa9d9ae48d8bf2f8a3eaeb1e31f : SUCCESS in 1h 01m 20s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/36f3247123cc445187b3382f6831b8ba : SUCCESS in 24m 35s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/f15e697f1bd34c31af8b8be07e023e86 : SUCCESS in 24m 20s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/bddea41fdc4f4cacbb954cef0e47a634 : SUCCESS in 22m 38s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36f43fb7b85b4842a15b5f3f933f07b4 : SUCCESS in 15m 45s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e86c08cc6434dae89a02da7abc02b98 : SUCCESS in 17m 48s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b8d7449e250e41e8bfc04012106c60d9 : SUCCESS in 1h 22m 50s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a3d309c530ae47599e6d81363f223c92 : SUCCESS in 51m 00s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fe881b954f3b49a9993c905f3a2527ad : SUCCESS in 2h 02m 20s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/bb20c7f09db84ea39c7bdfa32951bb39 : SUCCESS in 24m 26s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/e0f544fd089a43cd9e01da6e2f8c59db : SUCCESS in 21m 32s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4b24a241f8f44fbd920c99667cd95254 : SUCCESS in 17m 28s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9ce92b0955f49b9b7f5d8c93eb7db0d : SUCCESS in 21m 26s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e2d08c3de4a495b8e29136dba42a1fa : SUCCESS in 17m 39s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/78d51ab7b88b48f78dab2cfc3884d3d3 : SUCCESS in 27m 00s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31ce9da87b6648fa8d82ffde19c789bb : SUCCESS in 22m 58s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5fb6eded6a2c4066b09217185e72c98a : SUCCESS in 16m 57s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/89473fa71bdd46609125c6323ebb7d4d : SUCCESS in 14m 53s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0792d9844a98464089be902271af0850 : SUCCESS in 15m 09s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d53805ad9f854aaab7faef1b042d8207 : SUCCESS in 11m 43s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/031c3414f1454cf89bdb31426dde0254 : SUCCESS in 12m 34s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/c4411d32295a4bd2908c3dda9948944a : SUCCESS in 23m 08s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/ba8324c6241d4b49800a99dce00d8bf8 : SUCCESS in 19m 50s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/72fae555c0be4a9b83997612314d11a6 : SUCCESS in 18m 56s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/729c5f05a6044ffdbe86800b91e38288 : SUCCESS in 14m 08s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69b2881d50634dcf9376f7a85c3972f3 : SUCCESS in 14m 00s","accounts_in_message":[],"_revision_number":1},{"id":"0c4e847abe8735e8928720705e01b8476a163d32","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 15:19:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"383c8650e603a66797ba10d79f34b1cc5de3c94d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-21 15:20:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dffdf2714e4748618353356a441fcc68\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83a9889cbeeb47edb23e4d72dddb4716 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0a284020ad7530c87f06034f82075e918923d59f","revisions":{"0a284020ad7530c87f06034f82075e918923d59f":{"kind":"REWORK","_number":1,"created":"2023-05-30 12:39:46.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/41/884541/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/41/884541/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/884541/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/884541/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/884541/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/41/884541/1"}}},"commit":{"parents":[{"commit":"79159c897d46c9cb71c2fac2d429e6e5365e71cd","subject":"Merge \"Adds support for custom Multipathd configuration.\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/79159c897d46c9cb71c2fac2d429e6e5365e71cd"}]}],"author":{"name":"Erik Berg","email":"openstack@slipsprogrammor.no","date":"2023-01-10 16:33:53.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-05-30 12:39:46.000000000","tz":0},"subject":"Limit interface names to 15 characters","message":"Limit interface names to 15 characters\n\nIf you give your bridge a long enough name, eg. `br-external`. The\nextra characters added to the veth pairs can make the interface name\ngo beyond the 15 character limit. We can solve this by truncating the\nname of the bridge used in the veth names.\n\nChange-Id: I5b890e24195d033897a597a0a93a1cacfb2030d2\n(cherry picked from commit b9b90a5d5ff38e5a64b8dbdd8d3dea32d097b3b1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0a284020ad7530c87f06034f82075e918923d59f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0a284020ad7530c87f06034f82075e918923d59f"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
