)]}'
{"id":"openstack%2Fkayobe~884542","triplet_id":"openstack%2Fkayobe~stable%2Fxena~I5b890e24195d033897a597a0a93a1cacfb2030d2","project":"openstack/kayobe","branch":"stable/xena","topic":"truncate_veth-stable/xena","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-08-02 15:02:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5b890e24195d033897a597a0a93a1cacfb2030d2","subject":"Limit interface names to 15 characters","status":"MERGED","created":"2023-05-30 12:39:54.000000000","updated":"2023-08-02 15:04:00.000000000","submitted":"2023-08-02 15:02:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"884542-truncate_veth-stable/xena","cherry_pick_of_change":869744,"cherry_pick_of_patch_set":3,"meta_rev_id":"9a78778dc45c8f97c490105a17227760a2cad3d3","_number":884542,"virtual_id_number":884542,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-02 15:02:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"all":[{"value":1,"date":"2023-05-30 13:25:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":2,"date":"2023-08-02 13:13:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-05-30 13:17:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-06-01 11:08:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-05-30 13:43:17.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":1,"date":"2023-08-02 13:13:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-30 12:39:54.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"state":"CC"},{"updated":"2023-05-30 13:17:07.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-05-30 13:25:57.000000000","updated_by":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"reviewer":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"state":"REVIEWER"},{"updated":"2023-05-30 13:43:17.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2023-05-30 14:48:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-02 13:13:45.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"6e1b6f8a577abd7205f14a40ebcf04fbd6852542","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-05-30 12:39:54.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b3285c457a7321d40deb5e4e5f68a8b20966ef9c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-05-30 13:17:07.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"81fa0df698269127825e24d83def880c39961465","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2023-05-30 13:25:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f05943ea88339825b4299cce365aa2c33490a7d5","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-05-30 13:43:17.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"26bb961c1e4a67349c61e896c12bacdfdc0916db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 14:48:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/050c5c24bb3b4b2db33fb149ebe9442c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/663dc43963914e6f91e929c719d288d1 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1b40c3aff2244be846725ecc82150ad : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20530357679a4455bc800407305a049a : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a12c390049774b8aa6d5ec48f8c735e8 : SUCCESS in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7baec5e80c64d7a8099dd1071a4cc05 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b07badaac99646c0950caf38b9a5f76e : SUCCESS in 8m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a854db361f024896b3c1485ad3c2b8d8 : SUCCESS in 5m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/589f7d8ab08049e3885d179889286d3e : SUCCESS in 13m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/97f436fc8d2d48b2bdff4b81b124097d : SUCCESS in 16m 08s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/89b75cbfcde6473883e876961e0b9d4f : SUCCESS in 1h 05m 44s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/a840fb97c03e4769839f7e67cf2bd1f7 : SUCCESS in 1h 06m 24s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d71bbddae904d04a45ce69bf6e4ecad : SUCCESS in 46m 27s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/b0e3c45faf174c56a1850c6dbebebe89 : SUCCESS in 1h 09m 06s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/9b755bcc477d4e06ab79b5a22e13b193 : SUCCESS in 26m 36s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/b4ea1d2625fb4bfbbd2e50d0458622e2 : SUCCESS in 24m 53s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a44d3f86b98445bebd9cb49cdcfc734b : SUCCESS in 18m 40s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5665c604d6b448f68efda0b519c7ef8f : SUCCESS in 2h 03m 37s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8274011be4ac4a359387eee8e60365fa : SUCCESS in 1h 28m 26s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/b239afc9d45d49f6b7023c111378d4fd : SUCCESS in 20m 19s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/2d7ffb291b2146b3b0bb7c638f9da5f8 : SUCCESS in 20m 51s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a402581c9734735880dd369823acba6 : SUCCESS in 15m 58s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/9a6fbce6195942918b0b762d90908ba7 : SUCCESS in 41m 57s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e9ff3d27c469427387ca8f164bdeac72 : SUCCESS in 31m 26s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c65cc9945df474d80bd0733777ee16b : SUCCESS in 25m 47s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c95c68cdf5a145c395f8e9b02815451b : SUCCESS in 17m 01s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/d82a0b9a60684bd781b820388992c46c : SUCCESS in 19m 24s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e19686bce5944d95b3f7b938623ab40d : SUCCESS in 11m 15s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/0a3058c1d84d48c4833384299cf7a66d : SUCCESS in 13m 59s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/4e9921b76b3f4b68aa58fa227f0378d4 : SUCCESS in 18m 47s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d48594bf56034fc18296b6aeae2e5071 : SUCCESS in 13m 38s","accounts_in_message":[],"_revision_number":1},{"id":"70e3a2c6c4be06b46bcaacef71628c0fc59b6a07","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-06-01 11:08:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c79437723c5163f66efe0fba4aa68a0879a231b5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-02 13:13:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e5fae7565fd776220aba9be59ab84a705fbc61fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 13:14:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"45ee87d9a4709cb68c3cb15209760c27a0e09073","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 15:02:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a4c774b366540faa1df71901e5e9b89\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/952c0502480141c18f19ac3417f9e11a : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff1e9ed808a9465aaa402f55911abd14 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1ad031e0ea54de78b31004ccbc70de7 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/633e3c06e705421d8713723b3e87cf05 : SUCCESS in 9m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dc84f398cde94bbeb1a8bafb47b3305f : SUCCESS in 6m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/74a705dde62746a9842d7ab9ecb1b868 : SUCCESS in 12m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7b3b16a7a22641b3bbf431375c4f4745 : SUCCESS in 25m 20s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/09cbe0b860584638b111bb68016a855f : SUCCESS in 1h 20m 49s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/6a1c3fcfe9424f0babf04eddb538112f : SUCCESS in 1h 06m 49s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/bc09b6d1ed0348588f1c6a7c7017115e : SUCCESS in 1h 14m 02s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c1a780498c29407b8cbf774c5fc5fb29 : SUCCESS in 26m 36s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/363fa61cd19a48668e7b06b96eaf104b : SUCCESS in 29m 02s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90c92996b5e74a0ab6ce4f65c984d113 : SUCCESS in 16m 55s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e27b3cf3412c4493a3aa2002a67bffcc : SUCCESS in 1h 47m 20s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/87d8943974594fa29e7d5c6a2d74ff5a : SUCCESS in 27m 57s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/082511e3337f419e95271328d6b807e8 : SUCCESS in 17m 47s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/485ed35c47aa4f05a53a01ae33eb14fa : SUCCESS in 17m 40s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/15e1520e21654800b0fbb61ab98dbf2c : SUCCESS in 33m 03s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ade71ca615b4dc7a059f6b0e9456e77 : SUCCESS in 25m 36s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5cf5740dfe504cf3a33c26953e992fb0 : SUCCESS in 19m 47s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/8e41d6d84726417cb71495e71bee1901 : SUCCESS in 17m 44s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd83884363d04fba9682b1aa812d1132 : SUCCESS in 13m 49s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/a6ca9d84af274970b8ebc4fad118945f : SUCCESS in 25m 22s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/adb92190a95748489a7fe857e7d30cc6 : SUCCESS in 19m 13s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e7f2d39425d45058963d8c4366fa88f : SUCCESS in 13m 31s","accounts_in_message":[],"_revision_number":1},{"id":"7d53e814ddcc983c389f898799ecc54fa449c4d4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 15:02:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9a78778dc45c8f97c490105a17227760a2cad3d3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-02 15:04:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1185370e4a9143a2af4be32a145c2305\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/780bfbacb9be40efb52fe6055c5d61f4 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ca6309a38d761d0ae3c3fffe0c23ae50a270ad0c","revisions":{"ca6309a38d761d0ae3c3fffe0c23ae50a270ad0c":{"kind":"REWORK","_number":1,"created":"2023-05-30 12:39:54.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/42/884542/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/42/884542/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/884542/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/884542/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/884542/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/42/884542/1"}}},"commit":{"parents":[{"commit":"190611fbadb68a8631893625dcc5423590b0b54d","subject":"Merge \"Adds support for custom Multipathd configuration.\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/190611fbadb68a8631893625dcc5423590b0b54d"}]}],"author":{"name":"Erik Berg","email":"openstack@slipsprogrammor.no","date":"2023-01-10 16:33:53.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-05-30 12:39:54.000000000","tz":0},"subject":"Limit interface names to 15 characters","message":"Limit interface names to 15 characters\n\nIf you give your bridge a long enough name, eg. `br-external`. The\nextra characters added to the veth pairs can make the interface name\ngo beyond the 15 character limit. We can solve this by truncating the\nname of the bridge used in the veth names.\n\nChange-Id: I5b890e24195d033897a597a0a93a1cacfb2030d2\n(cherry picked from commit b9b90a5d5ff38e5a64b8dbdd8d3dea32d097b3b1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca6309a38d761d0ae3c3fffe0c23ae50a270ad0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca6309a38d761d0ae3c3fffe0c23ae50a270ad0c"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
