)]}'
{"id":"openstack%2Fkayobe~885014","triplet_id":"openstack%2Fkayobe~stable%2Fyoga~Ie272e701b8952ef106be99b8fe36eb04cb524898","project":"openstack/kayobe","branch":"stable/yoga","topic":"update-gitignore-vscode-stable/yoga","attention_set":{},"removed_from_attention_set":{"35264":{"account":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"last_update":"2023-06-05 12:19:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie272e701b8952ef106be99b8fe36eb04cb524898","subject":"Update gitignore for .vscode/ configuration files","status":"MERGED","created":"2023-06-01 13:38:15.000000000","updated":"2023-06-05 12:20:05.000000000","submitted":"2023-06-05 12:19:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"885014-update-gitignore-vscode-stable/yoga","cherry_pick_of_change":884694,"cherry_pick_of_patch_set":2,"meta_rev_id":"f7ad5a3bcbdf4a4209e80577d2a8667fa4588802","_number":885014,"virtual_id_number":885014,"owner":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-05 12:19:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2023-06-01 13:58:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-06-05 10:04:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-06-01 20:44:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-06-05 10:04:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-01 13:58:09.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-06-01 15:54:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-01 20:44:46.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2023-06-05 10:04:11.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"8810c5e31595db0e60961567000cb0d74dcf764d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2023-06-01 13:38:15.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9ae8ee5f4e669ff71d29a04cc405883cc19deca2","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-01 13:58:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c2ab9a16785fcac78fac6865a0f9bc7cf94750dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 15:54:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06ffa21a02d54c8b8c3f3a6e673a6d18\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c27f0c8fd56044e3bbebf82de5ac38a8 : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c397b460e07446078f0cbc9313a72f92 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca4df91e1f824baa940b8ec39cb63b6b : SUCCESS in 6m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a492db8ba91466f9d26050fcb7cc398 : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07cfca2be8e2478fbd3b063f5b686fd1 : SUCCESS in 11m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/250098e099134161bcb4c646fdbc6c43 : SUCCESS in 6m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b3b9eb23f4d844fa8bd7925a630809b2 : SUCCESS in 12m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f1874aedbd11440982c05a0f9620218a : SUCCESS in 17m 20s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/8cf3feb7fe414272ac4a9bb71bcf7eb7 : SUCCESS in 1h 05m 19s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/cd016b25d26a443bb5a2b55232492446 : SUCCESS in 47m 04s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/cba9a996d6b646a690984b88ea902656 : SUCCESS in 1h 40m 45s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/fb5b608b91f8467e8a9a09ccfe859355 : SUCCESS in 53m 44s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/687e033ffb11450bb861248112f46b17 : SUCCESS in 55m 52s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/34ee5a453337495187f6851d5a99be2e : SUCCESS in 50m 47s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/21e5be10eea5443ca9578954fad34694 : SUCCESS in 25m 35s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/dad9d0180acc4cd2968194a3cd7dd0a2 : SUCCESS in 23m 22s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/fdd97238d8ad4495be8224f888fbc6b4 : SUCCESS in 21m 37s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/66095bfbe48c4574a5883820835e5bee : SUCCESS in 21m 32s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c4a2d35e09f4d0d92882f9eaff0f426 : SUCCESS in 16m 17s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f3a931a4558e48b5a5774a6af26c5ee5 : SUCCESS in 15m 28s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/79f186b6266a415aae606e9b7ae7a272 : SUCCESS in 1h 12m 44s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/5cb2c14b77ac4b95b9e2eb9e3be6b990 : SUCCESS in 1h 01m 31s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0676f114692245aa9aebe2a4b5b3040e : SUCCESS in 2h 03m 38s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2cfd156be6ea4daf8db1dad1dc63131e : SUCCESS in 1h 40m 09s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/f6f09d1a7c25494389af1bd0890bb88c : SUCCESS in 24m 15s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/bb18286af52e48a99db5e5b5b29c406c : SUCCESS in 22m 34s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/7302c3ba99fa4455971be19c7432e618 : SUCCESS in 19m 14s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9cc8837e3a1e4345846f9159bb3e92db : SUCCESS in 16m 24s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c88cf969cd9f4a9cbe9c38fe15b1578b : SUCCESS in 16m 55s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c101ebe4e8ea4149881d724ab392c8af : SUCCESS in 33m 57s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bef5c11fcf7444c8bec8d93139464604 : SUCCESS in 30m 11s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c02ceb8578444390bc5975b4d609d832 : SUCCESS in 23m 41s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/f770feaec32e4c8099c3ed0ee47edce2 : SUCCESS in 19m 01s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d3c19a3339bb4b0095459d98269d3021 : SUCCESS in 14m 55s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d6c6bbd6bd44a7cbcf74c67863c4b67 : SUCCESS in 18m 49s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6934402cd5fa44e38d2f2c0dd849baee : SUCCESS in 16m 32s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/ab74f7b8b39e4ac7aba3991d9cf485c5 : SUCCESS in 21m 41s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/8830301987a54a3f98fea5321e305ec4 : SUCCESS in 20m 02s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/565690b4209d458298c5e6463d988b14 : SUCCESS in 17m 13s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01c4a672d60f4344908cf492bd530c69 : SUCCESS in 20m 39s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c878c07b8b9447be9ee34f254bea9f2d : SUCCESS in 13m 48s","accounts_in_message":[],"_revision_number":1},{"id":"adc490936428c28589182f07ea57da906b0d24bd","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-06-01 20:44:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b542ee9e46b7d31f2baa7de730f9bebcf4740723","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-06-05 10:04:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d1b75ead1a400bd1e24257c537a6abe52aabb9c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 10:04:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"28dd5f06aeae389205602f658454834ff27e4663","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 12:19:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2655b2d0f06f403993acb1add0e48c2c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17783141cbdb4447a7a452daebf828eb : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4ae48bfb9d9f473e80d6ea7ca712f40e : SUCCESS in 6m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55b9164c50034f118ab97f039252a88f : SUCCESS in 7m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14016d6618d640fdb26badbde1782ab7 : SUCCESS in 11m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7d51cef9e49c420d8f7005d76dca5be5 : SUCCESS in 6m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8356334177ff4c8194082e6b46725c22 : SUCCESS in 13m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2264ce515e3f4ab489665b8ae98e7263 : SUCCESS in 16m 02s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0ee45be44cce46e4aa7af4f5d30ca067 : SUCCESS in 1h 15m 23s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/e00ab698dbc648beb26c8e879863db3c : SUCCESS in 1h 10m 20s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/c9abf7fe33a8499cbd022c75a3c3b8f6 : SUCCESS in 51m 46s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed68e5d5373b4199b3babe58c39c89e3 : SUCCESS in 55m 28s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/352a429b65474506b7b45a21322d4a1f : SUCCESS in 58m 53s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/230bc689639f43ec974e0660b116b583 : SUCCESS in 28m 08s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/ba94191ffca84453861a15f402e7666a : SUCCESS in 25m 05s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c7fa5f53c06a4cacbfdf740c0b0674f0 : SUCCESS in 18m 52s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f3fde554ac04970b3d4b5f5043ac8b5 : SUCCESS in 14m 42s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c9e04cba39f463e82030e17c1e82e51 : SUCCESS in 18m 34s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/376a01abb6da4f1097f868ac86eaf786 : SUCCESS in 1h 26m 11s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/bd9cd565ebf941c895738141c01cb6e4 : SUCCESS in 57m 54s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2321337a537e45358cd6a999a7173a17 : SUCCESS in 1h 53m 34s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d65bea51b4e8447194a4f8ed454b9d3b : SUCCESS in 1h 43m 22s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/74e1eef066814b3f85918a966eb98e3d : SUCCESS in 23m 55s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/13fbf949768345238c61977db97c3ac0 : SUCCESS in 21m 42s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/82e44610a608411592a2d9e3fa006ac2 : SUCCESS in 22m 59s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c092e29bc9bd43eaa5cbf3ba8e7d0104 : SUCCESS in 24m 44s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7f1aeb9cda4741b39f256d07f86f5031 : SUCCESS in 17m 40s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/55f93f02ba3d48ddbc286e47a8f708b2 : SUCCESS in 38m 48s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/060287c7a6184b03bbd1a6be3580cfc6 : SUCCESS in 31m 00s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f84451a7ddff4360a94d50500f3e9ac2 : SUCCESS in 21m 12s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/7a81ae75999e49d393a21e9da1f4ffc7 : SUCCESS in 20m 56s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/19287285418648a98dea893071b5ae78 : SUCCESS in 16m 49s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/40f462b756564420ad3742e87565cbd6 : SUCCESS in 12m 38s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c1a2e8fd93b24fbf8c90864d272b53e1 : SUCCESS in 11m 17s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/9a2b1905c658464e90bcbed9aacecb32 : SUCCESS in 23m 35s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/cf5321d9c2da4fa6b231d748e8197366 : SUCCESS in 23m 16s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/afc50caba9d9438094fedf6cfb968769 : SUCCESS in 18m 55s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4a1be8bd83f450f8d53567936ed16e6 : SUCCESS in 12m 17s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1e18dca32c6c4c38b20588746e153aea : SUCCESS in 14m 47s","accounts_in_message":[],"_revision_number":1},{"id":"c06b85eed552a47d9808e2901c77bdba977cac8b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 12:19:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f7ad5a3bcbdf4a4209e80577d2a8667fa4588802","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 12:20:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79560e5fd58a4cf78877ef4c588c5474\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a72d156d9a51405c85799d8965291db5 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"64d2d6f2aeac7a6b85b9835d424c2b1720318f4a","revisions":{"64d2d6f2aeac7a6b85b9835d424c2b1720318f4a":{"kind":"REWORK","_number":1,"created":"2023-06-01 13:38:15.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/14/885014/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/14/885014/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/14/885014/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/14/885014/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/14/885014/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/14/885014/1"}}},"commit":{"parents":[{"commit":"74911c52daf74347cdc479c87bcd4a065a5bc535","subject":"Show an error when *_bridge_ports is not a list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/74911c52daf74347cdc479c87bcd4a065a5bc535"}]}],"author":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2023-05-30 13:15:57.000000000","tz":60},"committer":{"name":"Alex Welsh","email":"alex@stackhpc.com","date":"2023-06-01 13:38:15.000000000","tz":0},"subject":"Update gitignore for .vscode/ configuration files","message":"Update gitignore for .vscode/ configuration files\n\nChange-Id: Ie272e701b8952ef106be99b8fe36eb04cb524898\n(cherry picked from commit 47829142cd9ebc3ae7d7f440870491e81c3ddfad)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/64d2d6f2aeac7a6b85b9835d424c2b1720318f4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/64d2d6f2aeac7a6b85b9835d424c2b1720318f4a"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
