)]}'
{"id":"openstack%2Fkayobe~885015","triplet_id":"openstack%2Fkayobe~stable%2Fxena~Ie272e701b8952ef106be99b8fe36eb04cb524898","project":"openstack/kayobe","branch":"stable/xena","topic":"update-gitignore-vscode-stable/xena","attention_set":{},"removed_from_attention_set":{"35264":{"account":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"last_update":"2023-06-05 12:27:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie272e701b8952ef106be99b8fe36eb04cb524898","subject":"Update gitignore for .vscode/ configuration files","status":"MERGED","created":"2023-06-01 13:38:35.000000000","updated":"2023-06-05 12:28:47.000000000","submitted":"2023-06-05 12:27:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"885015-update-gitignore-vscode-stable/xena","cherry_pick_of_change":884694,"cherry_pick_of_patch_set":2,"meta_rev_id":"4e1fc6b5cce04ed7521cb69ccc3dda2b34741fb3","_number":885015,"virtual_id_number":885015,"owner":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-05 12:27:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2023-06-01 13:57:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-06-05 10:04:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-06-01 20:44:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-06-05 10:04:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-01 13:57:55.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-06-01 15:59:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-01 20:44:51.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2023-06-05 10:04:19.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"bdfcf55c13e8d5b5ef7f526a896e340707aa071d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2023-06-01 13:38:35.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f3dad97491d49ef97e7dbe1b116d4761bfda5513","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-01 13:57:55.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cd85eea8f3751a597d7b19859a132864bab2a5ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-01 15:59:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/804027ec83be4714a4b67db05c822b0b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8bb3908681cb4bf78d6a7972468a8c9c : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9975fa2a82824d9a817bd1cabcf9b4a5 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac6b5408295e4f1ea25135cb10b818cc : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a53322eff513440b9de4afae3c55c882 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da1bff8828e3491ebb2b116bb115715c : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db33cae2e8944e009674cea297577a5c : SUCCESS in 10m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/14a9530e8d70480eaf2370e81a218451 : SUCCESS in 6m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2179a6f4fe724abe9435910b17318597 : SUCCESS in 9m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2193b49be7a14323b712ed10fc088f34 : SUCCESS in 16m 53s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7db6e0576dbd4572b40023f873f94040 : SUCCESS in 1h 20m 25s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/669a6a5cfabc4c3bb0632a2acfc525e8 : SUCCESS in 1h 08m 28s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee29778b5bee490a92a6b3a6884c7b4e : SUCCESS in 55m 18s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/3a5a7b97f850472fbb4f70829fde1cb0 : SUCCESS in 1h 08m 49s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/3ab302aeec8b4e64ab9a30e7f51fd2cb : SUCCESS in 25m 19s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/3445a82c025842feb32e4af988b9118f : SUCCESS in 26m 03s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a5c09dc84214135adb820479c8d5615 : SUCCESS in 13m 53s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/920498f47dfe427f8cea98abed8d90ed : SUCCESS in 2h 05m 32s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5b5052efa28d4d12b68101e0ceee3316 : SUCCESS in 1h 29m 34s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/63e5f3df8a794c609b33f6815a593468 : SUCCESS in 25m 02s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/9d7a25b85874449d85fe229e32072dc2 : SUCCESS in 20m 00s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad6521ef07c44af08943d9104ca54329 : SUCCESS in 15m 28s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/98af920e97ff47fdafcbccd643c8ddd6 : SUCCESS in 40m 30s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/43f7a519d8ed454685eebc9d5f825d27 : SUCCESS in 29m 20s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a971099dc9cb477e82e32fd4fd2cabc1 : SUCCESS in 24m 11s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e7041040548f4c37b92648bd1e761a6c : SUCCESS in 21m 08s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/0af7449d6fd54b48a24f79ee50f64600 : SUCCESS in 16m 00s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ca7dc5a160824a2a8ea861e160106313 : SUCCESS in 11m 33s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/d66f76cbeff848cba6691454a97e8da9 : SUCCESS in 18m 35s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/f2bbf27c266d4ef09c2af919272eb7f6 : SUCCESS in 19m 37s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f4c3fa98b3bb4cae93d305a126933024 : SUCCESS in 15m 05s","accounts_in_message":[],"_revision_number":1},{"id":"8c4779ec2576baae407280dcc58b641ca5ca8635","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-06-01 20:44:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3bbb768cf23b030697760d56598d45184fbdb5a3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-06-05 10:04:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8563fe91236b9f9c0e5aaaa83772092c32301cb5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 10:04:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"316b2bf266a4bdd403aba7d3f4a505e0e559a507","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 12:27:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a5bde37ce2c4a44ba4ff193e69f15e0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7983edc61ff647b48380562ea9e2702a : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31e8b54e39fe4c2998fe7a982eda5ebd : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/daefcf94bb8e411f9edbabcc84aa9af2 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26c8a1830dea46ed93f3030dcd3d3bb6 : SUCCESS in 7m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c79b2734788d408cb64a9ae3d19cc966 : SUCCESS in 4m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b0e87a2e90cf4df49d82408288f05ba5 : SUCCESS in 13m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b350b915828649a1bca32c477df1d255 : SUCCESS in 17m 50s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/4486379793fa40db95a4b2ba8cfdfe3e : SUCCESS in 1h 05m 16s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/e0f54915713042bc99783fd1b9548d0c : SUCCESS in 1h 09m 18s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/985bba17024d42618ccb60e951012a51 : SUCCESS in 50m 01s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/df73a852faf846d2a587f35391ef1e9b : SUCCESS in 1h 08m 51s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/570583aa019a4fcea94858f4f63d7605 : SUCCESS in 24m 38s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/4899563ce02e4394ae4adfc8c67accba : SUCCESS in 20m 33s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/180fced89d8d466eac5d5b5837a0c4dc : SUCCESS in 12m 36s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/188df94f8ec345dd8a507b828cf4be9c : SUCCESS in 2h 12m 34s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/20793c8534c14297bd0cbda7ad9574d1 : SUCCESS in 1h 03m 41s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c813f4119acb4028a9c8c6ac3732955f : SUCCESS in 17m 50s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/dfbfdcce332a402bb96d8baebc6081b3 : SUCCESS in 22m 18s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/557e07f4a3d14bffa36127b9d25417f1 : SUCCESS in 17m 44s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3f43124bcb494ec6bd85f7ba2172c6e3 : SUCCESS in 34m 23s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/babb27f5d2ce4a4abb70ec2b6911b158 : SUCCESS in 20m 49s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e5294c26f36c4ec5a797e9261d6d4a07 : SUCCESS in 16m 48s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/dff2d70eab6f4f239ceaf98551972231 : SUCCESS in 16m 38s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46fc47b399ec4a3fb65a7412aba6be96 : SUCCESS in 12m 54s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/48998b8176c5416f878b7aedc6c1c478 : SUCCESS in 21m 13s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/e47b79d298804cc9bb760b48e74b435d : SUCCESS in 16m 32s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5800314154584c31b83cda13b8ac8548 : SUCCESS in 14m 40s","accounts_in_message":[],"_revision_number":1},{"id":"35826ed6eb0d543f2cf5bd5f13b493899a1317a1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 12:27:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4e1fc6b5cce04ed7521cb69ccc3dda2b34741fb3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 12:28:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c40001e6100248b59bb7bf0658542510\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a06f60950835409f863b136e48eb4fdd : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fc75a8d9555136f6d6fa73b24438d6609e55e7b4","revisions":{"fc75a8d9555136f6d6fa73b24438d6609e55e7b4":{"kind":"REWORK","_number":1,"created":"2023-06-01 13:38:35.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/15/885015/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/885015/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/885015/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/885015/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/885015/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/885015/1"}}},"commit":{"parents":[{"commit":"0bf45b190c62785514991866a38ca70344b006c6","subject":"Show an error when *_bridge_ports is not a list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0bf45b190c62785514991866a38ca70344b006c6"}]}],"author":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2023-05-30 13:15:57.000000000","tz":60},"committer":{"name":"Alex Welsh","email":"alex@stackhpc.com","date":"2023-06-01 13:38:35.000000000","tz":0},"subject":"Update gitignore for .vscode/ configuration files","message":"Update gitignore for .vscode/ configuration files\n\nChange-Id: Ie272e701b8952ef106be99b8fe36eb04cb524898\n(cherry picked from commit 47829142cd9ebc3ae7d7f440870491e81c3ddfad)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fc75a8d9555136f6d6fa73b24438d6609e55e7b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fc75a8d9555136f6d6fa73b24438d6609e55e7b4"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
