)]}'
{"id":"openstack%2Fkayobe~885253","triplet_id":"openstack%2Fkayobe~stable%2Fzed~I643f54a67edfe22a3fd5c86f157bd1246d617399","project":"openstack/kayobe","branch":"stable/zed","topic":"junos-check-mode-stable/zed","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-06-05 18:56:08.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}}},"hashtags":[],"change_id":"I643f54a67edfe22a3fd5c86f157bd1246d617399","subject":"Install ncclient even in check mode","status":"MERGED","created":"2023-06-05 10:01:36.000000000","updated":"2023-06-05 20:14:16.000000000","submitted":"2023-06-05 20:13:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"885253-junos-check-mode-stable/zed","cherry_pick_of_change":885053,"cherry_pick_of_patch_set":1,"meta_rev_id":"6c9787207c726a666cff4bd9b1d11d4950556250","_number":885253,"virtual_id_number":885253,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-05 20:13:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2023-06-05 10:11:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-06-05 18:56:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2023-06-05 10:26:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-06-05 18:56:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-05 10:01:36.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2023-06-05 10:11:24.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-06-05 10:26:47.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-06-05 11:56:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f001bf5fd0cfd87403eb7d48e1ec765831668630","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-06-05 10:01:36.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5d90e9076df49267e93b1a66d54eba9ec0999513","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-05 10:11:24.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4547b991bf4780c2d70bfc4227eae84b5a465692","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-06-05 10:26:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"10feb2890932cb7b81084b3cdfc8b93847a56f5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 11:56:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aea34ad8fec24eae997a2e7e7735dbed\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e4d50963a4441d2b0602ccb0d8e0759 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4192138194b45c0947805708098d4fe : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f84d3aebf9746f9ae88f1a7e5d92e4e : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0a212e6f22b14a8b98f6808f70417ef3 : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/453ee46403f044f280d1367d2586ebe7 : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/789bbf3ecd6a4b8caa3e6f93dc686709 : SUCCESS in 8m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1899a15627a48929102a633087f2028 : SUCCESS in 3m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/aaa79cd6df894b32ba2f6ef727d49ac7 : SUCCESS in 3m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/86bdab9e11be4746a1c5e8603a38c409 : SUCCESS in 12m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d5da1d7dc6bb4655994b3a772ca6a6f7 : SUCCESS in 16m 47s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/536ff54a650e4538bc7c0d6a541f64af : SUCCESS in 44m 17s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/0be8db84ccc14d1aa755d50afc308f50 : SUCCESS in 59m 44s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/190b88542a8544d6a77f08fe92def428 : SUCCESS in 49m 51s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/4f77894caee64f42a62d6c6cee8b9b13 : SUCCESS in 58m 02s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6f887fa303e841889d1cba3761ac916d : SUCCESS in 18m 41s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/e15f2c9980394226aa33dd2682553a8b : SUCCESS in 15m 34s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8831c90925cb421e9af242b2d374822c : SUCCESS in 17m 59s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aab659679aa7417e983f7d5ecdc8984f : SUCCESS in 1h 40m 19s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/fcdc9aad64c04ffb9a61709478e11f04 : SUCCESS in 19m 45s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cf01b5da940f47f1a84de8e2197cd2c7 : SUCCESS in 12m 22s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/17ed359fce08447984db65dc164362ce : SUCCESS in 35m 55s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5d015dbd6f145d085c3be9b6fc96c73 : SUCCESS in 26m 39s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/81b53cfbf0e54d798b3cc810b010d14d : SUCCESS in 12m 56s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/60048a66d91b4753a66151a923a5efff : SUCCESS in 11m 26s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5252f90a95514f1495c52782108ea771 : SUCCESS in 15m 31s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ed0741fa75648138491a097e338009c : SUCCESS in 13m 35s","accounts_in_message":[],"_revision_number":1},{"id":"ab13dd0fe333c477632652fb9886e445b66b7bc5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-06-05 18:56:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"92317f98776124e7bbb66b3e0b955c5b451dcf8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 18:56:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c07649f9b36c6430e0e0243d6e200ecaa79c1253","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 20:13:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52f5a55312e4410fbfffe29a6c8ee6a0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71637cf5159a49e1afb35557a4ba5100 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ed7a98f81e940c5a2f17d1024648c85 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/102256f64ef14970b2a946552d74d551 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b16f3111e074123bc782f982886a851 : SUCCESS in 7m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b6c24e3171e435cb3e88195fdbf3783 : SUCCESS in 3m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/69f5f693b20d4981946b59041766bc44 : SUCCESS in 3m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0f828c034e944a81a0aef6fab93388c5 : SUCCESS in 9m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5f6aa252d7734f90a60aba1831e9c6e6 : SUCCESS in 13m 09s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/978faa0e0b03459c9b241bdeee872c65 : SUCCESS in 39m 49s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c83dc61cb99b4a0db7f9ef961776e081 : SUCCESS in 46m 41s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ca8d3cae62004750983ff627af1e638d : SUCCESS in 52m 58s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/1eedd42fd9c84fa080074c3ad26a9d8e : SUCCESS in 16m 40s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/626581e750a74d668835bf48b5052f40 : SUCCESS in 13m 59s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/793d216ba9f1456c9aeae7f022f9b64d : SUCCESS in 1h 16m 11s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/6a613f553d844a9e9ee543598e5243a4 : SUCCESS in 16m 51s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/50948e271c08424d825d2728075f9c2f : SUCCESS in 15m 08s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/617a91bf0e684112800c2f06b828bd1e : SUCCESS in 26m 24s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/51b62c2a5f98421cb968115928bdef47 : SUCCESS in 12m 35s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7543c3ccd6b54898a18802e897619386 : SUCCESS in 11m 16s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f75cce4a2927472380922bbf12484626 : SUCCESS in 17m 09s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d90a1064b324269815984f2b220b364 : SUCCESS in 13m 32s","accounts_in_message":[],"_revision_number":1},{"id":"98b0915c5c978c50fd6dbfd3acc4ebefc333b1b7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 20:13:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6c9787207c726a666cff4bd9b1d11d4950556250","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 20:14:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a7aad64050843309142dd8613a01936\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93b948f0f31f44929447d6117ac6d084 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9efd228c0f0f40b6b1a56f5130d968c2 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a9de74812c996f977f519f3cb0c5979cec0d0524","revisions":{"a9de74812c996f977f519f3cb0c5979cec0d0524":{"kind":"REWORK","_number":1,"created":"2023-06-05 10:01:36.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/53/885253/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/53/885253/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/885253/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/885253/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/885253/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/53/885253/1"}}},"commit":{"parents":[{"commit":"586d23550224d1f74a97d6e204fb23c84421f826","subject":"Show an error when *_bridge_ports is not a list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/586d23550224d1f74a97d6e204fb23c84421f826"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-06-01 14:19:50.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-06-05 10:01:36.000000000","tz":0},"subject":"Install ncclient even in check mode","message":"Install ncclient even in check mode\n\nWhen running `kayobe physical network configure --check` against Juniper\nswitches, Kayobe may fail with the following error:\n\n    Failed to import the required Python library (ncclient)\n\nThis is because the ncclient installation task is skipped by check mode.\n\nChange-Id: I643f54a67edfe22a3fd5c86f157bd1246d617399\n(cherry picked from commit 7b8abfb6693f1b1683a72e3e128049df30a2c49a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a9de74812c996f977f519f3cb0c5979cec0d0524"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a9de74812c996f977f519f3cb0c5979cec0d0524"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
