)]}'
{"id":"openstack%2Fkayobe~885288","triplet_id":"openstack%2Fkayobe~stable%2F2023.1~I209459ecf5b1a6e5f541cbc72d74aa7defb459d4","project":"openstack/kayobe","branch":"stable/2023.1","topic":"kolla-build-repos-yaml","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-06-07 11:32:00.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-06-07 09:56:42.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I209459ecf5b1a6e5f541cbc72d74aa7defb459d4","subject":"CI: Fix kayobe-seed-images-* jobs","status":"MERGED","created":"2023-06-05 13:29:23.000000000","updated":"2023-06-07 11:33:12.000000000","submitted":"2023-06-07 11:32:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"885288-kolla-build-repos-yaml","meta_rev_id":"778826c9b39d6d2d64a7a4308c26828cb2d126d8","_number":885288,"virtual_id_number":885288,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-07 11:31:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2023-06-05 13:51:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2023-06-07 09:56:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2023-06-07 09:56:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-05 13:51:01.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-06-05 15:12:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-06 10:19:57.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"3ce367847684c4300d837afead32e129aac8146b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-05 13:29:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ef0aa325e91d68233d19e0b79f9a6d3fb848dff","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-06-05 13:51:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"59c23128cb35199a03721b3f7bd7471830f67eec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 15:12:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3c43ca043634a54ba60a7628b225a7f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f3d44f75c5644e05a2b4148f1749df0f : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad2673f3440642d3bd6feb2de8593cad : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a823418f24d948d1a19593f551ca8977 : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a63e66eec2fc4595b14786332aca5344 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ffe2c6966444c438b7c5f11afffeced : SUCCESS in 7m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2817ddbc10a947ecaad4a2376b065931 : SUCCESS in 3m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/eeecb590311a468eb84dc7233c08b587 : SUCCESS in 11m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a600c0a8c8d046fd8bffb609d9aa84c1 : SUCCESS in 12m 44s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/80e9d75f6c8740dbb7d3eef775e00891 : SUCCESS in 1h 00m 15s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/37c275f8364b47879c225a68efc885c6 : SUCCESS in 55m 01s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a61447d004a344fabad8f119b91565fa : SUCCESS in 48m 39s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/6ef4c8d104914acc9d0605ebaff109a4 : SUCCESS in 1h 02m 22s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/26df22c7b25841798ded242a5e7f3111 : SUCCESS in 16m 40s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/b29ab72fa90c4b908428dd45d39086e4 : SUCCESS in 15m 20s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a2365ac568ff4190aeb187eff648850a : SUCCESS in 12m 19s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b590561d36f14baf86fcd72ace25aa40 : SUCCESS in 1h 36m 27s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/411d6711e57d4470b12c8a5abf8d41c3 : SUCCESS in 1h 29m 29s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c521c15b934b48599f3878bc9511d209 : SUCCESS in 14m 42s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b075b22b5c1541da914d777543208b40 : SUCCESS in 14m 57s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/892b7dbb06894928b4974abb3c1f05d1 : SUCCESS in 34m 09s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/78aaf7f5db1949e687a39648fc187863 : SUCCESS in 26m 35s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66262d6cc21c443b9ef5fecf03bd2651 : SUCCESS in 25m 29s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/91a146e2c7c7449ebbbc8777d36c58df : SUCCESS in 14m 18s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b5cd5e3973444b03a82b6256b504c0e2 : SUCCESS in 11m 03s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ecb853023bc14b7eb93d643ddfe8feb1 : SUCCESS in 33m 03s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ee8cf24b3f44c94a045ec8eb4a1d95b : SUCCESS in 22m 34s","accounts_in_message":[],"_revision_number":1},{"id":"9f7529e4722afec3089ba9c2464e06457b83c096","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-06-07 09:56:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e39e946de8e41fa1de4f0fc035fddbd7c931bd8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 09:57:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"32c1e16b871ab6779c8db557a5bfa4666c2ad868","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 11:31:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7f488021ee1462cb5e007800cd52bdf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5e85047d25a44a6b87e5743e15470e4 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d818e70d48c6459f8d1549ad0eeaa34d : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1fcb9549a0574bed8bc4045c19d2947a : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b31a07ba62ef4b5f89a4d38147a1b4b2 : SUCCESS in 5m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8890cffc7b3d4ebeb20c88a35630fa59 : SUCCESS in 3m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/82976e6aea6d43d58373f23ed85afafd : SUCCESS in 10m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b224c4c2219f426c9351121a9f6e878d : SUCCESS in 11m 57s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/2092a40d0b084d36929f76dd319f7a9e : SUCCESS in 56m 20s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/43980626ca774c85b3e027969a5c8fa4 : SUCCESS in 43m 14s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/4b2e9d1adead41a8b11fc083b1b002cf : SUCCESS in 1h 03m 21s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/9a175cba264747139140e52867d9b234 : SUCCESS in 16m 51s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/445ab953c2524902b7f698860eef2614 : SUCCESS in 14m 12s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/466a2ae25d8449fdbb51878a09d1536b : SUCCESS in 1h 29m 36s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c7790090b034deaaf6aa1a3be86a3ef : SUCCESS in 1h 27m 39s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/0166237e3eae47cdafc3216c09d987b4 : SUCCESS in 17m 58s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e3234696f06d41c4b1df2753a14cb9ea : SUCCESS in 13m 39s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9641e6b5858d478f927b48236d35524c : SUCCESS in 29m 59s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d7bddc247db74783b7e65d983fa3d698 : SUCCESS in 24m 23s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/52ad079acd1e4615ac0a9f499ad468f6 : SUCCESS in 14m 59s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b2b76e38600047ad9a5b6e05be736532 : SUCCESS in 14m 51s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/77a6505f19b34853b7efd6fcd912d61a : SUCCESS in 35m 35s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2da50090a33a432580eecf9948b7c106 : SUCCESS in 29m 00s","accounts_in_message":[],"_revision_number":1},{"id":"ce9b2d76c81c0b79e1c27e84661da336a9ce81c9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 11:32:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"778826c9b39d6d2d64a7a4308c26828cb2d126d8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-07 11:33:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e473cb473ab45f5894914cdbcf5df3c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6af27ed39de44306a83b5e2c6a12f04d : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"48cd0eb6ed49f6c7519b11c2f6b0b40081659bc7","revisions":{"48cd0eb6ed49f6c7519b11c2f6b0b40081659bc7":{"kind":"REWORK","_number":1,"created":"2023-06-05 13:29:23.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/88/885288/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/88/885288/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/885288/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/885288/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/885288/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/88/885288/1"}}},"commit":{"parents":[{"commit":"f8cef91dc2ef36a14132d9c99f4ec47472e1dd16","subject":"Update .gitreview for stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8cef91dc2ef36a14132d9c99f4ec47472e1dd16"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-04-27 00:57:00.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2023-06-05 13:28:45.000000000","tz":180},"subject":"CI: Fix kayobe-seed-images-* jobs","message":"CI: Fix kayobe-seed-images-* jobs\n\nFollowup on I0d7ab0d8ff5b16ac5de8e50e63400bd455996555.\nFix issue with \u0027permission denied\u0027 while doing stat files.\nAlso we collect kolla-build configs for CI.\n\nChange-Id: I209459ecf5b1a6e5f541cbc72d74aa7defb459d4\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 7088ff2c2f10ba9897d1cd5424fea6d9268020a9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/48cd0eb6ed49f6c7519b11c2f6b0b40081659bc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/48cd0eb6ed49f6c7519b11c2f6b0b40081659bc7"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
