)]}'
{"id":"openstack%2Fkayobe~886274","triplet_id":"openstack%2Fkayobe~stable%2Fxena~I4ea8d8c3a822a7c217bcfcfd5027eecfd21beaed","project":"openstack/kayobe","branch":"stable/xena","topic":"bug/2024163","attention_set":{},"removed_from_attention_set":{"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2023-06-16 20:09:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4ea8d8c3a822a7c217bcfcfd5027eecfd21beaed","subject":"Correct name for role mrlesmithjr.manage_lvm","status":"MERGED","created":"2023-06-16 15:05:57.000000000","updated":"2023-06-16 20:11:46.000000000","submitted":"2023-06-16 20:09:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886274-bug/2024163","meta_rev_id":"a5db5c4bf2d438a5b43e1f5adbf53321e861156f","_number":886274,"virtual_id_number":886274,"owner":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-16 20:09:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-06-16 17:48:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-06-16 17:46:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-06-16 17:48:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-16 17:46:02.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2023-06-16 17:48:33.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-06-16 18:20:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"07b1717929bb4153c6329c55e6bb6ff47205bd47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-06-16 15:05:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ee3dbcac8461c505f0643da518cbca387252a4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2023-06-16 16:17:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f1ff6ee9ab0368978032a3117d5b3a13d84e932e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-06-16 17:46:02.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"91ed1d0426407d7f7c6a55c22917c193ef812d79","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-16 17:48:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5ddc1862406a513534543f48a236e0121eb6de54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 18:20:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2de41adc07d546c894144b6d2c9864c1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00628374ead646cba6a501e02cd969ea : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ef6e5d85a9e499b86667e6692f83743 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c8e1e8553ce4385aee41949307d4492 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bde56c157b59451ebc1eabec67b569ed : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e4c709c3262e463abcdfea60f7e0b37c : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c672686b95e14105adfaa1de3afd720d : SUCCESS in 8m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e6b61c8d68048769ca76b362b60b942 : SUCCESS in 3m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bd945064ad05490f83f155d6753591e7 : SUCCESS in 6m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/42ff1ba9e3b74b41ba7962fe182a0016 : SUCCESS in 10m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b5baef21686d4f61856f610c79e4f113 : SUCCESS in 18m 10s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a5e39a8bf9fd4857b37a65583e20da40 : SUCCESS in 1h 20m 11s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/edcfc86f7a4542c3ab147cc76dab5904 : SUCCESS in 53m 20s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/dc70fcb573684f88ab7d1c1ba7247692 : SUCCESS in 54m 20s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c7145c30d61548469f97dc32fed3e687 : SUCCESS in 26m 20s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/22262b98da854fa4b352957a623da8af : SUCCESS in 19m 20s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e204524c174144a29643dc20056c84f0 : SUCCESS in 15m 00s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4b2ab89b68d84b67a633c347217407f7 : SUCCESS in 1h 34m 51s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/54c3c5a23c4b4d5497e141772b0d757a : SUCCESS in 21m 40s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/8a55f5c051c24546b93a3fcc5fd7b158 : SUCCESS in 15m 15s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/07f32d8880554f29b4ff6f61c4292203 : SUCCESS in 15m 57s\n- kayobe-seed-images-centos8s https://zuul.opendev.org/t/openstack/build/a98e882c86524e789071606200865a7c : SUCCESS in 41m 05s (non-voting)\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e98cbed64d5b48f08cd18db605c06c66 : SUCCESS in 36m 24s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9bc72d0fc4d43169ab38aa8525f038f : SUCCESS in 24m 33s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ddd9519b83c34fd38bde8d88030d1bb9 : SUCCESS in 19m 17s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/d8e9682b864b4376a5071f02ebde179e : SUCCESS in 17m 09s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1eaa42c6491c49ea86498296d9e80e3c : SUCCESS in 11m 39s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/86faf5c1cfc9419583f758c197fb7b10 : SUCCESS in 19m 36s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/63c933555acd43c398b441ce91fa94d9 : SUCCESS in 18m 13s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/714d42c9150d4fc89836d96daa8af6d5 : SUCCESS in 13m 30s","accounts_in_message":[],"_revision_number":2},{"id":"7785b1ea6efc1044b3051caa069d15887e497075","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 18:21:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a1a5584813de791a18f6f079602118913f9185b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 20:09:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30b97160193b41d6bd27d070593a1015\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6888ae26ca4a4fc0a277e25775d4c2d7 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93a36f7cb7b9409c8860982f168c9fdc : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c374277835344d639f084540ffdd4c15 : SUCCESS in 6m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/519e82ec0c6c4625af553944d64db7b2 : SUCCESS in 7m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26e61c3797fb45f9ac0c936f1191329e : SUCCESS in 4m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/105698e503dd405eb29893cceb454dae : SUCCESS in 4m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/193af86c7b4e4746a5acb4c83fd1372f : SUCCESS in 12m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/47ddc98fb34a4f559939d7c3c198a46f : SUCCESS in 14m 08s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/84d401bdd3fd4175aaa707406118b907 : SUCCESS in 58m 25s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/4482ec14ed4b44aca5d8ff7969f22f6c : SUCCESS in 1h 09m 29s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/7b15b764dcc1484c938d83c2a54c3357 : SUCCESS in 55m 17s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e8585921f904423aaa157fdda26843af : SUCCESS in 19m 47s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/46b1d505c23a4fc886aaaa3d9f492760 : SUCCESS in 25m 06s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9486325b5f864f0b8fe4b7e95e55afd2 : SUCCESS in 12m 17s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/66871f36d5bc438b9aa9511472ea7c76 : SUCCESS in 1h 48m 20s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/93b729ef72cb4a978a69433f4a5e3f9e : SUCCESS in 23m 01s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/6f76ac619eab4203a64c092d387ff7af : SUCCESS in 16m 22s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9d726c3ac3f4b958079dcb9cc623c52 : SUCCESS in 12m 12s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d6d8e38431b847608db7f0aab1b907ed : SUCCESS in 33m 05s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/556b0b9b6b05442eaf03b0c15a0c6d58 : SUCCESS in 27m 00s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/4d17c3905a63443ea6ad6456892cd2de : SUCCESS in 15m 26s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/f2a338841a7d479cbf7e99b846d04cdb : SUCCESS in 16m 19s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df0aa477182947c88140b155b4c35bd6 : SUCCESS in 12m 00s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/e94dffd3c7a34f6c83256ab80512a8cc : SUCCESS in 19m 00s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/6a6fcba1bbdd44ecb3478e50daf7c490 : SUCCESS in 43m 25s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a2a31a53b69d4b4e8e4054bce8cfffb4 : SUCCESS in 10m 23s","accounts_in_message":[],"_revision_number":2},{"id":"b9811cb0f026646612e57c2baf681c9bb212e742","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 20:09:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"a5db5c4bf2d438a5b43e1f5adbf53321e861156f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 20:11:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94d5a2b9271e4487a7f6f53c0568df77\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/849bed7d6e0f4dfd9edb2664c6355cd3 : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/594eddc227984b6d8a58ad00bfb93fb4 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"98d8e04bb5645b25526bd3987629e6e8e4c4eebd","revisions":{"047982dc6b912640a5649c99562bb314a558736c":{"kind":"REWORK","_number":1,"created":"2023-06-16 15:05:57.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/74/886274/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/886274/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/886274/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/886274/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/886274/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/886274/1"}}},"commit":{"parents":[{"commit":"f57af430de2f1f9fd4e74dc32b90681bd36d4685","subject":"Merge \"Allow to use own repos.y[a]ml file to build containers\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f57af430de2f1f9fd4e74dc32b90681bd36d4685"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-06-16 09:13:54.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-06-16 15:05:48.000000000","tz":60},"subject":"Correct name for role mrlesmithjr.manage_lvm","message":"Correct name for role mrlesmithjr.manage_lvm\n\nThe name of ``mrlesmithjr.manage_lvm`` was changed yesterday in release\nv0.2.10 to use an underscore, instead of a hyphen. As this changes the\nname of the role on Ansible Galaxy, it needs to be updated in\n``requirements.yml``.\n\nCloses-Bug: #2024163\nChange-Id: I4ea8d8c3a822a7c217bcfcfd5027eecfd21beaed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/047982dc6b912640a5649c99562bb314a558736c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/047982dc6b912640a5649c99562bb314a558736c"}]},"branch":"refs/heads/stable/xena"},"98d8e04bb5645b25526bd3987629e6e8e4c4eebd":{"kind":"REWORK","_number":2,"created":"2023-06-16 16:17:01.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/74/886274/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/74/886274/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/886274/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/886274/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/74/886274/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/74/886274/2"}}},"commit":{"parents":[{"commit":"f57af430de2f1f9fd4e74dc32b90681bd36d4685","subject":"Merge \"Allow to use own repos.y[a]ml file to build containers\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f57af430de2f1f9fd4e74dc32b90681bd36d4685"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-06-16 09:13:54.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-06-16 16:16:53.000000000","tz":60},"subject":"Correct name for role mrlesmithjr.manage_lvm","message":"Correct name for role mrlesmithjr.manage_lvm\n\nThe name of ``mrlesmithjr.manage_lvm`` was changed yesterday in release\nv0.2.10 to use an underscore, instead of a hyphen. As this changes the\nname of the role on Ansible Galaxy, it needs to be updated in\n``requirements.yml``.\n\nCloses-Bug: #2024163\nChange-Id: I4ea8d8c3a822a7c217bcfcfd5027eecfd21beaed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/98d8e04bb5645b25526bd3987629e6e8e4c4eebd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/98d8e04bb5645b25526bd3987629e6e8e4c4eebd"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
