)]}'
{"id":"openstack%2Fkayobe~886293","triplet_id":"openstack%2Fkayobe~stable%2Fyoga~I4ea8d8c3a822a7c217bcfcfd5027eecfd21beaed","project":"openstack/kayobe","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-06-19 10:27:36.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-06-19 08:19:53.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I4ea8d8c3a822a7c217bcfcfd5027eecfd21beaed","subject":"Correct name for role mrlesmithjr.manage_lvm","status":"MERGED","created":"2023-06-16 20:37:17.000000000","updated":"2023-06-19 10:29:39.000000000","submitted":"2023-06-19 10:27:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886293","meta_rev_id":"aa2a2e9c3ffd1506690cccf665df9a1593a0dd36","_number":886293,"virtual_id_number":886293,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-19 10:27:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-06-17 21:04:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-06-19 08:13:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":2,"date":"2023-06-19 08:19:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":1,"date":"2023-06-19 08:19:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"CC":[{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-16 20:37:17.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"CC"},{"updated":"2023-06-16 22:00:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-17 04:00:08.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2023-06-17 04:00:11.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-06-17 21:04:33.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-06-19 08:13:46.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"906b1f0d7a6c94551387a408638164bc65fe962a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-06-16 20:37:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a1d2d1fd1368f4fdf39cc85b9508f9121185d6f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-16 22:00:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/62ae914492c140289939b75b6b43e825\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/67761e5c077043e2b0136a990e18876e : SUCCESS in 7m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44558935a9a041bf84cdd59cac3f0cf6 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8f75989bd2b460e88c8c6073d55fe0c : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b9eaa0f1d2b644168c85ce47b1be3056 : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0df186775484e8f9f8ec1d8d969c1a4 : SUCCESS in 11m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6420deca4822456396f93f046f2b4d64 : SUCCESS in 2m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b592dc29814a4a309253f8d3143810a8 : SUCCESS in 5m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cc7109c8e9cc433095494b433c798147 : SUCCESS in 10m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/841deafa0ae74788a120c128455970a6 : SUCCESS in 13m 09s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/faf0893655b04ca5a8e372d56c9141a3 : SUCCESS in 54m 04s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/4807086bfb744f11ad8bc2b615c76dcd : SUCCESS in 39m 05s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/76f3c321ab5745c59db70ed2efe9206e : SUCCESS in 50m 33s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/6be171006e49447eb597ae5476d2b967 : SUCCESS in 36m 06s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1b2ed720411d46849cc6a348d3dee6a1 : FAILURE in 28m 25s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a348ba3d0aac4252800b0b8f96f8ada4 : SUCCESS in 19m 06s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/3ab4aafc5a764957bdcb31159c1152b5 : SUCCESS in 16m 54s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/0107369f84494c1092d9f3727a199877 : SUCCESS in 17m 28s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4e5831ce5a114710a522b888fc4351c6 : SUCCESS in 15m 38s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/759dcf0b0164461ea55a506e97dc3da0 : SUCCESS in 13m 49s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9178941cfd1e4dcaa639404dca398ae0 : SUCCESS in 13m 43s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c7fad2e3091f4200816ba75937349e58 : SUCCESS in 55m 39s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ac004aa1d63345688b1abd44600969e7 : SUCCESS in 45m 04s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4015ae2caf344ee1bfec8ab0f4742b3d : SUCCESS in 1h 22m 07s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/997c3441a05b4003be9e1b8d4e0bdff5 : SUCCESS in 17m 37s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/bc7e325cf84b4bc6888eaa2c2b649540 : SUCCESS in 21m 57s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/2ceed139724c4f8fb24d45fce730fa20 : SUCCESS in 20m 22s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f67fcdb1dde4949aa947c77c72fe458 : SUCCESS in 19m 11s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a5cf56f747a64bf4a3fc00ad3cdf15a8 : SUCCESS in 19m 17s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/947bb016620144a9964c73427ebfd015 : SUCCESS in 28m 44s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2fd666ef869b4a06a0d08d1da9e91281 : SUCCESS in 25m 08s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/cb37330576784ebeaa127448665361c7 : SUCCESS in 17m 20s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/a41d2661c2574740a1f6b4cc1c72ea8a : SUCCESS in 14m 21s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0f09e78e323348858f62148dbfbd5414 : SUCCESS in 15m 26s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/48967dcebd814f93a89b5bcf4a06cff1 : SUCCESS in 12m 27s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1324b4e52c2c49a191ae46e520e5b472 : SUCCESS in 11m 55s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/c11ac36d558e4000b67016319aec0879 : SUCCESS in 22m 46s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/00ecc5ef638c4e4195904d6fd3f9909f : SUCCESS in 19m 53s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c8e16dd916e046238862098841456a1d : SUCCESS in 18m 02s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d8897039d774ef69127e57e6e4a13ea : SUCCESS in 13m 47s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d5bd931443314698ba17b8435a4fb17c : SUCCESS in 16m 16s","accounts_in_message":[],"_revision_number":1},{"id":"6bc7bfedf92f788e9ce7003f900bda5384ecdcac","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-17 04:00:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ef6eda48ea49e7bf1223054970dce19772246581","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-17 04:00:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e3851ff10dabeafb9521e06d1c220e5c183f426c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-17 06:10:43.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/29117246951d471c99786fb4157e127f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/304c20eaa49a42159df701a48fc1b2f0 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6720bc5b9a6f415cbc902407a1c66a41 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/393a0751e1ec476cbacc1642d9d582c9 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5643004882b841e7ad3d721e29e841d2 : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c896c220a3649eda1ae5f2da0197a87 : SUCCESS in 12m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a32f7f206304171afdff70151d2da63 : SUCCESS in 4m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f4c4fc2f001b4e60887a349ebd52dd8f : SUCCESS in 5m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cfdbd8a8ef334055afb2ebd772b6bfcd : SUCCESS in 13m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4de843587423466d8ea95fc271772467 : SUCCESS in 16m 46s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f85c4fe697ae4f35a7d98fd75554d0ff : SUCCESS in 1h 07m 59s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/4a105046d4df43c1ba628406f68a559f : SUCCESS in 1h 03m 29s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/bc7d486989064c89871da9a3e7d004e7 : SUCCESS in 57m 02s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/a5b326c1f13342d2af04173a5eac2d83 : SUCCESS in 45m 19s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d3c458955ca3469f9f621f66845e721a : FAILURE in 38m 14s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/8365a229812d4472bf009f3d37aa13ef : SUCCESS in 19m 01s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/7830ab7bf4904e509b19bdb7f1e67680 : SUCCESS in 26m 09s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/24cbd83d3e704c72b78d27fd1767e1f3 : SUCCESS in 16m 29s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/a37bd0e68b1a4ddf98a0f07fefd73267 : SUCCESS in 20m 32s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12ef325370ad4a1d870fa5e807793c20 : SUCCESS in 13m 18s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c8014c4bcbb449a0b1ff8da7edf6e513 : SUCCESS in 12m 59s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/8e2bc2de58f14e1db7295b659e7b3237 : SUCCESS in 54m 26s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/6fe504eaacf64c27981e61242da82e31 : SUCCESS in 1h 01m 12s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2f731804c9334190baf0541666a2aa11 : SUCCESS in 2h 05m 11s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/b267c85fe51b444b8c982007dcdc8791 : SUCCESS in 21m 36s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/65f34468aaae48059248ea9f99ce455c : SUCCESS in 19m 15s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/7b21e3c5b7f64842bdf2f1ffb4052b32 : SUCCESS in 17m 52s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ed8c325df44483b8ac724fae59499dd : SUCCESS in 15m 00s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/646a1027d8544aa284d48f201dc2d197 : SUCCESS in 14m 34s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/96e172d5b8574611a3305d8bc92c29f5 : SUCCESS in 24m 06s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/474aef9eb46846fe825507dec2ae5cfe : SUCCESS in 22m 24s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/36236c408f8440fb833262d0a6e70c30 : SUCCESS in 15m 35s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/e135c17f970c4c39b83b3d9e24b83d6e : SUCCESS in 14m 41s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1e0ee884eed7471f9eebd9662afbb5de : SUCCESS in 14m 15s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2dc47a0637414251aeb89f7e2be4a898 : SUCCESS in 11m 12s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/88daa6158bea4c9e827167d8b7ddbee8 : SUCCESS in 12m 54s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/1caec0bfa6b2478fb738c4966ab71dc6 : SUCCESS in 20m 29s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/2964c96c912d434ba4e7957b3d3f4a28 : SUCCESS in 18m 07s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3c86b0da1706434690cf82115b3b1a21 : SUCCESS in 16m 43s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/48bdcf32432045feb05f0cca3884a607 : SUCCESS in 1h 12m 32s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c0fb2d237b5b47da87efd17242b61681 : SUCCESS in 1h 07m 36s","accounts_in_message":[],"_revision_number":1},{"id":"10362124b9fd36a2fc7a596239c62b50621efd44","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-06-17 09:02:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0559325071d81e2e67d7b643c9974f51c646381e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-06-17 09:03:01.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"13cca73e034b3799e34b5268f152f302f87915a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-17 10:45:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/172f1c31a66942598d99751901a0eabd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1a7956df9114982847314b64396617f : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3e14c0143e24594886823a5f8919643 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0cfb6b6c8f6042d38278d4b8e08ed30e : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6930e5c10d2c4d9ab59512d3f14f0f46 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dde3debf5bd442380ee5af9a4381f8f : SUCCESS in 8m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a38c96ef2d304f9fa34a9eed818ff829 : SUCCESS in 3m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8c85b93901ec44ae91d3226402cf4ba3 : SUCCESS in 5m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dab2ff460c7e42b8ae955515ff55294f : SUCCESS in 13m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bf1020508e7043dcb4bf3d1bc4b8a28d : SUCCESS in 15m 23s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f9db31b4bec34c038aed31aa9095cab7 : SUCCESS in 49m 47s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/5e3d448fc3f14496986d74217d050cb9 : SUCCESS in 42m 17s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/9a730d0c019c457eb0002ee7bbca88cc : SUCCESS in 46m 58s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/8a2d9d704110475eb5b580830ffb73ea : SUCCESS in 37m 13s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/02db62d476f3402ca1b3d8e9b69f2b0e : SUCCESS in 22m 18s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/7a86de2cccaa4d519f63b633adda7212 : SUCCESS in 19m 24s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/c6b7557dde9d49d7a475d3af2ee18e26 : SUCCESS in 23m 43s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e930f983dc73414baf4ee485e8a7b1db : SUCCESS in 20m 00s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e43d5896e5c42c29c5d45a720ced296 : SUCCESS in 15m 22s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c69ea1cfa95b46b7af4cf69bb66624f4 : SUCCESS in 16m 49s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/2736c3af644a40b085e5bc7b9a4fd7df : SUCCESS in 1h 19m 03s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/199493e283074167b8d5b36d47c78fa8 : SUCCESS in 1h 00m 50s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c6e9f62a32484578be8592f528b1f713 : SUCCESS in 1h 21m 47s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/1450b42019e44f4caf93690727d4bff7 : SUCCESS in 15m 33s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/234e9e03a729462ab55c5af45d556906 : SUCCESS in 19m 07s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/714ebd5ec389459a8e685b29685edec2 : SUCCESS in 15m 07s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/63ed586abcc740d98ef6972fb6538a21 : SUCCESS in 15m 28s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9935219091154d36aaa6408d8b36c600 : SUCCESS in 14m 37s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e1562a34c9594adfa76e942f59bd6557 : SUCCESS in 22m 50s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5264428c0ed49029ec80886be2096b9 : SUCCESS in 26m 30s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e2846106a0ca4320b506d35414d8034e : SUCCESS in 13m 19s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/f5735a809d9846218dbfbcd6cc60d726 : SUCCESS in 17m 53s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/adfc7a32b4164ae1ae36121b8b131182 : SUCCESS in 17m 10s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c4a5aaff933240ea9a239f1e797b9a3a : SUCCESS in 9m 37s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/083a535e83c3430896b52cb816d01ff7 : SUCCESS in 15m 05s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/02b63fdb1cc34fe7a31f092eff18ee9a : SUCCESS in 21m 27s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/888a5014c2834d1a82347d0bc7717f35 : SUCCESS in 20m 00s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a3575678f03443768dde1f5f44e1c728 : SUCCESS in 13m 47s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9dcf51e593e74184bc90df327262cbe6 : SUCCESS in 11m 48s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/271a4d8d8b514900ad48b824fca8557b : SUCCESS in 12m 48s","accounts_in_message":[],"_revision_number":2},{"id":"5d3160ba53fe7c53280b89ad940cf76742913e13","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-17 21:04:33.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"6bf19c77ce993b56b02943942ffbefb23292eac0","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-06-19 08:13:46.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"434e4a940668636218ab45d8ab31afaa44ad03be","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-06-19 08:19:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b685f186c95ae9cbbf357bb3caee3304bf102f06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 08:20:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6bda13d5a8c510959133296521262037c5ae1594","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 10:27:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b01f1751762a43ff93a6a9664ccfd7df\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b74cc98d3157462aaa7a9ba0c2c67add : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b537261370d447f7806a434b81c8b4de : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/693fd42649b144a9a6b59a9e1e0eb145 : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f062a8de4c98410980c54761b13f748a : SUCCESS in 10m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5105d33fd1a43e397097351a10a5ca6 : SUCCESS in 2m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ff799dd0a36b4aa19c9af970612572b0 : SUCCESS in 6m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9a28439d35154c828412aec3ae0f0849 : SUCCESS in 11m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9b3b666d82744a10b46d766b8b480ce9 : SUCCESS in 17m 13s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0e32278bedad4034ba3aa7828e1e9b97 : SUCCESS in 50m 14s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/1398a7c9e6bb4eafb0dd53b50c62155c : SUCCESS in 57m 54s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/5511d9e9a0554d93bfec661da5e209e6 : SUCCESS in 38m 29s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/3c6c2e134e634d62802e68e1255467a7 : SUCCESS in 23m 45s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/5f305ae552514b2b865bb028b239b032 : SUCCESS in 26m 00s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4f3897dd4d81478897d3b0815b0d1e62 : SUCCESS in 21m 04s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b15d71756ba04d9b8608fbae9ce66bca : SUCCESS in 13m 49s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d938eda536764208862483d839e0fa33 : SUCCESS in 15m 32s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/090af52009ae4366b52938480d65f796 : SUCCESS in 1h 17m 49s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/9a8a27bfec09412fb72594b74e26a9c2 : SUCCESS in 58m 04s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ee2fff4174bd43588098fdc126125886 : SUCCESS in 2h 03m 19s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/a2d96d1086cc4d97ac7117717cc1493c : SUCCESS in 22m 23s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/4fea54b75f1f4fd18cd7293463e25042 : SUCCESS in 16m 10s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/87351ec4229d4a078bdf2ee9b0f257e1 : SUCCESS in 22m 12s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70bd89ebfe2c40cca49d090e1b804ae5 : SUCCESS in 17m 02s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fa64c7b725a74b8e8952a5d2661d02ea : SUCCESS in 18m 08s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/1ee6def6dfab4fba87ba1a03ac94dcd6 : SUCCESS in 23m 05s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21dc8733c5854a1c9c0e7a3023287336 : SUCCESS in 27m 35s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/98a756ec02be4e7eaad68fc3b18322ad : SUCCESS in 17m 37s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/eb131836f0d64d49b55e242fe9629d7f : SUCCESS in 12m 26s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2951f6319caf49999f2e46268dadf9c3 : SUCCESS in 16m 28s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc94e2b5921340eb98227c1590455783 : SUCCESS in 12m 06s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/efbf448ff5d94876bea69cb47f9dd632 : SUCCESS in 13m 48s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/c5d72b0dc82a4cf7929cae1e794915a4 : SUCCESS in 27m 18s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/4fdca9bc13c040c8b360c321b6b2ca51 : SUCCESS in 19m 52s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5998676fe2ec49138dfcfe64e502c40a : SUCCESS in 18m 35s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c91ae94ae754349a51d5bf5778c2140 : SUCCESS in 14m 35s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ad61b16594fd4435bbcff60c9e7c62ac : SUCCESS in 15m 22s","accounts_in_message":[],"_revision_number":2},{"id":"700cc24b7548cbcd4c9d27e8f8cf668f01056bec","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 10:27:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"aa2a2e9c3ffd1506690cccf665df9a1593a0dd36","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 10:29:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38a3b7ea4a7f4de792c355d74225a714\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e3390a0f98146e7a5fdc855add067ef : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f9d04d97e0d421486d2828cd36030a7 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c4fa262a53f3df6da5c5e7d1e5a3e3f33e5c3100","revisions":{"67dc480eca18345b656841eff9567774ca7df331":{"kind":"REWORK","_number":1,"created":"2023-06-16 20:37:17.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/93/886293/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/93/886293/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/886293/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/886293/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/886293/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/93/886293/1"}}},"commit":{"parents":[{"commit":"dc90bc50eb61046c2e8f1e592da4e9aa3f5afc2a","subject":"Merge \"Allow to use own repos.y[a]ml file to build containers\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dc90bc50eb61046c2e8f1e592da4e9aa3f5afc2a"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-06-16 09:13:54.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-06-16 20:37:04.000000000","tz":120},"subject":"Correct name for role mrlesmithjr.manage_lvm","message":"Correct name for role mrlesmithjr.manage_lvm\n\nThe name of ``mrlesmithjr.manage_lvm`` was changed yesterday in release\nv0.2.10 to use an underscore, instead of a hyphen. As this changes the\nname of the role on Ansible Galaxy, it needs to be updated in\n``requirements.yml``.\n\nCloses-Bug: #2024163\nChange-Id: I4ea8d8c3a822a7c217bcfcfd5027eecfd21beaed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/67dc480eca18345b656841eff9567774ca7df331"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/67dc480eca18345b656841eff9567774ca7df331"}]},"branch":"refs/heads/stable/yoga"},"c4fa262a53f3df6da5c5e7d1e5a3e3f33e5c3100":{"kind":"REWORK","_number":2,"created":"2023-06-17 09:03:01.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/93/886293/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/93/886293/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/886293/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/886293/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/886293/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/93/886293/2"}}},"commit":{"parents":[{"commit":"dc90bc50eb61046c2e8f1e592da4e9aa3f5afc2a","subject":"Merge \"Allow to use own repos.y[a]ml file to build containers\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dc90bc50eb61046c2e8f1e592da4e9aa3f5afc2a"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2023-06-16 09:13:54.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-06-17 09:02:55.000000000","tz":120},"subject":"Correct name for role mrlesmithjr.manage_lvm","message":"Correct name for role mrlesmithjr.manage_lvm\n\nThe name of ``mrlesmithjr.manage_lvm`` was changed yesterday in release\nv0.2.10 to use an underscore, instead of a hyphen. As this changes the\nname of the role on Ansible Galaxy, it needs to be updated in\n``requirements.yml``.\n\nCloses-Bug: #2024163\nChange-Id: I4ea8d8c3a822a7c217bcfcfd5027eecfd21beaed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c4fa262a53f3df6da5c5e7d1e5a3e3f33e5c3100"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c4fa262a53f3df6da5c5e7d1e5a3e3f33e5c3100"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
