)]}'
{"id":"openstack%2Fkayobe~886396","triplet_id":"openstack%2Fkayobe~master~I391d3387bd68e29a25689499bc4b3ebe8156a2fa","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-06-21 08:59:43.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-06-22 15:24:37.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-06-22 11:14:23.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-06-21 10:46:18.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"I391d3387bd68e29a25689499bc4b3ebe8156a2fa","subject":"set previous_release to 2023.1","status":"MERGED","created":"2023-06-19 12:28:49.000000000","updated":"2023-06-22 15:27:06.000000000","submitted":"2023-06-22 15:24:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"886396","meta_rev_id":"7d6159ce08ec83f2aeba58193c98f1b24ee82aa9","_number":886396,"virtual_id_number":886396,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-06-22 15:24:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-06-21 08:59:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-06-20 13:23:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-06-21 09:09:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":2,"date":"2023-06-21 10:46:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":1,"date":"2023-06-21 10:46:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-19 12:29:37.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-06-19 12:45:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-06-20 19:11:10.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-06-20 19:11:10.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2023-06-20 19:11:10.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2023-06-21 09:09:41.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"0c353be6e31aabc794ec38476a3e8416361ad7fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-06-19 12:28:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"86f6ce7a1e284da914b09e793cfc9997ef9b5a92","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-06-19 12:29:37.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"53b09d660233e2b0c1ffadcaad9a8992bd728b78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-19 12:45:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a4320ac39524299849636df35c7cfc5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a2f9f5066f94553b0d0d3f486c92cd8 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f824cee7c7c44ec82dc7915a731d7ac : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ddec149262448cb89313c4ee8eaf4d5 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68e85a5fd699432e97f15d256f799927 : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f5f5b1c242734f519bab9f131df920a0 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86fbd7aa84d544529e9f939bbb89ca9e : SUCCESS in 6m 06s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ccb6833492ac42838b4e7740385bb881 : FAILURE in 7m 00s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/77726e3a6bf44e23adbfc9906e3d6166 : FAILURE in 6m 25s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/eb7c854e695748f29976640162f60326 : FAILURE in 7m 33s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/98f7e85f0f774b4f876e4362f68299d0 : FAILURE in 6m 02s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3373a42df43f4508872e18d9f22797ec : FAILURE in 7m 26s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/d3a7624f2fab4024b762060558f9971d : FAILURE in 8m 26s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30cb72faf1d748848a38e268fd333964 : FAILURE in 6m 02s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c08cde033edf4879891175f5322d1495 : FAILURE in 6m 56s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bd5f4c28b11e41dcb8f09e8375471923 : FAILURE in 6m 32s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/c1d631ef09c04248893965162f7c0f61 : FAILURE in 7m 19s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/42fdbee5004d46238fa12d5a51965714 : FAILURE in 6m 11s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/8976d261d9ed46b7a4c23c45b6e92540 : FAILURE in 6m 55s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ccf49150ebd141b2a92033ec469f229e : FAILURE in 7m 01s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a5b8ab391074c9aa8e7219483d1f93f : FAILURE in 7m 21s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/3de0c7f805094d49a807a99685819222 : FAILURE in 7m 01s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ce1380cb46f4d189dff2db327bfc787 : FAILURE in 6m 13s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a7ea6e297eef4c879cd931c7d99e0a83 : FAILURE in 6m 53s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a4780fac36d4a9b9ad9b526f3517780 : FAILURE in 6m 52s","accounts_in_message":[],"_revision_number":1},{"id":"e8f8e0e41e664bced7a10939a935f85733bbd6f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-06-20 13:23:05.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c648b7b00a79f9939cada01a5f5000496d76c7c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 15:32:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98b91371b8e2403aa662016a675e8475\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fad3f380ef4943dc80db0e20e949b5db : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/156ba4f639ec494097d1003e193eeeab : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6cd367d745e44919c9e986f08161c27 : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5cec747743b04e86b2684d67fadd6f04 : SUCCESS in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2e5bf515f8e64c6ea6d9dee6e7f1c48f : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e44ee06c14b447f3a8c4dcf6d24d0e28 : SUCCESS in 6m 58s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/83d28b5f63544d95b0a217c46322a1ec : SUCCESS in 59m 08s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/404929a8b87e4c918f143875be3558cd : SUCCESS in 51m 57s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a7adf4f0802b43268e06b65bf5252f98 : SUCCESS in 54m 20s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/3277c8b082354924b085730c76bba4b3 : SUCCESS in 59m 38s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/7714a02395b54ed2bd10f5466e1a9d36 : SUCCESS in 18m 14s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/e724a154685e45aabee7e3e59283be93 : SUCCESS in 20m 13s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/27340526461d4554a1b5a63dc4b9adbd : SUCCESS in 16m 17s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2011e881f5374f29b8b909b72f229660 : SUCCESS in 1h 49m 32s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c177c134848441058c2c1ed7204a7c2f : SUCCESS in 1h 47m 27s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/a4c7ce0885744e0eb64a07c983463c0f : SUCCESS in 19m 36s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e2596dec339b42a0ab1a5d97d55f8d98 : SUCCESS in 15m 34s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/7db832194e94468e8ebabc408457c737 : SUCCESS in 33m 14s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c660306a6b524515bf9aaf7e9d32624e : SUCCESS in 30m 43s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a767e2d26e95447eaee6842f60bd63a9 : SUCCESS in 27m 51s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/bc57371f98134156a12df6fd91bef0ba : SUCCESS in 15m 02s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/48b5747609df4baa9b6f8ce43326fa72 : SUCCESS in 11m 58s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e93b62f3747b423e9dd4ae511fb609f9 : SUCCESS in 36m 08s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0471a9a3449b4a7095c3b24fb0314378 : SUCCESS in 28m 55s","accounts_in_message":[],"_revision_number":2},{"id":"b4f860da684b80ace48ad7a6f63e3655dbcf189e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-21 08:59:43.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7f708fe959d77b4217fd86c073032316e968a865","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-06-21 09:09:41.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8afc0755572d990d0a957bec5226f5752e0f0c91","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-06-21 10:46:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d76fb26333bf4ac8fca427d64b7b3823d29f0f94","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 10:46:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cc4d516d60758fedce7d47ed084f35967c99e184","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-21 13:03:57.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/886313 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"26d1afbcd48772d8ea406f4491fe1fd983062977","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-06-22 11:14:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d8e29262304c9092af57d4b3860be469b5a3878e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 13:21:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e41467dd3a445c4b56c9f8dce21917e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c86b24332c0b41a8bec0a9f59b36e5aa : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d059b15fbb32431c8eb078d9e38890f6 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/920a65ae7af145e6a3ec1be15ad2f358 : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a79164351924c73aca0509894000108 : SUCCESS in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/805ab2da6b0447fe861f38955556a149 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10dc72195ecd4a7b921d197c13636a19 : SUCCESS in 7m 20s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ef6f40718f3d42cbbb5061f33c52f0a3 : SUCCESS in 1h 06m 28s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/0b28bfa36eed4d41b539585a47379090 : SUCCESS in 56m 48s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/71da5abc67ab46e586e935762626d97a : SUCCESS in 1h 13m 25s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/fc8e32292d4b4bcb8fa971f74fe59845 : SUCCESS in 1h 05m 38s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ec753ee83ea14938a96648e2e29036dd : SUCCESS in 16m 52s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/c7a92f4dade84d3e80b6e9f10e162a02 : SUCCESS in 18m 03s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da4a7b14ba36493b87ed2d3834bb29ef : SUCCESS in 14m 00s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3b5de13e12f44d7e93f766ee088cd8d4 : SUCCESS in 1h 53m 57s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7b864629a88a401a942187f3e97b8880 : SUCCESS in 1h 27m 57s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/1df477237ca94c0b95b2d23353dfb5cd : SUCCESS in 16m 39s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cbe319aa8f97412b9e648ea908fe65e5 : SUCCESS in 18m 11s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/ee68c4b2676747ae9fc8a1e7d6293bc1 : SUCCESS in 39m 09s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d87abeba7f404b0d909e385d34a29e18 : SUCCESS in 22m 27s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/df8ac6786c804ebaa28c0ae39dc3446f : SUCCESS in 22m 37s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/166f3f19a31a46849a3817232379fe91 : SUCCESS in 14m 09s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff783d5aee6544a7b79fba691e3756c5 : SUCCESS in 11m 20s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/dbb38fb077b5414db1f414712662f441 : SUCCESS in 34m 27s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c12ad60e3e0f4f9c88c28c3d65451a45 : SUCCESS in 27m 46s","accounts_in_message":[],"_revision_number":2},{"id":"fbd8b78cb3c38c354688873130901331fb438987","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 13:22:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b6642a32dda93372b18d3d1bd1c78f58e6c33e2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 15:24:37.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12f259e471cd4bada07d8236bb23f59b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48a2b1e23deb461caab2d9df288e1eec : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd59bd99f9ae42eb94589d770d36d2ac : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c4e5c83c17e74824b676c0c8c0998da0 : SUCCESS in 7m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c82adc75ac254c1bb6c2726b261b4416 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a075e966588491da2b528434de7246d : SUCCESS in 6m 50s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/46d8b7195d9946a28c7609981cbe9737 : SUCCESS in 1h 05m 51s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78bcab1bd30e4004a1ab7d6a2e6f72e8 : SUCCESS in 53m 27s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/038a44b5cf3f4e01befc27031732c785 : SUCCESS in 1h 08m 13s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0a5fba5d6ec9489e8dd026467c1a474d : SUCCESS in 17m 07s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e738fb2895794e11ba8114ef58370e8c : SUCCESS in 13m 41s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/29baa59ef6854b03a1a9cbada0fc1124 : SUCCESS in 1h 24m 24s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/daac4f11b7124b33aab16e7ee530dd6f : SUCCESS in 1h 51m 16s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/4625a7e4231444d2a7230dc9528e86d8 : SUCCESS in 19m 00s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d5d14804ea8439989e384625c4b2a88 : SUCCESS in 15m 15s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/99c9c6ce8cb447c8b602f2e269cf2ed6 : SUCCESS in 35m 43s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07039bfdcb7f49af91e15b7e226382aa : SUCCESS in 25m 57s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9b739ba175074e919bfc9dd4b7a66b01 : SUCCESS in 15m 35s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/32c9abfcc2364232a9d6ecc085b931e9 : SUCCESS in 10m 01s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5cdac19db81b482dbe807e594ef62b2f : SUCCESS in 39m 20s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/24c4c3a596f74f70801bc749ff60cc14 : SUCCESS in 25m 08s","accounts_in_message":[],"_revision_number":2},{"id":"7ce75a9d05b48ae2a92bbcb3b6b47367e7eecad1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 15:24:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7d6159ce08ec83f2aeba58193c98f1b24ee82aa9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-22 15:27:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f48be889ca74c4b80c9937fdc9943c2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ac9ddaaf4094cdda0b4a9cf167babf9 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e87a7447340ea7b0e9a488d5158082e970815add","revisions":{"5e8bebce2562cdebb554fb2e281620fee785fc90":{"kind":"REWORK","_number":1,"created":"2023-06-19 12:28:49.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/96/886396/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/96/886396/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/886396/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/886396/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/886396/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/96/886396/1"}}},"commit":{"parents":[{"commit":"2d2ca347298a646ebfa86aaf2d88796beac0475b","subject":"Revert \"Use OpenStack 2023.1 release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2d2ca347298a646ebfa86aaf2d88796beac0475b"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-06-19 12:26:57.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-06-19 12:28:42.000000000","tz":120},"subject":"set previous_release to 2023.1","message":"set previous_release to 2023.1\n\nChange-Id: I391d3387bd68e29a25689499bc4b3ebe8156a2fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5e8bebce2562cdebb554fb2e281620fee785fc90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5e8bebce2562cdebb554fb2e281620fee785fc90"}]},"branch":"refs/heads/master"},"e87a7447340ea7b0e9a488d5158082e970815add":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-06-20 13:23:05.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/96/886396/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/96/886396/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/886396/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/886396/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/96/886396/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/96/886396/2"}}},"commit":{"parents":[{"commit":"d7b648a82ecd6b512de73f95911f8900d8757f69","subject":"Revert \"Use OpenStack 2023.1 release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d7b648a82ecd6b512de73f95911f8900d8757f69"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-06-19 12:26:57.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-06-20 13:23:05.000000000","tz":0},"subject":"set previous_release to 2023.1","message":"set previous_release to 2023.1\n\nChange-Id: I391d3387bd68e29a25689499bc4b3ebe8156a2fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e87a7447340ea7b0e9a488d5158082e970815add"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e87a7447340ea7b0e9a488d5158082e970815add"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
