)]}'
{"id":"openstack%2Fkayobe~886493","triplet_id":"openstack%2Fkayobe~master~I65f89cea5d70ba3be4828889026b893e1e93db8a","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I65f89cea5d70ba3be4828889026b893e1e93db8a","subject":"overcloud_test: add debug","status":"ABANDONED","created":"2023-06-20 13:32:05.000000000","updated":"2023-08-01 08:02:58.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"ba2c3a832f9f63f879452f6e9597af2c35cc1579","_number":886493,"virtual_id_number":886493,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-06-20 16:30:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2023-06-20 14:52:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"efe51b22699da9b718f1f25ff6bbf0817cf6ca3a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-20 13:32:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ea63cf19730b1c80bdc0455dddc2651ca6a9f7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 14:52:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f39defecf05547ce9ecc246866485044\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/68afcfadbb1345ebb1525dab6a0de187 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c68ebe3a25ec474fbb3ca271d1a1019a : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b2a625bb6b0411493cb776d83d75e22 : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f8b4709551ac4cd6863870e9a7f132cb : SUCCESS in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/be01c1d49b2a4c3e8b21494a350e19cf : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cf2061bab4d4a4aa5e24697737afc2b : SUCCESS in 9m 42s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6cae4516d5b041f398ab62ba97569133 : FAILURE in 9m 51s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9a3e1b68f9dc4cc88983f1f219336ce9 : FAILURE in 7m 01s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/86cacc9acef746d18f03730ee22b9528 : FAILURE in 58m 29s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0b24266ccce645e8b36ad74b0e6fa31c : FAILURE in 7m 28s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa129f9898d948968dddba1cbe402bf0 : FAILURE in 18m 35s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/daa6947050a140c0a4fbc497bad36522 : FAILURE in 8m 49s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78d22c72721a41d08767d268b22f9f2e : FAILURE in 7m 14s","accounts_in_message":[],"_revision_number":1},{"id":"23e0a9f8e7e1abe4782539bd95275b901e7f5bae","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-06-20 15:34:02.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9496295b33f3d7db00ae30f66af6cd3d332b8265","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-20 16:30:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f689e591c4ae43ae9506a9086f6431b8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3713c42f4a9e40d1ae2ab84c7752215e : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95e64a2fff8446c99bae1b30af7536bd : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8aaf918517f4bb083932fef222db00e : SUCCESS in 8m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da8c997c79e742269970bd65a4f5390a : SUCCESS in 4m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/16bd3612467544d3a2c3bc64ffca175d : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/040658a5541e4a17b0329f0e84de529b : SUCCESS in 8m 48s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4067b37e017e4328984ecdef3d6e2e44 : SUCCESS in 53m 12s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ab2ea30a4e1b4884b7615c48ac6ead61 : SUCCESS in 11m 55s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ad79e874473432b90d803a2a2cef039 : FAILURE in 44m 36s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7e343e227714dcca12c843a97176c59 : SUCCESS in 24m 51s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a4d46166f15e4e36b06769eeec14c0ed : SUCCESS in 22m 20s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a5b2505012f64384a8dbbe05c3c93260 : SUCCESS in 11m 22s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f6fdc465864438b9a6040a73297840c : SUCCESS in 26m 45s","accounts_in_message":[],"_revision_number":2},{"id":"ba2c3a832f9f63f879452f6e9597af2c35cc1579","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-01 08:02:58.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0366dbeb0570e31ed989120f7c9f7e8c05e22cc3","revisions":{"6d2bee9e319aa4480f659e23c32337c8bc06530f":{"kind":"REWORK","_number":1,"created":"2023-06-20 13:32:05.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/886493/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/93/886493/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/886493/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/886493/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/886493/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/93/886493/1"}}},"commit":{"parents":[{"commit":"cde32faa4e79228a90b35b7f5cbc0d872e0c7567","subject":"Merge \"CI: Fix kayobe-seed-images-* jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cde32faa4e79228a90b35b7f5cbc0d872e0c7567"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-20 13:30:54.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-20 13:32:03.000000000","tz":120},"subject":"overcloud_test: add debug","message":"overcloud_test: add debug\n\nLet\u0027s debug those Ubuntu glitches\n\nChange-Id: I65f89cea5d70ba3be4828889026b893e1e93db8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6d2bee9e319aa4480f659e23c32337c8bc06530f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6d2bee9e319aa4480f659e23c32337c8bc06530f"}]},"branch":"refs/heads/master"},"0366dbeb0570e31ed989120f7c9f7e8c05e22cc3":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-06-20 15:34:02.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/93/886493/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/93/886493/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/886493/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/886493/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/886493/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/93/886493/2"}}},"commit":{"parents":[{"commit":"505bb28b8d2c3582820ff736b168e64c74cb0a00","subject":"Correct name for role mrlesmithjr.manage_lvm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/505bb28b8d2c3582820ff736b168e64c74cb0a00"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-20 13:30:54.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-06-20 15:34:02.000000000","tz":0},"subject":"overcloud_test: add debug","message":"overcloud_test: add debug\n\nLet\u0027s debug those Ubuntu glitches\n\nChange-Id: I65f89cea5d70ba3be4828889026b893e1e93db8a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0366dbeb0570e31ed989120f7c9f7e8c05e22cc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0366dbeb0570e31ed989120f7c9f7e8c05e22cc3"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
