)]}'
{"id":"openstack%2Fkayobe~887792","triplet_id":"openstack%2Fkayobe~master~I2085f9d1fd457e5b54b17f37a5e725990a60e8bc","project":"openstack/kayobe","branch":"master","topic":"docker_login","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2023-08-23 10:05:33.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-08-23 08:15:37.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-08-22 14:44:48.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-08-23 08:17:24.000000000","reason":"\u003cGERRIT_ACCOUNT_14826\u003e replied on the change","reason_account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},"34940":{"account":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"last_update":"2023-08-23 10:05:33.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-08-23 10:05:33.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-08-23 10:05:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2085f9d1fd457e5b54b17f37a5e725990a60e8bc","subject":"Fix for seed-containers being unable to use password protected registry","status":"MERGED","created":"2023-07-06 11:10:54.000000000","updated":"2023-08-23 10:06:34.000000000","submitted":"2023-08-23 10:05:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"887792-docker_login","meta_rev_id":"7fd0f38a76f860863002acb38ea7ec897a2f263f","_number":887792,"virtual_id_number":887792,"owner":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-23 10:05:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-08-23 08:15:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-08-22 14:44:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-08-23 08:17:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-08-23 08:17:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-08-22 14:44:48.000000000","_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-07-06 13:29:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-07-10 09:51:15.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-07-10 13:06:12.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2023-07-20 11:56:10.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-08-18 14:55:11.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2023-08-21 08:53:15.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2023-08-22 10:31:49.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2023-08-23 08:15:37.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"bb1d3de518957b2aafb3906508c127361c5e42f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2023-07-06 11:10:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0ba764c1c5d7164e40f2b2ccf16c4069120ca895","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2023-07-06 11:29:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6b11430fc77c545d5d85fda005b089151160bd2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2023-07-06 11:31:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"672fa61b3d6a3f549697fdcefbf4ac53a129eb21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-06 13:29:55.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5f5fb95019754f938a83057169aec7fd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5bc434d8bbf74dd09eb7f929a346528e : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d573092f2994d50ae4e97b1c555f479 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e024bf8ddfc44a7848398afb5fd5104 : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/193974d223ef4fb5a7ff890ca5887414 : SUCCESS in 4m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/db66108e19394f84b06f070f98b5178b : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a0f9049d587420f99931c6632b18c6d : SUCCESS in 8m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2abc6fb244a9473e8ebe10c52bd52892 : SUCCESS in 3m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/82b5432d5ec64c9fa1d5236bf48090a0 : SUCCESS in 4m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/872ad769a1154c30ae05562a5736b9e2 : SUCCESS in 12m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7397d29986e34c1788946458a9d5bf1c : SUCCESS in 13m 17s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/132d6bef59da46bc8c17e48584a6e137 : SUCCESS in 1h 03m 40s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/bfc76a534a784f31ba2fadc381867836 : SUCCESS in 58m 31s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c27917b07c843c5b46cb867c48b4438 : SUCCESS in 51m 33s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/cd51564c63ef449e895359c4f2a8beff : SUCCESS in 1h 03m 11s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/0df2514a510848d491777c8f79206eda : SUCCESS in 17m 02s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/da5f286c128242d88398258f68d99bf1 : SUCCESS in 16m 33s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3672f5eea54a41a2bc8d04f509490511 : SUCCESS in 15m 44s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/069c2ab3df494f79b5237a8c51f07c00 : FAILURE in 1h 47m 27s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f6a5524f2233408fbf7fefd7843b057d : FAILURE in 1h 31m 01s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/6ecf912372e3431d90709365b17e5fb6 : SUCCESS in 19m 03s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/10588fec29e94ce58d24a81d1ab97711 : SUCCESS in 14m 38s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/3eb3e9dcfe444308ad717a66c161fd0d : SUCCESS in 33m 15s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b6644fe77e8848158c62aeef4dbabf2b : SUCCESS in 31m 22s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/afe47face2e8466887825afbf17513f4 : SUCCESS in 24m 12s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/da7617ee97604b99b424c9974d85e1b4 : SUCCESS in 11m 58s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1330b4c0c974404dbac2de419d4d3973 : SUCCESS in 9m 32s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a2c7d9e356014a2687ecf42dc0ba5293 : SUCCESS in 33m 32s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a94c463c8e64842bab6569c464297a6 : SUCCESS in 19m 23s","accounts_in_message":[],"_revision_number":3},{"id":"f13a72e922021e5cc5bf818f0261512f563a4cdd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-07-10 09:51:15.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"4805034be1e186c6cc9eab7b28ad294cb8b02eec","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-07-10 13:06:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9870a604ca63d0b746897bbcf3314124df2772f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2023-07-11 08:14:38.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"b5977ff6e3c73e98aa77df6f60c8281684cdd31e","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2023-07-11 08:27:44.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"dd1b945193a7edeaa73eeb043dc58917724a2c63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-11 10:03:32.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ff0c41866ad4da6894d7f3642fa6950\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5430df167244277b38dadccb4ddf488 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98a710275b9d494198075e1790234528 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ac01657181b4696848cfe072f645f39 : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a50b9836c5e64a0b9b5925887fd0fadb : SUCCESS in 5m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9ba258bdddfd43cf845fce75b313eec2 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/837af825bf6441debd19effddde6bd02 : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d3754c3188d444f83b3e9bfa2272e92 : SUCCESS in 4m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d1615295914c4d99bbb87704bc463279 : SUCCESS in 5m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5bd0fdfb7da6431584ce90955c408bb9 : SUCCESS in 12m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aee80232102c4c1c834ee98fb75609bd : SUCCESS in 14m 07s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/af30697d652c43e7b4fba2758fefb581 : SUCCESS in 1h 01m 01s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/d4591222a56243bbb572b1200dedb5b1 : SUCCESS in 50m 55s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f370054a2577413c9867c6092c153cff : SUCCESS in 1h 04m 54s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/10d429f24f0843d2bc37059bf67e8257 : SUCCESS in 1h 01m 57s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/061f9ac2c73c4406b92ecc6c22300264 : SUCCESS in 18m 47s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/e87b177a32d541af86168262862592de : SUCCESS in 18m 12s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/241674a1f160470c921a8fff263bad72 : SUCCESS in 15m 18s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b798a7fe0c664006a7f3b02d372c9863 : FAILURE in 1h 44m 06s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/91f555d77cd54b31b246850a6b4204ce : FAILURE in 1h 15m 58s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/2bd64e4236554418bd39fba2f9026b1f : SUCCESS in 17m 16s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/198073f8bbdd40519ebb905aed6f07f7 : SUCCESS in 14m 49s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/dab3f5fafec44a319eb80fcb539c9193 : SUCCESS in 36m 47s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/833dc8bc282442dc8fa5d79ff0934749 : SUCCESS in 22m 16s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b4cb0e6724ec4a9e96a7503dc9b276fe : SUCCESS in 20m 19s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6d366b93a1034e449469a26a0ec36690 : SUCCESS in 13m 05s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/120f0d5dc6e9456687c0769513c9f535 : SUCCESS in 10m 40s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c864c15dbb2f4dd3930d451bd180d6d7 : SUCCESS in 27m 21s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/771e3e721abe47889d6ead46ae593d58 : SUCCESS in 28m 49s","accounts_in_message":[],"_revision_number":4},{"id":"51ffef09c4812f715afb9b4f496a817a0b081bf9","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2023-07-11 11:34:33.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6084ab7c04e056200c46be75bb917dca86afddc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-11 13:23:18.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b1229007119d426892fd0d67d94624ab\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1707890bc3f74f13883a1a9420937cca : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d8505b9199440ed8beac3df8cca73df : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79e42bd911924067a033747da538ee3e : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e2756f87c994c03aa8702c61b102d41 : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/137a20317bfe4c0581c8bf868c69c750 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6df5a831a3994f13856c3c22d8032a9a : SUCCESS in 7m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/712ac82e55f84b5db2f7cede22b51974 : SUCCESS in 3m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e1fd1e07b668487d9adaae6611635ec4 : SUCCESS in 4m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/19f6f633b1ef42b89e1bf97786a71f64 : SUCCESS in 12m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/94afe49bede944f3bd49e6c31670f978 : SUCCESS in 15m 45s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/cff6339e467644baaa76174c3f9677ab : SUCCESS in 47m 32s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/2a413b26993b43c2b6f571f0816544a4 : SUCCESS in 53m 16s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fae67b4d0a9e4698b1a9cbbaf15173e3 : SUCCESS in 46m 33s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/6b2fbf9ee2ee457b80df0d6e7e6ea844 : SUCCESS in 50m 20s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/87a334743c8e4a6f885bc940ee7319d4 : SUCCESS in 13m 21s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/7444e28b96c540d0966fbe4674fe0ec3 : SUCCESS in 19m 12s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9ce227058de04895b5155f13215ae9fb : SUCCESS in 10m 40s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/171b1a7c93ef4203a55fd2e48ff1d20a : FAILURE in 1h 41m 44s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/568c8bb44d7b4438b0c51b586657a2a9 : FAILURE in 1h 27m 23s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/00576ca702044c75a17f4ab1f2399915 : SUCCESS in 17m 02s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b93e6a9fd92443aab06b3f9e390c1f2 : SUCCESS in 12m 13s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/cdb7fe158e714bbe961c63ebb1f3c267 : SUCCESS in 32m 12s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8fba3e2288104ff4a6893cbfaadaa770 : SUCCESS in 29m 45s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ae4d5ff938c14fcf8e87514730290ac2 : SUCCESS in 17m 39s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c56fcce5086040b98e5156359017455f : SUCCESS in 14m 18s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5392f3acce7d42d3b4d7f6b4597006b7 : SUCCESS in 8m 41s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/844e1217770e456ea631ff0e9af38818 : SUCCESS in 25m 31s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d5847e3951c4513aefc723f5f910243 : SUCCESS in 21m 05s","accounts_in_message":[],"_revision_number":4},{"id":"9630a4dcc94d3f0978fe04b55a0bbdb774599f2d","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2023-07-20 11:29:58.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8e311336d1e4d9ea8a265edec70967a84645a3e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2023-07-20 11:51:55.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"9bbbf9caa0116343132c6ba9029db00c2c91deec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-20 13:44:51.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1183788d48124aebb60fbdc9f4d30c9f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/179720ebc07f4abc9bf1e1e7de7b1436 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33ec2c5da9554acaac6446d7a28d615b : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f95cdb1aaec4095ac867e77ff6bd1d0 : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab49192935824e3388927e754cd9ee59 : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bba51d93b7584b3d9c3834237a161ebd : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb4e0709c03e4da496b0860fe787bb02 : SUCCESS in 7m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6fcc892eae848a8b0823595e94e4f1e : SUCCESS in 3m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2d545b7aa8bf499fb0c7add9c631422d : SUCCESS in 5m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4b872af34686404a9d9f0bc11974cf6d : SUCCESS in 10m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7cb3f5e8582845d3bc428249d50e55c8 : SUCCESS in 17m 30s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/14ad05dc6c7e4417b29d639b653c567d : SUCCESS in 52m 24s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/9a836b790341440a92ab189d0d3f75dd : SUCCESS in 59m 08s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/999ea20d0a364b728b35ed21648175d7 : SUCCESS in 1h 09m 35s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a5ff17934ccb46d4b00a7ba319c02468 : SUCCESS in 1h 12m 12s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/88a0952154a1422f9f7b498b98784516 : SUCCESS in 16m 44s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/fce2119ca79a430089710a284ed0b9f5 : SUCCESS in 19m 17s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/42226a1cfe2249e691a2552bbaed4207 : SUCCESS in 16m 27s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8dbe5669de11465c98b24b8c8848f3ca : SUCCESS in 1h 46m 37s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f7dc5a8ca26746b6945c6e5cc2eb0467 : FAILURE in 1h 40m 49s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/398495f396a4463ebe9a7cae055abd28 : SUCCESS in 20m 46s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67f7fe044100447cbcb1a6aff99bfe62 : SUCCESS in 16m 20s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/f5a99165ec874b07ab770566287e9e97 : SUCCESS in 32m 23s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/db8a479bac6e4c8ab97f4136ba8b49ee : SUCCESS in 31m 45s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/523e22ff06b1492781e38e828c3ba2fc : SUCCESS in 27m 40s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/98704a1463b8410489da9a2f42e93e29 : SUCCESS in 15m 56s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9ec93d2726f42a0a3bd0da0d09c8486 : SUCCESS in 11m 36s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7186e52d9059443e91d92839bec86acf : SUCCESS in 30m 56s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/02bbef3df45f473cacf716281a20dddf : SUCCESS in 25m 42s","accounts_in_message":[],"_revision_number":5},{"id":"6541fdd6575a7788ddef8fc26590edcd3ff78664","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2023-07-25 09:16:52.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"917e4ea80e9247370c98c4889d217547d72fcd41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 11:20:40.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ad43ceba13b044efb001641ee4fd48b2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d846ba5f7284448fa5fe0535bb2fa06f : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a87f2e638694cafa76f95f8473add46 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e126d36d5338494c944d7fd198cda8e1 : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eeefa01dcc0f4e8dadecd43bd19a63e8 : SUCCESS in 4m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1717fcb46d8c4766b34cea92f357bb5d : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41dd928223504ba9938fcd26520dbe80 : SUCCESS in 7m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a6de9a368c1480a838aa6d4264ebcae : SUCCESS in 3m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a67a0a482742424082ba13c45f9abd19 : SUCCESS in 4m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1e1c4c34abc04d9986aa35a68f5f16c1 : SUCCESS in 11m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d8bf1efe85b748e6b7da240da58d4e92 : SUCCESS in 15m 04s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/4c5ea3383842476281a8b1498e222d85 : SUCCESS in 1h 04m 24s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/f4f0b80602174224a6521f8ad900e35d : SUCCESS in 1h 06m 31s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/891652986dfe447c8019ce8c741873d0 : SUCCESS in 46m 14s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/0679aa8baf5643d08a7ac5b9d906d763 : SUCCESS in 1h 23m 51s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/615c6f08136a4cab8bfe6e2576c0d49f : SUCCESS in 19m 42s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/64430f24151e4a11a22dc847e9f4b425 : SUCCESS in 20m 31s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/025b1d6a432a45619f5178b4380510a6 : SUCCESS in 13m 02s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8c6b40a546264f848003a5427e28681a : SUCCESS in 1h 38m 31s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0861f861c07840e58c5b319914e0dc13 : FAILURE in 1h 33m 26s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/76ff7bc213ef4aeaa04c3b3a01665b84 : SUCCESS in 19m 46s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c472f3c09dbc4091b41a32c1dd39c715 : SUCCESS in 12m 50s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/e92c868417b24121b0b42f40d9f5eea5 : SUCCESS in 26m 28s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8f9a0dc387fc4662ab7445b21dd19a12 : SUCCESS in 36m 03s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e8998fa201e4e6eb215a7ef7d26842b : SUCCESS in 26m 21s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/2a6208ef6d0948969580d11c7c5e1d21 : SUCCESS in 14m 08s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/16b1b85e5b3c4546b95a0027d1e146e5 : SUCCESS in 12m 17s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ec13010e677e4e6e86240ab212b737bd : SUCCESS in 36m 08s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/879e8638318042e7bd1f794122154f7b : SUCCESS in 32m 52s","accounts_in_message":[],"_revision_number":5},{"id":"fb285c385297594d9c2188e7c6b89917e13aea94","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2023-07-28 08:16:24.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"16cfdb34999d43bdfb06f684920aead5aa728084","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-28 09:59:31.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9840a5231ad148a9ba4a5db8229d354d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a9b3a9d8a75451cb18a029e2554c7c4 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6539b8924b348ce80e3cec343e623e6 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e903d4e1db3a49a287458c54051bc5b9 : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/350c8a6bc5aa49e3a7a4e14c31d72618 : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9fdbcddcd4f6406a9184163eefc9f333 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42a7668db566400bb05eac2fed920dce : SUCCESS in 8m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b30c109729774cc48e335d4bf978ce90 : SUCCESS in 3m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6b5eb61f5b6b494bbe99d95595a3ff3e : SUCCESS in 3m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bba59f6fd0934dab9af4fbd53d69ec4c : SUCCESS in 10m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fb7d51a56b104a59a696fec7a677b2e1 : SUCCESS in 12m 57s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/35ba5de372b643caa98874ffe01719ca : SUCCESS in 59m 46s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/4f789180b6a84306b26805f3f2203047 : SUCCESS in 58m 13s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/06afc7da584147c896b37df44ebef592 : SUCCESS in 47m 01s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/1553797cedfa49039d52b867bf3f93dd : SUCCESS in 52m 53s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/3c8aefc2228d47b3afb90af707f3e169 : SUCCESS in 18m 21s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/0bee92b6e7294e57889e2c427c056dd7 : SUCCESS in 17m 15s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/51bf591ff20345bc856b606f717751f6 : SUCCESS in 14m 44s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5f0ab4d3afae4c2cb475415b9ce1e646 : FAILURE in 1h 34m 40s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7e824a8fbdf8464ba3e2da820e39380a : FAILURE in 1h 30m 59s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/2e770502a7164e6795e794b501f2bb05 : SUCCESS in 17m 48s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/11923569c34543e49c208f658d0446b6 : SUCCESS in 15m 01s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/2e0528e0c3c9445983f3b332dfd88a7b : SUCCESS in 35m 43s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ab9d9d309d694270a4cc6af6b732c690 : SUCCESS in 27m 41s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/28325d7fd4e64454a9971a9780162f14 : SUCCESS in 20m 47s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/f8f609fda5384ca6aea7d2491abaac7a : SUCCESS in 11m 01s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7f647f67f4bf4f698d14bf6c5c5bbdf3 : SUCCESS in 11m 42s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/dd9c364bc5774d6688a9f489a9fa6c7a : SUCCESS in 35m 20s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2e3aa80cc9f841939cac38bdfd511f96 : SUCCESS in 28m 07s","accounts_in_message":[],"_revision_number":5},{"id":"4a2786122ba05ffcdf98e77cb48f8e2908a48fcb","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2023-08-18 08:03:07.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a3c7f0ffd872288d951fdc5a2f5ded02b48ae287","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 10:08:55.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c7f203c5347472ca9dde39bcb8288bf\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc1427fcb9694a6db59a31681f9e914a : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca2a896829df459495b4c60c4bbf2fe2 : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44731d62ebee408583fd4054f9b33e09 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/80a68f0a97d041ecba6e8a087c07c4a6 : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1eaef54c2bc34862930332bed9467eb8 : SUCCESS in 4m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/766e1a9c88c542069487658848b3e617 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19ae46a3a6664e0b964106b14a50d208 : SUCCESS in 7m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6603a8b359bc40c5843c740aabbb7026 : SUCCESS in 2m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/439e418a6d28405aab7f08dc24f2fc3b : SUCCESS in 4m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5a2b42c1312948c5aeb69040a94cb2f5 : SUCCESS in 9m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7257a59bc81b4634b24440c85f226d02 : SUCCESS in 12m 32s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/caf2171048ff4258baefbbf4f4b757bf : SUCCESS in 1h 04m 49s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/c7e6de14909140cb80633a6563201b77 : SUCCESS in 58m 01s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe574f77541d4775a10b400e4bf5cf79 : SUCCESS in 48m 11s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/d6310a57667b400cb295a1877f4810b6 : SUCCESS in 1h 06m 03s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/180f5520b7bc48ecba9781d35d052d0d : SUCCESS in 17m 22s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/85f1032fa98e4fa292dfdcbaf272c97c : SUCCESS in 14m 14s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea2f3063bae147c28b188369ff418d71 : SUCCESS in 13m 25s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2700528a86034d50a0fed43c8e9198b6 : SUCCESS in 1h 59m 07s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/74a03095ad4f44d58ca87ed8021cd55e : FAILURE in 56m 01s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e38e79c9b333459f90873e2fc0cee831 : SUCCESS in 12m 50s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/156ca65395cc45dcaf02e2464a2ef584 : SUCCESS in 13m 34s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/cecd92267ff5481d8344ed592948668d : SUCCESS in 25m 05s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8e09b774cbd74b6ab4c12bb11c880b2f : SUCCESS in 30m 51s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c98b44feef6149318d9accf39f0e0035 : SUCCESS in 25m 23s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/020fb8426c5e4f1897badae9c00f0734 : SUCCESS in 11m 57s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fd16fde38e334865918d5122bb381156 : SUCCESS in 8m 57s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1e37d90072994dbc9795ebb19f74df01 : SUCCESS in 32m 49s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7325a87551764f7ead00cdcfd1d16fb4 : SUCCESS in 26m 34s","accounts_in_message":[],"_revision_number":5},{"id":"b06499c3374d3dc62dcf0c1539662900cb7580a5","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2023-08-18 12:26:08.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"eda9f4a4fce55a5a67843a4e018bad27e9a6e2a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-18 14:30:32.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36afcf205fb24e59b45a674bbd49f6b2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4af6a99c1d7e4789a9f1ca51042aac84 : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de6d6ca1989f434ab890901b5ee7335f : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04e62def66fa4340ac95111d15e5e296 : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc52cd11484f4020916fdd4a610e908e : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d4bd5eb20ff54963a8d5f290f9c10c4b : SUCCESS in 4m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/234fca2577ad439ca663287d5e0e1bd8 : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56e2947c187c4deca23af28410c046b3 : SUCCESS in 6m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e46dc332a41d4ecab269d202b39a36ed : SUCCESS in 3m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e2c4e687dc2c4679afba9329da036754 : SUCCESS in 4m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/78bebe91456248109d2bcbf005e69d68 : SUCCESS in 9m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/66f99586d5634d9bb1e494acab46dae7 : SUCCESS in 13m 26s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d6f1086c8779433ab909c987d870d6d2 : SUCCESS in 59m 48s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/4e4b361f8d304341a8ecd3328082b7d1 : SUCCESS in 53m 32s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52521091c4f14b9a96e649b8158caff1 : SUCCESS in 49m 51s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/6809001e8cf8481d84b97340785962c9 : SUCCESS in 1h 03m 03s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/7a27df9608fc40b8b214e1e0b58a472b : SUCCESS in 13m 57s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/995401e3094045ffb63db720b2862222 : SUCCESS in 16m 28s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2ae4cdb5fb9d49228a376accb71b81ce : SUCCESS in 14m 01s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/0671e252cc4d4803abe9621c8316ccee : SUCCESS in 1h 56m 31s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39f49db07a664145a7c1938a7d92e68f : SUCCESS in 1h 46m 22s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/32c530acc36e4d4f882de846085b10cd : SUCCESS in 17m 37s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a548cdfe74b4c96af4bde91b34d9ffa : SUCCESS in 15m 47s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/377f13e54d2045d98698ecedfbf7afbe : SUCCESS in 35m 22s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e18c8f488c504ec382e4bad1981c1e0e : SUCCESS in 28m 51s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db1ce6f4945f4b779c81bf497f623082 : SUCCESS in 22m 40s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5baacafae5f947acb0fcf60565b1c57a : SUCCESS in 13m 08s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b12725cbf2349798801ba6e144d4444 : SUCCESS in 11m 18s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/29cea812b4a44e498d9eb31a503dcd77 : SUCCESS in 30m 06s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8a44a99d29ff41d58f98d24ea6d17f90 : SUCCESS in 21m 04s","accounts_in_message":[],"_revision_number":5},{"id":"3dc1227712a39ee8e00c86091d303514aa05b210","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-18 14:55:11.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2aa7bd42e43f87873e5788c23213b0cdb327d4ba","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2023-08-21 05:47:05.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9ec3add714ec828f58356023bcc84d99e4927a8c","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-08-21 08:53:15.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"f39f0b7dbdfaffaec544c8a446095e29afe92b4b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-21 09:58:19.000000000","message":"Patch Set 5: -Code-Review","accounts_in_message":[],"_revision_number":5},{"id":"1f70f3b71b20943cb941bb7d5e3af130e61f798d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-21 09:58:24.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"c18eba5e359ed7bff8e83f5832b67a2dbf5b031f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2023-08-22 10:18:36.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"7645a28c67763037111d14525912bfc07167bb1c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-22 10:30:58.000000000","message":"Patch Set 6: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"83016994ba619aa4c879a6b7781b6e01fc9a6140","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-22 10:31:49.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"86293015861c2f273a13487a7bdcb26fd30c5453","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-22 10:41:15.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"831dde39b9b81131cf5fe923b84bf9374ebe5fb2","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-22 10:42:45.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b49a0a501b43c4e543ef0cfab028d75daf4b4f4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2023-08-22 12:07:31.000000000","message":"Uploaded patch set 7: New patch set was added with same tree, parent tree, and commit message as Patch Set 6.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":7},{"id":"442755523a09067049f305815de18996aa811a29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2023-08-22 12:16:20.000000000","message":"Uploaded patch set 8.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":8},{"id":"2616bad17cdab71c4e4142a78049e0cc57e79546","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2023-08-22 12:30:48.000000000","message":"Uploaded patch set 9.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n","accounts_in_message":[],"_revision_number":9},{"id":"18af012a4b92bcdedae98dec4656a10986fac09b","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2023-08-22 12:31:56.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"30d5a9d1c9d5eea1fe123347aaa63d1d50e4e2ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 14:34:37.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf8435a759c2404ea7d60047ffc7b915\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4ed655785d04811ac5356bfeab97811 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa42f9cb1d34407b80b9fda75e2b5592 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/07575f71453d4e4584eff0223ba20a0c : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c37621baf416419cbb355a54642d609d : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/db2bc45158664e9f8853434c06a7cb3f : SUCCESS in 4m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/26fd4f717ba0414a84e49792783ba500 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a9f2ce28b7641f48fe6ac83a5f453f9 : SUCCESS in 7m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a811a51d1c2453d98fb2c7a8545aca1 : SUCCESS in 3m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c7d6e3335694457694a735816501ab58 : SUCCESS in 4m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4ec30485f6d340b296567b3d191f074d : SUCCESS in 13m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9fbcfa53b4fa4e42a67996b2265941b3 : SUCCESS in 14m 46s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/74081c4f7e6a465794fcc4cd4cfc6f18 : SUCCESS in 1h 18m 13s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/5682f5d197e042b5a212af5c3bc1083d : SUCCESS in 59m 53s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/897fb21b8ed740a6bd7e74c18c75bcc6 : SUCCESS in 54m 29s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/cc61a9a50fe34a81851fdc55986e1b9b : SUCCESS in 1h 04m 29s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/05bf8b8416134eafa6ea1689bfcdcb8b : SUCCESS in 17m 39s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/a2fff8fc0aab4d9b96bbdbd584108ce0 : SUCCESS in 16m 47s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f58bc9d389204848ba2bb58307204fa1 : SUCCESS in 15m 10s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5180ff588ed44a36841f18fc6b5efb2c : SUCCESS in 1h 36m 01s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea0f6618eef74274b8b0c21c0fe2328a : SUCCESS in 1h 56m 02s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/586ff861a08d49e18e9bf8325dd6784e : SUCCESS in 16m 58s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/679e9d6d68974e70aea074c6aa4c964c : SUCCESS in 15m 01s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/6f2eaf01d2e34f18b33ed5af18fefe9a : SUCCESS in 31m 13s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/29dd586aaff84b08b21c4d155e4a5af5 : SUCCESS in 28m 40s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a979319d70247f6a8a1f9930d0dbb95 : SUCCESS in 25m 44s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ab1d88e0512e4f8c87ad0485205be073 : SUCCESS in 13m 14s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f64a6de5259c4c469bab79eafc41599e : SUCCESS in 10m 50s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7e403e23f0174fdeaa7a0b365cb96ccd : SUCCESS in 32m 59s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d90b3c532d14afa8bef8040c7fa385c : SUCCESS in 25m 53s","accounts_in_message":[],"_revision_number":9},{"id":"835ac025e18a3ca557031b5a0fe0e83a659bd8bf","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-22 14:44:48.000000000","message":"Patch Set 9: Backport-Candidate+1 Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"9e476938ef24d7c788e6d2dd1de96cbe4c231514","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-23 08:15:37.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"406176a390c2ac96a058544104a2803e11ba1a32","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-08-23 08:17:24.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"71268a71a0e44ca9eb8d7641545726509c670446","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 08:17:49.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"93bbd6b409ae271d1eaf0c8356ca9112994b01c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 10:05:32.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f58cefdcad94764a21425ee99a702a9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a79ca3e0ea2c4a1994bada0cbef79aeb : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36147669973b4bab9c6da7ec9412b1f7 : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e7f687940b6149cf8038eb6717231928 : SUCCESS in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b53ba7f32d5b4437b018ee63cffd2f27 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7452d143973e4de794ba1a8cbba47443 : SUCCESS in 7m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8bdccc61c26e4adaa0fd8b6fcf6e9ac3 : SUCCESS in 3m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/873dbf4de2db4a3e8db0f124d2ab9489 : SUCCESS in 4m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dbb2e75613384f3097e89e9dc18bac16 : SUCCESS in 13m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3d7ad7a02b214d46bfdc65dc1d14637f : SUCCESS in 14m 51s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/e921803801a9414dbadd0f72be4e5b51 : SUCCESS in 46m 10s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/866c88bace1f47f29a0705906073e74c : SUCCESS in 56m 31s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/6e7fa4fbbab046698622fc3909c72de4 : SUCCESS in 1h 07m 20s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e9caf83e3f2f4b4982936b4270d3418e : SUCCESS in 16m 50s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4d9e2ec1418a48c89db8d3b19c059eed : SUCCESS in 13m 06s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/e987d0df1581496182db935fc58f0426 : SUCCESS in 1h 46m 19s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a38e07e12f9e49239f8ba7a235a001d3 : SUCCESS in 1h 42m 12s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/5c6d74a88d674cbcb37c773f8a644d9e : SUCCESS in 14m 47s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/abcac6b2fc4e4c69a02da47359557766 : SUCCESS in 13m 11s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/fe4828244111406f9b847b30d621ea84 : SUCCESS in 31m 36s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2e5268c37444a8095c1f0f0f0c3886f : SUCCESS in 33m 05s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1ed18b7afb3d4d5aaee67a757ca6632c : SUCCESS in 13m 34s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ea9b1f74fa444907835d5bdf36b90ab2 : SUCCESS in 10m 53s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b88800ab8d1c48ea8f0c88e2ebebbcd1 : SUCCESS in 30m 41s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/48717c62a5544d52b55bb8f2203fc812 : SUCCESS in 21m 12s","accounts_in_message":[],"_revision_number":9},{"id":"da1a3547e909972a6a367d2e7fffe899b5e04a3b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 10:05:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"7fd0f38a76f860863002acb38ea7ec897a2f263f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 10:06:34.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2a4a761e7274cef9b67a17ea592811b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/472aa9f0aea14c468d3ba5c3233caa2d : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c59f020b19dc4185ad068c9542d69c13 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"ed263eeb4f8e85c2b33d3f92f5f711f725f5fdae","revisions":{"081b1e08f0ae7e131ccddfe31c30e19cf5e4a8c4":{"kind":"REWORK","_number":1,"created":"2023-07-06 11:10:54.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/92/887792/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/92/887792/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/887792/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/887792/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/887792/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/92/887792/1"}}},"commit":{"parents":[{"commit":"e87a7447340ea7b0e9a488d5158082e970815add","subject":"set previous_release to 2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e87a7447340ea7b0e9a488d5158082e970815add"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-07-06 11:08:49.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-07-06 11:08:49.000000000","tz":120},"subject":"Fix for seed-containers being unable to use password protected registry","message":"Fix for seed-containers being unable to use password protected registry\n\nCloses-bug: #2026276\nChange-Id: I2085f9d1fd457e5b54b17f37a5e725990a60e8bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/081b1e08f0ae7e131ccddfe31c30e19cf5e4a8c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/081b1e08f0ae7e131ccddfe31c30e19cf5e4a8c4"}]},"branch":"refs/heads/master"},"817ba2627eca543ea17a41cebd0ec392a9d54105":{"kind":"REWORK","_number":2,"created":"2023-07-06 11:29:55.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/92/887792/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/92/887792/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/887792/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/887792/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/887792/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/92/887792/2"}}},"commit":{"parents":[{"commit":"e87a7447340ea7b0e9a488d5158082e970815add","subject":"set previous_release to 2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e87a7447340ea7b0e9a488d5158082e970815add"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-07-06 11:08:49.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-07-06 11:29:11.000000000","tz":120},"subject":"Fix for seed-containers being unable to use password protected registry","message":"Fix for seed-containers being unable to use password protected registry\n\nCloses-bug: #2026276\nChange-Id: I2085f9d1fd457e5b54b17f37a5e725990a60e8bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/817ba2627eca543ea17a41cebd0ec392a9d54105"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/817ba2627eca543ea17a41cebd0ec392a9d54105"}]},"branch":"refs/heads/master"},"9957f8d06dd3ac58a423c3cfd86a55e567af6d28":{"kind":"REWORK","_number":3,"created":"2023-07-06 11:31:39.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/92/887792/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/92/887792/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/887792/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/887792/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/887792/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/92/887792/3"}}},"commit":{"parents":[{"commit":"e87a7447340ea7b0e9a488d5158082e970815add","subject":"set previous_release to 2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e87a7447340ea7b0e9a488d5158082e970815add"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-07-06 11:08:49.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-07-06 11:31:23.000000000","tz":120},"subject":"Fix for seed-containers being unable to use password protected registry","message":"Fix for seed-containers being unable to use password protected registry\n\nCloses-bug: #2026276\nChange-Id: I2085f9d1fd457e5b54b17f37a5e725990a60e8bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9957f8d06dd3ac58a423c3cfd86a55e567af6d28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9957f8d06dd3ac58a423c3cfd86a55e567af6d28"}]},"branch":"refs/heads/master"},"d7512724f7dd7bd681e6e5a2a0c6d09461236c30":{"kind":"REWORK","_number":4,"created":"2023-07-11 08:14:38.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/92/887792/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/92/887792/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/887792/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/887792/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/887792/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/92/887792/4"}}},"commit":{"parents":[{"commit":"e87a7447340ea7b0e9a488d5158082e970815add","subject":"set previous_release to 2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e87a7447340ea7b0e9a488d5158082e970815add"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-07-06 11:08:49.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-07-11 08:14:25.000000000","tz":120},"subject":"Fix for seed-containers being unable to use password protected registry","message":"Fix for seed-containers being unable to use password protected registry\n\nCloses-bug: #2026276\nChange-Id: I2085f9d1fd457e5b54b17f37a5e725990a60e8bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d7512724f7dd7bd681e6e5a2a0c6d09461236c30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d7512724f7dd7bd681e6e5a2a0c6d09461236c30"}]},"branch":"refs/heads/master"},"ce671c7e9fff862372c72773d69ba2c0de8973b7":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2023-07-20 11:51:55.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/92/887792/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/92/887792/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/887792/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/887792/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/887792/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/92/887792/5"}}},"commit":{"parents":[{"commit":"e87a7447340ea7b0e9a488d5158082e970815add","subject":"set previous_release to 2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e87a7447340ea7b0e9a488d5158082e970815add"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-07-06 11:08:49.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-07-20 11:51:55.000000000","tz":0},"subject":"Fix for seed-containers being unable to use password protected registry","message":"Fix for seed-containers being unable to use password protected registry\n\nCloses-Bug: #2026276\nChange-Id: I2085f9d1fd457e5b54b17f37a5e725990a60e8bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ce671c7e9fff862372c72773d69ba2c0de8973b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ce671c7e9fff862372c72773d69ba2c0de8973b7"}]},"branch":"refs/heads/master","description":"Edit commit message"},"5d37fd128636dd16e1c6c5d948e8c287f96edf99":{"kind":"REWORK","_number":6,"created":"2023-08-22 10:18:36.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/92/887792/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/92/887792/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/887792/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/887792/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/887792/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/92/887792/6"}}},"commit":{"parents":[{"commit":"e87a7447340ea7b0e9a488d5158082e970815add","subject":"set previous_release to 2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e87a7447340ea7b0e9a488d5158082e970815add"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-07-06 11:08:49.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-08-22 10:18:18.000000000","tz":120},"subject":"Fix for seed-containers being unable to use password protected registry","message":"Fix for seed-containers being unable to use password protected registry\n\nCloses-bug: #2026276\nChange-Id: I2085f9d1fd457e5b54b17f37a5e725990a60e8bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5d37fd128636dd16e1c6c5d948e8c287f96edf99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5d37fd128636dd16e1c6c5d948e8c287f96edf99"}]},"branch":"refs/heads/master"},"ec711f46f0b34e8044cea8e5c6bcebc82dbcc785":{"kind":"NO_CHANGE","_number":7,"created":"2023-08-22 12:07:31.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/92/887792/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/92/887792/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/887792/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/887792/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/887792/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/92/887792/7"}}},"commit":{"parents":[{"commit":"e87a7447340ea7b0e9a488d5158082e970815add","subject":"set previous_release to 2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e87a7447340ea7b0e9a488d5158082e970815add"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-07-06 11:08:49.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-08-22 12:06:38.000000000","tz":120},"subject":"Fix for seed-containers being unable to use password protected registry","message":"Fix for seed-containers being unable to use password protected registry\n\nCloses-bug: #2026276\nChange-Id: I2085f9d1fd457e5b54b17f37a5e725990a60e8bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ec711f46f0b34e8044cea8e5c6bcebc82dbcc785"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ec711f46f0b34e8044cea8e5c6bcebc82dbcc785"}]},"branch":"refs/heads/master"},"40c4ef0a0aadf4b6d196c85f5563501039923164":{"kind":"REWORK","_number":8,"created":"2023-08-22 12:16:20.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/92/887792/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/92/887792/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/887792/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/887792/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/887792/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/92/887792/8"}}},"commit":{"parents":[{"commit":"e87a7447340ea7b0e9a488d5158082e970815add","subject":"set previous_release to 2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e87a7447340ea7b0e9a488d5158082e970815add"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-07-06 11:08:49.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-08-22 12:16:04.000000000","tz":120},"subject":"Fix for seed-containers being unable to use password protected registry","message":"Fix for seed-containers being unable to use password protected registry\n\nCloses-bug: #2026276\nChange-Id: I2085f9d1fd457e5b54b17f37a5e725990a60e8bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/40c4ef0a0aadf4b6d196c85f5563501039923164"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/40c4ef0a0aadf4b6d196c85f5563501039923164"}]},"branch":"refs/heads/master"},"ed263eeb4f8e85c2b33d3f92f5f711f725f5fdae":{"kind":"REWORK","_number":9,"created":"2023-08-22 12:30:48.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/92/887792/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/92/887792/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/887792/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/887792/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/887792/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/92/887792/9"}}},"commit":{"parents":[{"commit":"e87a7447340ea7b0e9a488d5158082e970815add","subject":"set previous_release to 2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e87a7447340ea7b0e9a488d5158082e970815add"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-07-06 11:08:49.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-08-22 12:29:11.000000000","tz":120},"subject":"Fix for seed-containers being unable to use password protected registry","message":"Fix for seed-containers being unable to use password protected registry\n\nUpdate old code to use truthy filter\n\nCloses-bug: #2026276\nChange-Id: I2085f9d1fd457e5b54b17f37a5e725990a60e8bc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ed263eeb4f8e85c2b33d3f92f5f711f725f5fdae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ed263eeb4f8e85c2b33d3f92f5f711f725f5fdae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
