)]}'
{"id":"openstack%2Fkayobe~891399","triplet_id":"openstack%2Fkayobe~master~I4830500d3c927b0689b6f0bca32c28137916420b","project":"openstack/kayobe","branch":"master","topic":"bug/2031390","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2023-11-30 12:40:26.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-11-30 08:59:01.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-11-30 14:57:28.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-11-30 14:57:28.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-11-30 14:57:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4830500d3c927b0689b6f0bca32c28137916420b","subject":"Fix configuration dump with inline encrypted variables","status":"MERGED","created":"2023-08-15 10:28:58.000000000","updated":"2023-11-30 14:58:43.000000000","submitted":"2023-11-30 14:57:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"891399-bug/2031390","meta_rev_id":"914605e29a45b5b8177817e10706140035dd27fc","_number":891399,"virtual_id_number":891399,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-30 14:57:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2023-11-30 12:40:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2023-11-30 10:08:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2023-11-30 10:45:21.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-11-30 11:27:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2023-11-30 12:40:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2023-11-30 12:40:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2023-11-30 10:08:24.000000000","_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2023-11-30 12:40:26.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-15 11:56:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-16 15:14:34.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-11-08 14:32:29.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2023-11-30 08:34:44.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2023-11-30 11:27:13.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"0e0f986f62f2d287442de935cc8c1561d1de9ea9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-08-15 10:28:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ccb2d71e65d88b92e989db623089b3df0e26fd2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-15 11:56:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c91a0858910c4e0ca4e3b61fd346f5e8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b30fd995a6d94d2bb4fdf5e22845ac51 : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d0d1adcffd5472bae9a1037deccfd6f : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45af77f6cdcf49a8be2e12cc6e34dd75 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7610f8e993fe4fe99ed1a087f13ca8b6 : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e303737e406b4682b7ecec1e32f5d18d : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a17ec5aa391d4ed2a718f1c5329308fd : SUCCESS in 7m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50b1915c6319410899042376d441806a : SUCCESS in 3m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/75174e92caec47cc85d22a9071ca3abd : SUCCESS in 4m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/da3735451ceb44468c3477c14c87c2f6 : SUCCESS in 10m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b24c9924f8ba448eb14a8297e65c2ca1 : SUCCESS in 13m 42s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/7689b0d9124e4f57bdedab4d33a6835c : SUCCESS in 59m 27s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/f126c7f34ea14314884ea3a8f67aa204 : SUCCESS in 1h 12m 36s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a7d8d29bb07f4b539bc7aaed5278ec74 : SUCCESS in 46m 36s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/bb11f7f91e6d4558b8f819bac8ebeab3 : SUCCESS in 1h 20m 00s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f54e55e756c2481c8995b2c44a3adf71 : SUCCESS in 13m 37s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/a3baa703b49f44358d88871031416462 : SUCCESS in 17m 50s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9427a8bf05a244f7901a6a45b7058f82 : SUCCESS in 9m 49s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/978cd57436b741cf957f174093468386 : FAILURE in 1h 09m 43s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6f1bdbd3690a4861a335cd639344fa90 : FAILURE in 1h 12m 54s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/28dcabd6d179400c973c8365ad75572f : SUCCESS in 13m 00s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7a6b39381cb45ffb64e832211fad734 : SUCCESS in 13m 35s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/863a50ec5fe14bd9936f773f552d8f69 : SUCCESS in 25m 11s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/27d449ec20c04b71bdfca9dc0714f6b6 : SUCCESS in 21m 46s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/43671b5466664b34b64931e6f5a7be30 : SUCCESS in 21m 54s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b6846d5734e643d69f144755db0b2c86 : SUCCESS in 10m 19s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa39d06d116b4ddda1410f84c05e26c5 : SUCCESS in 10m 27s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/76278cbc1ec245cab868e132c7d1c1eb : SUCCESS in 23m 18s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/08cf6c90bacf48478460ad92fd9a19f9 : SUCCESS in 23m 23s","accounts_in_message":[],"_revision_number":1},{"id":"8adc43ac2d14dd369cd146964c109b3fe8b6ab27","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-16 15:20:10.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"de42ff8cd06a14fa7d5301a4fb168c13afe12bdc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-10-03 11:54:43.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR is:ANY\")\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"af7f8b37f11abab380b67c94a75f73c4991700c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-03 14:39:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/70b764eebaad4d0cab4c7094c3f8f225\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf20a5e793e44829ad224cf65c6c0e29 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2d6546e5f4d4b49b3f8abc786f45846 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7eaac6175334289949c7d3fe74e23a9 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/56d78e3f22ac4d9bb4e3a35e7731e523 : SUCCESS in 4m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bb8562c1de5a476e866d435590d504ce : SUCCESS in 3m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8d10b4d085ad48818e2e307476acb396 : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a566483ea1149fd9f8ab2f1a8e6f5b2 : SUCCESS in 8m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2729076bb8394b669e2d7006394fd246 : SUCCESS in 4m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4d3583970db04e97bec552e0df63682c : SUCCESS in 4m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e8fff1d43f5b4f478d90917534a869c4 : SUCCESS in 12m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/57189cc2668b4beebf299ec91b65fe24 : SUCCESS in 11m 15s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/2fcac1b4e9df4546bf91dc1e2d797f76 : SUCCESS in 1h 00m 15s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/e2b40e3e06ec4da98df506b3a3e4c1d5 : SUCCESS in 57m 16s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c4d88c2b5eee42c581a43fbfc1974a0e : SUCCESS in 45m 11s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a194856916674973b020ccb95d74331f : SUCCESS in 1h 03m 11s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4da41774548e41749c225a7507c990ac : SUCCESS in 19m 57s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/12692a07ca1f4a5db1221aac8f1511be : SUCCESS in 17m 01s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/11c5530121b3426cab933f50a60f0994 : SUCCESS in 16m 06s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/8559688f7ca24596ae29f1cfc001d554 : TIMED_OUT in 2h 37m 12s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe55efdabfcd43159b1b406c91dc12b1 : SUCCESS in 2h 22m 08s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/6ac909239e1c47509ed90ca58b1cc34a : SUCCESS in 20m 20s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be1574168f544a8294969ef2543a9600 : SUCCESS in 14m 15s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/ad201b25e5b0460b856438c4f0785dab : SUCCESS in 27m 32s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3d52dc5099d84888880649b53443a0f9 : SUCCESS in 29m 23s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/020b84e2a0a649b7a7d9ed73c11984de : SUCCESS in 26m 35s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/477421d4ccf645fd8eef31a0f800b651 : SUCCESS in 18m 14s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/43fe5b381fa34eceae0135405c767542 : SUCCESS in 11m 07s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9bdc1d77f52047899feefc23e258aac3 : SUCCESS in 36m 04s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a7e13a8905f44ee4aa0f715161ddce62 : SUCCESS in 27m 49s","accounts_in_message":[],"_revision_number":2},{"id":"49a707a64e9d45fe5d757b488dfa845708d8af8e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-11-01 17:17:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"82b33a0ae7f377114ffbe217320e65a720f2ea85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 19:46:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fdcf8214540451a9ed00858fa8c84e8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/622603edcf4844f586aac678a94bd653 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5fc8aedfbdd48c1b269a9cea048ce3e : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c803c4c5233344bb867bfa5ce0a4c512 : SUCCESS in 3m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8ca2a0d2dd66471face551f4b6dd992e : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a60803fce694f7189d207296b0e5d80 : SUCCESS in 7m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/227814ad60d34c1bb4387333f350173f : SUCCESS in 3m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4789400f3a154030a57f698e2d0e2a1d : SUCCESS in 4m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0244501bb50c40a8bbfc2be2cb18697a : SUCCESS in 14m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3f5b1e4a43c840fb8ac4f160cf37f479 : SUCCESS in 10m 20s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/b3070717b0ee4e6dadeb855848bc4805 : SUCCESS in 57m 48s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/70fd552a511641788fa3a01427eed4d8 : SUCCESS in 52m 06s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/88bcdbed89a74cacaf9a97c5feb76992 : SUCCESS in 53m 32s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/86ef3e7c8a9b4220bb4623f7e6500de0 : SUCCESS in 1h 06m 51s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/298721d56f0c49b3853fc84f60426a45 : SUCCESS in 13m 32s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/6160f92211bb454493aa2deb33f52ced : SUCCESS in 15m 40s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/490f36066c9246649f9fbfd88a149d68 : SUCCESS in 14m 00s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/a6f8a1721ce342bbb33956f8fd2725e6 : SUCCESS in 2h 24m 16s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97a2366afbda4e78a7e1eb5db7083990 : SUCCESS in 2h 05m 11s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/74d74589c474478db04194a347c42342 : SUCCESS in 17m 37s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e6e572506ca4e1bab69d7eb8bea3860 : SUCCESS in 18m 10s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/fe53d87978d94e7fa9ec688cfcfdeddf : SUCCESS in 33m 14s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7e641bd6c0304f36bd55cab9a36193d2 : SUCCESS in 30m 10s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/24c2122651bd4bbeac63038a5de1d9f2 : SUCCESS in 26m 50s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/25bdddd2f62545c3bc7df7ead55a379b : SUCCESS in 10m 04s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e78edbfa7d4c433096bdfce81c09fb10 : SUCCESS in 10m 26s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e1322a25715a48d8819eca3e964c962a : SUCCESS in 18m 52s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/38e7cf88d4f24a3ab77b523a3475336d : SUCCESS in 14m 38s","accounts_in_message":[],"_revision_number":2},{"id":"7dfe55c4b7ba5de5be6524e8e6b3770034b6a598","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-11-08 14:32:29.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a6e7a479d5fe2999db145213411ff8c735391624","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-11-15 14:54:09.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"026952e24f85a45fa25a73e3cdb29f8d228da073","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-11-15 14:54:24.000000000","message":"Patch Set 2: Backport-Candidate+2","accounts_in_message":[],"_revision_number":2},{"id":"be56bf2912cb2ff0407652452e49b8ae8d46d941","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-11-30 08:34:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a420824792033c9286118614ce74441f6bc05337","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-11-30 08:58:35.000000000","message":"Uploaded patch set 3.\n\nCopied Votes:\n* Backport-Candidate+1, Backport-Candidate+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"01fc0b3352d18bd07fd1bc608fb25ac65ac1fe62","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-11-30 08:59:01.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cb55f9e401d63739c2366b5fd2eac8cfd39242b0","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-11-30 09:14:44.000000000","message":"Patch Set 3: Backport-Candidate+1 Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1726bde2236bba1ecea35f1f7fcdaa6642dfdcb6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-11-30 09:55:58.000000000","message":"Uploaded patch set 4.\n\nCopied Votes:\n* Backport-Candidate+1, Backport-Candidate+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":4},{"id":"0bd3ed3b97e71008fa83d4a89efd7e92efa28f04","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-11-30 10:08:24.000000000","message":"Patch Set 4: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"aa1a6167a6c1016d44d3b248aa3f0e093d1fe239","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-11-30 10:45:21.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b71c0b421c133c00b438ff280c2f2dc093151f2e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-11-30 11:27:13.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"ddb15822c42ccfa8074cc5fa8b925b1942a884f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 12:17:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b934d0f1c589482699abc6c1053e6711\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d10800c9b94940188c8fc293c7d25748 : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afc31f6d6eb44e87aa5dea7e6af53f82 : SUCCESS in 9m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1176aadbf3c3482dad25db0cb7d2bb29 : SUCCESS in 4m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/32782b2eeae0440aac14046827cf9260 : SUCCESS in 5m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d296499e3e7c4c8abd77da91dcf8d130 : SUCCESS in 11m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3fe517cc41b249b59e14eda884105649 : SUCCESS in 11m 27s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/467f9be6af0749fca21de3fc193a18bb : SUCCESS in 1h 10m 07s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/5bd584b6bfc24d2b97f0b73af2e66c41 : SUCCESS in 59m 35s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cac7945cbeeb466286ca637a87dd566b : SUCCESS in 53m 53s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ff94bd13c77a4272bfcc8d2dc5985112 : SUCCESS in 1h 09m 36s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/4dd4522c2ae040a79640135a29735e0b : SUCCESS in 17m 27s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/b1c9051413e14bbb99340b4c7aa175cc : SUCCESS in 17m 18s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b20178c54cb6451093493fc6047c2169 : SUCCESS in 16m 16s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d9edd6fef80e4e91a6b40815e2ba28ba : SUCCESS in 2h 20m 32s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db177321c60f482490a6dbb52333813a : SUCCESS in 2h 04m 04s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/7bce399b0ea6479ca52392c8da66b834 : SUCCESS in 15m 58s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1762d7a5224e446dbcf8bf276945692f : SUCCESS in 15m 50s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/580dd1cfd117423894821f1ca85e0e11 : FAILURE in 29m 14s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/7fed8f9ce5f14fe7b44fa7b6dab05e03 : SUCCESS in 25m 32s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7ab057e47744b0f83d8efec57a2755a : SUCCESS in 24m 45s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7afe4d3ac7984bad818d79d43fd19853 : SUCCESS in 12m 16s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a442cc6a92d84e0f94088783d5b6ef3e : SUCCESS in 10m 16s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9a46acfa4e404541b9307c043cbbe989 : SUCCESS in 17m 31s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ee958234bcc4edf84b4f5721ae9c0b2 : SUCCESS in 13m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd79803a9fe24fe0a98c23343d48b82c : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4b0d4b27430b4a77bd9d2ae0ca397b6a : SUCCESS in 5m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7beac50383f8446e8b7d2947d4066593 : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":4},{"id":"46d4471fc77364e227a48b8fbe8fd70b47cd59af","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-11-30 12:40:26.000000000","message":"Patch Set 4: Backport-Candidate+2 Code-Review+2 Review-Priority+1 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"d3d9d66f6c5ceef474dd90f3bb27ca31d6982354","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 12:40:49.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7f04747347f19789c2c513227576432998f3b02c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 14:57:28.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb4847e0ad3541e197e90599cd9120e2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51d1c03dbecd466eac9ee526e4656097 : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6069e3b923d14524899762b7a0a196c4 : SUCCESS in 3m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/86a8ac49d9914c3bade32ad96661f5cf : SUCCESS in 4m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1e3b7217aea64747acbcc96f40c27565 : SUCCESS in 9m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5a6a26b2651a4d988325636adaf3109d : SUCCESS in 11m 31s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/573c2c085fbe482d8b6ea2d6f922be35 : SUCCESS in 1h 02m 30s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/615d2c4ef1554275baabc231f00ae71c : SUCCESS in 47m 43s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/a9f0926846fa4ac3821e9e0070e137ea : SUCCESS in 1h 11m 49s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6136bf8449cf4643a1bee71eb22cb356 : SUCCESS in 13m 26s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/14ce8ea07f15489d875cdebb15c32fac : SUCCESS in 10m 28s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9e566a8af3d545e4b49436a0253b44cf : SUCCESS in 2h 05m 04s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/960ad86b48244b578a8bfbc585cae61e : SUCCESS in 2h 11m 40s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/9bceb66f8bfb4af594c5d1084c75fe83 : SUCCESS in 17m 24s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/53ba9f113f224d6ba08d848967cbe64c : SUCCESS in 15m 19s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5eea7d0492944b3fa4e203c4377c9940 : SUCCESS in 32m 10s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b779cacf7a5e46028aa6ee58930ea087 : SUCCESS in 23m 41s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/75d13ff5d93a42bd8011b214b935516b : SUCCESS in 14m 19s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2da995d429004f14b7e9b74a3ec1c3a2 : SUCCESS in 13m 00s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9f016a272fe347f99b0d8196e885b3b9 : SUCCESS in 19m 35s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/712b8aa27730472ea879955f7fe8654e : SUCCESS in 13m 01s","accounts_in_message":[],"_revision_number":4},{"id":"303f9084bbf3f22393565dcb58ed6a23e83a08ab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 14:57:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"914605e29a45b5b8177817e10706140035dd27fc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 14:58:43.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcded44b4bbd434d9d30a117bd75851a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73b678d4fa684ed780cd505e944a6c62 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/543758d434a04e04864afe81d47828c3 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"78702d0e3094e6d6a16a31eaf2517d4e0f25d1c7","revisions":{"5121282002c6c6d7487bdcda35844404223adf69":{"kind":"REWORK","_number":1,"created":"2023-08-15 10:28:58.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/99/891399/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/99/891399/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/891399/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/891399/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/891399/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/99/891399/1"}}},"commit":{"parents":[{"commit":"c7b0ffd266c4d567ecee8c5b596f252e3e3c531e","subject":"docs: Add a warning on creating network loops","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c7b0ffd266c4d567ecee8c5b596f252e3e3c531e"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-08-15 10:19:41.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-08-15 10:26:26.000000000","tz":60},"subject":"Fix configuration dump with inline encrypted variables","message":"Fix configuration dump with inline encrypted variables\n\nIf inline Ansible vault encryption is used to define an encrypted\nvariable in kayobe-config, running \u0027kayobe configuration dump -l \u003chost\u003e\u0027\nfails with the following:\n\n  Failed to decode config dump YAML file /tmp/tmp_fg1bv_j/localhost.yml:\n  ConstructorError(None, None, \"could not determine a constructor for\n  the tag \u0027!vault\u0027\", \u003cyaml.error.Mark object at 0x7f1e5c7404c0\u003e)\n\nThis change fixes the error by using the Ansible YAML loader which\nsupports the vault tag. Any vault encrypted variables are sanitised in\nthe dump output. Note that variables in vault encrypted files are not\nsanitised.\n\nChange-Id: I4830500d3c927b0689b6f0bca32c28137916420b\nCloses-Bug: #2031390\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5121282002c6c6d7487bdcda35844404223adf69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5121282002c6c6d7487bdcda35844404223adf69"}]},"branch":"refs/heads/master"},"370713955a0efcc7db56458029eb88efaa59ff5a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-10-03 11:54:43.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/99/891399/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/99/891399/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/891399/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/891399/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/891399/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/99/891399/2"}}},"commit":{"parents":[{"commit":"1847ad3f17a039f29e22f465508991fcb6e1cca7","subject":"Fix data file path detection with new pip","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1847ad3f17a039f29e22f465508991fcb6e1cca7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-08-15 10:19:41.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2023-10-03 11:54:43.000000000","tz":0},"subject":"Fix configuration dump with inline encrypted variables","message":"Fix configuration dump with inline encrypted variables\n\nIf inline Ansible vault encryption is used to define an encrypted\nvariable in kayobe-config, running \u0027kayobe configuration dump -l \u003chost\u003e\u0027\nfails with the following:\n\n  Failed to decode config dump YAML file /tmp/tmp_fg1bv_j/localhost.yml:\n  ConstructorError(None, None, \"could not determine a constructor for\n  the tag \u0027!vault\u0027\", \u003cyaml.error.Mark object at 0x7f1e5c7404c0\u003e)\n\nThis change fixes the error by using the Ansible YAML loader which\nsupports the vault tag. Any vault encrypted variables are sanitised in\nthe dump output. Note that variables in vault encrypted files are not\nsanitised.\n\nChange-Id: I4830500d3c927b0689b6f0bca32c28137916420b\nCloses-Bug: #2031390\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/370713955a0efcc7db56458029eb88efaa59ff5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/370713955a0efcc7db56458029eb88efaa59ff5a"}]},"branch":"refs/heads/master","description":"Rebase"},"263be7241ced0d3de77c2738ac6df86a351a8935":{"kind":"REWORK","_number":3,"created":"2023-11-30 08:58:35.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/99/891399/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/99/891399/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/891399/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/891399/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/891399/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/99/891399/3"}}},"commit":{"parents":[{"commit":"323912d769c41bf7f99c7eabe1eec00574108541","subject":"Merge \"kolla-bifrost: Add precheck for inspection allocation pool\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/323912d769c41bf7f99c7eabe1eec00574108541"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-08-15 10:19:41.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-11-30 08:44:30.000000000","tz":60},"subject":"Fix configuration dump with inline encrypted variables","message":"Fix configuration dump with inline encrypted variables\n\nIf inline Ansible vault encryption is used to define an encrypted\nvariable in kayobe-config, running \u0027kayobe configuration dump -l \u003chost\u003e\u0027\nfails with the following:\n\n  Failed to decode config dump YAML file /tmp/tmp_fg1bv_j/localhost.yml:\n  ConstructorError(None, None, \"could not determine a constructor for\n  the tag \u0027!vault\u0027\", \u003cyaml.error.Mark object at 0x7f1e5c7404c0\u003e)\n\nThis change fixes the error by using the Ansible YAML loader which\nsupports the vault tag. Any vault encrypted variables are sanitised in\nthe dump output. Note that variables in vault encrypted files are not\nsanitised.\n\nChange-Id: I4830500d3c927b0689b6f0bca32c28137916420b\nCloses-Bug: #2031390\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/263be7241ced0d3de77c2738ac6df86a351a8935"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/263be7241ced0d3de77c2738ac6df86a351a8935"}]},"branch":"refs/heads/master"},"78702d0e3094e6d6a16a31eaf2517d4e0f25d1c7":{"kind":"REWORK","_number":4,"created":"2023-11-30 09:55:58.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/99/891399/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/99/891399/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/891399/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/891399/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/891399/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/99/891399/4"}}},"commit":{"parents":[{"commit":"323912d769c41bf7f99c7eabe1eec00574108541","subject":"Merge \"kolla-bifrost: Add precheck for inspection allocation pool\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/323912d769c41bf7f99c7eabe1eec00574108541"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-08-15 10:19:41.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-11-30 09:55:46.000000000","tz":60},"subject":"Fix configuration dump with inline encrypted variables","message":"Fix configuration dump with inline encrypted variables\n\nIf inline Ansible vault encryption is used to define an encrypted\nvariable in kayobe-config, running \u0027kayobe configuration dump -l \u003chost\u003e\u0027\nfails with the following:\n\n  Failed to decode config dump YAML file /tmp/tmp_fg1bv_j/localhost.yml:\n  ConstructorError(None, None, \"could not determine a constructor for\n  the tag \u0027!vault\u0027\", \u003cyaml.error.Mark object at 0x7f1e5c7404c0\u003e)\n\nThis change fixes the error by using the Ansible YAML loader which\nsupports the vault tag. Any vault encrypted variables are sanitised in\nthe dump output. Note that variables in vault encrypted files are not\nsanitised.\n\nChange-Id: I4830500d3c927b0689b6f0bca32c28137916420b\nCloses-Bug: #2031390\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/78702d0e3094e6d6a16a31eaf2517d4e0f25d1c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/78702d0e3094e6d6a16a31eaf2517d4e0f25d1c7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
