)]}'
{"id":"openstack%2Fkayobe~892091","triplet_id":"openstack%2Fkayobe~stable%2F2023.1~I6a9c521d43b21ab6b6c01e4688cab697608a667c","project":"openstack/kayobe","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2023-08-22 11:00:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6a9c521d43b21ab6b6c01e4688cab697608a667c","subject":"Fix CentOS / Rocky route options","status":"MERGED","created":"2023-08-21 08:46:29.000000000","updated":"2023-08-22 11:01:40.000000000","submitted":"2023-08-22 11:00:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892091","cherry_pick_of_change":889680,"cherry_pick_of_patch_set":2,"meta_rev_id":"59eaaf849c9ef465460a1c8d772633291c8b6ff8","_number":892091,"virtual_id_number":892091,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-22 11:00:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2023-08-22 09:13:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2023-08-21 10:07:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-08-21 10:07:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2023-08-22 09:13:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-21 08:46:29.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"CC"},{"updated":"2023-08-21 10:07:09.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-08-21 10:29:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-22 09:13:12.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"ba416936d5277792ea96dadc3061a8bb561ca8cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-08-21 08:46:29.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c3f13d79737c374b55d10a03b90bc827556ad724","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-21 10:07:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a52f056f8aa2013528d18fcb54dec6345c83287d","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-08-21 10:07:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a98697bede28e7bd89df0239bcdc880fd72f683f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 10:29:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02d415eb456042c9989d31047ebc45da\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b96084d403b4d67b711033f254f8ed2 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/713008b1ede7489e99140d263d66f33a : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6eba6927873a415eba5498a989ef062e : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/841281c0e6604b6c95c712c64a811942 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dcfce44a7754c9fbfe3496d1321e344 : SUCCESS in 9m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a564270121a49f6ac8853e997e0601a : SUCCESS in 3m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/17ab308499874d3c97772d052db03657 : SUCCESS in 4m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/442dea31ac4a4b2491b3206683f27a56 : SUCCESS in 11m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ea4a6337e8da4f019e8ec16446ae4309 : SUCCESS in 14m 06s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/032ac1723892493b9eb79807b532c775 : SUCCESS in 1h 00m 59s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/053c6e8cf5b74e62a40712d9d5ff56fb : SUCCESS in 56m 50s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6cd975113eaf4b838caf1a9be13631aa : SUCCESS in 58m 11s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/bd6864ddf4e24e4eb49b4f262e2be11a : SUCCESS in 1h 06m 37s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6e6572d6e70d4918b51052d5fb8a16cc : SUCCESS in 16m 39s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/afa0bbf0dc2a4a29a91874c53c9ce34b : SUCCESS in 18m 46s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/75b565ba4ac449b486bca121c86f04d3 : SUCCESS in 14m 12s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/10557cf782de4cca9d05305d07c0fe4f : SUCCESS in 1h 38m 21s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/25ba0523e4ce4f6aa58fa25a2ecf45e4 : SUCCESS in 1h 35m 36s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/7dfe3d5b52b64a8ab41a7bbfcd62334b : SUCCESS in 14m 05s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c2992c57b9e84a8aaf11b7aeda438589 : SUCCESS in 15m 00s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/bbbfdc71cfd74eddae0fadccc4a66a1e : SUCCESS in 31m 41s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/be6787a7463946b6910483791596b5ac : SUCCESS in 28m 46s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a12b6aa9f9284529a2257d28a5babe03 : SUCCESS in 22m 52s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/87be2bc1d7574611a4b3ab924a018241 : SUCCESS in 10m 22s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2bd0c13088c24118900f9430860bb414 : SUCCESS in 12m 14s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6da895a355c446cabbc0c36ea567d8b0 : SUCCESS in 31m 44s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b7c11eefd8b43d4bbb3455da11e5abe : SUCCESS in 30m 50s","accounts_in_message":[],"_revision_number":1},{"id":"3e77293c90dfbdc4261b7c51ca6c4b048eeb0996","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-08-22 09:13:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b47d2aaf80ba58397c96a23ff31da350ad0aef48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 09:14:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fb1732d0b54d853dccb99fc2441ffb8b8e494af9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 11:00:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7f0cb7353ad413baeec01c84e346ab1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47e0242aab8a4634869d3838003c4d46 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a5e230af9714cff97a39d3d512d53d0 : SUCCESS in 5m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a902490fb8c64ddcb412d63283839b09 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/864ec89b3978466b968f976a545ea46f : SUCCESS in 9m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61795741faea48bd95846159e945d60e : SUCCESS in 3m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/08ea07382274419facb1ab3abc254e69 : SUCCESS in 5m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2da33320275a4e80a1b866a5c456b499 : SUCCESS in 10m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6b59fb37804d47318b207fd8378f20bb : SUCCESS in 15m 04s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/2228d572320f466dbcb3aa6a4bc5cc74 : SUCCESS in 55m 48s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c0fb31bbb0447b88bb53108e35d9a73 : SUCCESS in 57m 30s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/53d1893130bc4ac29dec2b6d86946f86 : SUCCESS in 1h 05m 37s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/632b1f6b2c35462fac32c632da05461d : SUCCESS in 18m 07s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6b9dc7d5afe14aea8fa613db806507fa : SUCCESS in 14m 40s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/d6a17bb2922a46cb9531ef201b638a48 : SUCCESS in 1h 36m 28s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be37725db76c498d966d2e6d5c31fdb7 : SUCCESS in 1h 38m 58s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/465cac6d1760463c97a1cd4e2a818b8c : SUCCESS in 19m 25s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/34be797accd94de39512a30cf13e2a30 : SUCCESS in 15m 16s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2838da5e654b4c6ea317b3b43b197653 : SUCCESS in 31m 13s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/47d64683509d4cbeb0335f7573e5d9d8 : SUCCESS in 26m 39s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/87c560c49a974e9ba3ab83644d0eaa20 : SUCCESS in 14m 14s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1d8eec8d77424af98c471e83033463cd : SUCCESS in 11m 23s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/7b4d20e75460405d91f19793d3cb5a56 : SUCCESS in 34m 47s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/12cf053992424295b97ec4647dd8dadd : SUCCESS in 29m 49s","accounts_in_message":[],"_revision_number":1},{"id":"2c6bcd798dae91e6bb153a7f05d9915a0a5d8340","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 11:00:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"59eaaf849c9ef465460a1c8d772633291c8b6ff8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 11:01:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/326665ed641b4c22b48d503074a49056\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5aae657ac6ec4ad9a41abb656c3a6fa9 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9ecd14ce692432c999af51c8813e696 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"624819945f46e8cda70bb1e6262fccc2644d3b57","revisions":{"624819945f46e8cda70bb1e6262fccc2644d3b57":{"kind":"REWORK","_number":1,"created":"2023-08-21 08:46:29.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/91/892091/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/91/892091/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/892091/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/892091/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/91/892091/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/91/892091/1"}}},"commit":{"parents":[{"commit":"68bdfcec6ae661c6c13f178bdb5e7ff859c3b785","subject":"docs: Add a warning on creating network loops","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/68bdfcec6ae661c6c13f178bdb5e7ff859c3b785"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2023-07-26 08:27:48.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2023-08-21 08:46:29.000000000","tz":0},"subject":"Fix CentOS / Rocky route options","message":"Fix CentOS / Rocky route options\n\nUpdates interfaces role to a fixed version.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/890198\n\nChange-Id: I6a9c521d43b21ab6b6c01e4688cab697608a667c\n(cherry picked from commit cf6f12b26d02d51bb92d81f6e398d7b11ae14783)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/624819945f46e8cda70bb1e6262fccc2644d3b57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/624819945f46e8cda70bb1e6262fccc2644d3b57"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
