)]}'
{"id":"openstack%2Fkayobe~892478","triplet_id":"openstack%2Fkayobe~stable%2Fzed~I2085f9d1fd457e5b54b17f37a5e725990a60e8bc","project":"openstack/kayobe","branch":"stable/zed","topic":"docker_login-stable/zed","attention_set":{},"removed_from_attention_set":{"34940":{"account":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"last_update":"2023-08-25 12:42:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2085f9d1fd457e5b54b17f37a5e725990a60e8bc","subject":"Fix for seed-containers being unable to use password protected registry","status":"MERGED","created":"2023-08-23 10:20:52.000000000","updated":"2023-08-25 12:44:49.000000000","submitted":"2023-08-25 12:42:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892478-docker_login-stable/zed","cherry_pick_of_change":887792,"cherry_pick_of_patch_set":9,"meta_rev_id":"9695dd8e8a9e538863f01b4d1ac4613641be3e19","_number":892478,"virtual_id_number":892478,"owner":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-25 12:42:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2023-08-23 10:57:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-08-24 08:33:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2023-08-25 11:01:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2023-08-25 11:01:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-23 10:57:09.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-08-23 11:20:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-24 08:33:39.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-08-25 11:01:36.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"8b9411de943d69b253268fee80120d73c230369f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2023-08-23 10:20:52.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"143203e822889f43aa4b52fc583b582661c622e6","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-23 10:57:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7a2787898e9682bba88800bbce46f551430ae94f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 11:20:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9f7b6d065c048908cd59a8207c4dfdd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/831e8b8046934eb09ad2b8005e958209 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8626c6c63284405690a1a7cf776a5ca1 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/137f0aed952c4953903670e80896a5eb : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc8017d3ac504b13b23d08901904a436 : SUCCESS in 5m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/84982bfe32954e42afbade1ae91add43 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ab18a527d3f49dfa79228af8c338719 : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c174a6399d448068f3524429287f65f : SUCCESS in 3m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ae09ad503e3e4afe867b3fb7c99e6be1 : SUCCESS in 4m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fc9015587f8f40c1b76c66ffcd6a00c0 : SUCCESS in 10m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aaaf722dacb148a38725784456ecb846 : SUCCESS in 20m 22s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/736bb4a867b74adb8dcebcd78ab1a7d8 : SUCCESS in 44m 16s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/e77e58215d2d4cdba9b45fd15509523e : SUCCESS in 50m 21s (non-voting)\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ce3174b04a9749abb01720d2c0031e28 : SUCCESS in 54m 37s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/76d6fe3792874704a6942e42d601a6ae : SUCCESS in 18m 27s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/cc190ba80091476f9b180f3a1762a03a : SUCCESS in 18m 14s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b450ed099bae4c2ba9ef22656ea74406 : SUCCESS in 12m 09s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/5b0aa91a17154a7586b62ceb606be607 : SUCCESS in 16m 12s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d34e3f2e02874a2a931acf5949122e53 : SUCCESS in 16m 09s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/bbf81f293c3b495ea09fcf9fe0100897 : SUCCESS in 31m 29s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7e40ea374da48539b6dd6e579996582 : SUCCESS in 24m 43s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4972816e32594dcdba9825ac7377402b : SUCCESS in 10m 26s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fa588268a425457a830a1cc05cc1f05a : SUCCESS in 9m 58s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/6a8cf7a91d7a41a1944bf0f08fd3e063 : SUCCESS in 17m 13s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/875d80a540574d549bc6044a1f587c05 : SUCCESS in 15m 01s","accounts_in_message":[],"_revision_number":1},{"id":"8830a373e3e9d04d710c4ff59858328a5c0741bd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-08-24 08:33:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a88314258e44f18f0672978a7af83d50754d8199","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-08-25 11:01:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"92fa9bbb085cb46f61a353df1917226a13003f23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 11:02:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b23eb26b972c95777fe0d0e79542588993287003","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 12:42:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5c77fc43ba1495fad99a9844fa3a619\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a922592702bf4261b42a33e2db267603 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6250a1d57fa7424d828b5bee23c2480e : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6642af1d8f3b4a51a5318933be3784a0 : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/237fa89aa6de4b398a94b226b738af92 : SUCCESS in 7m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bea2edc1c20942a880e0704688ba08ba : SUCCESS in 3m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9eb498281ab54ec98a4bfc6cc813f748 : SUCCESS in 4m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/41360b416f2c44e89c22da2bc0bb79fb : SUCCESS in 10m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c172bf129b2248f9ba51b28b1d15df1b : SUCCESS in 34m 33s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ddbd01adb84b45239b8e085281d1d47f : SUCCESS in 49m 05s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/25b102ef83a14817a00eb19c4d713820 : SUCCESS in 56m 24s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/d98edc8a741a46c796e69824dc329841 : SUCCESS in 17m 56s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0e94794e7269482f9f2dffb624abaee6 : SUCCESS in 13m 14s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3bb449c2df94441681cb395684d85802 : SUCCESS in 16m 54s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8c6d0ccb8ca94b23819210ab987cb627 : SUCCESS in 15m 42s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cbd73e1673d843a392ccb7a682f96d81 : SUCCESS in 26m 55s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/670a25dd1fde42e0994b06807b52dbb7 : SUCCESS in 10m 47s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1fff9a1cdfe64e7b94dc7a98aa2ae62f : SUCCESS in 10m 21s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/094a860eb60443cdaecefebcb5c22719 : SUCCESS in 14m 25s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1438aa7d52f44e39a94454053ba29287 : SUCCESS in 13m 41s","accounts_in_message":[],"_revision_number":1},{"id":"0b3492456d7f5c6e1167d468221690c7aea69cba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 12:42:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9695dd8e8a9e538863f01b4d1ac4613641be3e19","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 12:44:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f535672b92c4d18b75b57ad987b0e34\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33785d0f906047ff92c8f071cf445ce4 : SUCCESS in 1m 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72f639caff7e47c59a46c338fd2cecb4 : SUCCESS in 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fbe1bd4f2daff277dbf790f2af5e1d075b1bb930","revisions":{"fbe1bd4f2daff277dbf790f2af5e1d075b1bb930":{"kind":"REWORK","_number":1,"created":"2023-08-23 10:20:52.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/78/892478/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/78/892478/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/892478/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/892478/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/892478/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/78/892478/1"}}},"commit":{"parents":[{"commit":"ea03f0f8cc74eeff2997561c028a1e9e5976d919","subject":"Merge \"Fix CentOS / Rocky route options\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ea03f0f8cc74eeff2997561c028a1e9e5976d919"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-07-06 11:08:49.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-08-23 10:20:52.000000000","tz":0},"subject":"Fix for seed-containers being unable to use password protected registry","message":"Fix for seed-containers being unable to use password protected registry\n\nUpdate old code to use truthy filter\n\nCloses-bug: #2026276\nChange-Id: I2085f9d1fd457e5b54b17f37a5e725990a60e8bc\n(cherry picked from commit ed263eeb4f8e85c2b33d3f92f5f711f725f5fdae)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fbe1bd4f2daff277dbf790f2af5e1d075b1bb930"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fbe1bd4f2daff277dbf790f2af5e1d075b1bb930"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
