)]}'
{"id":"openstack%2Fkayobe~892479","triplet_id":"openstack%2Fkayobe~stable%2Fyoga~I2085f9d1fd457e5b54b17f37a5e725990a60e8bc","project":"openstack/kayobe","branch":"stable/yoga","topic":"docker_login-stable/yoga","attention_set":{},"removed_from_attention_set":{"34940":{"account":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"last_update":"2023-08-25 12:42:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2085f9d1fd457e5b54b17f37a5e725990a60e8bc","subject":"Fix for seed-containers being unable to use password protected registry","status":"MERGED","created":"2023-08-23 10:21:28.000000000","updated":"2023-08-25 12:43:47.000000000","submitted":"2023-08-25 12:42:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892479-docker_login-stable/yoga","cherry_pick_of_change":887792,"cherry_pick_of_patch_set":9,"meta_rev_id":"9750a295b7f7d9a8dfa003b6f8abebc943248c6f","_number":892479,"virtual_id_number":892479,"owner":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-25 12:42:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2023-08-23 10:57:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-08-24 08:35:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2023-08-25 11:01:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2023-08-25 11:01:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-23 10:57:16.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-08-23 12:29:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-24 08:35:54.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2023-08-25 11:01:30.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"58ad086a7b71786bf12c3d51e267c84e3fcf1895","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"date":"2023-08-23 10:21:28.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"3f1993bb647bcd016d7c44a3babb722c4657ffba","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-23 10:57:16.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bf4bb2a7206028b3f1eddd7580e105d9c33f3afc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-23 12:29:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cbb6d9b0d8043fe9dbb9e6c7da84876\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb1e367b82b6439198456a5990e24a5d : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/733d0ffd014c47bfbec768e45c5bc038 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5394cfc796047ea87b90ba3797bb188 : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d6c25ace9d04d90b865f89f24902dc5 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34880a46522f4d36ab853dc4f66fef1b : SUCCESS in 9m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59bc6037042544e5bb15266d3feacb9a : SUCCESS in 4m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/35e27e52dcd845e18341c4c2a8357f74 : SUCCESS in 5m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7d18715c7b6843ae89e35f1def898f48 : SUCCESS in 14m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fe2151731e484d79b1c163d96e18204c : SUCCESS in 21m 46s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7aada9ee92c44140b9659fc6b7d5c86f : SUCCESS in 1h 09m 12s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/07234b3436cd4957a6f9974d39c029b3 : SUCCESS in 45m 56s (non-voting)\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/281829c268124f6d8d25e205a1a40a87 : SUCCESS in 1h 05m 31s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/eef2edbbc74745769c10cd34368bd4c4 : SUCCESS in 1h 01m 30s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0cf033b7ea0540369ac4065475099e1a : SUCCESS in 25m 22s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/d6b069f5438f40abad25e3ba7223ecba : SUCCESS in 21m 00s (non-voting)\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/bb4314d078b34ca1bc4fbf2dc902633e : SUCCESS in 19m 31s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/6945544b424e485fa114fdf12f3f02b5 : SUCCESS in 20m 13s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5f784c54c7943f0afaad191141b6e64 : SUCCESS in 16m 47s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/57559bc4863c42839548258bea4ce94d : SUCCESS in 16m 24s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/fbefeaf8f6764756bfd1eeef21fe77f2 : SUCCESS in 1h 15m 05s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/33493d455ca44bde9f55c4ffafbafc89 : SUCCESS in 53m 55s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/79827b94047e4290afa5274e1c21ce8c : SUCCESS in 2h 03m 22s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/6d8ede4462a54997a1607811bcc3464a : SUCCESS in 18m 22s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/6cf75fb0214b41cd8520d7b8b9cf6ade : SUCCESS in 21m 15s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/331533f4bcb649cba39fe08d5d63380c : SUCCESS in 22m 01s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34eada7a55f24422bb0ff80d1087c1a7 : SUCCESS in 16m 52s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/65bfda26220b469f84510406f43c61e4 : SUCCESS in 18m 45s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f28f5021bdb340288004c8b4736a4ae5 : SUCCESS in 30m 31s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f976d8125d34d7eb135f879a5f554e6 : SUCCESS in 25m 17s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f89b9d2ac337466a890f873f01cbff47 : SUCCESS in 19m 08s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/2ba3cd06de7c4052b44574f2d49265e7 : SUCCESS in 17m 32s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/b00e65850769468abf7f9c3a8aaaed11 : SUCCESS in 16m 31s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ca8eb540a8a4190bc9a169a92527629 : SUCCESS in 12m 19s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f648fa316919473c89d040b998d8e1f4 : SUCCESS in 12m 10s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/e415c8915dbd40728015e0ce5349f860 : SUCCESS in 18m 18s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/919c948afb944db883e81c08f06a7fd3 : SUCCESS in 17m 45s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0ee0811d24504e79b48a2e3bfedb1e43 : SUCCESS in 18m 18s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f623d09e996d452e88e453ce9574a014 : SUCCESS in 16m 42s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c021ce30a0f54167b1ee505d770e68a7 : SUCCESS in 12m 23s","accounts_in_message":[],"_revision_number":1},{"id":"5d51fa8ad82e1233eeea0a6514768829a545515a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-08-24 08:35:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6105897e9208604bf6eba8bbd5eef0a017978b93","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2023-08-25 11:01:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ccd8b7a17214fc3dbd8571229365728456b1f8bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 11:01:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7ceba9a1f972cf0e79f384c5d931a8d18230af4e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 12:42:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9f5b41c42e344818a3bc35d68330ee4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd52afd0a4b740ebb2c7ca991f2c7bd5 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a2b32224e4f413d94594132c6efaa92 : SUCCESS in 6m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab669dccae9d41ea87ca1a4b6488ff80 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcbf2528583e4e5681e67c80e5192516 : SUCCESS in 9m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb3802eef0624bc39d392e08b444628c : SUCCESS in 4m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dae7d288ab294e169036982b70b27877 : SUCCESS in 6m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d06bea525fc8403380ca0c66a4fbc0c3 : SUCCESS in 13m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0ac5f11628894b70ac5c16f9de811131 : SUCCESS in 18m 54s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6120f2c20963476d99cf9953fdaa8746 : SUCCESS in 1h 04m 02s\n- kayobe-overcloud-rocky8 https://zuul.opendev.org/t/openstack/build/74019c9c0d6a4e8b8b377c7e0dd21824 : SUCCESS in 1h 06m 25s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/90d1b0cc9f654936a7e4d0b8637e989d : SUCCESS in 54m 54s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/eeac694be6534132bf8ca272edb042ad : SUCCESS in 24m 53s\n- kayobe-overcloud-host-configure-rocky8 https://zuul.opendev.org/t/openstack/build/914db29913f445818caf135a03cb07de : SUCCESS in 24m 33s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/59bf6b7f9c9d4474933ce63e522554a0 : SUCCESS in 20m 33s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7a6bca0fc934be68a272b9332b78338 : SUCCESS in 16m 11s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8f4c83a9bb634c2a93dc883cebf01e58 : SUCCESS in 17m 56s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/38977c0c6c3e464e8330e4d4039c7dce : SUCCESS in 1h 13m 24s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/6c1933d589e64a85ba915303b5ae58a9 : SUCCESS in 52m 29s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4ee87fb2a5414b99a71999d7ab7fcd0f : SUCCESS in 1h 35m 46s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/9718dd5b793346ec8fe4ddef83391c3a : SUCCESS in 22m 24s\n- kayobe-seed-rocky8 https://zuul.opendev.org/t/openstack/build/2051b07b77e54d3ea030463b0acac514 : SUCCESS in 18m 21s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/dfdcc0c399724712956415ccdda9ca18 : SUCCESS in 21m 21s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d6959d00d2448eebc3d814bea78380e : SUCCESS in 14m 39s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/10f1371f7b1749cbbf2219a22b04c204 : SUCCESS in 17m 31s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0ef0aa1909e945c9b583c4a23a2897e3 : SUCCESS in 25m 29s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd4a6fec7449428886913b51874ff617 : SUCCESS in 24m 44s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/2e50341f550a443b92b4e76c6f4f040a : SUCCESS in 16m 14s\n- kayobe-seed-vm-rocky8 https://zuul.opendev.org/t/openstack/build/13a5c49de4fe42bd9d64f51a5f16c953 : SUCCESS in 12m 48s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5ee58d1e1d8e446f87fad89cd7fc35f0 : SUCCESS in 15m 45s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75467a198cb54e2bbc2a39fee4ea5d0b : SUCCESS in 11m 14s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1d57338bacc84bcf87d9bdd308c83f5c : SUCCESS in 11m 54s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/c68916bc24444ddd9e052e089b764bb2 : SUCCESS in 21m 30s\n- kayobe-infra-vm-rocky8 https://zuul.opendev.org/t/openstack/build/161871b57a314d3dad40e991005d02fb : SUCCESS in 18m 06s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9d558d64f5614e5aa95cdcdccf80bc2f : SUCCESS in 17m 59s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/88a8f7f439064f40904bd06342031dea : SUCCESS in 14m 23s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/33aab8e48dcc431fb77e9a688f6b5654 : SUCCESS in 14m 38s","accounts_in_message":[],"_revision_number":1},{"id":"9ad929261518f9ac532dd01375c9a6ba7f04874d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 12:42:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9750a295b7f7d9a8dfa003b6f8abebc943248c6f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-25 12:43:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b554c6655b14714897d33794106cfcb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6b2fc90664d45c28eabaef06498cd53 : SUCCESS in 37s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da419c25c8dd41ec9af4d6bb0a177fa5 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4132c190a8dd979e704597a9cbd66eb7c1e07dc9","revisions":{"4132c190a8dd979e704597a9cbd66eb7c1e07dc9":{"kind":"REWORK","_number":1,"created":"2023-08-23 10:21:28.000000000","uploader":{"_account_id":34940,"name":"Rafal Lewandowski","display_name":"Rafal Lewandowski","email":"rafal@stackhpc.com","username":"rav"},"ref":"refs/changes/79/892479/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/79/892479/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/892479/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/892479/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/79/892479/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/79/892479/1"}}},"commit":{"parents":[{"commit":"b9b2e71d3a3aab402d3e6a9b6ad052536bc03b50","subject":"Fix CentOS / Rocky route options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b9b2e71d3a3aab402d3e6a9b6ad052536bc03b50"}]}],"author":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-07-06 11:08:49.000000000","tz":120},"committer":{"name":"Rafal Lewandowski","email":"rafal@stackhpc.com","date":"2023-08-23 10:21:28.000000000","tz":0},"subject":"Fix for seed-containers being unable to use password protected registry","message":"Fix for seed-containers being unable to use password protected registry\n\nUpdate old code to use truthy filter\n\nCloses-bug: #2026276\nChange-Id: I2085f9d1fd457e5b54b17f37a5e725990a60e8bc\n(cherry picked from commit ed263eeb4f8e85c2b33d3f92f5f711f725f5fdae)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4132c190a8dd979e704597a9cbd66eb7c1e07dc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4132c190a8dd979e704597a9cbd66eb7c1e07dc9"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
